)]}'
{"id":"openstack%2Fneutron~834650","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I5b96b3a91aacffe118ddbb91a75c4892818ba97a","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/1965807","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-24 09:30:53.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-03-23 09:28:49.000000000","reason":"Slawek Kaplonski replied on the change"}},"hashtags":[],"change_id":"I5b96b3a91aacffe118ddbb91a75c4892818ba97a","subject":"Remove exception ``IpAddressAllocationNotFound``","status":"MERGED","created":"2022-03-22 11:21:40.000000000","updated":"2022-03-24 09:32:18.000000000","submitted":"2022-03-24 09:30:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834650-bug/1965807","meta_rev_id":"7109ecff5ff972f66b2d817495e6546ad32328f3","_number":834650,"virtual_id_number":834650,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-24 09:30:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-03-24 08:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-03-23 09:28:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-03-24 08:56:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-22 12:03:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 08:36:05.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-03-23 08:39:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"a047fa14c1fcc8cde9590f6357f85206c53b46d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-22 11:21:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61302f83d603462ad07963a92f14d1e8e2325d16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:03:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9098ed46af03406fa3655e290e9c6861 : SUCCESS in 36m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a361c0455bd64e839eb0d596c0711dbd : SUCCESS in 33m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7a970209bda4358b0d9b87d2e06ddb8 : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bfbdbd32f574c0a879bf5f672cb2804 : SUCCESS in 30m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ba3949b9fb84ecdbea22b6778ca62c9 : SUCCESS in 30m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906193c4d6254850a6090ef80a750919 : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"e5f204ec39acc9ee502ac6aacf20c8665db4d928","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:50:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a197d6d5fa37489ba54a3f9e33329c38 : TIMED_OUT in 1h 22m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cced9717f76c4cc4ba3531ac8ef562bd : TIMED_OUT in 1h 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a852fca25dc46dc9b93420f716c0c25c9987dd5d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-23 08:36:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f9b960e2df34c26f429bafa64a1d5c8c2f64a20","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-23 08:39:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b1aebde94e077fd185c960b5dce8827e89772d7b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-23 09:21:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"733626e15fcab506aea02a6cf112a4ac9460d85e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-23 09:27:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5a632ed0af5f0ad70428f699e9a377c4220c1f9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-23 09:28:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a0144489fb5c5f7f53f90f668cb1d01a6a4e0dd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-23 09:28:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"efbc2ebbe4f2f5b881ca88155f315bc18f75a8eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 10:18:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ded17a0188d6465897fe53aa59c66f72 : SUCCESS in 43m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed2963beab7b498aa7d423a0957245d9 : SUCCESS in 34m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532dac13e4f1428daef6893cac5b2a3e : SUCCESS in 16m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8d05383e95b48799dc28ba127d43863 : SUCCESS in 29m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af305c6c19db45e29c44b16043aaae62 : SUCCESS in 28m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02d5e3e6c790441886034ab099554de9 : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":2},{"id":"fcfa808fadcd19fdc60d2c801184dd57cc4a69bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 10:33:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a903b30cec2476e878c162fd4b34835 : SUCCESS in 1h 01m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f577f47788643afbf79eac8a8832c67 : SUCCESS in 1h 02m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c06f61b2589dd1102243f209ffb1d8c19b89b3a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-03-24 08:56:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5a7f7b8392de6016fc60891ee862721c1a909248","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 08:57:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f6c593d9942f5e8ea1640d833bb486847cd44621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 09:30:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2402890d7d6445fb9a657de57f4f477 : SUCCESS in 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac4c29e8e2b54a769986e07ceca6c8a8 : SUCCESS in 14m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5e5261d40fa490b919c40c40f6ed506 : SUCCESS in 30m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e67abdb882c2444b81df89e993e42381 : SUCCESS in 29m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c21c2982d74e179e39b6983ab04371 : SUCCESS in 10m 48s","accounts_in_message":[],"_revision_number":2},{"id":"0690db2b7dd0f131c19dab3d5c548327f282f70a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 09:30:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7109ecff5ff972f66b2d817495e6546ad32328f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 09:32:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7aaac6a5224997a053f66563cd4ea7 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a6a4ab5dacc2e92c24d68774bf286c9e6ae2d94","revisions":{"412371ab5c1568b9b46353b56a41d77576bffcec":{"kind":"REWORK","_number":1,"created":"2022-03-22 11:21:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/834650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/834650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/834650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/834650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/834650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/834650/1"}}},"commit":{"parents":[{"commit":"962d075fb80acdf94a63981f385b8e8e120f5497","subject":"Merge \"Enable sctp module in the fullstack Centos node\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/962d075fb80acdf94a63981f385b8e8e120f5497"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 16:32:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 20:38:44.000000000","tz":0},"subject":"Remove exception ``IpAddressAllocationNotFound``","message":"Remove exception ``IpAddressAllocationNotFound``\n\nThis patch removes the ``IpAddressAllocationNotFound`` exception. This\nexception was raised when a IPAM register was called to be deleted\nbut not found.\n\nAs reported in the LP bug, this IPAM register deletion can be called\nseveral times if a port fails during the creation. The IPAM register\ndeletion calls the DB deletion but doesn\u0027t raise any exception if the\nregister does not exist. The code ensures the IPAM register is\ndeleted and there is no need to fail if it is not present anymore.\n\nThis patch also removes the exception catch and try in \"update_port\",\nthat was added in [0] as a fix for [1]. That was added because the\nsubnet deletion code involved a port update call [2] during the\nIP allocation deletion, if any port was still present in the subnet.\nSince [3], this code is not needed because the subnet deletion does\nnot call a port update anymore.\n\n[0]https://review.opendev.org/c/openstack/neutron/+/373536\n[1]https://bugs.launchpad.net/neutron/+bug/1622616\n[2]https://github.com/openstack/neutron/blob/pike-em/neutron/db/db_base_plugin_v2.py#L1017-L1018\n[3]https://review.opendev.org/c/openstack/neutron/+/713045\n\nCloses-Bug: #1965807\nRelated-Bug: #1954763\nRelated-Bug: #1622616\n\nChange-Id: I5b96b3a91aacffe118ddbb91a75c4892818ba97a\n(cherry picked from commit 83b6ce9e9ec2f0261fed630e18cbe9cfa363a0dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/412371ab5c1568b9b46353b56a41d77576bffcec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/412371ab5c1568b9b46353b56a41d77576bffcec"}]},"branch":"refs/heads/stable/yoga"},"2a6a4ab5dacc2e92c24d68774bf286c9e6ae2d94":{"kind":"REWORK","_number":2,"created":"2022-03-23 09:27:37.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/50/834650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/834650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/834650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/834650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/834650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/834650/2"}}},"commit":{"parents":[{"commit":"962d075fb80acdf94a63981f385b8e8e120f5497","subject":"Merge \"Enable sctp module in the fullstack Centos node\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/962d075fb80acdf94a63981f385b8e8e120f5497"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-16 16:32:31.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-17 06:25:30.000000000","tz":0},"subject":"Remove exception ``IpAddressAllocationNotFound``","message":"Remove exception ``IpAddressAllocationNotFound``\n\nThis patch removes the ``IpAddressAllocationNotFound`` exception. This\nexception was raised when a IPAM register was called to be deleted\nbut not found.\n\nAs reported in the LP bug, this IPAM register deletion can be called\nseveral times if a port fails during the creation. The IPAM register\ndeletion calls the DB deletion but doesn\u0027t raise any exception if the\nregister does not exist. The code ensures the IPAM register is\ndeleted and there is no need to fail if it is not present anymore.\n\nThis patch also removes the exception catch and try in \"update_port\",\nthat was added in [0] as a fix for [1]. That was added because the\nsubnet deletion code involved a port update call [2] during the\nIP allocation deletion, if any port was still present in the subnet.\nSince [3], this code is not needed because the subnet deletion does\nnot call a port update anymore.\n\n[0]https://review.opendev.org/c/openstack/neutron/+/373536\n[1]https://bugs.launchpad.net/neutron/+bug/1622616\n[2]https://github.com/openstack/neutron/blob/pike-em/neutron/db/db_base_plugin_v2.py#L1017-L1018\n[3]https://review.opendev.org/c/openstack/neutron/+/713045\n\nCloses-Bug: #1965807\nRelated-Bug: #1954763\nRelated-Bug: #1622616\n\nConflicts:\n    neutron/ipam/exceptions.py\n\nChange-Id: I5b96b3a91aacffe118ddbb91a75c4892818ba97a\n(cherry picked from commit 83b6ce9e9ec2f0261fed630e18cbe9cfa363a0dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a6a4ab5dacc2e92c24d68774bf286c9e6ae2d94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a6a4ab5dacc2e92c24d68774bf286c9e6ae2d94"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
