)]}'
{"id":"openstack%2Fneutron~834740","triplet_id":"openstack%2Fneutron~master~I569401fc91d67ce358c83183de3816c70471cc4b","project":"openstack/neutron","branch":"master","topic":"remove-get_security_group_rules_count","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-23 08:30:18.000000000","reason":"Rodolfo Alonso replied on the change"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2022-03-23 08:30:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I569401fc91d67ce358c83183de3816c70471cc4b","subject":"Remove unneeded method \"get_security_group_rules_count\"","status":"ABANDONED","created":"2022-03-22 16:05:46.000000000","updated":"2022-03-23 08:30:21.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0aa5f15b90083493df7b8856af4792902ad3b757","_number":834740,"virtual_id_number":834740,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-03-22 17:17:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-03-22 17:46:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-03-22 20:27:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-22 17:17:27.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 17:32:35.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-22 17:46:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-22 18:24:59.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-03-22 20:27:20.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"}],"messages":[{"id":"2338b3b3048a6dbf6a23836e4d84d0d64b7b20f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-22 16:05:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd76a70bef556803caabcef8618595c4c26da672","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-22 17:17:27.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/40/834740/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9e0a20ff0a9c1ea45110a4351654fd7cded14314","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-22 17:32:35.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/40/834740/1/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/40/834740/1/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"3282d44a9def7102bd2569455396217843f1c43f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 17:46:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/fb4c9c49e37a437bb0ec571708dc9127 : SUCCESS in 48m 46s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4d2e14e0d69544ba94e3d8d4bf819ee4 : SUCCESS in 56m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/589dea7d70bc42b1a189ac8eb3d6c4a9 : SUCCESS in 1h 11m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e9d0c33abfff4719bd356a4ba3183459 : SUCCESS in 48m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d6f7982f2bbd4138a2ae3b2448e6b380 : SUCCESS in 58m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26ef33593d6b420e883c4b46363f831e : SUCCESS in 27m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca2b01b5e9e347629895cdf713f15784 : FAILURE in 29m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f18e52e2bd0479bb40e427ce2ffa9d1 : FAILURE in 37m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61b6eac405fa41ac81d1004a84a70aea : SUCCESS in 15m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/625aed163ad24807932fe838d1168e85 : FAILURE in 26m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1bfc7b464d84045b307a2347274f0dc : FAILURE in 27m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98e21d1d77594da488d82cf4c62a123d : FAILURE in 32m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3c49fb4035457893350fa5280e7bb0 : SUCCESS in 11m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a56ab486b194694b78e205b4e55d49d : SUCCESS in 53m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4f86b5ca3e054ef0970d4f9a4c5ccc7f : SUCCESS in 1h 33m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f1080b790e624bbe8cbf088448ee85cb : SUCCESS in 1h 17m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c52348725b8a4ba58d83a1d864a292ee : SUCCESS in 58m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fcef86bcf91a40f081b0f0e3f67fabae : SUCCESS in 1h 06m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/75d95e82b6fd446a9064a3b64ab2cfc6 : SUCCESS in 1h 19m 03s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b852f5f00aab43bdb2c2438379603753 : SUCCESS in 1h 12m 31s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a40e5a3147db40e585464efb145bbe0d : SUCCESS in 1h 00m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/832e74f72c464971891d85774664bf22 : SUCCESS in 1h 15m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65c29ce21455b33419250dd0cd1680c374ae215f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 17:46:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6de05eefae1f471db2be1e4fdeb60af5 : FAILURE in 52m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e16368601d974a3b876e51324d4373b0 : FAILURE in 53m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c353833aea8ca13f5f5164c34bf56090efb8523","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-22 18:24:59.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/834740/1 : FAILURE in 2h 18m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ff7abd9942a79285b9f72b63ef840cc0aa2c2a2f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-22 19:37:29.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/40/834740/1/openstack-sriov-neutron-master-direct : FAILURE\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/40/834740/1/openstack-sriov-neutron-master-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7c0581d25cbba5ab6da851a5d39ee396efa070fe","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-03-22 20:27:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e205a7a9fe0d04feea88778114b9e2e35250cf4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-23 08:30:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0aa5f15b90083493df7b8856af4792902ad3b757","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-23 08:30:21.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"59a82581e70420fabe2754fce97c7395d279049e","revisions":{"59a82581e70420fabe2754fce97c7395d279049e":{"kind":"REWORK","_number":1,"created":"2022-03-22 16:05:46.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/40/834740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/834740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/834740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/834740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/834740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/834740/1"}}},"commit":{"parents":[{"commit":"dd55f1acd36dee4ab4ec806a694b5efcc8e53cb8","subject":"Merge \"Remove exception ``IpAddressAllocationNotFound``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dd55f1acd36dee4ab4ec806a694b5efcc8e53cb8"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-17 01:46:00.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-17 01:46:00.000000000","tz":0},"subject":"Remove unneeded method \"get_security_group_rules_count\"","message":"Remove unneeded method \"get_security_group_rules_count\"\n\nTrivial-Fix\n\nChange-Id: I569401fc91d67ce358c83183de3816c70471cc4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59a82581e70420fabe2754fce97c7395d279049e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59a82581e70420fabe2754fce97c7395d279049e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
