)]}'
{"id":"openstack%2Fneutron~834852","triplet_id":"openstack%2Fneutron~master~I0df376a80140ead7ff1fbf7f5ffef08a999dbe0b","project":"openstack/neutron","branch":"master","topic":"bug/1963899","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-03-25 09:43:18.000000000","reason":"Rodolfo Alonso replied on the change"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-04-01 11:50:39.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-01 11:50:39.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-04-01 10:17:26.000000000","reason":"Lajos Katona replied on the change"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-04-01 11:50:39.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-04-01 11:50:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0df376a80140ead7ff1fbf7f5ffef08a999dbe0b","subject":"Ensure that re_added ports are DOWN before set back to UP","status":"MERGED","created":"2022-03-23 11:30:57.000000000","updated":"2022-04-01 11:52:20.000000000","submitted":"2022-04-01 11:50:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"834852-bug/1963899","meta_rev_id":"439862a6a305a8e7f32c9634cd0c3203f01df19e","_number":834852,"virtual_id_number":834852,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-03-24 16:49:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-01 11:50:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-03-25 09:43:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-01 10:17:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-03-28 07:27:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-01 10:17:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-23 12:22:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-23 12:36:41.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 12:40:31.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2022-03-23 13:11:09.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-03-23 13:37:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-23 16:46:36.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-03-25 09:43:18.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-03-28 07:27:50.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-03-28 07:52:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-03-28 07:52:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-04-01 10:17:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"4d010bef1152e86d195f94919cbe1c08ce95588b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-23 11:30:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04bd5bda3d37103081471f3155cce4c22a309fde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 12:22:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5cda2f3614043a5ae00c47c7677664d : SUCCESS in 47m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/371293dec7a84f77af05491589c8c63d : SUCCESS in 49m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c5ba557fd8f2cf9b65a81e10387aeb2bb46526e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-23 12:36:41.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/52/834852/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9a1ca702ea1168dcee3aacdee0faaafed354c196","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-03-23 12:40:31.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"76e6fcc391687aa34c56ab0044591f2da515b76d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-23 13:11:09.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/834852/1 : FAILURE in 1h 39m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4fc08481b9f04a8bbfbea9d7507fa52b10bc3ed7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 13:37:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bd8627b96ae54781b36337f056279435 : SUCCESS in 53m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/96ee4364498447c8998b2e13591c492f : SUCCESS in 1h 59m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ed02d0d3659e4b168254fe7be142ed73 : SUCCESS in 1h 00m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04c17b3e0b324f4c9b21af9d09cfaae0 : SUCCESS in 42m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f3d83930c824fd9b9232c29d0f0b465 : SUCCESS in 41m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/516685c2ed3045558eaa3af3157f9f53 : SUCCESS in 17m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/680622e448cc4603897806628126a30a : SUCCESS in 27m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bc0624c9b1e4e7686e193d2d9e7374e : SUCCESS in 28m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a309e06d06ef486597fd21e330a109e5 : SUCCESS in 34m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d05fd58ec974eeaa35bbccd8f744e3a : SUCCESS in 11m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd93dc4412fb41d399a0a137a9ea5e44 : SUCCESS in 1h 04m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8aadcf746ba04f08ac9837a39fc3977b : SUCCESS in 1h 39m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/33b82f1d0baa4e979bb3b56df77af8fa : SUCCESS in 1h 15m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/207f2ffc4cdc473a827f0d79f35f8d60 : SUCCESS in 1h 02m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/98073fb6215c429ba7baf4a7033dcc3d : SUCCESS in 1h 01m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1f319074b5864e08b86cca4d2abd623f : SUCCESS in 1h 33m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e90493585a8e49f2a286ff4146b58211 : SUCCESS in 1h 20m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4a9f36d65e37ace91cf1dd160f1f733a1352b3da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-23 14:17:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"376abcd3b4ac757ff144808455d70eee95af2cf5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-23 14:17:15.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c4596c826e73008aeede92a841ff551bb4cc160d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 15:04:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffc4ee4af8d74708838be0c108ddc935 : SUCCESS in 44m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/713e6122fefd46b9a79e06685367f5a4 : SUCCESS in 44m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"05620d61565839c42bacfd20f8792d1b366d34ac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-23 15:54:47.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/52/834852/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"2e90a214b6485376cebfbd688f1315d23d661309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-23 15:59:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d0b958d4f247480da077c27ba341502e : SUCCESS in 45m 35s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/07612a49feab42bfbbdf2b9760c24a29 : SUCCESS in 1h 25m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1c2012ff84348d8874ab86f1ff08aa5 : SUCCESS in 1h 17m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52858c6d43c240a6a6a8926095b82f2e : SUCCESS in 36m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29c84f5eb2d2490c8baf450b4c621a76 : SUCCESS in 33m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce23c92d860b468096ac756ac72964e6 : SUCCESS in 12m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/846ffc97ccd349b2bff2b5ee7f14570d : SUCCESS in 27m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12fca7fb33424f808b9bcc8f188c0fe6 : SUCCESS in 23m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82305bfde94f4e62bcde283917024cad : SUCCESS in 35m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01ca550887b0433fa289525db77187be : SUCCESS in 11m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b3da10fca2dc4faea9dda7a34688f40a : FAILURE in 56m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8795bed3fff54da384a52fc8b6c924a4 : SUCCESS in 1h 36m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/60d2d3b5aa1e42ecaa114e1d25a7b7fb : SUCCESS in 1h 14m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb6a58e1aaec4cf6af08cb855bf71abc : SUCCESS in 1h 01m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/96992d0f0de449cb88ac6dde5f1d7531 : SUCCESS in 1h 07m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/16aa1ab785f64e20a0890bce2139c65e : SUCCESS in 1h 18m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0b69c2bfe5724b9f9f17a98054e731e6 : SUCCESS in 1h 11m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b1a4b977c0bac83acffff2ca5b3d9d14cc85a270","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-23 16:08:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/834852/2 : SUCCESS in 1h 50m 17s","accounts_in_message":[],"_revision_number":2},{"id":"7289004cf470766846d018015454de83bc242096","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-23 16:46:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2275421b3a26dc2fba83a5a79df49957c54c0ee8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-24 15:55:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"647ddcd8d9bff5d77cec522ca87d768ac1c7da32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-24 15:55:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cb380c9142bd2867c214033eaa4efe4217557540","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-24 16:49:20.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/52/834852/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"e81504f2c0013cc241310306ed52297927dd68ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 16:59:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/457e668b9f8c42e8b8d826876c760a60 : SUCCESS in 1h 01m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e465d6dfe934641925ea5ae3ba34db5 : SUCCESS in 1h 01m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54b97a5a12bd878d1d1c625ae7e902289d98255f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-24 17:30:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/834852/3 : SUCCESS in 1h 34m 28s","accounts_in_message":[],"_revision_number":3},{"id":"41c373f0ba4f82d65d8821ce7a57e03e0f7218ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 17:37:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/31195cf4ced844db937ec03dc8dd3c72 : SUCCESS in 46m 43s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/96200c27453b4878a1260421991f3314 : SUCCESS in 1h 05m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/34147185825544918309fb796d047d03 : SUCCESS in 50m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c9fce48848847809a09cf6f475da831 : SUCCESS in 41m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1b827fbae0b46f695ada6547d37fbb6 : SUCCESS in 31m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f699948c609d42a6856e2aa7a38dcbef : SUCCESS in 15m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fd445fa5e564475bb0d887070dd2a7b : SUCCESS in 32m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b29893fc22a4734a281db9f884dad17 : SUCCESS in 26m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/387ef1718e264935a6ce35217aed94ca : SUCCESS in 28m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/492a633c81db4acca09eef172fd2b107 : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c275279c901b40bb9baa0755a24d3d4f : SUCCESS in 55m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e535d5002194b88bd78413442d1a136 : SUCCESS in 1h 34m 48s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/74246a5d5c1747efa06209cbed386c1e : SUCCESS in 1h 18m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ebb6f0b878da4ed1a51f1af9a3d26dbb : SUCCESS in 57m 50s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/32596f65f8f041f78569934efee50670 : SUCCESS in 53m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/404fb6eb5ce8415a819d8e14977a59f2 : SUCCESS in 1h 20m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6b6931a3e5fb4c17964c445460874271 : SUCCESS in 1h 28m 11s","accounts_in_message":[],"_revision_number":3},{"id":"7e259289533e4c6f295686d2833b6ad2b83aadc9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-25 09:43:18.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"09e466e6ed20742acabaa22412705568a9210b3d","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-03-28 07:27:50.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f958cb8ab0eed4eb6edd49cffdac6a90357d2a6d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-28 07:52:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6116dc4c26570aeee896c8d455be2beb8635ef66","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-04-01 10:17:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6f4fec59c6e721fc29a8250e7d5931907781e09d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 10:18:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3db644c5837b292fb8c82eb3b4d17169068ed1e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:50:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/871fe26cbd9749de91194d3f9c04831b : SUCCESS in 44m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c319138d9a4c490783ab432ef74cc1f6 : SUCCESS in 1h 15m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3b255412af5049e19068bd62165c1788 : SUCCESS in 47m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea379f3cde864c2782ce6d8aa42c638d : SUCCESS in 32m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12538b9ec8c5431d96d01ffe45ee1906 : SUCCESS in 12m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/981b9d7f825f469cb817f1571c5fd397 : SUCCESS in 26m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cde207b636114764903a37adf9c0108e : SUCCESS in 25m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74c6e355b3b44c3ab034e02a0c6c004d : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e8e35751bbe74affad806e969bbfa077 : SUCCESS in 1h 04m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4628bbf9fb64f2f86a019e0649ebf50 : SUCCESS in 1h 28m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/837fbba536fa4c90a2e3fedc3442dec5 : SUCCESS in 1h 30m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b33a7a3ddb8b43f79a25efb3aae57c6d : SUCCESS in 1h 04m 30s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a07bfe8c402b4e21ae82790f26b59dc6 : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":3},{"id":"95023c6d99a2db77cc8e9399ba774dbb816ab2ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:50:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"439862a6a305a8e7f32c9634cd0c3203f01df19e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 11:52:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee3e167a775443f89dce091fea36972 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e7edcec2608fe1cb30d7458a882eb81e058eb76e","revisions":{"e76634abfd11decf3ea3ef97ffc1b59e2c4e7602":{"kind":"REWORK","_number":1,"created":"2022-03-23 11:30:57.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/834852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/834852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/834852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/834852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/834852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/834852/1"}}},"commit":{"parents":[{"commit":"b50183bc421214e7a360a6e70b27a77d7477468a","subject":"Merge \"Log request IDs for matched Nova external events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b50183bc421214e7a360a6e70b27a77d7477468a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-23 11:25:26.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-23 11:30:48.000000000","tz":60},"subject":"Ensure that re_added ports are DOWN before set back to UP","message":"Ensure that re_added ports are DOWN before set back to UP\n\nDuring e.g. rebuild of the server by Nova, ports plugged to such server\nare quickly removed and added again into br-int. In such case, ports are\nin the \"re_added\" ports set in the neutron-ovs-agent.\nBut it seems that in some cases it may happen that such port isn\u0027t\nswitched to be DOWN first and then, when neutron-ovs-agent treats port\nas added/updated and reports to the server that port is UP, there is no\nnotification to nova-compute send (because port\u0027s status was UP and new\nstatus is still UP in the Neutron DB).\nAs Nova waits for the notification from Neutron in such case server\ncould ends up in the ERROR state.\n\nTo avoid such issue, all ports which are treated as \"re_added\" by the\nneutron-ovs-agent are now first switched to be DOWN on the server side.\nThat way, when those ports are treated as added/updated in the same\nrpc_loop iteration, switching their status to UP will for sure trigger\nnotification to nova.\n\nCloses-Bug: #1963899\nChange-Id: I0df376a80140ead7ff1fbf7f5ffef08a999dbe0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e76634abfd11decf3ea3ef97ffc1b59e2c4e7602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e76634abfd11decf3ea3ef97ffc1b59e2c4e7602"}]},"branch":"refs/heads/master"},"95c9db9da66a5b1f6b9eca326506677088e6a45e":{"kind":"REWORK","_number":2,"created":"2022-03-23 14:17:02.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/834852/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/834852/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/834852/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/834852/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/834852/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/834852/2"}}},"commit":{"parents":[{"commit":"b50183bc421214e7a360a6e70b27a77d7477468a","subject":"Merge \"Log request IDs for matched Nova external events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b50183bc421214e7a360a6e70b27a77d7477468a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-23 11:25:26.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-23 14:10:30.000000000","tz":60},"subject":"Ensure that re_added ports are DOWN before set back to UP","message":"Ensure that re_added ports are DOWN before set back to UP\n\nDuring e.g. rebuild of the server by Nova, ports plugged to such server\nare quickly removed and added again into br-int. In such case, ports are\nin the \"re_added\" ports set in the neutron-ovs-agent.\nBut it seems that in some cases it may happen that such port isn\u0027t\nswitched to be DOWN first and then, when neutron-ovs-agent treats port\nas added/updated and reports to the server that port is UP, there is no\nnotification to nova-compute send (because port\u0027s status was UP and new\nstatus is still UP in the Neutron DB).\nAs Nova waits for the notification from Neutron in such case server\ncould ends up in the ERROR state.\n\nTo avoid such issue, all ports which are treated as \"re_added\" by the\nneutron-ovs-agent are now first switched to be DOWN on the server side.\nThat way, when those ports are treated as added/updated in the same\nrpc_loop iteration, switching their status to UP will for sure trigger\nnotification to nova.\n\nCloses-Bug: #1963899\nChange-Id: I0df376a80140ead7ff1fbf7f5ffef08a999dbe0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95c9db9da66a5b1f6b9eca326506677088e6a45e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95c9db9da66a5b1f6b9eca326506677088e6a45e"}]},"branch":"refs/heads/master"},"e7edcec2608fe1cb30d7458a882eb81e058eb76e":{"kind":"REWORK","_number":3,"created":"2022-03-24 15:55:20.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/52/834852/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/834852/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/834852/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/834852/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/834852/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/834852/3"}}},"commit":{"parents":[{"commit":"b50183bc421214e7a360a6e70b27a77d7477468a","subject":"Merge \"Log request IDs for matched Nova external events\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b50183bc421214e7a360a6e70b27a77d7477468a"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-23 11:25:26.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-24 15:54:26.000000000","tz":60},"subject":"Ensure that re_added ports are DOWN before set back to UP","message":"Ensure that re_added ports are DOWN before set back to UP\n\nDuring e.g. rebuild of the server by Nova, ports plugged to such server\nare quickly removed and added again into br-int. In such case, ports are\nin the \"re_added\" ports set in the neutron-ovs-agent.\nBut it seems that in some cases it may happen that such port isn\u0027t\nswitched to be DOWN first and then, when neutron-ovs-agent treats port\nas added/updated and reports to the server that port is UP, there is no\nnotification to nova-compute send (because port\u0027s status was UP and new\nstatus is still UP in the Neutron DB).\nAs Nova waits for the notification from Neutron in such case server\ncould ends up in the ERROR state.\n\nTo avoid such issue, all ports which are treated as \"re_added\" by the\nneutron-ovs-agent are now first switched to be DOWN on the server side.\nThat way, when those ports are treated as added/updated in the same\nrpc_loop iteration, switching their status to UP will for sure trigger\nnotification to nova.\n\nCloses-Bug: #1963899\nChange-Id: I0df376a80140ead7ff1fbf7f5ffef08a999dbe0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7edcec2608fe1cb30d7458a882eb81e058eb76e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e7edcec2608fe1cb30d7458a882eb81e058eb76e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
