)]}'
{"id":"openstack%2Fneutron~835014","triplet_id":"openstack%2Fneutron~master~I0df9e9e6ef927ada4d8047c810774c2632f003bb","project":"openstack/neutron","branch":"master","topic":"sgl_drop_per_sg","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-13 10:18:49.000000000","reason":"Change was abandoned"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2022-07-26 10:20:08.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I0df9e9e6ef927ada4d8047c810774c2632f003bb","subject":"[OVN] Log drop events per security group","status":"ABANDONED","created":"2022-03-24 09:08:39.000000000","updated":"2023-09-13 10:18:49.000000000","total_comment_count":14,"unresolved_comment_count":4,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"0bfd85de2fc8a14b73409cc5dd0c58605288b69d","_number":835014,"virtual_id_number":835014,"owner":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-26 12:01:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-03-24 09:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-03-24 09:57:16.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-24 10:50:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-24 10:56:50.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-03-24 10:57:43.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-03 16:53:20.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-05-10 10:03:42.000000000","updated_by":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"CC"},{"updated":"2022-07-12 10:05:53.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"8468ea01ad417db6cbb9bf2bf00da620673aeb69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-03-24 09:08:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9114be1de52672a524f94d4cce4a252971c019a7","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-03-24 09:09:05.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"54cf77d03076a814c8939ed25ea95fca4d00e8de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 09:57:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49bfc752fc174edd8d9ba409f821c015 : FAILURE in 44m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab3ce193e3d3420aaabf4033a736719c : FAILURE in 46m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a7a790a30007cd3e2384827a3f3122a177aea6a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-03-24 09:57:16.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/March-2022/Arista_Tempest_Testrefs/changes/14/835014/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"d6d6bca2f253e579f8b2dcc7b525417bfa9898ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-24 10:50:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d3839fb24ac0401fa57a81a3df02fd89 : SUCCESS in 42m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/7db07b1653e34dec955c30c968bed675 : SUCCESS in 52m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a40dd9ca01d34ffe910f111d94a92ac0 : SUCCESS in 42m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7578dcdb06fa435f9312aa11cfdc59af : FAILURE in 28m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f19c71c2f9b421093585028ebe46af7 : FAILURE in 27m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2ec70dd6e4c485a9e6587fe718bcba9 : SUCCESS in 13m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb8a68c0cff64b348a3074067426fc1f : FAILURE in 22m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11e6bacea6904effa7e0568aa65fce5f : FAILURE in 22m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fd5e53b68fb4fbd9c82e0ecd6c03cec : FAILURE in 33m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e7a61d00424b6c909a22b4ec6eea37 : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fd605ebf4f0b48f397e5570c1b072382 : FAILURE in 51m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/247a598bf51d4c96aac985715f20b348 : SUCCESS in 1h 36m 22s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aca0480b919a42d088c281a82265008a : SUCCESS in 48m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/90878b691f534e2b89849faff6617336 : SUCCESS in 1h 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a4b4c323f5745936f59d420dbeaef8e5f067f54","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-24 10:56:50.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/835014/1 : FAILURE in 1h 47m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a00319442b8f86b8bcbbdd8511c01d2f9b56c165","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-24 10:57:43.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/14/835014/1/openstack-sriov-neutron-master-direct : ABORTED\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/14/835014/1/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"3cac5aef337d45f8fdc9e857f036748b7751c8bc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-03-24 12:01:52.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/14/835014/1/openstack-sriov-neutron-master-macvtap : FAILURE\n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/14/835014/1/openstack-sriov-neutron-master-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ce184f9a299722811a55ac767a0a973d94938fe6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-05-03 15:50:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17e754b2e5609d3a31d90c26383503e3bee11460","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-05-03 16:16:22.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c9fcc9c3062ee22fde5897625df43d6df0c47fb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 16:54:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/965b2967c348432e9c64f303a432ba20 : FAILURE in 49m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ec32ca4510641689506f29b6e484896 : FAILURE in 45m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ce51cca42a842a15749f986c1d24aea65c1f2bf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:17:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bf739198df9e4838b4ab47a276b3dc3a : SUCCESS in 35m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a9b994177c0240d9952fa7058f6d5a9a : SUCCESS in 44m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fefda98767c541fc859ff79a1971dded : SUCCESS in 27m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fad0367eb9d40b6918e7b98b5098d90 : FAILURE in 39m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68faf294cf71407ab4e5e0e55f689bb3 : SUCCESS in 16m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11793588d04442bc8a00d3a8cf8847db : FAILURE in 25m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96f2a518513245bcafe319a8e586c60d : FAILURE in 27m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6399a004ec34f1f962d412cc8d55bce : FAILURE in 27m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f3a5ccc3324de48c2e465de06a2fbd : SUCCESS in 9m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c0c7b102f8c402d963c99aaac218e0c : FAILURE in 53m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f03d4a419878469e9de0fde9279888af : SUCCESS in 1h 24m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02e508942d1a4e9daa90235509059ccb : SUCCESS in 59m 28s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/726a9a6119c0403088a6db8d40347a5a : SUCCESS in 1h 17m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be45b65a49b53a00083fe1da5e858a10e0cd36e8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-03 17:26:29.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/835014/2 : FAILURE in 1h 35m 22s","accounts_in_message":[],"_revision_number":2},{"id":"fa41d38bbb2073c2babce3513d6b522c779c013c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-03 18:00:53.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/14/835014/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"04b63857f959954b26ecc48cce1a7a4615a0b565","author":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"date":"2022-05-10 10:03:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4e8aeff71033051d927daa4b16a013eab6a9faa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-05-12 12:53:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b580cbdc9da1ef384a1379f8d23629f7d944bb34","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-05-12 12:53:46.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"79691b548d2ce2b25bd6e7c4875a88b779ca0740","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-12 13:51:02.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/14/835014/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"bed4ff7aadb3694df45c68783a7fb762fb5611f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 14:08:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2727a8c4bbdf4895afe9506dba988613 : SUCCESS in 49m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f7ca0482a9a4c9983e09928debca88c : SUCCESS in 43m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"baf250998659feed726f90718ac78c8081ef71ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 14:25:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0afa26fa10cd4144abb75ba156ed12c0 : SUCCESS in 54m 25s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a47e4237b9124a93b2996ca376485ced : SUCCESS in 1h 10m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7620af90a0b743e4a523a7f3f02c9bb3 : SUCCESS in 32m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39e9fe8ddc0e48068d944e6fadb5d920 : SUCCESS in 34m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e502bed5fb04b7eba45a77e2711e237 : TIMED_OUT in 30m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f3884e64b1941c8be18f6b37e6dda91 : SUCCESS in 29m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/509394874ccf43dfb0c0a4f7029e8137 : SUCCESS in 25m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/044830e373a8420db18522834c353c3c : SUCCESS in 28m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c25e77f2a214f9894bdc61d9f82d8aa : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/79e71e25fd744ac0a08aa03527dead3f : FAILURE in 53m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/16b75d6bb96e42f0ae9f76a5ca53dd78 : FAILURE in 1h 24m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd383eba4566413a99e818c3c2fd036b : SUCCESS in 1h 05m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a74dcbd2f20f4718900aed3cf948173a : SUCCESS in 1h 02m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b988dbdfda02fc8fd5a094a1aef24375a9c46c7f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-12 14:32:53.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/835014/3 : SUCCESS in 1h 38m 42s","accounts_in_message":[],"_revision_number":3},{"id":"25a998922a66c2698b920b0c13c94178c7033bcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-05-13 16:14:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a7cf5006f68649ad322e45bf7ec49ca5ef9be5a5","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-05-13 16:14:19.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"0a888191d4469796ee2060d36caa64baaf2fa9bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 16:56:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8733129ad734b1fb827813a3a508fe7 : SUCCESS in 39m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0492dcde90d4288a758afdb15f68955 : SUCCESS in 39m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"75affbc7a3d2c5a3279f7332addc28d13525964d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-13 17:38:57.000000000","message":"Patch Set 4:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/14/835014/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"8fb0ef8dc2b79a10bf8b49395a8120a3826af168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 17:49:14.000000000","message":"Patch Set 4: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4ce026d0c38e4617a4716baec2f546cb : SUCCESS in 38m 29s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/229506e00be74ebc8cb1157ac491437f : SUCCESS in 1h 05m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/07133f4ebd9240f1a14a0e096c951d81 : SUCCESS in 34m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cf9ba833ef14501a9aa5de50e5930f8 : SUCCESS in 36m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fde68f7adc14add9b9dd94114d6c174 : FAILURE in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e0a5a568cf544afba6026d588f08004 : SUCCESS in 28m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/877377114fdf49afa360c34d2745d0d5 : SUCCESS in 21m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11f7720375834fe09b427b7ece4d1b2b : SUCCESS in 20m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8fb5e9f10046ff931fad88849ee4d0 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f2549f1373a4c559a7eef14e97bfd27 : FAILURE in 46m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/418524834c494d458378523b3bdbd215 : SUCCESS in 1h 33m 43s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f82ab6b459145c99cf085c1dd1c3935 : SUCCESS in 52m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2a8f3ee813d14591adc6fb867dd51c15 : SUCCESS in 1h 09m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"05a877a67cf0508c4055d70bc56b25c9917be204","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-13 18:21:22.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/835014/4 : FAILURE in 2h 06m 40s","accounts_in_message":[],"_revision_number":4},{"id":"102ca247030892f16ccae55cfa7de1cb4f1c1cd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-05-19 17:18:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8aff954d6874c79713280d53c333e8c5cf76c8dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 18:14:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b6c81267bd84030b0da942e99227cf8 : SUCCESS in 40m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6375a2f0fa3442ec828e83a392217fa3 : SUCCESS in 54m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d04e9fe4f51e95e842bd0e5ff8bc77cb535398e8","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-19 18:46:53.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/14/835014/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"5e691b8bf672123173df103b378d21e8db9aff68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-19 19:01:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/da4788b475e8406fba7492df3bfa6497 : SUCCESS in 42m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/58c430b4f22741659edd9a26d8ccf915 : SUCCESS in 1h 17m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ffed920afdd4c259a367a55bcf275da : SUCCESS in 38m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cb7634f3e0a483088c5cfbd71960284 : SUCCESS in 33m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbfa40b7002d4c5b98df019d8c98b7ba : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e86f5d7f8ab94cfa8a099a9bc6cab703 : SUCCESS in 25m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fed4631568048eca5f55a127650d788 : SUCCESS in 28m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e0ec9a037e94dc48128e05e71079925 : SUCCESS in 24m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc608dd67b648b889c0a769600f38a6 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/60d1fcabcb2049dda6e2bc378ff41068 : FAILURE in 46m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/89eb6f416ddb4082b08e297054106fac : SUCCESS in 1h 36m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/163905dfde9548548b3604c1eb022028 : SUCCESS in 53m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/feba3b5c43034a08a993d0ac9da3bf41 : SUCCESS in 1h 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"650528489c6462a6c97d88f10199a1d03e493896","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-19 19:35:22.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/835014/5 : SUCCESS in 2h 15m 40s","accounts_in_message":[],"_revision_number":5},{"id":"db8d87d442dd545aa145f86f933a28d8f089c8f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-05-24 09:32:04.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1f496552102a3ab695270d423bdcfe194773ce16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 10:15:44.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89bc12cb3b8c45589cbbd8dd55a2ebef : SUCCESS in 40m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8eabd27ff9d4c998d159e38a5ce4abf : SUCCESS in 41m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"565adcad7b22195895e163f1c67d960b3e30029d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-24 11:09:21.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/835014/6 : SUCCESS in 1h 36m 34s","accounts_in_message":[],"_revision_number":6},{"id":"4b210aa26ca22c3bf7b26ac39fdee6137f31da7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-24 11:20:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bcc4da36d04a44f6b496bc40a4840ff6 : SUCCESS in 40m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6ca68bb5ac5940bc9f756d7a3c9ad15a : SUCCESS in 49m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/316b6a3baa1e43ad8c86f40c37e55cde : SUCCESS in 34m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa5151e5249c4797b02c0a6bfb5252ea : SUCCESS in 29m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b519902bfb4aea9588b4f8cf9a33cb : SUCCESS in 11m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a18ceeb14a2c4e349fac4cbf25b4d450 : SUCCESS in 20m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f08e61711d314b788adc44d1786f7d28 : SUCCESS in 21m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98f85b58259a4f6b9ebeee5264f647d4 : SUCCESS in 27m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a52cf683174fe39699b62c125ab2f3 : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7aa9cbc317184a80bbc4163291b5f9ab : SUCCESS in 47m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e1b3e5451494c7b89b87d2974d90cda : SUCCESS in 1h 41m 41s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12e9d269640b48a48268a1d236f1ffec : SUCCESS in 1h 04m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2a8f1a93e769438a8171680048ce3356 : SUCCESS in 1h 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"af06581a6a2037691e51148b25f8f02d27f81dc2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-24 14:14:24.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/14/835014/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"c5a536d6ce3bf0dedad68a41b6df3ea9447163bb","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-12 10:05:53.000000000","message":"Patch Set 6: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"221125aa97cd7f780c98377dc70123539bffe03c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-07-26 10:20:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4b1788458c906fde8c3b18bec82664c8db130e57","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-07-26 10:21:31.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e2795a916dbd06d222ff87da1d78ba6b18b4f2b9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-26 10:45:37.000000000","message":"Patch Set 7:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/14/835014/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"ed566c2bd47bbca0e41b6f8eb0b68736c8891889","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 11:01:27.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/352db389ad91440ea740a1fb5321db9a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a764080667584419ab93c5e1dccf2e09 : SUCCESS in 39m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea1c897dc3bc40589b4305de112e178b : SUCCESS in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a4249d2ed5d5853253db77e4cc085c2e32a6968e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-26 11:59:11.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/835014/7 : SUCCESS in 1h 38m 25s","accounts_in_message":[],"_revision_number":7},{"id":"94c038fd304e9dd771ebd67a203bcbe5a9ec28e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 12:01:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b6771e8624a4668b2ff54cbd1ae4326\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/435e63b0fd564487aa97a8f49f2487fa : SUCCESS in 1h 39m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/14a61b4caceb4e56822be453171bc99f : SUCCESS in 31m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3de3cafb2bff477c8ba84a0e6561a037 : SUCCESS in 58m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5855cd58fd432aadd75c190e979b87 : SUCCESS in 14m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e99324461a2a4937b868b96d2c959e91 : SUCCESS in 26m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9f1786b7b79409d989db6dd198c4370 : SUCCESS in 25m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca4b6ab3ed874b84baec4ca9da9c675b : SUCCESS in 24m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2b994cb835446d88a88dd48173f73c8 : SUCCESS in 9m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d40bc33715cd4c0e9b24267df5ba9f17 : SUCCESS in 53m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/24260ac4bfaa44478eca2c3be88f9645 : SUCCESS in 1h 24m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3bd9870456740b590ce7c44d5e85bf4 : SUCCESS in 55m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/77f861442b3847c8b841db7a8c2b3f82 : SUCCESS in 1h 09m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0bfd85de2fc8a14b73409cc5dd0c58605288b69d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2023-09-13 10:18:49.000000000","message":"Abandoned\n\nDropped for now","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f928b80853bdffab5a4e98128858352a270132ce","revisions":{"762de41f65c7344fce261aa7b34329e042e4eee8":{"kind":"REWORK","_number":1,"created":"2022-03-24 09:08:39.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/14/835014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/835014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/835014/1"}}},"commit":{"parents":[{"commit":"9c9dfb1ca6c58398a9c216a853a7594e3be4748f","subject":"Merge \"[FT] Run all .*MySQL tests in an isolated thread\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c9dfb1ca6c58398a9c216a853a7594e3be4748f"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-03-24 09:04:12.000000000","tz":60},"committer":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-03-24 09:04:12.000000000","tz":60},"subject":"[DNM][OVN] Log drop events per security group","message":"[DNM][OVN] Log drop events per security group\n\nRight now dropped traffic is logged for all security groups\nbecause there\u0027s only 1 general ACL for all of the port groups.\nThis patch creates extra ACLs for each SG that are associated to their\nport group and collect all of the dropped traffic for them.\n\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I0df9e9e6ef927ada4d8047c810774c2632f003bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/762de41f65c7344fce261aa7b34329e042e4eee8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/762de41f65c7344fce261aa7b34329e042e4eee8"}]},"branch":"refs/heads/master"},"29d2252b444e254cad183d8aa2420425eab92a48":{"kind":"REWORK","_number":2,"created":"2022-05-03 15:50:30.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/14/835014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/835014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/835014/2"}}},"commit":{"parents":[{"commit":"bc8e512b4a7b68f2e5c31efd320d8b347d06bba0","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc8e512b4a7b68f2e5c31efd320d8b347d06bba0"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-03-24 09:04:12.000000000","tz":60},"committer":{"name":"elvira","email":"egarciar@redhat.com","date":"2022-05-03 15:50:09.000000000","tz":0},"subject":"[OVN] Log drop events per security group","message":"[OVN] Log drop events per security group\n\nRight now dropped traffic is logged for all security groups\nbecause there\u0027s only 1 general ACL for all of the port groups.\nThis patch creates extra ACLs for each SG that are associated to their\nport group and collect all of their dropped traffic.\n\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I0df9e9e6ef927ada4d8047c810774c2632f003bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29d2252b444e254cad183d8aa2420425eab92a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29d2252b444e254cad183d8aa2420425eab92a48"}]},"branch":"refs/heads/master"},"fd2f7ef8d9aaa14a7c87d0a3a3cb7abc9ebe3a60":{"kind":"REWORK","_number":3,"created":"2022-05-12 12:53:34.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/14/835014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/835014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/835014/3"}}},"commit":{"parents":[{"commit":"cbb4c2274a19af436651ecb9c36096ed541c1cb8","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbb4c2274a19af436651ecb9c36096ed541c1cb8"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-03-24 09:04:12.000000000","tz":60},"committer":{"name":"elvira","email":"egarciar@redhat.com","date":"2022-05-12 12:47:53.000000000","tz":0},"subject":"[OVN] Log drop events per security group","message":"[OVN] Log drop events per security group\n\nRight now dropped traffic is logged for all security groups\nbecause there\u0027s only 1 general ACL for all of the port groups.\nThis patch creates extra ACLs for each SG that are associated to their\nport group and collect all of their dropped traffic.\n\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I0df9e9e6ef927ada4d8047c810774c2632f003bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd2f7ef8d9aaa14a7c87d0a3a3cb7abc9ebe3a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd2f7ef8d9aaa14a7c87d0a3a3cb7abc9ebe3a60"}]},"branch":"refs/heads/master"},"6ea590ac4c348ec0afb2ba30fff2c113ec6b3ce7":{"kind":"REWORK","_number":4,"created":"2022-05-13 16:14:05.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/14/835014/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/835014/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/835014/4"}}},"commit":{"parents":[{"commit":"0f3f94d05a9a610d684beba5b369ae4de0f58602","subject":"Merge \"Update python testing as per zed cycle teting runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f3f94d05a9a610d684beba5b369ae4de0f58602"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-03-24 09:04:12.000000000","tz":60},"committer":{"name":"elvira","email":"egarciar@redhat.com","date":"2022-05-13 16:13:47.000000000","tz":0},"subject":"[OVN] Log drop events per security group","message":"[OVN] Log drop events per security group\n\nRight now dropped traffic is logged for all security groups\nbecause there\u0027s only 1 general ACL for all of the port groups.\nThis patch creates extra ACLs for each SG that are associated to their\nport group and collect all of their dropped traffic.\n\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I0df9e9e6ef927ada4d8047c810774c2632f003bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ea590ac4c348ec0afb2ba30fff2c113ec6b3ce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ea590ac4c348ec0afb2ba30fff2c113ec6b3ce7"}]},"branch":"refs/heads/master"},"167d8569c7a77f04e5c10bb4f6475f057c544fa6":{"kind":"REWORK","_number":5,"created":"2022-05-19 17:18:43.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/14/835014/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/835014/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/835014/5"}}},"commit":{"parents":[{"commit":"5c6eff7e15d7b5bf7f0f175293adcec5d953fd6b","subject":"Merge \"Update OVN gaps document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c6eff7e15d7b5bf7f0f175293adcec5d953fd6b"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-03-24 09:04:12.000000000","tz":60},"committer":{"name":"elvira","email":"egarciar@redhat.com","date":"2022-05-19 17:06:49.000000000","tz":0},"subject":"[OVN] Log drop events per security group","message":"[OVN] Log drop events per security group\n\nRight now dropped traffic is logged for all security groups\nbecause there\u0027s only 1 general ACL for all of the port groups.\nThis patch creates extra ACLs for each SG that are associated to their\nport group and collect all of their dropped traffic.\n\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I0df9e9e6ef927ada4d8047c810774c2632f003bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/167d8569c7a77f04e5c10bb4f6475f057c544fa6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/167d8569c7a77f04e5c10bb4f6475f057c544fa6"}]},"branch":"refs/heads/master"},"71e9db7373fb33d80f5bacf7b97f39a6ae12a1f8":{"kind":"REWORK","_number":6,"created":"2022-05-24 09:32:04.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/14/835014/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/835014/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/835014/6"}}},"commit":{"parents":[{"commit":"5c6eff7e15d7b5bf7f0f175293adcec5d953fd6b","subject":"Merge \"Update OVN gaps document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5c6eff7e15d7b5bf7f0f175293adcec5d953fd6b"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-03-24 09:04:12.000000000","tz":60},"committer":{"name":"elvira","email":"egarciar@redhat.com","date":"2022-05-24 09:31:03.000000000","tz":0},"subject":"[OVN] Log drop events per security group","message":"[OVN] Log drop events per security group\n\nRight now dropped traffic is logged for all security groups\nbecause there\u0027s only 1 general ACL for all of the port groups.\nThis patch creates extra ACLs for each SG that are associated to their\nport group and collect all of their dropped traffic.\n\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I0df9e9e6ef927ada4d8047c810774c2632f003bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71e9db7373fb33d80f5bacf7b97f39a6ae12a1f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/71e9db7373fb33d80f5bacf7b97f39a6ae12a1f8"}]},"branch":"refs/heads/master"},"f928b80853bdffab5a4e98128858352a270132ce":{"kind":"REWORK","_number":7,"created":"2022-07-26 10:20:08.000000000","uploader":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"ref":"refs/changes/14/835014/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/835014/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/835014/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/835014/7"}}},"commit":{"parents":[{"commit":"fb1c84c23a95ed7c044716b77e753234fda8032c","subject":"Merge \"QoS minimum bandwidth documentation: OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb1c84c23a95ed7c044716b77e753234fda8032c"}]}],"author":{"name":"Elvira García","email":"egarciar@redhat.com","date":"2022-03-24 09:04:12.000000000","tz":60},"committer":{"name":"elvira","email":"egarciar@redhat.com","date":"2022-07-26 09:43:05.000000000","tz":0},"subject":"[OVN] Log drop events per security group","message":"[OVN] Log drop events per security group\n\nRight now dropped traffic is logged for all security groups\nbecause there\u0027s only 1 general ACL for all of the port groups.\nThis patch creates extra ACLs for each SG that are associated to their\nport group and collect all of their dropped traffic.\n\nSigned-off-by: Elvira García \u003cegarciar@redhat.com\u003e\nChange-Id: I0df9e9e6ef927ada4d8047c810774c2632f003bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f928b80853bdffab5a4e98128858352a270132ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f928b80853bdffab5a4e98128858352a270132ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
