)]}'
{"id":"openstack%2Fneutron~835340","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Id050ad6d95e2c4f3a6eb61a4cd88c9ce6e6dad1a","project":"openstack/neutron","branch":"stable/wallaby","topic":"wallaby","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-05-30 07:32:05.000000000","reason":"Change was abandoned"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-05-30 07:32:05.000000000","reason":"Change was abandoned"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-05-30 07:32:05.000000000","reason":"Change was abandoned"},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2022-05-30 07:32:05.000000000","reason":"Change was abandoned"},"23567":{"account":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"last_update":"2022-05-30 07:32:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id050ad6d95e2c4f3a6eb61a4cd88c9ce6e6dad1a","subject":"Revert \"Use Port_Binding up column to set Neutron port status\"","status":"ABANDONED","created":"2022-03-28 11:07:43.000000000","updated":"2022-05-30 07:32:05.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":823269,"meta_rev_id":"f09230af4516febe38ea86b6ae070ed9bbf93331","_number":835340,"virtual_id_number":835340,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-13 15:31:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2022-05-13 14:01:32.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-28 11:07:43.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-03-28 11:07:43.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-03-28 11:07:43.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-03-28 11:07:43.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":23567,"name":"Luis Tomas Bolivar","email":"ltomasbo@redhat.com","username":"ltomasbo"},"state":"REVIEWER"},{"updated":"2022-03-28 11:07:43.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-28 11:08:10.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"d58f98026817963dcdacee704d56d0e8ad785138","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-28 11:07:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5338ab6646969e521067d878536bf4ad361b2cab","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-03-28 11:08:10.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5b316c4f2dea1f14f1f738673757b5ec95bfa859","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 11:08:15.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 835340,1","accounts_in_message":[],"_revision_number":1},{"id":"a90585d94ee571581267780d6c2d0febc629ded9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 11:08:28.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 835340,1","accounts_in_message":[],"_revision_number":1},{"id":"e8c619837d8390e55b264b6f4d44dfb6780fe26a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-28 11:10:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d4b34263c05cbdb323413c85fb483214f29d5989","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-03-28 11:10:59.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8968bf05dd1546ed6228ae16c385b940c06ff56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 12:00:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca42cbebb6764137adfa62e75a2b8fe1 : SUCCESS in 44m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d49e8bccde424d99909db93dd177ab6e : SUCCESS in 47m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65d0cd9b5b85b94063ccff477e1f336117add331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-28 12:35:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/6eafa0f925364ce49cb4b1edec8ac3c9 : SUCCESS in 46m 00s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/e8307f77a94249baa0ab6988b1e5d182 : SUCCESS in 1h 01m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4526e7aa29e442a1912a05f212e85a3e : SUCCESS in 31m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d2e7b80004543f786d45ca5c154f5cc : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34c05e69318247e6b4744b38db49a245 : SUCCESS in 26m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72b38c57dbb04d7b8d624809a749fb4d : SUCCESS in 25m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c1e1dd60aa14b8db84f539334c42609 : SUCCESS in 26m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58fa6e1626ff4e7aaa0bd470cd56cda6 : SUCCESS in 11m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5f6a8fcad5e74f379b5a3dd8cdb9ae44 : SUCCESS in 45m 46s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ecf10d0de97b4947950a3a80aea584f0 : SUCCESS in 1h 02m 25s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/faf78f32b9a9408aa22606ad310f6e18 : SUCCESS in 35m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3faeaf67e4a14a978d4a886fb91b9817 : SUCCESS in 1h 22m 29s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3395bd4610ec45dd8ee19fc80a862bb4 : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":2},{"id":"7e7b44e9f1c13d3ac8980af74b78e740bd1a45e3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-03-28 12:47:49.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/835340/2 : SUCCESS in 1h 36m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0a5ee89efe4215d87a641be3c84cf54e7264d372","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-05-13 14:00:30.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f5e80dd68414e612a83170a0a456a913546ed126","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-05-13 14:01:32.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"247990a8273f92e446ea5d0da5f51f1d36040eb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 14:51:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9dce720abfec4340a8f4358db8fcd65b : SUCCESS in 48m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b45639be300e4ec0b3682cd4b0009bd8 : SUCCESS in 43m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aeead0cc647343cbb56a0fc03d0a0af00dadfa6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 15:31:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/f3882366d95847299caa8602b3120f66 : SUCCESS in 50m 57s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/28c7a6dacfa54720b29f7b8b84e864e1 : SUCCESS in 56m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d5bc52eef8146459c5fa08c595e81f0 : SUCCESS in 36m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7fa2b9a567a4e21b46018ce0c647f34 : SUCCESS in 14m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e5ac210e5374c778553f3c16f4563d3 : SUCCESS in 33m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de356727c5b248c3a6db1380ad0980fa : SUCCESS in 30m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04ed5d4887014ee385b90709c358ecb0 : SUCCESS in 24m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd113fdb2b54479280739d3fa90a5054 : SUCCESS in 9m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0fb54876f5f5406681a60d9b37920ce2 : SUCCESS in 43m 48s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/b38dff6033434b8ca330a3aec66efa8c : SUCCESS in 57m 55s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/484d9b7477e44322ad399637fb8be98d : SUCCESS in 29m 56s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d268d2bcd2f745308706fadf240a6c9b : SUCCESS in 1h 28m 46s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/91aa06a9766a42ff8224a243611d3bcc : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f16a15c42f869d9cd84fe9e1c930bc0590246016","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-13 16:11:07.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/835340/3 : FAILURE in 2h 09m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f09230af4516febe38ea86b6ae070ed9bbf93331","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-05-30 07:32:05.000000000","message":"Abandoned\n\nBeing reverted with https://review.opendev.org/c/openstack/neutron/+/843763 with more context","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7ebc7275493dfa8d7faa542788bd3c070e1f1a33","revisions":{"aeda014bf5c6bda514548c49632dd9f9e559294e":{"kind":"REWORK","_number":1,"created":"2022-03-28 11:07:43.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/40/835340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/835340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/835340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/835340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/835340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/835340/1"}}},"commit":{"parents":[{"commit":"f2ddb6741ead160773076f1092b8bd1487736eae","subject":"Use Port_Binding up column to set Neutron port status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f2ddb6741ead160773076f1092b8bd1487736eae"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-03-28 11:07:43.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-03-28 11:07:43.000000000","tz":0},"subject":"Revert \"Use Port_Binding up column to set Neutron port status\"","message":"Revert \"Use Port_Binding up column to set Neutron port status\"\n\nThis reverts commit f2ddb6741ead160773076f1092b8bd1487736eae.\n\nReason for revert: https://bugs.launchpad.net/tripleo/+bug/1964940\n\nChange-Id: Id050ad6d95e2c4f3a6eb61a4cd88c9ce6e6dad1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeda014bf5c6bda514548c49632dd9f9e559294e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aeda014bf5c6bda514548c49632dd9f9e559294e"}]},"branch":"refs/heads/stable/wallaby"},"ecca14e9f181350147005c0dc17b07ccb97ea78c":{"kind":"REWORK","_number":2,"created":"2022-03-28 11:10:31.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/40/835340/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/835340/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/835340/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/835340/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/835340/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/835340/2"}}},"commit":{"parents":[{"commit":"b053dfe3844aaf024668847aab3498c3af559dc0","subject":"Remove exception ``IpAddressAllocationNotFound``","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b053dfe3844aaf024668847aab3498c3af559dc0"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-03-28 11:07:43.000000000","tz":0},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-03-28 11:10:07.000000000","tz":330},"subject":"Revert \"Use Port_Binding up column to set Neutron port status\"","message":"Revert \"Use Port_Binding up column to set Neutron port status\"\n\nThis reverts commit f2ddb6741ead160773076f1092b8bd1487736eae.\n\nReason for revert: https://bugs.launchpad.net/tripleo/+bug/1964940\n\nChange-Id: Id050ad6d95e2c4f3a6eb61a4cd88c9ce6e6dad1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecca14e9f181350147005c0dc17b07ccb97ea78c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ecca14e9f181350147005c0dc17b07ccb97ea78c"}]},"branch":"refs/heads/stable/wallaby"},"7ebc7275493dfa8d7faa542788bd3c070e1f1a33":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-13 14:00:30.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/40/835340/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/835340/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/835340/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/835340/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/835340/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/835340/3"}}},"commit":{"parents":[{"commit":"751c854dae4990543b2e33531aaa94a051ef68cf","subject":"Merge \"Add extra debug log with current and new port statuses in ML2 plugin\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/751c854dae4990543b2e33531aaa94a051ef68cf"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-03-28 11:07:43.000000000","tz":0},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-05-13 14:00:30.000000000","tz":0},"subject":"Revert \"Use Port_Binding up column to set Neutron port status\"","message":"Revert \"Use Port_Binding up column to set Neutron port status\"\n\nThis reverts commit f2ddb6741ead160773076f1092b8bd1487736eae.\n\nReason for revert: https://bugs.launchpad.net/tripleo/+bug/1964940\n\nChange-Id: Id050ad6d95e2c4f3a6eb61a4cd88c9ce6e6dad1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ebc7275493dfa8d7faa542788bd3c070e1f1a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ebc7275493dfa8d7faa542788bd3c070e1f1a33"}]},"branch":"refs/heads/stable/wallaby","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
