)]}'
{"id":"openstack%2Fneutron~835866","triplet_id":"openstack%2Fneutron~stable%2Fxena~I1e31565475f38c6ad817268699b165759ac05410","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1959567-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-05 15:12:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e31565475f38c6ad817268699b165759ac05410","subject":"Fix ingress bandwidth limit in the openvswitch agent","status":"MERGED","created":"2022-03-31 08:37:47.000000000","updated":"2022-04-05 15:14:54.000000000","submitted":"2022-04-05 15:12:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835866-bug/1959567-stable/yoga-stable/xena","cherry_pick_of_change":835864,"cherry_pick_of_patch_set":1,"meta_rev_id":"cc860530b19a728e9d116d4c772d73a6d376138f","_number":835866,"virtual_id_number":835866,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 15:12:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-04-04 10:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-04-05 11:30:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-04-05 11:30:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-31 10:12:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 10:51:09.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-04 12:39:38.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-04-05 11:30:59.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"32823c457ad2db93a03672cc84a3560a780559d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-31 08:37:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"c27950b2eda5f2ce14ab545d52d43c01ecd43b41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 10:12:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/431036238de74161888c65f192659382 : SUCCESS in 46m 06s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/dadb8d516498445787a0b3786eb2ed4e : SUCCESS in 50m 49s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/3ede3e9ec08e4e45864060933e9a8242 : SUCCESS in 1h 00m 58s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/1b960868d6604dabb1f621ba8755b1dd : SUCCESS in 47m 45s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/383187e013c94bb2b8ec325f20f9879b : FAILURE in 12m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6671c5a1980e47328214e39d939eff13 : SUCCESS in 40m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/032eb0bad75946adbf23ef2fbb46d089 : SUCCESS in 14m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2de735efdcbc4321b9c9259f4136a079 : SUCCESS in 28m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a4a1d65fc954ce1ba326998950aab41 : SUCCESS in 24m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03322dccf1384aee846e665b0e384a37 : SUCCESS in 24m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a34a767340445de8245be9c084c6889 : SUCCESS in 9m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77aa68aa7d1e461c9f6a58861925abe5 : SUCCESS in 48m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e7d8725747ea44d587015d26afd30d38 : SUCCESS in 1h 22m 00s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f78c2367cb4d4a92ab20d5a42f044b41 : SUCCESS in 1h 16m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb737c81eaec492493a36944866f0159 : SUCCESS in 1h 03m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d5fa212795964b88a65c89f49e7b2e43 : POST_FAILURE in 1h 04m 57s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cd3238a3864d4f78a7d2952f37971e26 : SUCCESS in 1h 09m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0aec3594ca3e4d84b254706887663031 : SUCCESS in 1h 11m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c3cca60ab914984abdaed0cb00a1fb1 : SUCCESS in 1h 28m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5b5a2d3136b94b23aa8af79875a3d47b : SUCCESS in 1h 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b75684e074d7561d8afa7c56935ea4c526cde44b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 10:15:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4668b84238a84668b036bc674c8b8040 : SUCCESS in 52m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82c5104e74944c149be02613d065f4dc : SUCCESS in 52m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0dde14e24e62621678f914fc36a89f660ac3498f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-04 10:51:09.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8dcc6ef4aed76463335c7a0526881d1797306fe0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 11:54:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06901e43bb57407ca30e1218d894d6e7 : SUCCESS in 44m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb4abb0643124c10bb555c7888c82035 : SUCCESS in 1h 01m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2cfa4b0b275f36af06fc78c7602d8586cfde69b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 12:10:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/f6727e6f285144509c8b7cbb570261ac : SUCCESS in 47m 27s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/5717530bd58e4e72a9a2b6e1f421e7ff : SUCCESS in 55m 42s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/191c3d7948c24af4a2e1bfd164aa359d : SUCCESS in 1h 16m 53s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/e4d72045997d4844836c51f95337abbf : SUCCESS in 1h 07m 59s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/e9edafc54acd40f8acc4e5156c56d303 : SUCCESS in 58m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09b58e52088c43d0a9b80bcabac0c091 : SUCCESS in 39m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d5233c87fa416297101d377237ce7b : SUCCESS in 15m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8faa80c5d5af462387405e52f3781bc2 : SUCCESS in 32m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce6e32f8678e4274b376bafc424b2a4a : SUCCESS in 27m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a78468703cbf4d6884ac4378cae09dd9 : SUCCESS in 24m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a79cd7c6521a4d448b42eaebedaddb93 : SUCCESS in 8m 47s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf1af077ca2f4c20b49b35c991c01ad8 : SUCCESS in 40m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7cb6507464944705b2854a342397b5df : SUCCESS in 1h 09m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5626d4cfb5a94e3dbf95870b6b4c7f86 : SUCCESS in 58m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/86426d85eaef41f186b370735622fe21 : SUCCESS in 52m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c98c25bc2cb246b99c9a36bb31b18f74 : SUCCESS in 55m 07s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ff06f90a96574f67993ec0e976bf4c56 : SUCCESS in 57m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/53b1573bf9bf4aa5900aeb956e5057fc : SUCCESS in 1h 05m 44s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99bf10389daa45cbb8aef3631345c184 : SUCCESS in 44m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/84f496783e4e46199b0f8d7c1e7c8251 : SUCCESS in 1h 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa1ba30f1ee1346f806487b1451877d0c883a0b2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-04 12:39:38.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/835866/1 : SUCCESS in 1h 47m 47s","accounts_in_message":[],"_revision_number":1},{"id":"b791f67a24cfe5fb6c3146ce62dcbd85bd0621a2","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-05 11:30:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"861d9e3cc8ded3f4b53ebf47f67311999ea454de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 11:32:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"438bd5fbd1ea19a798c2f5a3c1ab47fbbd767857","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:12:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/5c26dca84071462e9934d2120ef30786 : SUCCESS in 56m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/900df72e11b44d0ba9e9e206f78cdc23 : SUCCESS in 15m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae419294016d499db77b1557306bcaeb : SUCCESS in 28m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b0db37abf8042b29caf85494b432b60 : SUCCESS in 27m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4413aff96d05465eb0352222579f71e1 : SUCCESS in 9m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fe919ec2b6744a67b4830547c2f274cf : SUCCESS in 54m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b052463fec047f6abd28cc5e0f4d726 : SUCCESS in 1h 34m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9aec4646ca3b4502b884bf85b9b4cafa : SUCCESS in 58m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/df7e276b5a7e435c96b6d50e8325228b : SUCCESS in 1h 12m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d965f8b0550c4bd782bab291e1d0edf6 : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":1},{"id":"315a5ad6c51c7f0471fd4a0d32b9ff9c412b4693","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:12:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cc860530b19a728e9d116d4c772d73a6d376138f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:14:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37fc45b72c9e4721a0519dcf734e7c95 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5573230d30a7beb242ab54eb85b8e36f33f79403","revisions":{"5573230d30a7beb242ab54eb85b8e36f33f79403":{"kind":"REWORK","_number":1,"created":"2022-03-31 08:37:47.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/66/835866/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/835866/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/835866/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/835866/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/835866/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/835866/1"}}},"commit":{"parents":[{"commit":"18e05b5b2aad44cc53ea8b3d1f9c2c1c4bc08571","subject":"Clean duplicated QoS bandwidth related methods in ovs_lib module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18e05b5b2aad44cc53ea8b3d1f9c2c1c4bc08571"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-08 15:44:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-31 08:37:47.000000000","tz":0},"subject":"Fix ingress bandwidth limit in the openvswitch agent","message":"Fix ingress bandwidth limit in the openvswitch agent\n\nFor ingress bandwidth limiting openvswitch agent is using QoS and queues\nfrom the Open vSwitch. There is always queue 0 used for that purpose.\nInitially, when this feature was implemented, we assumed that queue 0 is\nkind of the \"default\" queue to which all traffic will be send if there\nare no other queues. But that\u0027s not true thus ingress bandwidth limiting\nwasn\u0027t working properly with this agent.\n\nThis patch fixes that issue but adding in the table\u003d0 of the br-int\nadditional OF rule to send all traffic to the queue 0.\nIn this queue for some ports there can be QoS configured\nand then it will be applied for the port. If port don\u0027t have any QoS\nconfigured, nothing will happen and all will work like before this\npatch.\n\nBiggest problem with that solution was the case when also ports with\nminimum bandwidth are on the same node becuase such ports are using\ndifferent queues (queue number is the same as ofport number of the tap\ninterface).\nIn case when traffic is going from the port with minimum bandwidth QoS\nto the port which has ingress bw limit configured, traffic is going only\nthrough br-int and will use queue 0 to apply ingress bw limit properly.\nIn case when traffic from port with minimum bandwidth set needs to go\nout from the host, it will always use physical bridge (minimum bandwidth\nis only supported for the provider networks) and proper queue will be\nset for such traffic in the physical bridge.\nTo be able to set proper queue in the physical bridge, this patch adds\nadditional OF rule to the br-int to set queue_num value in the pkt_mark\nfield [1] as this seems to be only field which can \"survive\" passing\nbridges.\n\n[1] https://man7.org/linux/man-pages/man7/ovs-fields.7.html\n\nCloses-Bug: #1959567\nChange-Id: I1e31565475f38c6ad817268699b165759ac05410\n(cherry picked from commit f7ab90baad83823cfb94bc8b9450cd915ea49c03)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5573230d30a7beb242ab54eb85b8e36f33f79403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5573230d30a7beb242ab54eb85b8e36f33f79403"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
