)]}'
{"id":"openstack%2Fneutron~835946","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I848af01c8fe3a08b26d05e37d225c944ea080f03","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1959567-stable/ussuri","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-05 13:07:35.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-04-05 10:22:51.000000000","reason":"Bernard Cafarelli replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-05 13:07:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I848af01c8fe3a08b26d05e37d225c944ea080f03","subject":"Clean duplicated QoS bandwidth related methods in ovs_lib module","status":"MERGED","created":"2022-03-31 08:48:12.000000000","updated":"2022-04-05 13:09:49.000000000","submitted":"2022-04-05 13:07:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"835946-bug/1959567-stable/ussuri","meta_rev_id":"613321e4744a5413555f2ed05f122af75cd30230","_number":835946,"virtual_id_number":835946,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-05 13:07:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-04-04 10:51:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-04-05 10:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-04-01 15:13:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-04-05 10:22:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-31 14:10:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-01 15:13:19.000000000","updated_by":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2022-04-04 10:51:51.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-05 09:29:55.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"5214e0f3fcc0ea38db4cac7532a7c042a4c67e5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-31 08:48:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"156b4902523fe2ad32c71fc367665d7abb1b8e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 14:10:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/4f7982d32a024a3d8aebd7ed70cb7636 : SUCCESS in 47m 52s\n- neutron-tempest-plugin-scenario-linuxbridge-ussuri https://zuul.opendev.org/t/openstack/build/88272896e73c4c01b37ac19b4f3a803b : SUCCESS in 2h 34m 45s\n- neutron-tempest-plugin-scenario-openvswitch-ussuri https://zuul.opendev.org/t/openstack/build/66526d8afd2d4e34a7cf7e69ec1b5d97 : SUCCESS in 1h 59m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-ussuri https://zuul.opendev.org/t/openstack/build/637d0f8ca75e4f35b786a48c6845a04b : SUCCESS in 1h 42m 14s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/6f35511d7c034d948577cda8ff7e8d7c : SUCCESS in 1h 58m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4e07302c86442bdb2e8ba907aec92f0 : SUCCESS in 33m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e59badd0e1de42c88d2ad04c677394f7 : SUCCESS in 13m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d73f9c7a510c43129a17eb5d6e30d516 : SUCCESS in 28m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/669c4820d8c1482c8c1c34a781c169e7 : SUCCESS in 24m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24bad1e5918b48a1a18f2b52d95566e3 : SUCCESS in 26m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa740f9eac634e6fae2f8ec943be2dec : SUCCESS in 9m 10s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d3530f1cebd74baab43bcaafb49b8d81 : SUCCESS in 53m 57s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/2e000fa976f94966ae7b2dd3d532d695 : SUCCESS in 2h 09m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb27307acae74b30b5071806be520e36 : SUCCESS in 1h 52m 18s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/db92bb4266e149d5ba39fe60f79d9252 : SUCCESS in 1h 32m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5866dedacd974ca6b6ac9c5e72531309 : SUCCESS in 1h 19m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/063e000a5ea04635b1465d945532341a : SUCCESS in 1h 11m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/08fa9b847fb7478daa21de8565a5c6f5 : SUCCESS in 1h 12m 02s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f609903387fa449e9b8c0b8a96000e3e : SUCCESS in 1h 10m 24s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/8804ab80faca4461bf13ea415d7fc1b1 : SUCCESS in 1h 10m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cdc73605d0884e85b45821ed45ef2fca : SUCCESS in 2h 01m 54s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/937fa3cf7d8e42d1b5283459eb3cdeef : SUCCESS in 1h 27m 05s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4f78a3ab0ad40299d7c44870609232c : SUCCESS in 1h 08m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7fc64ef91eff41628dd5a0e0b67ffb1f : FAILURE in 1h 04m 31s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/76e72d16294b4882ae82a51d694ea894 : SUCCESS in 1h 42m 58s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41f321d252c844d5a9a9e002d3cb0eeb : SUCCESS in 1h 06m 07s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d519e5a33f534aa6ae2f969a16e974ee : SUCCESS in 1h 17m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/f73b80595f174db4af35d5d3e4e5ef1b : SUCCESS in 50m 20s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/7b6cf7a0a14344f9b38e684634b48bdf : SUCCESS in 1h 21m 59s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d0854be88e3434ab499420581c71666 : SUCCESS in 1h 04m 24s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/767363d1c43a4cd5b2df9b20f4a023bb : SUCCESS in 4h 01m 24s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/678a5f35f3c74e77a9c78772de96b3c3 : SUCCESS in 1h 25m 33s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/796cebe1fb6043d587d0fbb52bd72590 : SUCCESS in 1h 44m 23s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0ef32915045946489040dbbccc4529d8 : SUCCESS in 2h 15m 04s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/3c5bdaae04a54de4899440746584bb0b : FAILURE in 1h 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7777715fb1bba318499ac47d766e3d0b87446ed3","author":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"date":"2022-04-01 15:13:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"18b80ba6f377cdef5acf0c035425ddfe9441528e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-04 10:51:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"de3e076ff0ba023e8c5a0908a10366a361180c8f","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-05 09:29:55.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6a7cac53092fa5911b7b578b0dfe6b6064cb7cf7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-05 10:16:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c26fcb12b3e77b7bcf8539c872ee9418b6c0b0b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-05 10:22:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1031f321777a45664c24d0c4a7f24b07505c7858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 10:24:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0744696af98af1fbd5b02fcb502aac10382e4ec3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:07:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/6b81ce2999834e57a35ba80c8c999810 : SUCCESS in 55m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b177f4999043eca39a1eee21386867 : SUCCESS in 16m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14b73d7dfadc4d9fa97c55460a61a62c : SUCCESS in 26m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f00ada596ba436fa8e29755d02d4567 : SUCCESS in 29m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f61fc24bce4dc18901b959a58af633 : SUCCESS in 10m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/8969300f05044889831496303a1a6a56 : SUCCESS in 55m 15s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c0390f598d77449b98a66dea3aa1c1c0 : SUCCESS in 1h 25m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6fd15aff85724c74bf3726e99bf97f40 : SUCCESS in 1h 09m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c6ad671108484812909a92fc6abbb66c : SUCCESS in 1h 46m 14s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/53df49a0257e4ff6b3be69760d112ff9 : SUCCESS in 1h 50m 07s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f3009177718453d9f229b71ba376377 : SUCCESS in 1h 25m 37s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cf92ab11277a471bab841538ef5341ed : SUCCESS in 1h 18m 42s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/275d66e663ee4ebc8b9e6bdcce10370c : SUCCESS in 2h 07m 01s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a35bb0100c754de8b33bfd93cada0779 : SUCCESS in 1h 21m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/48a4c894943e40919cdf75dd9e446d0a : SUCCESS in 1h 33m 49s","accounts_in_message":[],"_revision_number":1},{"id":"6653039b9b150824700ce0519137ccb8c18bde59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:07:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"613321e4744a5413555f2ed05f122af75cd30230","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:09:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2514d6ea12ca4193b73994c38d4b8a31 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44a22bb15c7e62e443ae440c0e85c8ef0c620b96","revisions":{"44a22bb15c7e62e443ae440c0e85c8ef0c620b96":{"kind":"REWORK","_number":1,"created":"2022-03-31 08:48:12.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/46/835946/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/46/835946/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/835946/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/835946/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/46/835946/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/46/835946/1"}}},"commit":{"parents":[{"commit":"146977e3251dc608c9b21295671934d7094de3ce","subject":"Merge \"Fix a error when l3-agent gets filter id for ip\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/146977e3251dc608c9b21295671934d7094de3ce"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-01-31 14:42:04.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-31 08:47:19.000000000","tz":120},"subject":"Clean duplicated QoS bandwidth related methods in ovs_lib module","message":"Clean duplicated QoS bandwidth related methods in ovs_lib module\n\nThis patch also some helper methods used in the\nminimum bandwidth qos methods as it seems that we had things almost\nduplicated in methods like _find/update/delete_{qos,queue} and\nfind/update/delete_{qos,queue}.\n\nIt also moves functional tests for the ingress bandwidth limit rules\nmethods to more appropriate module.\n\nConflicts:\n    neutron/agent/common/ovs_lib.py\n    neutron/tests/functional/agent/common/test_ovs_lib.py\n\nRelated-bug: 1959567\nChange-Id: I848af01c8fe3a08b26d05e37d225c944ea080f03\n(cherry picked from commit 0255f41ad0405eac8b2a4e0870513846e2660f7f)\n(cherry picked from commit 18e05b5b2aad44cc53ea8b3d1f9c2c1c4bc08571)\n(cherry picked from commit 9e3aa7e3d66f8def5055a8088f4baa5bcbd70328)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44a22bb15c7e62e443ae440c0e85c8ef0c620b96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/44a22bb15c7e62e443ae440c0e85c8ef0c620b96"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
