)]}'
{"id":"openstack%2Fneutron~835949","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I1e31565475f38c6ad817268699b165759ac05410","project":"openstack/neutron","branch":"stable/train","topic":"bug/1959567-stable/train","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-12 13:05:36.000000000","reason":"Slawek Kaplonski replied on the change"}},"hashtags":[],"change_id":"I1e31565475f38c6ad817268699b165759ac05410","subject":"Fix ingress bandwidth limit in the openvswitch agent","status":"MERGED","created":"2022-03-31 08:53:55.000000000","updated":"2022-04-12 19:59:19.000000000","submitted":"2022-04-12 19:57:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"835949-bug/1959567-stable/train","meta_rev_id":"9cffa32d1f9b8e9073ae79204ceedd14c7f378ec","_number":835949,"virtual_id_number":835949,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-12 19:57:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-04-07 14:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-04-12 09:02:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-04-12 09:02:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-31 13:13:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-31 17:47:37.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-04-05 09:15:54.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2022-04-07 14:54:22.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-12 09:02:36.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"f18f93ecedcb9c91868673b11711a2d033f04874","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-03-31 08:53:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c79e9d91320d2c366bdef973a6a68eb60efa1868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 13:13:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/06b909dedd6643638b6474c3dd4af506 : SUCCESS in 48m 40s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/16680c77137e40da99b614b33dc9be0a : SUCCESS in 2h 20m 54s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/d085d6632f3049c6b0f373d136bed724 : SUCCESS in 1h 56m 30s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/a7f4ebbe32924cffbb9c05cf643dfcdf : SUCCESS in 1h 32m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9017bb1527844402b5ee8f187afe72df : SUCCESS in 29m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18028e2afa9d4ab78823ef7c661b7e67 : SUCCESS in 13m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8147af29e63a4f3d93cfc6af5cd4ff0b : SUCCESS in 22m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06a5ef9f7091402cb4572bd5fa0732b7 : SUCCESS in 21m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94c10d38c7334b9f97b3382f3aa28758 : SUCCESS in 23m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f62c38dae5b423f8ebcf9d3441dbc47 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/fbc560700a604df58173d77ab51febdd : SUCCESS in 1h 09m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/83ad18af18c1475e99cd1d82f78f9b8c : FAILURE in 23m 39s (non-voting)\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d701deca8fb14e71aca25441e4636fc0 : SUCCESS in 1h 44m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aee30ded213d4621aba3196cb77b2e84 : SUCCESS in 59m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/907d46f21eda4a1ea77d420e902e4681 : SUCCESS in 49m 23s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/0d7bfc7ce8624ab697b747e4e2ad5e81 : SUCCESS in 50m 48s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/d9fd2b5b29ba45cda767d93006703654 : SUCCESS in 1h 18m 15s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/27bea7cac1c8488f8e37b1474c6a2e9b : SUCCESS in 1h 39m 58s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bb894216dabc440c9ff5a209493d4ffe : SUCCESS in 1h 24m 57s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6300504fd242479d8d4fffd8aab83eea : SUCCESS in 1h 15m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7229feee22c4b0da19f132d9b9e35e8 : SUCCESS in 1h 06m 34s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/44a695c2c0144822b910d163998d1532 : SUCCESS in 1h 08m 45s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/345f5ba9df82412794e215f5aebb79a3 : SUCCESS in 1h 10m 55s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/d44bbd105ae34c8faed169ef4596a3f9 : SUCCESS in 1h 12m 55s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/01277259d18c402f97616cea8d734d03 : SUCCESS in 1h 37m 22s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4c712b4e1c2548959b466999f77b7566 : FAILURE in 2h 28m 05s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/50f3a40cd8ca4bcd8c2c3e02d2246453 : SUCCESS in 1h 06m 32s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44dffc956d1d42bc90d3f6298bea9257 : SUCCESS in 1h 01m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5bad2f56d84cadf776c57826629943c872a124f4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-03-31 17:47:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f9f79821fe48a627048322f0d8804130db0b07d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-31 20:21:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/b446c339f51e412da581ff49bd085b26 : SUCCESS in 50m 01s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/24f6e045ec0a43f2b95785be227b5a51 : SUCCESS in 2h 14m 45s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/f71ce21fc9e4426e88981ac8baf588f8 : SUCCESS in 1h 37m 36s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/01c1625fe63948daac4b31d76fb28dbc : SUCCESS in 1h 34m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37b1535217bf474ebd70fd017d9616ab : SUCCESS in 26m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5edeac1b10646168df2fb1760018241 : SUCCESS in 13m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa6c540654eb4ba588f83f1ed3659d7d : SUCCESS in 19m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bfb37f083534d1ab9acf4977b2263fb : SUCCESS in 22m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a520ce64064043b6a77a4758cc1716d7 : SUCCESS in 21m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09ea1f772235404d9274f055dbbec661 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/ee61273283e6471f8c96fe7a53083be7 : SUCCESS in 1h 13m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7de9fe7df95146ed922bad1652cbef01 : FAILURE in 16m 50s (non-voting)\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e7ba3d62e8d649f6bbbae20b82872b5d : SUCCESS in 1h 20m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/084af54c46da43749b3fad23210b8424 : SUCCESS in 59m 35s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d3256fa3397f4b518685e3b78c1151ed : SUCCESS in 42m 28s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/5db852f771604f29920f7ffe657c6b67 : SUCCESS in 57m 26s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/4a757eede6cd4a9b8dd10c3b49b24702 : SUCCESS in 1h 07m 20s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/95a9c115f003466aa1bc20189fe86683 : SUCCESS in 1h 38m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/69266069fd934cf98852c5ae95cb3540 : SUCCESS in 1h 06m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2a01586f305c48f5a8e154487ce62d80 : SUCCESS in 1h 03m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9729a45be66a46279c772aa22d2d87a7 : SUCCESS in 56m 46s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/81df30ba31824042981becc86a588a3f : SUCCESS in 49m 01s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/adb356bfde1f4529913756c097b35fe1 : SUCCESS in 1h 10m 53s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0eb99c8f78f84f96b15ca0ceaeedc7f8 : SUCCESS in 1h 04m 23s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e722f30439254282ad6285653618005b : SUCCESS in 1h 44m 19s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/57d7c34b876041c3a9db67aced3a0a6c : FAILURE in 2h 30m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1586abb14be9442c9c5b85d607feda23 : SUCCESS in 56m 40s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0501854bcd35463693079598136dc76d : SUCCESS in 1h 08m 36s","accounts_in_message":[],"_revision_number":1},{"id":"fc736d4fd8c48fbc6b8463d82e598506c7ac2770","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-01 14:30:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"55bd556f0e8de9aafe014a94bd44ae6b2e4d5b14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 17:03:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/e7886cd231c54e35a6e339fd8e7094ce : SUCCESS in 46m 12s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/c91f725caac849ec9e52b04aa8152a23 : SUCCESS in 1h 55m 48s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/99d4267975dd4fef84f8e9aa3c3a457d : SUCCESS in 1h 50m 35s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/563e532f21ad40f8a3fb9bbb80a04aab : SUCCESS in 1h 35m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/154ad4ea59444868943ed6e006c52caf : SUCCESS in 27m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbf1c133ba2b4bdf9da1d4a640710769 : SUCCESS in 13m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/965d3bbbd95549e8ae4f1d20443b210b : SUCCESS in 19m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b844ee4e4ad844f99d03fd6534eec6f5 : SUCCESS in 19m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/951ac8d27a7148cb810bd9f8abaaf587 : SUCCESS in 30m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4228166ecbcd4910bb21e995596a6e6d : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/6e40e927db724b8ca60ea2c205cb9d56 : SUCCESS in 1h 12m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5f2167cc5db74d49a1557ef71a0e771c : FAILURE in 18m 57s (non-voting)\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/73981b2312de47d599bd0ee953ef4dc9 : TIMED_OUT in 2h 07m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7a99e3f71474aaf902e1becaf04bec6 : SUCCESS in 50m 36s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d143e89f1c96440088797c506f3023e3 : FAILURE in 57m 00s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/bc5381fd9bf34d009a4c77fcb33eede1 : SUCCESS in 34m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/358ce2aa36e44f27ab44255e7c70d66f : SUCCESS in 1h 40m 34s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c987b4e8ea7e4c1da42f55cd6680a15b : SUCCESS in 1h 41m 04s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d685f8c02ed741db916aef4b2dd2fc4f : SUCCESS in 1h 12m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5ffa608ef6aa459293d0f7cc418732de : SUCCESS in 1h 23m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8d62f8cc54741ed94cd9edaec823f36 : SUCCESS in 1h 19m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6b1150c50b694a7d8c8e4fec86bc2a7f : SUCCESS in 1h 02m 52s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6d9d031d108644faa36ce7be3f5378f1 : SUCCESS in 1h 01m 34s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/73d166a128844ba1a759a10f58b81b3d : SUCCESS in 55m 27s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/27e223cb089444eb82be3c4f52c917ca : SUCCESS in 1h 40m 01s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/86fe42ef45284c26b02a0ac635191bdc : SUCCESS in 2h 12m 39s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/98eb1746b5d84068a4d8b5561f032736 : SUCCESS in 1h 08m 09s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b47c3e4fb4f438fbf597a42f99dd8d2 : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":1},{"id":"648fbd9ceb3e592f67c123fc1c78779eb6c6999f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-04 10:52:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e47d16ba5b7aabe0abafe84631cbe07e84ef452","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 13:12:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/82b1719cab344aa7a2f5b03e486b220e : SUCCESS in 52m 31s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/d2e8e5a09dc44dedb2f36f06dc8d9215 : SUCCESS in 2h 04m 56s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/cfd2bc88b6c44f78a794b9e33174bbde : SUCCESS in 1h 54m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/4a35eda470844b27968607997ddd5df5 : SUCCESS in 1h 28m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d47b6f5cfe554aadba07d2342cd6f851 : SUCCESS in 25m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082736ace6c44ef59446fd226c2b7401 : SUCCESS in 13m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/445490bdec31419b98009299dbd92d3e : SUCCESS in 21m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d11af458e1f4cdbaa806bcc132fff00 : SUCCESS in 19m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9cbbe9360e643b1b6b37a7f27a921b5 : SUCCESS in 24m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfab79abcac04d28aa94c8d3bde9e028 : SUCCESS in 10m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/cf1cf18e4d5248d58e9c9273a4a62ca2 : SUCCESS in 56m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4dd7bbd22a4042c79ac3f3809ac9354b : FAILURE in 17m 36s (non-voting)\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/d80e7b08637540b58ff1ddf8380ff44d : SUCCESS in 1h 09m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fc00ac1bd7d4af2a41ef1a6aa5efe04 : SUCCESS in 56m 52s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/052e6e3a9ad14233b6f9e8608018c8f1 : SUCCESS in 46m 44s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/06a2914973d04594ba2dc90674bc73b0 : SUCCESS in 45m 23s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/279cd5aa12e04b568252f7e01ecb3b81 : FAILURE in 1h 11m 21s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/6af7256e32864a489fc1d8d6d4324ce2 : SUCCESS in 1h 15m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/b39526d37c454781bcc038ca5c60f283 : SUCCESS in 1h 11m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c6ed8fe283e0479fafc8aa040a5b8f84 : SUCCESS in 1h 13m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3eb4149da1c34dd09c1c00f1f36137cd : SUCCESS in 54m 28s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/319756ff949345e2892c653b67cb99c6 : SUCCESS in 1h 07m 19s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ea3d3fed5c9142acaa8b14b6618decd1 : SUCCESS in 1h 17m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/3e473c7a580c404a8e61c82d9e4cf069 : SUCCESS in 1h 10m 36s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/39984fd478b84d63a4689d72b1996898 : SUCCESS in 1h 31m 27s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/fe059672bf504c75a22a2bae23cf2505 : SUCCESS in 2h 17m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b7f6c38d499a44a39efdf6ebab11e8a8 : SUCCESS in 52m 34s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b9b2895d4ff42bd9eb5eb57be90316a : SUCCESS in 1h 00m 08s","accounts_in_message":[],"_revision_number":1},{"id":"d63abe660439e7d8954995890b90ecb77b6a059a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-05 09:15:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67ab2fada687bbb92fc6b575efbe00491527579f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 12:11:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/c50b9fc35e644e499c68ddfbf1234f10 : SUCCESS in 50m 05s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/9198e6d7d09345b8829ec8a293e286bd : SUCCESS in 2h 28m 48s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/376ed03808974d8f9add398440231e8d : SUCCESS in 1h 37m 17s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/899132f733cb4ecd8ee3f55e29d4e59c : SUCCESS in 1h 28m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb79faddf6ed404786ea997593fe8541 : SUCCESS in 27m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9b382924e1e403fbdb49953a5ef4bd4 : SUCCESS in 13m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de2d1f21bfd64c129c37f1bb124d23e6 : SUCCESS in 22m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d33d57f58d5744aaa5e32048954acdb2 : SUCCESS in 21m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c19fc18b68b4fc08df4254d121aa37e : SUCCESS in 23m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5c18cb030bd4cf68cb8aea858c41869 : SUCCESS in 11m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/85fde9d6811f4fc795053de933683662 : SUCCESS in 1h 05m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5b0f64dbd001481db09a36cea79818f0 : FAILURE in 23m 17s (non-voting)\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/717f30891cc0464d883a03245601b5e7 : SUCCESS in 1h 28m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7043ae612b842acab921877de5a0041 : SUCCESS in 55m 41s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/68d57aa08365479aa816b589995f10ba : SUCCESS in 39m 51s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/a3a78f61c92248daa71c639047bcfb38 : SUCCESS in 47m 05s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/05894538be1248df878c4cd1609254f2 : SUCCESS in 1h 10m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/26490f7f65434356943f05393b6edbed : SUCCESS in 1h 23m 06s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/47fb64552c4645958d12da242bc69202 : SUCCESS in 1h 17m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/48ae53e01f864446a92cf939698e637a : SUCCESS in 1h 14m 14s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4985ccd01337490b8fe50f6a6bc659dc : SUCCESS in 54m 39s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/32390785c8064fd38d882b1e75db13a4 : SUCCESS in 51m 40s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8744c56dc2e2476ab39b798945787588 : SUCCESS in 1h 11m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/042f0dcd872949ea8e5d58dc9f731472 : SUCCESS in 1h 21m 05s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ea12a8620a8f41bba34d5cb6d2a8e206 : SUCCESS in 1h 32m 23s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/19ed709ba41a4623a4d632e0cc80be08 : SUCCESS in 2h 32m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/07414dfe6f50467b9c154987028d5eb9 : SUCCESS in 1h 01m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ee260871445477da7a247ee9c4d362c : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0f9aefcdfd0b142aab030cec34a2f84149b1da10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-05 14:10:46.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2f34f1c1db295b9673dd2defce54e46a39ca964a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 17:30:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/7fbe19cc099342b1a6932bf74d01d074 : SUCCESS in 55m 38s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/80bff90533394124b4bb1787a79afc8e : SUCCESS in 2h 19m 43s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/097995fba88c42c99bd0b19bf1a838de : SUCCESS in 1h 27m 56s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/bb079fd25e7343238e2fc18f48c4f651 : SUCCESS in 1h 14m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66d78ccc46f14149b8b41b0b2f43430b : SUCCESS in 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44c6d4087cdb4b0ab5ccf2e6f0b65110 : SUCCESS in 12m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ddd10eb2c09428b8cd1fe0328dc0ca9 : SUCCESS in 21m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85d38d2d8ea34d209ce41ba75d9f0c07 : SUCCESS in 29m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b96656ba10c46d3b397c59f3d3edbc7 : SUCCESS in 23m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b95cb10d42472a9f5f049d65dbdbc7 : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/81a35da18a28486e986f937553a81379 : SUCCESS in 1h 00m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/96744dc5f86e4c7d8b1f7f1c31a874b7 : FAILURE in 17m 14s (non-voting)\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/464b5e0d021e46309124e9bb29726109 : SUCCESS in 1h 02m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a447aa6a607466fbe55a7f39626e9a7 : SUCCESS in 53m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/254ac1e0c98a469591803c4447e22b7d : SUCCESS in 46m 05s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/d54b0801133b46eda66a0622d4750e10 : SUCCESS in 38m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/34fbd614e80b46db91a5402d2ae71ba2 : SUCCESS in 1h 08m 42s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/602e4a59f660413bba4de33b0ca94ea3 : SUCCESS in 1h 27m 15s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1cc69dabd9d94db79e62b5259ed7de65 : SUCCESS in 1h 17m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e1e011d816e24d529c4cfde5bf4dae45 : SUCCESS in 1h 04m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/291d8342270e46d3ac2d8608f885e381 : SUCCESS in 59m 00s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7730e3592b6744449a2a0db449b8fad0 : SUCCESS in 57m 11s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d4fca1c8d4f740128a331a99b573bf3b : SUCCESS in 1h 04m 54s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/e54f074a1ca7432cb5f4d8744fa6159e : SUCCESS in 57m 23s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3114e439f68642eda3d340f333e38029 : SUCCESS in 1h 35m 29s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a7c8df4c7d7d45259317ccb757260a95 : FAILURE in 2h 55m 58s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/951ba4d28ba0422aba29bca8cc31f256 : SUCCESS in 1h 19m 00s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16342cf0d30346d4913ea036f052f01a : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":2},{"id":"75b2fd98ea712208fe8b76b34d4d91a51ee43968","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-06 07:38:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"78c9642e8537aa8352ef3aae378a1058c36f46dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 10:19:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/aaa5ddc5a171422396da8b7c1f9c28e2 : SUCCESS in 56m 03s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/673087c5651e421a9181eb6dfcb32b77 : SUCCESS in 2h 34m 30s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/80ff463e332b4250959d95822ec63a9c : SUCCESS in 1h 43m 29s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/1571f68a1c784609a0eb5f464de8d42c : SUCCESS in 1h 46m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63ecc80898794f09b173f8513cfefee9 : SUCCESS in 28m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9643abf3e7346979c866a152aa5f22f : SUCCESS in 10m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57f5853e48964121a5e0bc27b21b435d : SUCCESS in 23m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/35adea35f6f8468d99690056a822a57c : SUCCESS in 25m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3c0cfabc2ee24b4c945b47a28340ea03 : SUCCESS in 24m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2c95d9924124f1abe9447105970f9eb : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/b70f0e485bc14764b1ac1aab366f8c75 : SUCCESS in 53m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ae7ea25e5a8e4218a398cb098e9b2163 : FAILURE in 25m 37s (non-voting)\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/84bc7c1ca3c044e483559c1ca7e8da7b : SUCCESS in 1h 15m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b7f8b8c111645dd843c226142caebdb : SUCCESS in 50m 24s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/6c9ae81fbf384fea93128d0f5526762f : SUCCESS in 42m 22s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/dd6c96e915184a0aa7670be50f30cdd5 : FAILURE in 43m 31s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f4de2b4b00164ce6a0a35904d3fa6bca : SUCCESS in 1h 06m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/353cc93d86844268985656c56fcc73a4 : SUCCESS in 1h 19m 29s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/66750d1e7fe54708bce43b5ad9b59810 : SUCCESS in 1h 00m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/83e0b2df61f5493fbd582bb09e97fe79 : SUCCESS in 1h 10m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e482b50c78d4426db5632a07320e4327 : SUCCESS in 55m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/47c3418334504df0bb77f229784ddc74 : SUCCESS in 55m 04s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6ce584a024fd445b954c858676d99916 : SUCCESS in 1h 15m 27s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/4b02851b092842978d2244a6349693fe : SUCCESS in 1h 13m 36s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d6248f11cc8f43b784430ce19fd8e7ae : SUCCESS in 1h 28m 06s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/1a7c45623cf34872856f66dbd42b0cef : SUCCESS in 2h 09m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4d541317186049fe871037908085159a : SUCCESS in 41m 53s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cece7dbee814f41a821680b1f5de8d4 : SUCCESS in 1h 05m 25s","accounts_in_message":[],"_revision_number":2},{"id":"e5dac92e9bdaa3258eb79210b14f9089f5156ec3","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-06 13:26:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ae9672712e66a6b5894ec9a7cc5835a9a4993c14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 15:42:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/97e6ba49d6c54035b39309a85b07fe78 : SUCCESS in 34m 24s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/eff84ba853894aef803f5bdfdd9d7c9a : SUCCESS in 1h 39m 30s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/41bae06d2ac545ed8bbb5a58a0780024 : SUCCESS in 1h 12m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/3a565b3900224a0498e39803d2fbbcfe : SUCCESS in 1h 15m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b66f50428e74c369a3ee7dd1d2a129c : SUCCESS in 24m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e77a96cdb94cd0844aec4947d8dee9 : SUCCESS in 13m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9636e693b2864583996386d076feecb9 : SUCCESS in 23m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1668a9a08044e9e99eca4f1c1bb2eda : SUCCESS in 25m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22a1c56095db4441ada2c62d3d186d45 : SUCCESS in 19m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbc647c236204a38bd497275bf408a49 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/1153e2791e3c46c1abbcb95b4efa95d6 : SUCCESS in 1h 17m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/41151b3db77b46dab0c1f1b3572327e2 : FAILURE in 16m 57s (non-voting)\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/ed72688bac0a4f7e9696325423581252 : SUCCESS in 1h 33m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8f50a9acfee4c10922d725b204d6594 : SUCCESS in 50m 20s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/50280fa666c64ad597a5ab57f737ff7b : TIMED_OUT in 2h 13m 18s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/8e4609eed0a04f1f97a70daeac2e15ef : SUCCESS in 52m 30s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b2da8b1ee8674530b7a2f252b3f2c728 : SUCCESS in 1h 07m 51s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/e812af7f5c984402811a722dcdc6110a : SUCCESS in 1h 18m 49s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/99e9ab70ef8c4df2b40065a266a3f393 : SUCCESS in 1h 37m 04s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6027405416e34382b80ef9ba5e332264 : SUCCESS in 1h 04m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/08a9951673874d9b964deb9f08e68f42 : SUCCESS in 57m 57s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/832cc13f521e435b9cf615baa03ac448 : SUCCESS in 59m 30s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/886d6ea7145344c98a23bd6e5cc7556e : SUCCESS in 1h 12m 08s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/19671501c53a42aaaa3a83a0ff654f4c : SUCCESS in 1h 00m 05s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c93839da88d9493c9b2a13cfcca6e5fe : SUCCESS in 1h 27m 20s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/dd238b3e9e094286bec3843ba01f4c7c : SUCCESS in 2h 06m 03s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2c66de19ea3946069c642879283937eb : SUCCESS in 54m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d64c903d95445759ab7b58d2c9965c0 : SUCCESS in 1h 03m 19s","accounts_in_message":[],"_revision_number":2},{"id":"f5f70be94a9981cfede5601d11ed30af899aee6d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-07 07:51:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a20ed48bedbfd793e85cbb49eb1321bc2869c540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 10:13:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/519a5e1756424e8da78fe31f8c610fe3 : SUCCESS in 53m 03s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/8ab0f843c3e14b6082867394143f7576 : SUCCESS in 2h 11m 57s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/8a4b015e64464e6395dedbec5c0b75f2 : SUCCESS in 1h 35m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/e70dff1d9bfb420285f175f2204f2385 : SUCCESS in 1h 22m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98c572755b9640c188c003e0c3c11149 : SUCCESS in 27m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc0a0aa0ded47c4a82895d92a91d8cb : SUCCESS in 13m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3092b396362e4ad0b241df1e8700ac50 : SUCCESS in 19m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca70e8c2db184fd98bf60a132eb6c979 : SUCCESS in 22m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1008da6794cf41e8bb2538122f488388 : SUCCESS in 24m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fea70e1829ae4e3581cd8658c758773e : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/4f32943e900a42a7a65a9f577418be38 : SUCCESS in 57m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/90a6435e6aeb4d5282cdca995116b042 : FAILURE in 15m 25s (non-voting)\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/707dae25ad734adfb5aa348b0eb1a8df : SUCCESS in 1h 06m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52fdd46f97e04884a1b6ce5a8dec57eb : SUCCESS in 1h 03m 29s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/455de6cf2a5043c1b32db9c77d200a37 : SUCCESS in 46m 38s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/3795fd17f56f4b349ec3be9c83212d0a : SUCCESS in 48m 47s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/11f151bb93554547a20baf7e5bd96d34 : SUCCESS in 1h 11m 37s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/a4b99bc8a1d24aafaa4af51e0c1a64d8 : SUCCESS in 1h 33m 09s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/de7825a6ba434d27ab77b60d51cad3dc : SUCCESS in 1h 04m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f42122fa1b774871820ef418923c54be : SUCCESS in 1h 15m 44s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e036fe1c4f6e4cae91e657a1639798f4 : SUCCESS in 1h 05m 59s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cd656ff4176749b7a43ab1e6d047c255 : SUCCESS in 56m 44s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/858bdca6cf3b4e53b1547dcb474c9222 : SUCCESS in 1h 15m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/101ad0187e0c4d41ba451264bd466446 : SUCCESS in 1h 08m 50s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/acefa65963844540b1fb704e3a0c8ad9 : SUCCESS in 1h 44m 23s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/192b8e3a3b3b4c08befc745584d415f8 : SUCCESS in 2h 16m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1a084055fbd84ca393edced54f29fed6 : SUCCESS in 1h 10m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bc8a6e225e041f9832fb70c362efeab : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":2},{"id":"26f5ca6ffebd0e6ab3b51a2385ac7b49aba54cce","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-07 14:54:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"67d5e6a1179f13cc225e680911436fcc06d4df19","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-12 09:02:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"66cc7c487f6ce305c16d80c4955bc016dad38c5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 09:03:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4d610cc179f66eb6b6fd3e24e656f073b86835f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:33:21.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"0e375d40bc8241ba93f98b1d08fa35d2e386417c","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-12 12:01:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"feb443ca6acb601173829124c804f4364c57a29d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-12 13:05:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4c3e81bd153c84aba6c8f96a45effd0bc18cdb29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 14:32:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/a6c3e35c971b41f98319ae500fbade19 : SUCCESS in 43m 45s\n- neutron-tempest-plugin-scenario-linuxbridge-train https://zuul.opendev.org/t/openstack/build/e447c6271bcb41bcb76af963c865ba8a : SUCCESS in 1h 37m 06s\n- neutron-tempest-plugin-scenario-openvswitch-train https://zuul.opendev.org/t/openstack/build/f0fb086dc40f48238fe7641f485d1153 : SUCCESS in 1h 40m 20s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-train https://zuul.opendev.org/t/openstack/build/647594abfa0e4ef2b2b4d12a4a8b8077 : SUCCESS in 1h 54m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c725b876af7443379d9a85d2eb264581 : SUCCESS in 27m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4009c215aba9411aa373408e98f70e19 : SUCCESS in 13m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1b1bda5a4cb448482914e6d11aec068 : SUCCESS in 24m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4eb21155a9a415bbedc8c658a6b2979 : SUCCESS in 20m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77d81a9d57d84f5b9becc09b01b86309 : SUCCESS in 22m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a5410095fde4256be1a6a7679b166b7 : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/bd84cb9bc4be4e06bcf0b28991662df7 : SUCCESS in 53m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e04f8f1bfed64dd38704854c15e750c5 : FAILURE in 14m 38s (non-voting)\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4c1d68dad35c4c11a33507e00955e6db : SUCCESS in 1h 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f37d63c4cfda41b2bc52f2496d49d62a : SUCCESS in 53m 17s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/b1cfcd2141154204b75093901365e05d : SUCCESS in 50m 26s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/ea77440e1a99435aa4d3c640a079e9f2 : SUCCESS in 42m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/39ddcfddea5d4a8982fb10d055c553ad : SUCCESS in 1h 12m 24s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/948803d2ff134a2c871d40d15a5d30e9 : SUCCESS in 1h 22m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/bde57d8b852d441d8a0c265f40749b49 : SUCCESS in 1h 13m 56s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/19b8b27666684b49991e2a1bf9561f7f : SUCCESS in 1h 08m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/04667cfb30a540e1861de695d5fbfe99 : SUCCESS in 1h 00m 47s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c908646d1e90427cb271153e61922c81 : SUCCESS in 1h 03m 32s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/35ad0c6602344c00b36ee66a35dc76c9 : SUCCESS in 1h 12m 26s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/dbd824c36707461dacf4b26de44a8414 : SUCCESS in 1h 14m 47s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/91dcebcb8b6e460fb302564f22dc4666 : SUCCESS in 1h 25m 56s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/4f07b1ff37e24125b27858c70b901931 : SUCCESS in 2h 12m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b46bfe8d14184efd806a3ee6690f0ae0 : SUCCESS in 1h 09m 30s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b21c0af11e442a6838d43bcd816fdfb : SUCCESS in 1h 12m 42s","accounts_in_message":[],"_revision_number":2},{"id":"1c9f3cfbd1b06c8c4107a90e22bb4c293c44976c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 14:34:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"245370e905e0d1dc1451a37b6b498326be7aee02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:57:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/57b1b7084da0485187d98686ea83707f : SUCCESS in 49m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d82027b3304412815f20760ff77eaa : SUCCESS in 13m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b6b4ad645e349d5a5b8ad6c6f716107 : SUCCESS in 23m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/756f999b6231479e8d99421a55dc4908 : SUCCESS in 23m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0cb3cc2e218a47ae9a6bd0918d548995 : SUCCESS in 24m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab092a1262784901a5c4a302b5a0fcdd : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/2cf92639c5f44cefb2d4a7d16dbf383e : SUCCESS in 48m 37s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/6d85e10dd19f4a27925f2bc302c49d83 : SUCCESS in 1h 13m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76ea3065f6b5414ba554fe8a6beefafe : SUCCESS in 58m 27s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e8692a870bc74f8d8915f40739b64ee5 : SUCCESS in 54m 47s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/53ba5d6bd839492f808da6f177487a2a : SUCCESS in 48m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/82aed39d55fb4a9d8a68181320058b7b : SUCCESS in 1h 07m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/8dfd1f83ad174a818ae24036f11c21ca : SUCCESS in 1h 10m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b372a93ae55a44ae9d35750ed357a6a6 : SUCCESS in 1h 06m 23s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/af8b0f1c3ca84849becc4860f2da0c7b : SUCCESS in 1h 06m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f5ed1cb1fc04150ac2e84b11da1f906 : SUCCESS in 1h 01m 04s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4a8758f7443f4b1f8974a10187a34a5e : SUCCESS in 1h 01m 16s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/0e36673713f94bb9a31a6489d944e1c2 : SUCCESS in 1h 06m 44s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b61895d97e242bf97fef17a66cbc91b : SUCCESS in 1h 34m 35s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/897f335ac3704d3c8d1859d18a7b355c : SUCCESS in 59m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf8de964afbf486895a2e3c2dbbce1c9 : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":2},{"id":"49da9183473c4dfc288bd4b708561d751c1021f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:57:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9cffa32d1f9b8e9073ae79204ceedd14c7f378ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 19:59:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa5bf8f6fd6471b8d8d1b57e58a80a0 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"196ab6bea12f9486efc110a8b37ccc8b495ef12d","revisions":{"59a1ef2a108c2de099eafeb0694b1580be3bfac8":{"kind":"REWORK","_number":1,"created":"2022-03-31 08:53:55.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/49/835949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/835949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/835949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/835949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/835949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/835949/1"}}},"commit":{"parents":[{"commit":"264f30b7c7ada899e3c154fabd71dfdfc84bf328","subject":"Clean duplicated QoS bandwidth related methods in ovs_lib module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/264f30b7c7ada899e3c154fabd71dfdfc84bf328"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-08 15:44:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-31 08:53:22.000000000","tz":120},"subject":"Fix ingress bandwidth limit in the openvswitch agent","message":"Fix ingress bandwidth limit in the openvswitch agent\n\nFor ingress bandwidth limiting openvswitch agent is using QoS and queues\nfrom the Open vSwitch. There is always queue 0 used for that purpose.\nInitially, when this feature was implemented, we assumed that queue 0 is\nkind of the \"default\" queue to which all traffic will be send if there\nare no other queues. But that\u0027s not true thus ingress bandwidth limiting\nwasn\u0027t working properly with this agent.\n\nThis patch fixes that issue but adding in the table\u003d0 of the br-int\nadditional OF rule to send all traffic to the queue 0.\nIn this queue for some ports there can be QoS configured\nand then it will be applied for the port. If port don\u0027t have any QoS\nconfigured, nothing will happen and all will work like before this\npatch.\n\nBiggest problem with that solution was the case when also ports with\nminimum bandwidth are on the same node becuase such ports are using\ndifferent queues (queue number is the same as ofport number of the tap\ninterface).\nIn case when traffic is going from the port with minimum bandwidth QoS\nto the port which has ingress bw limit configured, traffic is going only\nthrough br-int and will use queue 0 to apply ingress bw limit properly.\nIn case when traffic from port with minimum bandwidth set needs to go\nout from the host, it will always use physical bridge (minimum bandwidth\nis only supported for the provider networks) and proper queue will be\nset for such traffic in the physical bridge.\nTo be able to set proper queue in the physical bridge, this patch adds\nadditional OF rule to the br-int to set queue_num value in the pkt_mark\nfield [1] as this seems to be only field which can \"survive\" passing\nbridges.\n\n[1] https://man7.org/linux/man-pages/man7/ovs-fields.7.html\n\nConflicts:\n    neutron/plugins/ml2/drivers/openvswitch/agent/extension_drivers/qos_driver.py\n\nCloses-Bug: #1959567\nChange-Id: I1e31565475f38c6ad817268699b165759ac05410\n(cherry picked from commit f7ab90baad83823cfb94bc8b9450cd915ea49c03)\n(cherry picked from commit 5573230d30a7beb242ab54eb85b8e36f33f79403)\n(cherry picked from commit 4a0ca56fe57dbab08face37a22b9c3273141667d)\n(cherry picked from commit e59a7d5c3d4b29b73973ba8ceb8faf6751ef4d37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59a1ef2a108c2de099eafeb0694b1580be3bfac8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/59a1ef2a108c2de099eafeb0694b1580be3bfac8"}]},"branch":"refs/heads/stable/train"},"196ab6bea12f9486efc110a8b37ccc8b495ef12d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-04-05 14:10:46.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/49/835949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/835949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/835949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/835949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/835949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/835949/2"}}},"commit":{"parents":[{"commit":"ec08253f81367fe4602e5cf4b64fafd4dde55b24","subject":"Clean duplicated QoS bandwidth related methods in ovs_lib module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec08253f81367fe4602e5cf4b64fafd4dde55b24"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-08 15:44:59.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-05 14:10:46.000000000","tz":0},"subject":"Fix ingress bandwidth limit in the openvswitch agent","message":"Fix ingress bandwidth limit in the openvswitch agent\n\nFor ingress bandwidth limiting openvswitch agent is using QoS and queues\nfrom the Open vSwitch. There is always queue 0 used for that purpose.\nInitially, when this feature was implemented, we assumed that queue 0 is\nkind of the \"default\" queue to which all traffic will be send if there\nare no other queues. But that\u0027s not true thus ingress bandwidth limiting\nwasn\u0027t working properly with this agent.\n\nThis patch fixes that issue but adding in the table\u003d0 of the br-int\nadditional OF rule to send all traffic to the queue 0.\nIn this queue for some ports there can be QoS configured\nand then it will be applied for the port. If port don\u0027t have any QoS\nconfigured, nothing will happen and all will work like before this\npatch.\n\nBiggest problem with that solution was the case when also ports with\nminimum bandwidth are on the same node becuase such ports are using\ndifferent queues (queue number is the same as ofport number of the tap\ninterface).\nIn case when traffic is going from the port with minimum bandwidth QoS\nto the port which has ingress bw limit configured, traffic is going only\nthrough br-int and will use queue 0 to apply ingress bw limit properly.\nIn case when traffic from port with minimum bandwidth set needs to go\nout from the host, it will always use physical bridge (minimum bandwidth\nis only supported for the provider networks) and proper queue will be\nset for such traffic in the physical bridge.\nTo be able to set proper queue in the physical bridge, this patch adds\nadditional OF rule to the br-int to set queue_num value in the pkt_mark\nfield [1] as this seems to be only field which can \"survive\" passing\nbridges.\n\n[1] https://man7.org/linux/man-pages/man7/ovs-fields.7.html\n\nConflicts:\n    neutron/plugins/ml2/drivers/openvswitch/agent/extension_drivers/qos_driver.py\n\nCloses-Bug: #1959567\nChange-Id: I1e31565475f38c6ad817268699b165759ac05410\n(cherry picked from commit f7ab90baad83823cfb94bc8b9450cd915ea49c03)\n(cherry picked from commit 5573230d30a7beb242ab54eb85b8e36f33f79403)\n(cherry picked from commit 4a0ca56fe57dbab08face37a22b9c3273141667d)\n(cherry picked from commit e59a7d5c3d4b29b73973ba8ceb8faf6751ef4d37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/196ab6bea12f9486efc110a8b37ccc8b495ef12d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/196ab6bea12f9486efc110a8b37ccc8b495ef12d"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
