)]}'
{"id":"openstack%2Fneutron~836178","triplet_id":"openstack%2Fneutron~stable%2Fxena~I0df376a80140ead7ff1fbf7f5ffef08a999dbe0b","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1963899","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-07 17:10:46.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-04-05 11:27:43.000000000","reason":"Bernard Cafarelli replied on the change"}},"hashtags":[],"change_id":"I0df376a80140ead7ff1fbf7f5ffef08a999dbe0b","subject":"Ensure that re_added ports are DOWN before set back to UP","status":"MERGED","created":"2022-04-01 12:46:33.000000000","updated":"2022-04-07 17:12:24.000000000","submitted":"2022-04-07 17:10:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836178-bug/1963899","meta_rev_id":"2403b45d388afc67cf4cc9cdccadf87fd7500c85","_number":836178,"virtual_id_number":836178,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-07 17:10:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-04-05 11:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-04-04 12:30:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-04-07 14:53:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-01 12:46:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2022-04-01 13:51:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-01 14:11:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-01 14:34:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-04-04 10:42:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-04-04 12:30:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"d55a7547274e086ada6e7bb72b697e5145ac62ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-01 12:46:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"635aa98bdf68fa246972bc744e053b0d27feb2c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 13:51:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/435081b8ca03470da1c9a857b1e1ae2b : SUCCESS in 43m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb977d5d803f48e99b5aab6691936722 : SUCCESS in 49m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d769339443a79869b0066f1dce76418529d1d7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-01 14:11:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/468b375a0a28440dbae4ca0cf8ed90b1 : SUCCESS in 31m 51s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/9cd4f0ba3a07405e81179739d5e465ff : SUCCESS in 56m 44s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/0b52d06dbe50414b89ec5e63c6124e36 : SUCCESS in 51m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a195058b11a4c9bbe4a325fd54f850d : SUCCESS in 29m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/844b23499479423ba350d6475ae95413 : SUCCESS in 12m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3466e4723ce14711aea608b1e7d053ea : SUCCESS in 25m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81c49f09f77e4dc0b48992a9bd75749b : SUCCESS in 22m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f394b9c370344dfda3b6d9a177c6cb97 : SUCCESS in 21m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/719dcb07515d4c729fca6eac45a6e608 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c4fdaff34dc74b1c9fb36ace121f5c48 : SUCCESS in 50m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c8c6a9ae3c1e498e99f4e710950f502e : SUCCESS in 1h 17m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c4a5e2cf61d845a9835bc3a034bf5695 : SUCCESS in 1h 16m 49s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3990eee8af2d46ac8640c6a3a1d2f1ab : SUCCESS in 58m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f856bf9ce0a14511b9882d316741fc9f : SUCCESS in 1h 00m 40s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fb88f175c2724a5782a2c4ce25673528 : SUCCESS in 1h 03m 45s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c29c036ff4354ddeab4d431285f39eec : SUCCESS in 1h 17m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c4c063dbc67f82bbb5ce1cd457d9e96dfb15c880","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-01 14:34:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/836178/1 : SUCCESS in 1h 45m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d18b8aefeed686e034a891f56994ab24f31c7652","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-04 12:30:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"caaad54375874e292656c0ee73a6e6d5adfd560b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-05 11:27:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"86ca48bdbb601817be41197fc667ee8d7dc56286","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-07 14:53:01.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f83cfaaa9c81c1a81c91a71ad6f3eb18a6e41db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 14:53:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89f48345c2872438ec14b8c6ae23aa5e22f08f2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 17:10:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/b29c128b192041a3a7cdb0dc3c959449 : SUCCESS in 48m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff4568caf0f48b98c2bd3f7db51cb25 : SUCCESS in 12m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/399cd100c4c34f9684e0b7a5b68a436c : SUCCESS in 25m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f47647685020424fa1250b61125e0788 : SUCCESS in 23m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9e195f14b114a41881588f6a8cdc322 : SUCCESS in 11m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/550989d1dba64058aa8372f862d4d5fb : SUCCESS in 52m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bb6bd14e2e1046ba8aa7d4fc4c2f64b6 : SUCCESS in 1h 36m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7aeb24927b4947a9b158d259795d3477 : SUCCESS in 1h 12m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6488fe0b8b3b4c5fa387e101c8de2f24 : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5e6ad618c0b23da2e8b0efa6662c95698a34b38d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 17:10:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2403b45d388afc67cf4cc9cdccadf87fd7500c85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 17:12:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08b11f9a50a3403aa3047a24963c6549 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7342d1142728ffce2aa418b8ee28677335b16d57","revisions":{"7342d1142728ffce2aa418b8ee28677335b16d57":{"kind":"REWORK","_number":1,"created":"2022-04-01 12:46:33.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/78/836178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/78/836178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/836178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/836178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/78/836178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/78/836178/1"}}},"commit":{"parents":[{"commit":"8c997bcc37a7c0f97a0e00389126b44bfcd24c26","subject":"Merge \"Log request IDs for matched Nova external events\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c997bcc37a7c0f97a0e00389126b44bfcd24c26"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-23 11:25:26.000000000","tz":60},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-03-20 11:35:23.000000000","tz":0},"subject":"Ensure that re_added ports are DOWN before set back to UP","message":"Ensure that re_added ports are DOWN before set back to UP\n\nDuring e.g. rebuild of the server by Nova, ports plugged to such server\nare quickly removed and added again into br-int. In such case, ports are\nin the \"re_added\" ports set in the neutron-ovs-agent.\nBut it seems that in some cases it may happen that such port isn\u0027t\nswitched to be DOWN first and then, when neutron-ovs-agent treats port\nas added/updated and reports to the server that port is UP, there is no\nnotification to nova-compute send (because port\u0027s status was UP and new\nstatus is still UP in the Neutron DB).\nAs Nova waits for the notification from Neutron in such case server\ncould ends up in the ERROR state.\n\nTo avoid such issue, all ports which are treated as \"re_added\" by the\nneutron-ovs-agent are now first switched to be DOWN on the server side.\nThat way, when those ports are treated as added/updated in the same\nrpc_loop iteration, switching their status to UP will for sure trigger\nnotification to nova.\n\nCloses-Bug: #1963899\nChange-Id: I0df376a80140ead7ff1fbf7f5ffef08a999dbe0b\n(cherry picked from commit e7edcec2608fe1cb30d7458a882eb81e058eb76e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7342d1142728ffce2aa418b8ee28677335b16d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7342d1142728ffce2aa418b8ee28677335b16d57"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
