)]}'
{"id":"openstack%2Fneutron~836608","triplet_id":"openstack%2Fneutron~master~I3390b8c65f4b8572be0ab766024ed4b4ec4aa4ae","project":"openstack/neutron","branch":"master","topic":"bug/1933207-mcast_flood","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2023-09-22 08:02:19.000000000","reason":"Change was abandoned"},"29889":{"account":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"last_update":"2023-09-22 08:02:19.000000000","reason":"Change was abandoned"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2022-04-06 09:29:46.000000000","reason":"Lucas Alvares Gomes replied on the change"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2023-09-22 08:02:19.000000000","reason":"Change was abandoned"},"23804":{"account":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"last_update":"2023-09-22 08:02:19.000000000","reason":"Change was abandoned"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2023-09-22 08:02:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3390b8c65f4b8572be0ab766024ed4b4ec4aa4ae","subject":"Revert \"[OVN] Disable mcast_flood on localnet ports\"","status":"ABANDONED","created":"2022-04-06 07:21:26.000000000","updated":"2023-09-22 08:02:19.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"revert_of":797418,"meta_rev_id":"aa31dc460be506563ed0aa829303fafb710c3695","_number":836608,"virtual_id_number":836608,"owner":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":1,"date":"2022-04-06 16:48:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-06 17:51:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"disliked":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":-1,"date":"2022-08-18 13:13:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-27 14:28:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-07-29 07:17:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 07:21:26.000000000","updated_by":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-04-06 07:21:26.000000000","updated_by":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2022-04-06 07:21:26.000000000","updated_by":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2022-04-06 07:21:26.000000000","updated_by":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"},{"updated":"2022-04-06 07:21:26.000000000","updated_by":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"reviewer":{"_account_id":23804,"name":"Daniel Alvarez","email":"dalvarez@redhat.com","username":"dalvarez"},"state":"REVIEWER"},{"updated":"2022-04-06 07:21:26.000000000","updated_by":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 07:21:56.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 07:59:59.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 08:42:58.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 09:53:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-07-27 14:28:06.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-07-29 07:17:10.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"4b1c6c9af1c0c1b2a202682070d8ba1804dbccad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"date":"2022-04-06 07:21:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22eec6f6d54d953a5674ff1f29fb13d4c72ba56c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 07:21:48.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 836608,1","accounts_in_message":[],"_revision_number":1},{"id":"fa8b5452241df6a8d685b1cd594df3699a034cce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 07:21:51.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/neutron for 836608,1","accounts_in_message":[],"_revision_number":1},{"id":"df8ddb3187b3c624cb1cf33b45e459fc5ce9dd63","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-04-06 07:21:56.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"6f28b43fed40de9853e59ce3ffc79bf188827cf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"date":"2022-04-06 07:55:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"062ad89fd6c3d88705cddf9fe53ec397fce4fe58","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-06 07:59:59.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/08/836608/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"f017327e2e56a63e1b184ee517e3e07e0b0a0b9e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-06 08:42:58.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/08/836608/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"493b1ee2ad6604330c3fcf4b65e15254e749191c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 08:52:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0545888d83e342feb88ea8f1bd0909fb : SUCCESS in 54m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24fecd60bd6740fea59573b9396c8d02 : SUCCESS in 51m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f91a686ebf72844b713f2e21b85e4d9587d91dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"date":"2022-04-06 09:03:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"96a8d1d885f2d2a5708f2856c7ef44d4161e5cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"date":"2022-04-06 09:05:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"de289cad2d57d1101dfb5d34c77b54b3feb6bd04","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-04-06 09:29:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"18d9f19891133530c13b39659b6ac9cdb2afb88e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 09:53:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/db14dde3d7a5417ba21d2a17ff71ea5b : FAILURE in 45m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e994d5a330f4cfb92d0f09b9948387d : FAILURE in 46m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8c89af7a89323602216060970c6ddb642f8dc32d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 10:35:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4626cc24e8d44110b9cde7bcece51d87 : SUCCESS in 54m 50s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/16ca5ba10f8a48eb98396e0173bdc1b4 : SUCCESS in 44m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a421bd6a18d74f89a21080e34a18b915 : FAILURE in 36m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90b985a053ef4da9a41815c79441c0c0 : FAILURE in 32m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a2e3d4af1104fd9b30132484323ca40 : SUCCESS in 12m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3df624daa3ef4f2797017e3125b6aa7e : FAILURE in 27m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f773ef2f5d5b448da99bc251bca747c7 : FAILURE in 32m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd68eb66001d44d4877e137a456272c4 : FAILURE in 28m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c19357c8e7c64978b05824e445b587a8 : SUCCESS in 9m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6e7b8012d4074232957903e28595abc3 : SUCCESS in 1h 22m 08s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cd29a3103654ccf92d5790b137bb0ea : SUCCESS in 52m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c036ee50ca234c4aa30002f8e8802ab1 : SUCCESS in 1h 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9d0fea6b2fef75806bd5dfddfe186c4674a6885a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-06 10:35:52.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/08/836608/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"e6d6e234f10266b3722ee5ed05205fd619df681d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-06 10:52:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/836608/4 : SUCCESS in 1h 46m 35s","accounts_in_message":[],"_revision_number":4},{"id":"10d336ef646e3ee8dd123b8c02726a22118e3c1f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-06 11:20:19.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/08/836608/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"a6432869257e3ff024617a34976b5eab7d3217f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"date":"2022-04-06 14:29:56.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6aa1682dcca4b409a806bda800ba44b6d6d65acc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 15:17:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a7b9fad740e4936909b0937ff7b8851 : FAILURE in 45m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/945a4daaf72243f6954d6efb8a2aa227 : FAILURE in 45m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"feb12de322d5601adfcfde34d9c47fe3288905ce","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-06 15:24:00.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/08/836608/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"6fe5aec8eec16ae2ed04392c9d6c249871e36a0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 15:43:41.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/1a9aa12c01384116bbb5a1ab01475428 : SUCCESS in 1h 07m 13s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d3fcbe07475a471b9a4f8d6ce3bfdcc5 : SUCCESS in 51m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/653f549903044664bf256bc3ccb22a22 : FAILURE in 31m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9737f8532e3b450c82ce022291cfbb7a : FAILURE in 25m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad7f99b96e54186b7e9f823a3af48e9 : SUCCESS in 11m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cf6631544c045dea8fb7a2110498689 : FAILURE in 22m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c066eb1bcf2435f9d206cbe87e1bd88 : FAILURE in 23m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fdd27be807c3449f9ead3163c1b4c495 : FAILURE in 28m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98e6ae820b284714bde748f6ec0570f3 : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/04ef57b5a59f489a99fe93432886b7b9 : SUCCESS in 45m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/494fd59db699435fbbf5b2a39957fa14 : SUCCESS in 39m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f4b73ffb156b4ef78f7c396b8685b86a : SUCCESS in 1h 09m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b2bfe246dfbabdef7cbc77057722ca4b9016ac5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"date":"2022-04-06 15:59:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"91b005fc3d664025f0c187a8dd4d7c5a32fa37bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 16:47:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9319d2db617242cd88db86cc5033fc5c : SUCCESS in 45m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2bbf3f8fd9a6477ab38358f2b2f14266 : SUCCESS in 43m 03s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"da2a18fb88705a451c12b6dfa3e3f0404baddf34","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-06 16:48:27.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/08/836608/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"00c7aa1a2027e5135c643033abc42f196f90244e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-06 17:39:29.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/836608/6 : SUCCESS in 1h 38m 44s","accounts_in_message":[],"_revision_number":6},{"id":"a1843ff459e19924428fce00a4e14abe6f2455ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 17:51:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/f8eeeaab605e4ea19a4ad5314eac4798 : SUCCESS in 54m 40s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a98610a16ced4d2caa2af3e7ac697ea5 : SUCCESS in 45m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3285e398d4994f8382450e8d3c319229 : SUCCESS in 46m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c475b907dab6442292a1e6c2d20c7093 : SUCCESS in 40m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e2fdedca0949a2a64e778a61a783ab : SUCCESS in 12m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b59f46d8cfc45759dab47c223b269a5 : SUCCESS in 23m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b3ef77444c04bd5ace838c084891e4f : SUCCESS in 23m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2d4e0fbd95a438ea039b61f1e4ad6af : SUCCESS in 23m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a05e46f746df42148c820a78d8fb3666 : SUCCESS in 10m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c269f78d763d4c3cbe1d631a29d5c0b0 : SUCCESS in 59m 38s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aaded3720b28477da8dc3047b5973566 : SUCCESS in 52m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d3d6ae388df4b8898ed7775001710c5 : SUCCESS in 1h 44m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"80beb1a2badd4f84a555b3ec916c7f9f4e8e3bb8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-27 14:28:06.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"e7d6b79363246ad8c33691e38327d565f80d94bb","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-07-29 07:17:10.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"43558d0a88071ad5440323d7683d78d97018f87a","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-08-18 13:13:37.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"aa31dc460be506563ed0aa829303fafb710c3695","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2023-09-22 08:02:19.000000000","message":"Abandoned\n\nPlease, feel free to restore the patch, address the comments and rebase the patch.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"dc83924d23557a5077139b441c365a538aac0dcc","revisions":{"ca70b09aade42ae5544cd5bffeef001fc8355098":{"kind":"REWORK","_number":1,"created":"2022-04-06 07:21:26.000000000","uploader":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"ref":"refs/changes/08/836608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/836608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/836608/1"}}},"commit":{"parents":[{"commit":"953eb92a4f1d686a45d91100c8b02f9fe0771ed9","subject":"[OVN] Disable mcast_flood on localnet ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/953eb92a4f1d686a45d91100c8b02f9fe0771ed9"}]}],"author":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 07:21:26.000000000","tz":0},"committer":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 07:21:26.000000000","tz":0},"subject":"Revert \"[OVN] Disable mcast_flood on localnet ports\"","message":"Revert \"[OVN] Disable mcast_flood on localnet ports\"\n\nThis reverts commit 953eb92a4f1d686a45d91100c8b02f9fe0771ed9.\n\nReason for revert:\n\nCommit d02fb560aa35c4931d40000e8ca4c0e4105299b7 to Neutron disabled \u0027mcast_flood\u0027 on localnet ports, in order to avoid multicast packet duplication. This introduces a regression because outbound multicast traffic from VMs is no longer sent to the provider network unless it is part of a multicast group that OVN has snooped. This makes it impossible for something external to OpenStack to subscribe to a multicast group running on a VM (in OpenStack) unless another VM (in OpenStack) subscribes to it first (because the multicast traffic never leaves the OVS).\n\nOpened issue on ovn: https://github.com/ovn-org/ovn/issues/125\nOpened bug to track progress: https://bugs.launchpad.net/neutron/+bug/1967996\n\nChange-Id: I3390b8c65f4b8572be0ab766024ed4b4ec4aa4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca70b09aade42ae5544cd5bffeef001fc8355098"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca70b09aade42ae5544cd5bffeef001fc8355098"}]},"branch":"refs/heads/master"},"ea201efab3a0064b50517e01c951d5f76dbd2165":{"kind":"REWORK","_number":2,"created":"2022-04-06 07:55:49.000000000","uploader":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"ref":"refs/changes/08/836608/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/836608/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/836608/2"}}},"commit":{"parents":[{"commit":"b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc","subject":"Set bigger swap in the functional and fullstack jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc"}]}],"author":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 07:21:26.000000000","tz":0},"committer":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 07:55:46.000000000","tz":180},"subject":"Revert \"[OVN] Disable mcast_flood on localnet ports\"","message":"Revert \"[OVN] Disable mcast_flood on localnet ports\"\n\nThis reverts commit 953eb92a4f1d686a45d91100c8b02f9fe0771ed9.\n\nReason for revert:\n\nCommit d02fb560aa35c4931d40000e8ca4c0e4105299b7 to Neutron disabled \u0027mcast_flood\u0027 on localnet ports, in order to avoid multicast packet duplication. This introduces a regression because outbound multicast traffic from VMs is no longer sent to the provider network unless it is part of a multicast group that OVN has snooped. This makes it impossible for something external to OpenStack to subscribe to a multicast group running on a VM (in OpenStack) unless another VM (in OpenStack) subscribes to it first (because the multicast traffic never leaves the OVS).\n\nOpened issue on ovn: https://github.com/ovn-org/ovn/issues/125\nOpened bug to track progress: https://bugs.launchpad.net/neutron/+bug/1967996\n\nChange-Id: I3390b8c65f4b8572be0ab766024ed4b4ec4aa4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea201efab3a0064b50517e01c951d5f76dbd2165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea201efab3a0064b50517e01c951d5f76dbd2165"}]},"branch":"refs/heads/master"},"0a38b633e23c63b41818f58face3f28db3527c74":{"kind":"REWORK","_number":3,"created":"2022-04-06 09:03:46.000000000","uploader":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"ref":"refs/changes/08/836608/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/836608/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/836608/3"}}},"commit":{"parents":[{"commit":"b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc","subject":"Set bigger swap in the functional and fullstack jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc"}]}],"author":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 07:21:26.000000000","tz":0},"committer":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 09:03:26.000000000","tz":180},"subject":"Revert \"[OVN] Disable mcast_flood on localnet ports\"","message":"Revert \"[OVN] Disable mcast_flood on localnet ports\"\n\nThis reverts commit 953eb92a4f1d686a45d91100c8b02f9fe0771ed9.\n\nReason for revert:\n\nCommit d02fb560aa35c4931d40000e8ca4c0e4105299b7 to Neutron disabled \u0027mcast_flood\u0027 on localnet ports, in order to avoid multicast packet duplication. This introduces a regression because outbound multicast traffic from VMs is no longer sent to the provider network unless it is part of a multicast group that OVN has snooped. This makes it impossible for something external to OpenStack to subscribe to a multicast group running on a VM (in OpenStack) unless another VM (in OpenStack) subscribes to it first (because the multicast traffic never leaves the OVS).\n\nOpened issue on ovn: https://github.com/ovn-org/ovn/issues/125\nOpened bug to track progress: https://bugs.launchpad.net/neutron/+bug/1967996\n\nChange-Id: I3390b8c65f4b8572be0ab766024ed4b4ec4aa4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a38b633e23c63b41818f58face3f28db3527c74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a38b633e23c63b41818f58face3f28db3527c74"}]},"branch":"refs/heads/master"},"7ac2a320cf365d18c3324833dda9ee34977bf7af":{"kind":"REWORK","_number":4,"created":"2022-04-06 09:05:38.000000000","uploader":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"ref":"refs/changes/08/836608/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/836608/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/836608/4"}}},"commit":{"parents":[{"commit":"b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc","subject":"Set bigger swap in the functional and fullstack jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc"}]}],"author":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 07:21:26.000000000","tz":0},"committer":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 09:05:20.000000000","tz":180},"subject":"Revert \"[OVN] Disable mcast_flood on localnet ports\"","message":"Revert \"[OVN] Disable mcast_flood on localnet ports\"\n\nThis reverts commit 953eb92a4f1d686a45d91100c8b02f9fe0771ed9.\n\nReason for revert:\n\nCommit d02fb560aa35c4931d40000e8ca4c0e4105299b7 to Neutron disabled \u0027mcast_flood\u0027 on localnet ports, in order to avoid multicast packet duplication. This introduces a regression because outbound multicast traffic from VMs is no longer sent to the provider network unless it is part of a multicast group that OVN has snooped. This makes it impossible for something external to OpenStack to subscribe to a multicast group running on a VM (in OpenStack) unless another VM (in OpenStack) subscribes to it first (because the multicast traffic never leaves the OVS).\n\nOpened issue on ovn: https://github.com/ovn-org/ovn/issues/125\nOpened bug to track progress: https://bugs.launchpad.net/neutron/+bug/1967996\n\nChange-Id: I3390b8c65f4b8572be0ab766024ed4b4ec4aa4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ac2a320cf365d18c3324833dda9ee34977bf7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7ac2a320cf365d18c3324833dda9ee34977bf7af"}]},"branch":"refs/heads/master"},"72c53425ecea8f58f6a6bccd20e31e850e0a7b21":{"kind":"REWORK","_number":5,"created":"2022-04-06 14:29:56.000000000","uploader":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"ref":"refs/changes/08/836608/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/836608/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/836608/5"}}},"commit":{"parents":[{"commit":"b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc","subject":"Set bigger swap in the functional and fullstack jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc"}]}],"author":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 07:21:26.000000000","tz":0},"committer":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 14:26:28.000000000","tz":180},"subject":"Revert \"[OVN] Disable mcast_flood on localnet ports\"","message":"Revert \"[OVN] Disable mcast_flood on localnet ports\"\n\nThis reverts commit 953eb92a4f1d686a45d91100c8b02f9fe0771ed9.\n\nReason for revert:\n\nCommit d02fb560aa35c4931d40000e8ca4c0e4105299b7 to Neutron disabled \u0027mcast_flood\u0027 on localnet ports, in order to avoid multicast packet duplication. This introduces a regression because outbound multicast traffic from VMs is no longer sent to the provider network unless it is part of a multicast group that OVN has snooped. This makes it impossible for something external to OpenStack to subscribe to a multicast group running on a VM (in OpenStack) unless another VM (in OpenStack) subscribes to it first (because the multicast traffic never leaves the OVS).\n\nOpened issue on ovn: https://github.com/ovn-org/ovn/issues/125\nOpened bug to track progress: https://bugs.launchpad.net/neutron/+bug/1967996\n\nChange-Id: I3390b8c65f4b8572be0ab766024ed4b4ec4aa4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72c53425ecea8f58f6a6bccd20e31e850e0a7b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72c53425ecea8f58f6a6bccd20e31e850e0a7b21"}]},"branch":"refs/heads/master"},"dc83924d23557a5077139b441c365a538aac0dcc":{"kind":"REWORK","_number":6,"created":"2022-04-06 15:59:44.000000000","uploader":{"_account_id":29889,"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","username":"dparv"},"ref":"refs/changes/08/836608/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/08/836608/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/08/836608/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/08/836608/6"}}},"commit":{"parents":[{"commit":"b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc","subject":"Set bigger swap in the functional and fullstack jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc"}]}],"author":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 07:21:26.000000000","tz":0},"committer":{"name":"Diko Parvanov","email":"diko.parvanov@canonical.com","date":"2022-04-06 15:59:33.000000000","tz":180},"subject":"Revert \"[OVN] Disable mcast_flood on localnet ports\"","message":"Revert \"[OVN] Disable mcast_flood on localnet ports\"\n\nThis reverts commit 953eb92a4f1d686a45d91100c8b02f9fe0771ed9.\n\nReason for revert:\n\nCommit d02fb560aa35c4931d40000e8ca4c0e4105299b7 to Neutron disabled \u0027mcast_flood\u0027 on localnet ports, in order to avoid multicast packet duplication. This introduces a regression because outbound multicast traffic from VMs is no longer sent to the provider network unless it is part of a multicast group that OVN has snooped. This makes it impossible for something external to OpenStack to subscribe to a multicast group running on a VM (in OpenStack) unless another VM (in OpenStack) subscribes to it first (because the multicast traffic never leaves the OVS).\n\nOpened issue on ovn: https://github.com/ovn-org/ovn/issues/125\nOpened bug to track progress: https://bugs.launchpad.net/neutron/+bug/1967996\n\nChange-Id: I3390b8c65f4b8572be0ab766024ed4b4ec4aa4ae\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc83924d23557a5077139b441c365a538aac0dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc83924d23557a5077139b441c365a538aac0dcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
