)]}'
{"id":"openstack%2Fneutron~836618","triplet_id":"openstack%2Fneutron~master~If6e1c6e0fc772101bcd3427601800aaae84381dd","project":"openstack/neutron","branch":"master","topic":"bug/1967144","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-11 14:02:39.000000000","reason":"Change was submitted"},"6773":{"account":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"last_update":"2022-04-11 12:24:36.000000000","reason":"Lucas Alvares Gomes replied on the change"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-04-11 14:02:39.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-11 14:02:39.000000000","reason":"Change was submitted"},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-04-11 14:02:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If6e1c6e0fc772101bcd3427601800aaae84381dd","subject":"Retry port_update in the OVN if revision mismatch during live-migration","status":"MERGED","created":"2022-04-05 11:03:39.000000000","updated":"2022-04-11 14:04:38.000000000","submitted":"2022-04-11 14:02:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"836618-bug/1967144","meta_rev_id":"046c75ee8172d6b33e429bc4ac24145c63e952bd","_number":836618,"virtual_id_number":836618,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-04-05 12:01:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-11 14:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-04-11 11:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-11 12:24:36.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-11 12:24:36.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"}],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 12:01:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 12:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 14:56:04.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-04-11 08:27:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-04-11 10:42:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2022-04-11 10:42:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"REVIEWER"},{"updated":"2022-04-11 10:42:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"629f1c75a9b18f5f93d0338d91a320a648abf2c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-05 11:03:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f3063bcd284eb2d33c3a8e2bec00f23f251283d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-05 12:01:40.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/18/836618/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"ba4bb27ec9978c02512de84fd3aa799c3e1906a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 12:19:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/99ae5a8b55994fd584ab8b3f25681444 : SUCCESS in 54m 41s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8e67f7efeb14496f881b0fdf9924903d : SUCCESS in 1h 12m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35e5a11821d7456e98f9c3858a1afee5 : SUCCESS in 35m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18909320c6e34a60823032520b981dfa : SUCCESS in 31m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e4a9a2276245a0913a89239578e837 : SUCCESS in 14m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d4b4e887d5043bb888b7fd3ea72542f : SUCCESS in 29m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f9a0d90eb2b475f9e9a877452210c3a : SUCCESS in 27m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6c5a5671340450a8ea2b8713eaff596 : SUCCESS in 28m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/906990065f234698871e28e403ba6cbf : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4be890387cdd463bb63cc29b5cb86b8c : FAILURE in 44m 52s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f04f45d62644dc1847694f6597acf3f : SUCCESS in 47m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/23ee632020c44178a7c5d76be026f508 : SUCCESS in 1h 08m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63a82e403ef60e3312a77933ff95c796ad5434dd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-05 12:58:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"460623aa0d93a7caa4b1be238bdd34a1b33ab08b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:41:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0114058e1bbd440c9f01c5840a85a9bd : SUCCESS in 52m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb6244a1025d426ab40cd15bfd270e06 : SUCCESS in 43m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d569af420dc0b86a29c4cc20882a9c948715b5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 14:20:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/4db65a89cf25484289c3bdcb201a6da4 : SUCCESS in 47m 18s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/cd0abc9495434307bff061dfb148163f : SUCCESS in 1h 04m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db9f50e889204332a0ef1637be15679b : SUCCESS in 42m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f41964c0309448b8c0c57fb6b7bc2a4 : SUCCESS in 23m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb09381d81ce4b3198403bb6e2373a05 : SUCCESS in 12m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70ed4fd721d14388b3069ee6df5d0010 : SUCCESS in 30m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a5deb825d0c40d2a5f66837fb77033f : SUCCESS in 28m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5731a270fd124c1388374efacf1cfe4a : SUCCESS in 23m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1344ea30c3ea475aa3cea4c65db882f7 : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/423dfb1cd8c24965b0b52f27a66c5d20 : SUCCESS in 52m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b2cbc80c5a84d97b9943de0491446c0 : SUCCESS in 41m 49s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3a269c772ac2452ab2a555fe338a820f : SUCCESS in 1h 16m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82d274c73a7d9073a99dc9f8ed5cd6441a920b99","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-05 14:56:04.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/836618/1 : SUCCESS in 1h 56m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9e9f6c67fa5d5b9e94ec49ed2d99f75ee6edc5ed","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-11 11:48:28.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7b31633fd9efbaecf51e9facb9b6deeeec085e29","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-04-11 12:24:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a0389a4b9a4568fe6d3e1db96c75c1a4a42152d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:26:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f00b0e0a2f3b29f471d0a4081d104bc192a7d003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:02:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a8fb0450fa0548c791ff1c66d175b0d0 : SUCCESS in 1h 05m 06s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/8d7276e83928430fb16279fb071137da : SUCCESS in 1h 00m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c889c78fff8e41b89e1b839fb4fffa31 : SUCCESS in 37m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b11792dd83a649078c2dd504c087093e : SUCCESS in 16m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40711481d968451dbfba66eb724ccbcb : SUCCESS in 35m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/341fe27fccae4e8787c98b0442f389a4 : SUCCESS in 31m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa14c50af454767b9ef1e1c1d2ebde4 : SUCCESS in 10m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/35e70f09081548ba9b623b9a030de7a9 : SUCCESS in 57m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11fcbad26d3e456bb99f4569090d6bd6 : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8026ce2f630c02f564889b7217834d583920eb2c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:02:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"046c75ee8172d6b33e429bc4ac24145c63e952bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 14:04:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/159defa017a8452790066d20b5e7ca02 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f75c6a616d3cb549153fcc496926358dfc9178a","revisions":{"4f75c6a616d3cb549153fcc496926358dfc9178a":{"kind":"REWORK","_number":1,"created":"2022-04-05 11:03:39.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/18/836618/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/836618/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/836618/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/836618/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/836618/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/836618/1"}}},"commit":{"parents":[{"commit":"3615cd85a4cc6aeecf7f066c4eb21c3cdca71d4c","subject":"Merge \"Add \"qos_policy_id\" field to \"Router\" OVO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3615cd85a4cc6aeecf7f066c4eb21c3cdca71d4c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-05 09:26:32.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-05 11:03:35.000000000","tz":120},"subject":"Retry port_update in the OVN if revision mismatch during live-migration","message":"Retry port_update in the OVN if revision mismatch during live-migration\n\nThis is terrible hack but it seems that there is no other way to\nfix/workaround the race which may happen during live-migration between:\n- port update event comming from the OVN db (port DOWN on the src node),\n- API call from nova-compute to activate port binding on the destination\nnode.\n\nIf those 2 events will be executed in specific order by different\nworkers it may happen that port binding activation will not update\n\"requested_chassis\" of the port in OVN northd due to revision mismatch\n(ovn_revision and neutron_revision will be already bumped by the worker\nwhich processes \"port update\" OVN event).\nIf \"requested_chassis\" will not be updated, OVN will not claim port on\nthe dest node thus connectivity to the vm will be broken.\n\nTo workaround that issue, port_update_postcommit method from the OVN\nmechanism driver will catch RevisionMismatch exception raised by the\novn_client and in case that this was port_update after live_migration,\nwill get port data from neutron db and try to update port in the OVN\nnorthd once again.\n\nCloses-bug: #1967144\nChange-Id: If6e1c6e0fc772101bcd3427601800aaae84381dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f75c6a616d3cb549153fcc496926358dfc9178a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f75c6a616d3cb549153fcc496926358dfc9178a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
