)]}'
{"id":"openstack%2Fneutron~836643","triplet_id":"openstack%2Fneutron~master~I1124af51b286999c633fd4ed347da71382741b1c","project":"openstack/neutron","branch":"master","topic":"bp/qos-minimum-guaranteed-packet-rate","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-04-05 16:03:37.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I1124af51b286999c633fd4ed347da71382741b1c","subject":"tests: add port-resource-request-groups to NETWORK_API_EXTENSIONS","status":"ABANDONED","created":"2022-04-05 12:19:35.000000000","updated":"2022-04-05 16:03:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2b5177b71aff3a9a1c6f050a73fdf400228f4436","_number":836643,"virtual_id_number":836643,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-04-05 13:40:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-05 13:42:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-05 13:40:53.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 13:42:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-05 14:31:07.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"}],"messages":[{"id":"f2b1993873cd36ce5327f1441109dd7fb22ac10e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-04-05 12:19:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6475a2e26ed6cb8a391a668789a689f5d1d701f4","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-05 13:40:53.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/43/836643/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b230e163ab0c05a532124dc7c544e5be3dc46d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:42:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/7bcdb693f1cd4c8496a0509b73f68629 : SUCCESS in 46m 48s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b009c90c500542ce8116a8e785fdc19e : SUCCESS in 1h 13m 15s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/6da388a2cc404a6ebda6c40c1f1f982b : SUCCESS in 1h 02m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/087caf55f4814ae18a059f372c2fa3e7 : SUCCESS in 58m 53s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e48f432f8a964d649f8f08fbf38ef742 : SUCCESS in 1h 01m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/28b8bd2546324ddeb1fd13d7973a5921 : SUCCESS in 33m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b6a395a9b6d4b4f94e7d4ec8c1e7c47 : SUCCESS in 40m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7c6bfe3fdfd4b80ada944a1aed83a62 : SUCCESS in 25m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e2f3edd71dc490eb605746a4338575d : SUCCESS in 14m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0da8855dbbd4438ebf37838216d9e889 : SUCCESS in 27m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8347aa80f03e4c0f849f863b7e2c3975 : SUCCESS in 35m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f34e1f7c15140ab9852359ecab0eff1 : SUCCESS in 29m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cbdfcbbbcfc4469908d59717b9d4ba1 : SUCCESS in 10m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/61e1ad7a7ff542019386544fd2a5bf7f : SUCCESS in 1h 04m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/237ddfe106c84a0f9bcf5fac2501f17b : SUCCESS in 1h 20m 12s","accounts_in_message":[],"_revision_number":1},{"id":"dcf6af1815ae178b2bee79f582422d2cf0ef2e22","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-05 14:31:07.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/836643/1 : SUCCESS in 2h 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9f6db0c4b76fb92ce8ad26527929a888d95f9c79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 15:56:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ac3b920f3514ebca66a99ced713a257 : FAILURE in 8m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d12865608c98415cb88d00a68ea2d58e : FAILURE in 9m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b5177b71aff3a9a1c6f050a73fdf400228f4436","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-04-05 16:03:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c252da5e075995cf09171eb7da578d4fc6a0d6c8","revisions":{"c252da5e075995cf09171eb7da578d4fc6a0d6c8":{"kind":"REWORK","_number":1,"created":"2022-04-05 12:19:35.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/43/836643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/43/836643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/836643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/836643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/43/836643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/43/836643/1"}}},"commit":{"parents":[{"commit":"3615cd85a4cc6aeecf7f066c4eb21c3cdca71d4c","subject":"Merge \"Add \"qos_policy_id\" field to \"Router\" OVO\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3615cd85a4cc6aeecf7f066c4eb21c3cdca71d4c"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-04-05 12:16:51.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-04-05 12:16:51.000000000","tz":120},"subject":"tests: add port-resource-request-groups to NETWORK_API_EXTENSIONS","message":"tests: add port-resource-request-groups to NETWORK_API_EXTENSIONS\n\nChange-Id: I1124af51b286999c633fd4ed347da71382741b1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c252da5e075995cf09171eb7da578d4fc6a0d6c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c252da5e075995cf09171eb7da578d4fc6a0d6c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
