)]}'
{"id":"openstack%2Fneutron~836725","triplet_id":"openstack%2Fneutron~master~I7fb5ff72ea0d4d485161f9f02714d1e5c1d32c6a","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-14 09:33:47.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-04-14 09:33:47.000000000","reason":"Change was submitted"},"30380":{"account":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"last_update":"2022-04-14 09:33:47.000000000","reason":"Change was submitted"},"21757":{"account":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"last_update":"2022-04-14 09:33:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7fb5ff72ea0d4d485161f9f02714d1e5c1d32c6a","subject":"Fix i18n missing in ipam_backend_mixin","status":"MERGED","created":"2022-04-06 03:11:35.000000000","updated":"2022-04-14 09:35:41.000000000","submitted":"2022-04-14 09:33:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836725","meta_rev_id":"f3efa6e0c782b2e39bad3f2a3f528aded8d4293a","_number":836725,"virtual_id_number":836725,"owner":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-04-06 04:43:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-14 09:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-12 10:58:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-04-12 08:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-04-12 11:12:07.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-04-12 11:12:07.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-06 03:27:21.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"REVIEWER"},{"updated":"2022-04-06 03:34:19.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 03:54:57.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 04:19:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-06 04:59:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-06 05:10:13.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-04-12 08:23:34.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-04-12 10:58:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-04-12 10:58:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-12 10:58:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-04-12 11:12:07.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"f9caf9026f4581377c67e32269f04ef06266824d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2022-04-06 03:11:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09d35e0a70b731edcdda4eacd080735ae6e9776f","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-04-06 03:27:21.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f596c1935c7a1f16fa2dd6019598761cf0571bff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2022-04-06 03:33:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a5af5f1fb0f19087fcb5fef7048b4e1b8764f1e9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-04-06 03:34:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/25/836725/1/openstack-sriov-neutron-master-direct : ABORTED\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/25/836725/1/openstack-sriov-neutron-master-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"00dc87e769c79dcf4d5b621b2800520e16d6fe60","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-06 03:54:57.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/25/836725/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"b92747a72034f05c318d5d30f93abab235e8ca80","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-04-06 04:18:13.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"70ca6b4cd81cb682c83bf447048a478754409a57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 04:19:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d846d1e699145c38e5f7afae2abc9be : SUCCESS in 40m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a9f9f83dcee42eab722e41f9a5bada0 : SUCCESS in 41m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f603061c4361aa0200c23374828f2eddf5ae6d9f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-06 04:43:45.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/25/836725/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8683ef7e7788e4add789fab1ff9f3767e4882b1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 04:59:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9ac70a13765c4cd0aca3553022b4f7f7 : SUCCESS in 1h 04m 44s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c089fc51e2c94a718b9945ad0770e7a9 : SUCCESS in 1h 06m 24s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/427899f7d1e5489ca6d4854a153b8de9 : SUCCESS in 1h 10m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/af5a75e7288a4b71a4ea6ed72e286f4a : SUCCESS in 1h 05m 52s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d656d84d16a24851b305a5117e303840 : SUCCESS in 59m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2c352e7825104db79372ca1b31839f29 : SUCCESS in 30m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02e058b2fba140bb9837efc8604f3bc3 : SUCCESS in 48m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ef2007bea964c1989b0b30bcf82d08c : SUCCESS in 34m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1388cb920804553b646e4c58944ccd7 : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c18abe60aa9d43ab86ddb1d4311e2c24 : SUCCESS in 29m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de34bb72ed6c4450baf757e14e0e35ea : SUCCESS in 22m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5753a8c783494e1fa7f0ea8a747ff95c : SUCCESS in 38m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5786ed710df4f5cb321dde276eca851 : SUCCESS in 9m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa57f361bb0343a08c17e7efa22cd220 : FAILURE in 55m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3dcbe81c2f0b4664ae71136715d2eea5 : SUCCESS in 1h 23m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a1b534ab76db44e7a1667ef1c5d72bd0 : SUCCESS in 1h 14m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0faedcfeddf40df8af399d411b1367c : SUCCESS in 1h 06m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8a3cecb76f31486ab7c4d39026bbeaf5 : SUCCESS in 53m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/42ea66caf78b48b691df674f6e0c37f2 : SUCCESS in 1h 17m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/14d567c42d5744499bb93c9f83b24465 : SUCCESS in 1h 07m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4668e363b08494a9e177b0d8c68974f : SUCCESS in 55m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ba04744f8bd9434ab22b8abe29cbc286 : SUCCESS in 1h 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0996fb503ec134bbccc2570e2c422b16c90c89dc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-06 05:10:13.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/836725/2 : SUCCESS in 1h 35m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b0fabdaf4a35ad4a080ad50cf018aea28282acae","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2022-04-06 05:45:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e97d586b66c753974a4481c420d935043af561eb","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2022-04-06 05:45:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bafb92de7932c9aeb6dff6fa2c4559d28b1dd8fa","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-04-06 06:37:38.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"22bd7cc31802b6aaf8ec95fab6de0e37bd1b4144","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 06:56:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c93ac0e3b8e44b6c964d50334cd7f81e : SUCCESS in 42m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90c5fdc66b9745b0bf3cdfc86b90b5c8 : SUCCESS in 1h 08m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"95565e16c5114f6eab551f36f51a6c13c26dcdc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 07:10:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/40dd30ba936d42409d5442c62d3a3235 : SUCCESS in 50m 52s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7860da9cd7cd4383be1ccd33537addc7 : SUCCESS in 1h 04m 46s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9763b52588aa428cade5eb3bc23da462 : SUCCESS in 1h 04m 42s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/94ddd037c87247ce98057f62f41c0d5a : SUCCESS in 54m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b153863c7ef443688897c34942b5f247 : SUCCESS in 54m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/41864a48ea1642abaa47255e8f067934 : SUCCESS in 26m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b480c6e8780e48018ac8f8fdb23d878d : SUCCESS in 31m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b93f1e8abfe4983bbeae246926326c7 : SUCCESS in 25m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4a172bd81548529235ec5fba35a076 : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/633a2037f3354417b81e5d2bb9f0e030 : SUCCESS in 23m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d026fd5163c4434d94434538df905848 : SUCCESS in 21m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/519757fe5f5d464da422c4933f51cca1 : SUCCESS in 28m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fd697561b0c4d2985cc5a8faf91a338 : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ad2bdce19da54ff79fe31a638395672e : FAILURE in 45m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cb512bbaffc14bccafc67bc96aec4be1 : SUCCESS in 1h 23m 11s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5ab24571e15946929c8efdaeec7468d1 : SUCCESS in 1h 01m 28s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/82d7a0fb57d6426aa23045f3287983a5 : SUCCESS in 53m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6d5edd40a5f0498fb7c310a2ad89d11a : SUCCESS in 56m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7d7c2a22310040ee943521218a3dc7de : SUCCESS in 1h 18m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/10d73cc74a7e4ef0bc29d5b191fb23b0 : SUCCESS in 1h 03m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cc302f204714c3faebb4879cbf2a461 : SUCCESS in 54m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/bc174d1389654d119747a7aad7e052a9 : SUCCESS in 1h 03m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f424242aa19e410e84ee18a54d1d4540a9d194bd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-06 07:50:52.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/836725/2 : FAILURE in 2h 04m 43s","accounts_in_message":[],"_revision_number":2},{"id":"bac0cc76246cd30af6c04f2b56aae384e6b46cf7","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-04-06 08:35:44.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"f690a1dd57f6b1aacd667e56974e399572606c16","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-12 08:23:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23e49d39675cb50ab2f8c651c7563dfff6310599","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 08:39:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/582c9188c8a948739bc55d3a5ace4432 : RETRY_LIMIT in 3m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f85371cdd699482f8bf146ff65ea0d29 : RETRY_LIMIT in 3m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"29195a5db76db14658188539ea90b85a53e9d24a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-04-12 09:35:23.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"2770f9701eef759d97989a94974e422bf24e42f8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-12 10:13:06.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/836725/2 : FAILURE in 1h 48m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a3ebbdf43bb5c6f6f8b07e16104dbd620a5b3dc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 10:26:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/474a43d8fbcf424186a7d5ed1d96ef58 : SUCCESS in 59m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0307bce2e4b841919c476820b762754b : SUCCESS in 1h 06m 10s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d4ef29a8ab3d4233a49e9be4a22dd4c8 : SUCCESS in 54m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b23ea240e4d5466abc72348cfa16c9dc : SUCCESS in 48m 44s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fac2802cd28240ed89eeb59f4d5c452e : SUCCESS in 47m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c8c8fb543eac4ef8a5fcde4e261bc17a : SUCCESS in 27m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcd4e408ad624e84bb112288241c501c : SUCCESS in 29m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79eec0093a394a9880b336e96f6ba0a7 : SUCCESS in 28m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c34c7bb4ca0d4426ad226edbd77b7f31 : SUCCESS in 14m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78ae747e11ca49c3a7f819b6b54820e1 : SUCCESS in 23m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/175f07ee58a34253a1fb15406a1062a1 : SUCCESS in 22m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/229ee3b17a3c46d2bfde4c4b6c056bb1 : RETRY_LIMIT in 2m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/617023d2b65d4068baed7bf7c83ce6ce : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b38a854e6deb46cf864e56dece0a10e6 : SUCCESS in 50m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/373d8f60d2af4465b63633b25115c3ea : SUCCESS in 1h 25m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8af3e347fced420b9c611917f583564f : SUCCESS in 1h 04m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/55866356ceb4441e99bd267ae71eb171 : SUCCESS in 53m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/242baca227f7480985bc0ee8637c3268 : SUCCESS in 58m 56s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7bf26bb97e824dabb6b8b88874dad147 : SUCCESS in 1h 15m 59s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b277eaa845444a52b51a028bfd2a3605 : SUCCESS in 1h 13m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7a2e683e178485c9396f0d6cea2940d : SUCCESS in 54m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9d53b11c85b540b4b633ef5d92696d42 : SUCCESS in 1h 03m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8b337403649a89b8993a4740281867a02f3cc54","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-12 10:58:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8a41c6b624f2825709f5a26e5587d587b7bd8495","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-04-12 11:09:13.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-direct : FAILURE\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ea0a9f32b29a74c24f6a93b4fbca51cec216ab7b","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-04-12 11:12:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f4feeca4af3c2ff0ae3e2aaf5040ada45bb1d92b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 11:15:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1384399dd4f820cee926407f0e0d3260cc70f31c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 15:03:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d04bcdba100e4639b2e9449609d81b4f : SUCCESS in 55m 04s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/2b9a8fd317804dd2a9d1fa9f0c2530f3 : SUCCESS in 1h 09m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/48ca6f65964849bf8828bbf41fe42fe4 : SUCCESS in 1h 04m 00s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/3b20ca766fea484d9a00ceb88e7b7d04 : SUCCESS in 1h 07m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/da25af17f031406ab8f82430d4d3b003 : SUCCESS in 1h 10m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4fb7166a09e46a2b50ebb5bb91a4a92 : SUCCESS in 34m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21e2f7a3996a4ff4a425f4e324216e01 : SUCCESS in 16m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db4431e27a374e4bac51c59f6beb0514 : SUCCESS in 25m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd29f5fd64014d279d4737d68bd98dc9 : SUCCESS in 21m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfef70a197e7483cb39c6e26a654f6dc : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/85288e52e1aa4c1e94d4335dddd83c81 : FAILURE in 47m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/906bf2f3d04a4f24a08445fb167d0ca0 : SUCCESS in 1h 48m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/182e8a4b331d4ecdb14c6cfc2a1e8e46 : SUCCESS in 1h 18m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/48b053b25fe64b68bb9a6f9ad0956ea9 : SUCCESS in 1h 10m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b6031a94b92649a69688778e0737252c : SUCCESS in 1h 04m 00s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d40dc8ee620847b894606c10f44008b8 : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e732944339ed8308814163d2c2328781620f975e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-14 06:27:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"26df674efb0cee0ae69248eb2c96709bac811cc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 06:50:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e87550872e6a4fb1980ea7566c8b317d : RETRY_LIMIT in 6m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19d89329e3df45e588fec4941b321173 : RETRY_LIMIT in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e228f34320fd76bcffa650245c4314cc3ecb908","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-04-14 07:19:25.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"78b7cff829c3a57a4225ef5b73359eb50b5b6224","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-14 07:45:49.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/836725/2 : FAILURE in 1h 17m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8ead7b0716d4c1d8f5527c9fc5acf8ca1ee5f184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 08:13:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/6c97968f04ed4c7a9c18810fe408ab88 : SUCCESS in 45m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/9b3ccb2e53fb4d4690fa11d66d572578 : SUCCESS in 1h 04m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/64aecfb944a54c00bb5d2fd555e46b4f : SUCCESS in 1h 03m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3856ffaa54fe41dfa986889608ee7beb : SUCCESS in 1h 05m 55s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/b94950d1b4ce4f0eb269cc020c26226f : SUCCESS in 55m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a957a6af9b324e44b3de992a3cdf85ef : SUCCESS in 32m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec4079b32c1f4e83907c3b65e76bd0fc : SUCCESS in 31m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1d70766eb694835b2745ae19db1636f : SUCCESS in 31m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b559d456a5f4316b37dfecc0339cc7a : SUCCESS in 14m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cb197e1717e46d59f71fc76dbbab062 : SUCCESS in 27m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2b96d8146fd4a5f9a3c1ccd31a6f115 : SUCCESS in 22m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4009ed1d43aa4171916e3de3c32ff60c : SUCCESS in 32m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b02ca1e89e545e594a4e1e3ad2496e1 : SUCCESS in 10m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ac113e221b01456e85cfe15a2e4300b5 : SUCCESS in 53m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6992ed0b1f644069de625a5eadb655f : SUCCESS in 1h 43m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/37619eed38934a8cab605ad5d29cf457 : SUCCESS in 1h 07m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0168e829f6d8496598bb15105e751ab5 : SUCCESS in 56m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f652529b88e94f198457d1796b424c3d : SUCCESS in 54m 55s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6f71a5f40d8140cb860f424ec2ce8f44 : SUCCESS in 1h 06m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d8629417fe494201a385190bcbbff31f : SUCCESS in 1h 26m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8008b6c4481047faa7da659da5a0d326 : SUCCESS in 50m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/eb447b2c25d14cd98945cba441ed56ac : SUCCESS in 1h 03m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e525cf7ec4bbecdc7d86b5be12cbff72de6f2a35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 08:14:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d17612530902463b1d3e7c52301bfde89c94f1e3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-04-14 08:42:48.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/25/836725/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"a25371b12cfda79916324b4838e50c62e6d7f539","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 09:33:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/07cae3d8a82941de97d7f33b69e4b5df : SUCCESS in 34m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c4ad5c5cb4524deca4853bf4ff4fc868 : SUCCESS in 1h 09m 03s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/754caf9b3d014fb18865c05fa62cce22 : SUCCESS in 1h 14m 12s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/a2025f4ee2a44c2a9c5a7fa7ee4d6119 : SUCCESS in 52m 39s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/cfeb581a00c84d05a8175d01116cdc18 : SUCCESS in 59m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/66947a29396143549a5291e53ce3e33b : SUCCESS in 35m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fab1dc4cc3ef42a2b13b7a3cd5a7a324 : SUCCESS in 16m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/992df3faaabd4f4ca283f1807f101f63 : SUCCESS in 24m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07b1d248d78146fbb2640763017604a6 : SUCCESS in 23m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b5f4496d6ed4930b3c2398fbb8f28eb : SUCCESS in 8m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4aeb30abbf3b4eecb6c034dbf8ad3b74 : SUCCESS in 49m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3b0601a7438f41bd902762b18ebb39ae : SUCCESS in 1h 16m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e85b94042e02467e84ab853421642f97 : SUCCESS in 1h 02m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0abd40a45c564cd9afe60dc78dc00273 : SUCCESS in 41m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/dc44e53265064f159a92ad48b3d162aa : SUCCESS in 48m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fde44ae7c2b4d21ab94f2e1c9e410b3 : SUCCESS in 49m 42s","accounts_in_message":[],"_revision_number":2},{"id":"c7c93392d4ef93e158646d24fa1b4f38d658f39c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 09:33:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f3efa6e0c782b2e39bad3f2a3f528aded8d4293a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 09:35:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4da1988f6324e6d8373023b6b113eef : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"416e0445a8d86c9d7e28b7a4272f525c40060a3c","revisions":{"51533c10caf894869f618af944d7df39cd40f092":{"kind":"REWORK","_number":1,"created":"2022-04-06 03:11:35.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/25/836725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/25/836725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/25/836725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/25/836725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/25/836725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/25/836725/1"}}},"commit":{"parents":[{"commit":"b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc","subject":"Set bigger swap in the functional and fullstack jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2022-04-06 02:39:49.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2022-04-06 03:11:32.000000000","tz":480},"subject":"Fix i18n missing in ipam_backend_mixin","message":"Fix i18n missing in ipam_backend_mixin\n\nWithout the \u0027_\u0027 in err_msg, we can\u0027t translate in zanata.\n\nChange-Id: I7fb5ff72ea0d4d485161f9f02714d1e5c1d32c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51533c10caf894869f618af944d7df39cd40f092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/51533c10caf894869f618af944d7df39cd40f092"}]},"branch":"refs/heads/master"},"416e0445a8d86c9d7e28b7a4272f525c40060a3c":{"kind":"REWORK","_number":2,"created":"2022-04-06 03:33:42.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/25/836725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/25/836725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/25/836725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/25/836725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/25/836725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/25/836725/2"}}},"commit":{"parents":[{"commit":"b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc","subject":"Set bigger swap in the functional and fullstack jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8dcb0b7afe06f92090d5d2c366b9aae3a532ebc"}]}],"author":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2022-04-06 02:39:49.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2022-04-06 03:32:44.000000000","tz":480},"subject":"Fix i18n missing in ipam_backend_mixin","message":"Fix i18n missing in ipam_backend_mixin\n\nWithout the \u0027_\u0027 in err_msg, we can\u0027t translate in zanata.\n\nChange-Id: I7fb5ff72ea0d4d485161f9f02714d1e5c1d32c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/416e0445a8d86c9d7e28b7a4272f525c40060a3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/416e0445a8d86c9d7e28b7a4272f525c40060a3c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
