)]}'
{"id":"openstack%2Fneutron~837140","triplet_id":"openstack%2Fneutron~master~Iafa8ec1d9710ff404450ad0a64baf56a0f993756","project":"openstack/neutron","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-15 06:25:31.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-15 06:25:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iafa8ec1d9710ff404450ad0a64baf56a0f993756","subject":"Add ADMIN role and remove scope from the check strings","status":"MERGED","created":"2022-04-08 14:47:05.000000000","updated":"2022-04-15 06:27:17.000000000","submitted":"2022-04-15 06:25:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837140-secure-rbac","meta_rev_id":"a498ee7edba3ec2aa55c1e19dbac54ab8617000b","_number":837140,"virtual_id_number":837140,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-04-11 12:11:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-15 06:25:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-11 14:24:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-12 09:17:25.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":1,"date":"2022-04-14 00:50:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-12 09:17:25.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 15:36:13.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 15:38:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-08 16:16:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-08 16:25:22.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-04-11 10:44:03.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-11 14:24:01.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-04-12 09:17:25.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-04-14 00:50:55.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"97e91c9b16e8730e48ae29898b9e79d23ddd09c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-08 14:47:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9681d5e067059032e11b7daabdeb94c9a45e761f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-08 14:48:43.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3b100eff33820eaa28d36437534c8b05f41f9532","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-08 15:36:13.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/40/837140/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"3befec072543beec324349af8c0d1e5345916fff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 15:38:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/081a5595d18e4cd0ae05d7dd28db51a0 : SUCCESS in 47m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4f161f667784f0bae056e8f22a8979a : SUCCESS in 47m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac52133fcdfbfe0bea3ceb49835eb7b73d06c19c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 16:16:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0e7b0c759af141d5bc9aa54e940cbb08 : SUCCESS in 38m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f0e6b7af240c4e2682db3088095fa744 : SUCCESS in 1h 16m 16s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/79dd05abb0264a03b7415b4b0b661e30 : SUCCESS in 1h 11m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/303904e4194e4a16b1d5c23aff46146f : SUCCESS in 1h 07m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/06fdd0519d4841f3ab25c3ebb823319c : SUCCESS in 1h 07m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b18aa4cf82c48169f72b8ae8c95c947 : SUCCESS in 31m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54d94cbd8a254c7d81a5365ce74faa87 : SUCCESS in 28m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf0eba913c224904a74782bd59a4a3ef : SUCCESS in 27m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/825aa84508fc42c58cbfe394b9fecfe7 : SUCCESS in 14m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dae730026574036b99e0375bd8bfdac : SUCCESS in 21m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/734712cc6ee94fd7ae2d25335579e3ca : SUCCESS in 30m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45c67783b0b74843ac837bc74311783b : SUCCESS in 41m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a664ce471a2c4805a6732cde083da1ff : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9510ae0f97e047338a99ab965e18c8e0 : FAILURE in 43m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a94f655e06c4126b7c74fecd9bf52d5 : SUCCESS in 1h 26m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/235a9de493354d87896203e1b21b0616 : SUCCESS in 1h 12m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6c77a8552eae4826b8704d7a43f495ef : SUCCESS in 1h 02m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ab514f361a44434f81738f251079da9e : SUCCESS in 54m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8059207d1efb4cec886bdb499cbba268 : FAILURE in 1h 15m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0dd98ce4dc61450f851dd8ec6f096773 : SUCCESS in 1h 12m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad9f5de0ce34421088db06ecdb405503 : SUCCESS in 50m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6ea0fbed87804c7eb6a3af85ea2686a9 : SUCCESS in 1h 05m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46d67b993c168d4767864ce3f621dc82b270fee6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-08 16:25:22.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/837140/2 : SUCCESS in 1h 35m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f90756e6f4ec6fcf694fa21f0a3a955813a8b0bd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-08 16:28:37.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/40/837140/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"225a5f1555e790a9f00f34cda3cc6940807ffbd0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-09 06:11:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d230aa23fdbfa2fcc21cf27fce06b0a1d85b1f64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-09 06:55:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be1d7c48182845bb8c8159f253211313 : SUCCESS in 42m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e8e63e50b1b423e8351b7cfb2a04539 : SUCCESS in 41m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e75b5c84fec2d36677ab7a442c9e8fb82caeb38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-09 07:46:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/410f1d06f52747b89ced8c2a53d2a452 : SUCCESS in 41m 00s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c33690cdbeff4b388002b55447de0f01 : SUCCESS in 1h 02m 41s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/3600f3d1b4c24100aa6fad173024d429 : SUCCESS in 57m 59s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/de0ed12517da42c29416e29e2d6c0eff : SUCCESS in 50m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2f2c6ccfe9d54d0892ec19713fbca12f : SUCCESS in 56m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ed79d1510374ab4a5e6990abade722b : SUCCESS in 30m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6f689c5438746bf9f5929366cd310ab : SUCCESS in 25m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1330cf5f28924aea9daf04b05c078cf5 : SUCCESS in 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab952bc7d79744278e47a0b3fc3e2e28 : SUCCESS in 10m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38ac15795f45466d91c880f3a5ea61e0 : SUCCESS in 19m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e83f1aead0454098ae2dd2d6bd07a4d4 : SUCCESS in 20m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29c9cf3d47b94f709b77c4b701514e18 : SUCCESS in 21m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d496e6d9f984ff0aba50943b0a6a49b : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9221496f70194dab810fe92fc9b57a01 : SUCCESS in 44m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a00f0b970424da8ba18f2564181d1b9 : SUCCESS in 1h 34m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0f815f207bfb4e008d3daf2a5b86a343 : SUCCESS in 49m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/aab4e2c19e904416aa5fa52ffb80bdca : SUCCESS in 51m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a78c8f4f94ad4e32aaa4ff578ce802ec : SUCCESS in 41m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/616e1de8b5fd4da6bd2b6db1badcfa7f : SUCCESS in 1h 09m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5343ad5bc7db4649af004b65b8d571e8 : SUCCESS in 53m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7812b907893648e098c49d6d21fe4121 : SUCCESS in 55m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aa8f59c8338b48ba9252b2b7837a4c47 : SUCCESS in 1h 08m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d48dee5e998c3272f2ee4250f3803e6de29da95","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-09 07:48:19.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/837140/2 : SUCCESS in 1h 36m 27s","accounts_in_message":[],"_revision_number":2},{"id":"b61ca226beebbda2d570b578898af2c5acb6bd97","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-11 10:44:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3d8c2ceb0752df19107d1c7755e379c75a88e9bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-11 10:48:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0a47aacd010eeb45743b69e3358a12633f7f3ae3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-11 10:48:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"802c85f07f07593e1ee18aff747bb8b8f397ac3b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-04-11 12:11:57.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/April-2022/Arista_Tempest_Testrefs/changes/40/837140/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"fb4d3053b159e8e3c559452a7a4f4c194d05ec20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:26:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/75d6fa1f3e524dc1b2f209d72b78742e : SUCCESS in 37m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f3949dda23f04caf841932723d568a31 : SUCCESS in 1h 05m 23s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5456a0d363d74537b52c55104f3b3d3b : SUCCESS in 1h 01m 22s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ddb58af34e24ecab493f21d13bd4547 : SUCCESS in 58m 51s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/e11e0a3364db42ababc54baf7e6be4c2 : SUCCESS in 1h 11m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f885fdcdc1a94dc9bf8161bf95476d46 : SUCCESS in 32m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/257cf269c9ea48399cf4ccbafecdd6dd : SUCCESS in 34m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a71beb103b04f1099f8604d5e4f10fb : SUCCESS in 32m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c28a0c14d0f747dfb3dc57ef1935d1c5 : SUCCESS in 16m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d9632035e8f4c1e8f31fa2999db1690 : SUCCESS in 26m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8928af311a3e4a969b31587697388295 : SUCCESS in 27m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b008ac821694077aee8bf511bb982ef : SUCCESS in 50m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1f50f3206954d09ab6c3a9481102720 : SUCCESS in 11m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f75eb3d6be54231952f5f9c7756b1b0 : SUCCESS in 59m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf8766cfcc934b068ab4f84609c146bf : SUCCESS in 1h 37m 12s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9a572447099c479ea30ff976fc950b22 : SUCCESS in 1h 10m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5082a0c7d86f4b0a9a535e8cc9022258 : SUCCESS in 1h 00m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a477f5bbe21d4bd9aaabac6a4db25c92 : SUCCESS in 1h 02m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f1178cb3cfab49728a9fd98844b0ff9c : SUCCESS in 1h 13m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/437ef1b945a0421b941b32ec424fe6fd : SUCCESS in 1h 16m 09s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/947b46c621f1461bb52ab72479bb115e : SUCCESS in 1h 04m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/772a801916014cb68c4cc76386337560 : SUCCESS in 1h 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"76426d5c0cf34f7871867b53fdf99d14fded31df","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-11 12:35:54.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/837140/3 : SUCCESS in 1h 46m 59s","accounts_in_message":[],"_revision_number":3},{"id":"da3b43ed32741e0e1498e0c208333fa1d0a4627e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 13:10:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1aaef23407964bb6a0506bdb805bb3ed : SUCCESS in 45m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97ef7db4a90a48b1b2d8feff3848d660 : SUCCESS in 49m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09f89f72240c7b74108b2e544494a09129d117f0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-04-11 14:24:01.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f3c6e05cb4190c8fc25174fc50562901faea3cc2","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-04-12 09:17:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ad02bbec4692e0c7fdd2a0841190b0ce53403b45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 09:18:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7a710415f01b3af23a8bac289bbe156ac6f31654","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 13:21:54.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/37f84f31c5cc4e1d87243e59bcf3aa0f : SUCCESS in 52m 09s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/631b21529bb54a669c9061d68b5f8ff4 : SUCCESS in 53m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/e0ecfb3b67f14869bde10fbc60f920e8 : SUCCESS in 1h 04m 45s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/fe23f7cfff6d43d084e08d5e5828e5fa : SUCCESS in 48m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8a523fb53b2b4e4e997af9e5f0cb270f : SUCCESS in 47m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/044c80f3ce1243dfbdab189c4d7a5769 : SUCCESS in 31m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/111ea25971094c8195fe17c8793907e7 : SUCCESS in 14m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/253acdafe88146e5b95a0e06ec880869 : SUCCESS in 26m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f96fc06cf5645b7b1bfa0fa745dce64 : SUCCESS in 27m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12935962e9b24277879e5c3e1cdbe2f2 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4b54125e942a4033b4c31e6c9abb80dd : FAILURE in 46m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/980ab843a38949bdbed8812843551689 : SUCCESS in 1h 58m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/79a91ffe959c441caf95b42aac33db63 : SUCCESS in 1h 29m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef8bec73764442a3b2a09eca2267332b : SUCCESS in 48m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eddab3d1f48f48e9a71960947f798955 : SUCCESS in 47m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89a5616d03c746bebed3a1113a7773e1 : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":3},{"id":"cfb7f2febfb7335af733870064daebf268135ef7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-13 06:06:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f5d6448bf6f125e7168cbca5dcd9da4e642ef0a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 06:30:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d54661d482c0484fa41cf07652af7a84 : RETRY_LIMIT in 3m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74135107f0ff45c586e7d7bb05204e64 : RETRY_LIMIT in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c831b1812acbed512c4236f0776e5ab3d1d26f77","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-13 06:45:36.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/837140/3 : FAILURE in 38m 01s","accounts_in_message":[],"_revision_number":3},{"id":"f4c2ac3f5318be1dfe8c810cf9af9434666409a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 07:37:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c6bc6b429a5b47809bc291bdce46b677 : FAILURE in 11m 53s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c114bd00bff142518ba16d65b497d750 : FAILURE in 10m 04s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/311b0b25d00248488af6885be5b8765a : FAILURE in 9m 00s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e441fb9c5d074c6f8e52628d86fb368d : FAILURE in 9m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/249030633dfe48799128f1577be6941b : FAILURE in 9m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc0de860c45542f28192543c63cede6c : FAILURE in 9m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89ea05615451438487f29432532778ea : SUCCESS in 41m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b034b2cb9f648879a8b4b628de40dbf : SUCCESS in 31m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e84cd3f833f4c9d803c9ef1e0af6205 : SUCCESS in 15m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1b91d06ef504b8394d9454a22d4a6de : SUCCESS in 30m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc14d60dc11044e2a1f71a84fd69a7e5 : SUCCESS in 26m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f20b025c2744f2795e9c0acadd61cc2 : SUCCESS in 22m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b61993e1524fa6a7f8d3aac9d05e13 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2205af4be21c4941818f9b3a602f9b58 : SUCCESS in 57m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/82d6476a4f5f4a9b90dcc360b9d6928f : SUCCESS in 1h 27m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/14b2df3e060546d7a86ec46d9831425c : FAILURE in 11m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/737d05abb84e4e92a5f25f5810f29986 : FAILURE in 23m 19s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d1fe22a0337f484c88686e802a80a12d : FAILURE in 28m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/738c8348c3624f4aa9f867dee0ca4f4d : FAILURE in 19m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0c8c79d44e5442e4a82ce81ded6f187d : FAILURE in 25m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccac4745d53544009fb66c96e3511cdf : FAILURE in 18m 05s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/801c0b4e23404e7396a33320be03cf16 : FAILURE in 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a55c53ecf9e437405bdadc31ee44f8513c7a28ec","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-14 00:50:55.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4ca829b6b8e47b88600577dd8794897ae637bfa9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-14 00:51:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"544581f566a2c6695e94c6af7ed533e79499eedf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 01:10:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc997be865114ef5b30f625d7ac410a6 : RETRY_LIMIT in 3m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8cabb177ceeb426dbbb7daeac79a51ee : RETRY_LIMIT in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"478bc27ba960077ab8ae72df865aa9de1bb4e858","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-14 02:05:57.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/837140/3 : FAILURE in 1h 14m 12s","accounts_in_message":[],"_revision_number":3},{"id":"59074fc84b1764600c222ac3f44ee3ae7d12db76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 02:32:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/05f4f76bcdd84bd28418458a7aec717b : SUCCESS in 46m 36s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/dc8e7ee0af6a427bb2fa74ee142efd0d : SUCCESS in 1h 03m 45s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/d7362b2fa2414f9e8bde591dc18f60d7 : SUCCESS in 57m 07s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fdce189178c24df6bf250d47b2a311a2 : SUCCESS in 51m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/89f30597b0974bd6ba2862f4740400a0 : SUCCESS in 58m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4726f1f02f2f494da3e7f2af22326415 : SUCCESS in 29m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c5be7ed2b714e90ad20efe25b350224 : SUCCESS in 33m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ad02ad0a0874f59b1d5a83b93b91696 : SUCCESS in 30m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/479fe0526fa94af7a3732e9f4ba80cd4 : SUCCESS in 15m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3790f930012040d1a826f872317e6d3a : SUCCESS in 27m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc168c57d45d4633bd6b812b6e811180 : SUCCESS in 27m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1432d7980d20424c89ae7a6cdfc4197f : SUCCESS in 23m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c14dc486e84da080b8703ec35156b5 : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6c1f6175a7b488bbfc0b299c887d256 : SUCCESS in 57m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3faec3203cf4aee90aa6adcf98bf672 : SUCCESS in 1h 38m 26s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f16320038e464531aa9276fe85c311fe : SUCCESS in 1h 05m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e15abf76c4e47809b90a83d0bf5c4e4 : SUCCESS in 56m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d5a00907f74a4d99bfa2970ea3133a6a : SUCCESS in 1h 15m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4a8cb1ca0e7649ee92137724391600ee : SUCCESS in 1h 11m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e24c7a7e8a6c4630acb8a92343e1752d : SUCCESS in 1h 10m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10d4d948909442448c6320e0d9cd88a0 : SUCCESS in 56m 13s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/35e739aecb2f4db0aa5a6f8a658b817c : SUCCESS in 1h 09m 41s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d07799f5b2a6bf3d8b08de68a92c686842e51c25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 02:33:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3377ac7f3b32a0bf9ec9763824af38531adb9586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-14 04:07:53.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/78e5032889e44ec7abf8b84c4dd818ce : SUCCESS in 57m 01s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/b2c3e4f59b5a4bbe97ba3e73fdc3cdff : SUCCESS in 1h 20m 58s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/f6ec787d5cf14a2aa5127a8e8a5a69d7 : SUCCESS in 1h 13m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/4077f87815844dd2957692e4df36ce24 : SUCCESS in 1h 07m 41s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/75dfd7b427d74754be96f3a5afe0d880 : SUCCESS in 1h 06m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14d2286a4b9d4125998284189ca6e7c0 : SUCCESS in 28m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b98e60ca6aa477aa8c1f84ab4cea95d : SUCCESS in 15m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f599974afa1b4ed692f5b57c8dcd4140 : SUCCESS in 29m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86e0fc1e3b5a4b61a249a3f674c1045e : SUCCESS in 31m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e3f363a712f4730a69cc5f44e8f1b6a : SUCCESS in 11m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5cd5823b62a04bf8b3881746c3e1b234 : FAILURE in 51m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8f686b965990481985804c23b3838224 : SUCCESS in 1h 28m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/610427868f4540f78be652a1624eceb0 : SUCCESS in 1h 24m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/799ee7489c75461d8629915e3cb6cd00 : SUCCESS in 59m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4eab30bf15544a04b06ee9d88f64e3b3 : SUCCESS in 1h 08m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a660da0e57342f79b3e3ea5f66cba54 : SUCCESS in 1h 08m 52s","accounts_in_message":[],"_revision_number":3},{"id":"6ffc84f3a7b62671a76982c4db58166c67803d2e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-15 02:38:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bf3b99aabc3acfb126e11cb137f5e00277ec7abb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 03:27:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/670df75e39df47d2a914db1ca9696412 : SUCCESS in 45m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0d293618ef146f8a90798059b485fd9 : SUCCESS in 41m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d7073e5f7b4cf0a2e6383d002b16d853456979b7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-15 04:39:41.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/837140/3 : SUCCESS in 2h 00m 28s","accounts_in_message":[],"_revision_number":3},{"id":"416243ff5173f0f767e22c0f8d567cc303f86e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 04:49:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/d6a0cda802a6465ea88660d11e9a9bed : SUCCESS in 57m 14s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/8d8d880524754c29b3af054b652173d4 : SUCCESS in 1h 09m 50s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b50e7dc9a25e45af860dcf39be15bd3a : SUCCESS in 57m 23s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/00c2d608affe47d189941db67ed01197 : SUCCESS in 1h 03m 43s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/78c4c0a778c140d5bc5789b2dc338fa2 : SUCCESS in 1h 08m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/575e0311a047483bbe032bf240ae6d55 : SUCCESS in 31m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ba6054a64db4e18a4b71029de909aef : SUCCESS in 41m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0758e23de46b42db870a440cecb531b3 : SUCCESS in 34m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42da027dd2e046e4a33b6933a95b33a0 : SUCCESS in 16m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2f0c1adb7e443cda177bff7c437ed06 : SUCCESS in 25m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e60ca3ad42fd41229e1101692ee21470 : SUCCESS in 26m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5d8424e35ce4a8ea2b33477c0019c72 : SUCCESS in 31m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92f555f0016b46ddb2aa8dc33680ecd7 : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3a99951f0a3f4837b602ef1691c1b338 : SUCCESS in 51m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/49b99a8504e3429f96ac07a6f19bfe8f : SUCCESS in 1h 36m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/68eb55ac00b048efbd4a9450815b2be1 : SUCCESS in 1h 01m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/097fd9714e0445e78c74e0ada4d1fb1d : SUCCESS in 1h 01m 27s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b82470a5b9b744fab1fe81c9f6225047 : SUCCESS in 1h 01m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3e5b2eb67917432c8e0d81be662a3cac : SUCCESS in 1h 25m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c6ebdaadfb5e4e789fd466cb4373dc85 : SUCCESS in 1h 21m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af91553af8c94c1799ea1c4d89aeebcf : SUCCESS in 1h 03m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/188d20e9f0c242faad70221c292a41f2 : SUCCESS in 1h 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e2046cb5e75b2524ae9b2b1cbec7d63baf3aff60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 04:49:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"254542aa131adb3f24c140ffea99b8b49e7184a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 06:25:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9e6116bfa1384f7ca348f4105a0cb2b9 : SUCCESS in 50m 18s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/c8411685cac1466bbf2c1aa7d2b78094 : SUCCESS in 1h 00m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/7d505b1abb2946489349234c5b0c71b6 : SUCCESS in 53m 23s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/89702ad07e1549d79a297f81e3ea33c5 : SUCCESS in 55m 08s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0786a0dbae954f628db6bbf56a2254af : SUCCESS in 51m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c11cef3db7b4ba295945586b2ec5c96 : SUCCESS in 30m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838e87e2a4284ff798ee82e7e2bbaf84 : SUCCESS in 15m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f50155a135764fe2ba8c33dabd6465e3 : SUCCESS in 26m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90336bba53f74d26ad611d230abc59fd : SUCCESS in 34m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9958cf93d3df41828f538a0eaa4eac10 : SUCCESS in 10m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6e0749d5e18488fb2398b8ffe32d1f1 : SUCCESS in 56m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bba18db3d8bb427291ef3c8051b4b2fe : SUCCESS in 1h 29m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/023707938a894436a05716401ab7200e : SUCCESS in 1h 22m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/45161299407e4ac2a404654f10017cd1 : SUCCESS in 1h 02m 04s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3c8783a20b2f4f08bee7f363f44a59bc : SUCCESS in 1h 00m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/928ea3f3f4d74b9f999bf70d332b88a3 : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":3},{"id":"4b3103e5f28851857d75faa2192c355de6106254","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 06:25:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a498ee7edba3ec2aa55c1e19dbac54ab8617000b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 06:27:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94e0ca05686f45789821dd68afbc2271 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c4618857b0249535eeed28f0c7a0abf5dbdbc9d0","revisions":{"7f99f1f5fb8fe363adc06817c37062132a2fe650":{"kind":"REWORK","_number":1,"created":"2022-04-08 14:47:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/837140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/837140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/837140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/837140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/837140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/837140/1"}}},"commit":{"parents":[{"commit":"4f9313fca1f9c6809fb49d652e5319b9d5f566ae","subject":"Merge \"Add tag to port more earlier\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f9313fca1f9c6809fb49d652e5319b9d5f566ae"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-08 12:31:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-08 14:43:08.000000000","tz":120},"subject":"Add ADMIN role and remove scope from the check strings","message":"Add ADMIN role and remove scope from the check strings\n\nThis patch should finally complete phase 1 of Secure RBAC community\ngoal [1] where it was agreed that there will be just 4 personas:\n* ADMIN\n* PROJECT_ADMIN\n* PROJECT_MEMBER\n* PROJECT_READER\n\nSystem scope personas will be added in next phases of the community\ngoal.\nTo complete phase 1 in Neutron, this patch removes hardcoded system scope\nfrom the check strings in system scope personas - that way it can be\ncontrolled by the config option and enforced by oslo_policy only when\nenabled.\nIt also skips all SYSTEM_MEMBER and SYSTEM_READER related unit tests. We\ncan re-enable them in the future where we will have those personas\nworking in Neutron.\n\nChange-Id: Iafa8ec1d9710ff404450ad0a64baf56a0f993756\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f99f1f5fb8fe363adc06817c37062132a2fe650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f99f1f5fb8fe363adc06817c37062132a2fe650"}]},"branch":"refs/heads/master"},"35637470bcbb334a46e2a7f935ae741accaa936b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-04-08 14:48:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/837140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/837140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/837140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/837140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/837140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/837140/2"}}},"commit":{"parents":[{"commit":"4f9313fca1f9c6809fb49d652e5319b9d5f566ae","subject":"Merge \"Add tag to port more earlier\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f9313fca1f9c6809fb49d652e5319b9d5f566ae"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-08 12:31:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-08 14:48:27.000000000","tz":120},"subject":"Add ADMIN role and remove scope from the check strings","message":"Add ADMIN role and remove scope from the check strings\n\nThis patch should finally complete phase 1 of Secure RBAC community\ngoal [1] where it was agreed that there will be just 4 personas:\n* ADMIN\n* PROJECT_ADMIN\n* PROJECT_MEMBER\n* PROJECT_READER\n\nSystem scope personas will be added in next phases of the community\ngoal.\nTo complete phase 1 in Neutron, this patch removes hardcoded system scope\nfrom the check strings in system scope personas - that way it can be\ncontrolled by the config option and enforced by oslo_policy only when\nenabled.\nIt also skips all SYSTEM_MEMBER and SYSTEM_READER related unit tests. We\ncan re-enable them in the future where we will have those personas\nworking in Neutron.\n\n[1] https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#phase-1\n\nChange-Id: Iafa8ec1d9710ff404450ad0a64baf56a0f993756\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35637470bcbb334a46e2a7f935ae741accaa936b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35637470bcbb334a46e2a7f935ae741accaa936b"}]},"branch":"refs/heads/master"},"c4618857b0249535eeed28f0c7a0abf5dbdbc9d0":{"kind":"REWORK","_number":3,"created":"2022-04-11 10:48:05.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/40/837140/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/837140/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/837140/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/837140/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/837140/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/837140/3"}}},"commit":{"parents":[{"commit":"4f9313fca1f9c6809fb49d652e5319b9d5f566ae","subject":"Merge \"Add tag to port more earlier\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f9313fca1f9c6809fb49d652e5319b9d5f566ae"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-08 12:31:57.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-11 10:48:05.000000000","tz":0},"subject":"Add ADMIN role and remove scope from the check strings","message":"Add ADMIN role and remove scope from the check strings\n\nThis patch should finally complete phase 1 of Secure RBAC community\ngoal [1] where it was agreed that there will be just 4 personas:\n* ADMIN\n* PROJECT_ADMIN\n* PROJECT_MEMBER\n* PROJECT_READER\n\nSystem scope personas will be added in next phases of the community\ngoal.\nTo complete phase 1 in Neutron, this patch removes hardcoded system scope\nfrom the check strings in system scope personas - that way it can be\ncontrolled by the config option and enforced by oslo_policy only when\nenabled.\nIt also skips all SYSTEM_MEMBER and SYSTEM_READER related unit tests. We\ncan re-enable them in the future where we will have those personas\nworking in Neutron.\n\n[1] https://governance.openstack.org/tc/goals/selected/consistent-and-secure-rbac.html#phase-1\n\nChange-Id: Iafa8ec1d9710ff404450ad0a64baf56a0f993756\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4618857b0249535eeed28f0c7a0abf5dbdbc9d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c4618857b0249535eeed28f0c7a0abf5dbdbc9d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
