)]}'
{"id":"openstack%2Fneutron~837488","triplet_id":"openstack%2Fneutron~stable%2Fxena~I80b0616c335134a564361137b2a00ff86dcbdf1c","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1965294-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-20 10:48:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I80b0616c335134a564361137b2a00ff86dcbdf1c","subject":"[API] Return 403 for POST requests when user is not authorized","status":"MERGED","created":"2022-04-13 06:01:07.000000000","updated":"2022-04-20 10:50:55.000000000","submitted":"2022-04-20 10:48:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837488-bug/1965294-stable/yoga-stable/xena","cherry_pick_of_change":837487,"cherry_pick_of_patch_set":1,"meta_rev_id":"122f7dcdc87e5f82623ac1420198adabd784ce60","_number":837488,"virtual_id_number":837488,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-20 10:48:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-04-19 14:03:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-04-20 09:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-04-20 09:22:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-13 06:38:08.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-04-13 06:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-13 07:27:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-18 05:01:06.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2022-04-19 14:03:59.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-20 09:22:24.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"b6c39fc9032f09c32cc654bd991c249e6fcb6a05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-13 06:01:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"6eea1c3fe63add5aa69ca527fb5cce4d99a1b5ed","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-13 06:38:08.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/837488/1 : FAILURE in 35m 58s","accounts_in_message":[],"_revision_number":1},{"id":"624f3b63dd1e765793722052cc0ed2dbdd9e88b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 06:59:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c522ce2e637043bd876b513c8e82c758 : SUCCESS in 55m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/17afe1c92c0c48e192140c64a0482afc : SUCCESS in 40m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5fe81c3952ac96399ab9a295b9844f84f149dbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-13 07:27:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/cceac4e36952446194ac5fbe33660a7c : FAILURE in 11m 13s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/a7e85b61c1a94672b68a78ee490bf549 : FAILURE in 9m 20s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/3c94a258788f4b4ab650b7ba85f6c72d : FAILURE in 10m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/df2ecdb0c1904c0bb838caf36d23b82f : FAILURE in 8m 52s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/081ca59a09c846e285e1680cc2744ce2 : FAILURE in 8m 25s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/cae3b33a66234f368060d4cbc2f8b6ec : FAILURE in 10m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ae7822e307b470eb0e983013f9910ad : SUCCESS in 38m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb29a5ba2ed44b88b67595f5d16ab26d : SUCCESS in 12m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28e344b3cb5e43c8b45c5a1cf06a4ba2 : SUCCESS in 32m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36b7b92a5b43475bb24f4c1380075a5b : SUCCESS in 21m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85864f411139428d9069281206e3e352 : SUCCESS in 31m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cdd5d763b7b4597953a9f778023e557 : SUCCESS in 9m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c3a62370233a47969f71c4a37ec6f2f5 : SUCCESS in 58m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b9f1e0f76d641b9afb0a313fe71446a : SUCCESS in 1h 19m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4f989412db204b4daea70b805887f464 : FAILURE in 13m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/688ba7a249684d29aff2b6bd17f04e59 : FAILURE in 15m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5e0f5058b35f40cf827e0d79838f5165 : FAILURE in 24m 32s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c67a69840bde4b7f8b630f6d92da669a : FAILURE in 14m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5adaa8dcffd34a639f17c11dbb84e3a5 : FAILURE in 14m 57s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cdb93d740aa4b0c9c1eea8806cc92ef : FAILURE in 18m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/05acb85e22304e99b5ae13b94d8ee57a : FAILURE in 16m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3872adb710465c61c80fea218407ccbf56f7abc4","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-18 05:01:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95bd8ddebf5554ead131779e356650bf6886b709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 06:26:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/9f33e0a3660749fe85cfef916c052cbd : SUCCESS in 38m 08s\n- neutron-tempest-plugin-scenario-linuxbridge-xena https://zuul.opendev.org/t/openstack/build/0158abb715c1484d9d45e15b61015883 : SUCCESS in 1h 10m 54s\n- neutron-tempest-plugin-scenario-openvswitch-xena https://zuul.opendev.org/t/openstack/build/a14f1679453d4319b0c5e7fd0d936462 : SUCCESS in 53m 33s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-xena https://zuul.opendev.org/t/openstack/build/897a27a6259642969ba3cb5db2e88665 : SUCCESS in 48m 24s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/3ee37a3ddd4543248c22f237794b06f5 : SUCCESS in 1h 02m 33s\n- neutron-tempest-plugin-designate-scenario-xena https://zuul.opendev.org/t/openstack/build/980eafa2dc674e3fa97ca28a387b4144 : SUCCESS in 30m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a29d2d26bd24417960bc5f30190c64d : SUCCESS in 35m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c257f39a1ab1414299a0c7f5f84a8c79 : SUCCESS in 11m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9643aa3ff870464ab8464d9fa17cb4a2 : SUCCESS in 23m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c9e9d4781d8496e9bba11102b82be5d : SUCCESS in 26m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44aa163b04be47bf938574e5242201eb : SUCCESS in 25m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7fc1d3e35f0498f82d5b1f01f2ab285 : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d20937cea40401986bd256f16fcc44d : SUCCESS in 41m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/db8f626d03e64fec9a1f1e6da7326cde : SUCCESS in 1h 19m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/069f60a62c2c462580f42d2be00b963f : SUCCESS in 1h 03m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d988f82adfed4953bf9f92f05a646f77 : SUCCESS in 59m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e8d73bd3b9fe4e57bb5a2b83402e0045 : SUCCESS in 56m 23s (non-voting)\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5997e53aa01841a2adc514d26af2cde3 : SUCCESS in 1h 05m 52s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a27c949b61b24b6cbdb572c3268dd049 : SUCCESS in 1h 06m 47s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ee1d1675bde4e75913060b8182ce13e : SUCCESS in 52m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a140217324904162b4ac256d24b5d212 : SUCCESS in 1h 04m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffc727fd4250f3ef639adbf080076a37511030ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-18 06:29:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14a0d859a8b24d7e977b65c6ca9ee8cd : SUCCESS in 40m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33eed13f08a34948b00e890d17324f24 : SUCCESS in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f85738315c78264335f74c26dc411b1a43b6d16","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-18 07:10:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/837488/1 : SUCCESS in 2h 08m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e2ed6c68caa248c8013c566b216269178aaf8311","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-19 14:03:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b680c7adaf90b39b29bedd94f5d69f2617ef1300","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-20 09:22:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36b8430fff7bbf088ff8e0c2c6fca059106e2b65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 09:23:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f256329ddb85966b0172458e04083d2c3dc1ca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 10:48:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/af5e5a696351425bb6a9e456c787ee95 : SUCCESS in 38m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eb1d0daab67438a9ed8212f5d4b525a : SUCCESS in 11m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21f51909bd3840a088bfed882b222632 : SUCCESS in 24m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79f442e175b14d9fbd64cc0e6cb79877 : SUCCESS in 22m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab118962ba2c4426ba1356d0689f510d : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5b68084ce36496d98c6a4bcd11b1155 : SUCCESS in 49m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3ba928137095463ca3593ad4a01e0534 : SUCCESS in 1h 22m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/43a4b4c45a3e4b86ae0bfd638c3eacfe : SUCCESS in 1h 04m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/10c612b9fdf544cca298f06d6cc2edc5 : SUCCESS in 57m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a7df3ae83b54a3c9ceb401eff592ec0 : SUCCESS in 54m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d4c4b9ff547fa5520b730cdcab671a85c8538412","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 10:48:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"122f7dcdc87e5f82623ac1420198adabd784ce60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-20 10:50:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a701c533aa445d3838d3270288e8e57 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"909f3bd3c5585e68da62b511af24f862c7863d6b","revisions":{"909f3bd3c5585e68da62b511af24f862c7863d6b":{"kind":"REWORK","_number":1,"created":"2022-04-13 06:01:07.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/88/837488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/88/837488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/837488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/837488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/88/837488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/88/837488/1"}}},"commit":{"parents":[{"commit":"afe3ba73ea1b2c3c244050d672c23134fa4bc08c","subject":"Force security_group_id uuid validation of sg rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/afe3ba73ea1b2c3c244050d672c23134fa4bc08c"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-03-17 13:33:41.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-04-13 06:01:07.000000000","tz":0},"subject":"[API] Return 403 for POST requests when user is not authorized","message":"[API] Return 403 for POST requests when user is not authorized\n\nIn the policy_enforcement module if policy.enforce() will raise\nPolicyNotAuthorized exception, there is additional check if user is\ntrying to modify own or someone else resource. In case when user is not\nallowed to show resource even, error 404 is raised to \"hide\" any\ninformation about requested resource.\nBut that was also the case for POST (create resource) requests and 404\nerror when user is trying e.g. create network is confusing.\nSo this patch modifies that logic and in case of \"create_\" actions it\nwill return 403 if user was not authorized to do such operation.\n\nCloses-Bug: #1965294\nChange-Id: I80b0616c335134a564361137b2a00ff86dcbdf1c\n(cherry picked from commit 60bc6c7a992383cecaf7dcf425668a6ea92b151b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/909f3bd3c5585e68da62b511af24f862c7863d6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/909f3bd3c5585e68da62b511af24f862c7863d6b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
