)]}'
{"id":"openstack%2Fneutron~839201","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I0c4d492887216cad7a8155dceb738389f2886376","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/1957185","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-04-28 12:25:23.000000000","reason":"Rodolfo Alonso replied on the change"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-04-29 13:48:26.000000000","reason":"Change was submitted"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-29 13:48:26.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-04-28 12:01:00.000000000","reason":"Lajos Katona replied on the change"}},"hashtags":[],"change_id":"I0c4d492887216cad7a8155dceb738389f2886376","subject":"[OVN][port-forwarding] Add lb also to logical_switches","status":"MERGED","created":"2022-04-25 12:55:56.000000000","updated":"2022-04-29 13:51:04.000000000","submitted":"2022-04-29 13:48:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"839201-bug/1957185","meta_rev_id":"46625fffb53d33a4eb35cca94ff75b65374ad215","_number":839201,"virtual_id_number":839201,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 13:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-04-28 12:25:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-04-28 13:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-04-28 12:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-04-28 13:19:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-25 14:08:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-25 15:01:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-25 15:03:44.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-04-26 07:17:04.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-26 07:17:04.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-04-26 07:17:04.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-04-28 13:19:11.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"8aa7dd51d0a13d0687a2d7bacd9a1b1163a23b37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-25 12:55:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2105bc97c51f0041935eef3bcae4e09e3574a99d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 14:08:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa83be80b69f421fab981cf40bcaa921 : SUCCESS in 1h 10m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71163c83d67c46f090e05674558a4dd2 : SUCCESS in 1h 09m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9bdd46977863a58666916a2fca7478c6e3aa980d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 15:01:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/adba20dc7d8a41eb8d2bdf0b56f703e2 : SUCCESS in 50m 09s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/b344233b4d8f41048bf878ef3a5a7d18 : SUCCESS in 55m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ab4cc54af49491d988ae74d1fa49336 : SUCCESS in 42m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/997da5c6512341cbbaf0b141454d7764 : SUCCESS in 45m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f046e779c65141f7959d3a70bdaa4bf7 : SUCCESS in 14m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b95e04a02ae846119587937e4cc311d7 : SUCCESS in 28m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4129edf8d39e4881a95cef658fae941d : SUCCESS in 26m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a8a43fb1b34f7ea83d8310e6b3c582 : SUCCESS in 10m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/61197389f87e48acb676b266d5c822be : FAILURE in 42m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4e35c5dd87b0474b8da4f1cf80196b4c : SUCCESS in 2h 03m 09s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2db307eff5c540a5a840b83fd7e2e30f : SUCCESS in 52m 26s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2966247817ee4eaabfbf7ae9cfeaad85 : SUCCESS in 1h 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bacc6bcf075d1e273637f0c3d3ec02ad884eae1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-25 15:03:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/839201/1 : SUCCESS in 2h 07m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ad71584424fc06041a862b6f5d28a8030321f43b","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-25 15:03:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dfee8ccbe1e603c206f9794ee9072ff9bb6b342e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-25 15:05:08.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/839201/1 : FAILURE in 38s","accounts_in_message":[],"_revision_number":1},{"id":"a4ec0b6993c4a09a2939ef2f8debbc2ed5a84ccc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 15:49:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e1f9ea07ab64a2096f5ee8ef324db18 : SUCCESS in 43m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a62e6364a2d494b9dac4d8e3002a17b : SUCCESS in 43m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dae74477ac143ed7a72cd6e516a475bc4441b2ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 16:38:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/cd89ba6f63aa44eb90a765a454ec6824 : SUCCESS in 55m 22s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/97119e70888940539ea8f102b122885d : SUCCESS in 57m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5dbc1a8477940caabeeb685b8cd8b67 : SUCCESS in 37m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75f5e3d2116043a495b6de2f55b629b2 : SUCCESS in 29m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1779f8e8e92c46fda7ce113e417398cf : SUCCESS in 11m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97afa008b210491f8b424d22a9bd9e0a : SUCCESS in 27m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af0d6d00045e4fe9aa4397ee75e69ad7 : SUCCESS in 28m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b139450d1741467db9f9cf3486c58738 : SUCCESS in 9m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/21939ca3f28546078b7b0f06d52a703d : FAILURE in 40m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ecff804bd4bd47049635e5f4754f7a2c : SUCCESS in 1h 27m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff46f96324e142bb937d8865851d9153 : SUCCESS in 48m 38s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f987ecf730bc4683b219c2742bbc007d : SUCCESS in 1h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3501077dc2976eae874fb7d182c87f8ceff65e52","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-26 05:36:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02db5e8329aaeb41a60159d77588c066bae8ca0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 06:38:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/720a4ba07a4044e48fbcfed71075df7a : SUCCESS in 45m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2a4cba6312942bc89ee07bca4d97369 : SUCCESS in 46m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ba26f3c6aa4ca009eb9a352755384fb88c54883","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-26 07:05:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/05aa54da6f8a4d488ff7803f81356b2d : SUCCESS in 31m 24s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/6097d1a9f1cc4e0ba799403ff43179f6 : SUCCESS in 57m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e74444bb25345b0abf7fb6774d4a10e : SUCCESS in 40m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75ee6b56841a4d4d8f1a3f565770b8b5 : SUCCESS in 33m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a447130ae4af41089dcefc350beee5ed : SUCCESS in 14m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/005e47da1a374596bbe1d4baf08bb129 : SUCCESS in 30m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ea1bd068c5847c78df97967610d3460 : SUCCESS in 25m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50992ecfb704ab8a2b9c9dcca80e343 : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba161c3df33248b99ba02631ad94df3c : SUCCESS in 54m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f99e394af0f437c9c4d527b35614a41 : SUCCESS in 1h 27m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/190c7632da9d4d339477417a5dedf0a1 : SUCCESS in 44m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/999b1867052a412685e5901e84fb4b95 : SUCCESS in 1h 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7f25eee458bd5be9f47d28c3bf718f7b4900ec2f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-26 07:25:16.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/839201/1 : SUCCESS in 1h 48m 36s","accounts_in_message":[],"_revision_number":1},{"id":"72756aa4c4f4a14111607806b56ac929d82b56d1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-04-28 12:01:00.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57c8bc7548331ade3e38a8d742019a8ef7e8788d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-04-28 12:04:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0cbd43e32a6b663c9276f1fd391c3720047be6a8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-28 12:25:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f6531c4a4709bc4f3ebc94b397ae1fad5aea24b5","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-04-28 13:19:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da5d7a85252ee49939836d33b21c309ed4ba0efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 13:20:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"502fefcace3e45fc605d3c640d1cd62af62a998b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 15:13:27.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/e0632dff8d78424eb14cceedea50be9c : SUCCESS in 46m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6eba8f1a9d4368a2d62ced38d97e09 : SUCCESS in 16m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6290c8c7a0704f45b9555fadcbee238d : SUCCESS in 38m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae3a9992712e4f6cb66ad2f10c28f66c : SUCCESS in 26m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27dedcf56fbb4ca783c338c955acd746 : FAILURE in 5m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8d9c4d491f0d44388a734eee3dbec333 : SUCCESS in 54m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/bfaca207fd0e46c2a032ca18efad1b39 : SUCCESS in 1h 32m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cc1e962b9254447ab0d9c3e481ec5f6 : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":1},{"id":"60355389a884777692efcc4d706d17ffd533cc1e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-29 10:44:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c116aa8088a49971298ab482836541b651df184","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:02:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/884bd0f621424460bd16b8710c9673c8 : FAILURE in 1h 14m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/685834cd9512434aa447b68a8bc684a4 : FAILURE in 1h 16m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e351c660727d32870f9e46b55367e3e46390cad3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:15:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/76747e71d6d74ac9b3d60c50652f4689 : SUCCESS in 48m 13s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/02ed282ffcc246fbad64af5ff661d9e5 : SUCCESS in 59m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7c43664cfe84735828cecf67e5f8d68 : SUCCESS in 37m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbe555148c8a423ebf3c266abce39cd4 : SUCCESS in 11m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1ba6bb8c7ac4cd0919d655b31d5ed5f : SUCCESS in 24m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b415113ff2d43818580c0cec53c17ac : SUCCESS in 22m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a4b8dae50ae4c1b935cd98db74b2390 : SUCCESS in 10m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f427ec568ad4d0ea1433cf9741d3926 : SUCCESS in 49m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/193a17c7c40d43fe9e86f65e25410264 : SUCCESS in 1h 28m 51s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/617fae70dac9408a9d680146ce06671c : SUCCESS in 57m 01s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7ca9e04e2ee24ca3a3ee3f3e0f3cab97 : SUCCESS in 1h 05m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58863413771d2563fd94b93efefc1bb23652d56e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:16:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"21eea79ec2826861a8e914ffc24c897489286d86","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-29 12:44:20.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/839201/1 : SUCCESS in 1h 59m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0537c111cce3007ed29cad1b8f5313676baf7a33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 13:48:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/381a550c1e074682a0064a6d3a1c179f : SUCCESS in 1h 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2197399dc3584fe9a900e4e3caed1987 : SUCCESS in 16m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab8712d764b149fc9ebb54a9baedbc1d : SUCCESS in 22m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51c85cad2270447881a47192f0d6c6fc : SUCCESS in 23m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6b1d196c774615964bca432076bb3a : SUCCESS in 9m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/009ce4f6e2fc42d99cf0d5430c24847a : SUCCESS in 49m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/454ccfc0388e4a1ba53e7317e573bae7 : SUCCESS in 1h 31m 39s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/198313b23b374425acc8a9a8bc1e928b : SUCCESS in 59m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e40ceb6c5b20873473a5054211f17f90c70a6160","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 13:48:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"46625fffb53d33a4eb35cca94ff75b65374ad215","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 13:51:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6766ce97e47465e9047b4756259adb1 : SUCCESS in 2m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82671b55024ddff8b9593ca9a76898c67468fcea","revisions":{"82671b55024ddff8b9593ca9a76898c67468fcea":{"kind":"REWORK","_number":1,"created":"2022-04-25 12:55:56.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/01/839201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/839201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/839201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/839201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/839201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/839201/1"}}},"commit":{"parents":[{"commit":"0f84be37204dc837d4b2908cba5dca9f129b8bc1","subject":"Merge \"Force security_group_id uuid validation of sg rules\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0f84be37204dc837d4b2908cba5dca9f129b8bc1"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-03-14 12:19:51.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-04-25 12:54:22.000000000","tz":330},"subject":"[OVN][port-forwarding] Add lb also to logical_switches","message":"[OVN][port-forwarding] Add lb also to logical_switches\n\nCurrently we are adding lbs to logical_routers, but\nto get the lbs accessible from the vms we need to\nadd lbs also to all the logical_switches connected\nto the logical router as suggested in the Related\nBug rhbz#2043543.\nIf in future ovn automatically handles addition of\nlbs to logical_switches if they exist on associated\nlogical_router, then we can remove this handling\nfrom port-forwarding service plugin.\n\nAlso subscribed to router_interface\u0027s after create\nand delete events to handle the case when logical\nswitches are added/removed from router after intial\nport forwarding create.\n\nRelated-Bug: #1957185\nRelated-Bug: rhbz#2043543\nChange-Id: I0c4d492887216cad7a8155dceb738389f2886376\n(cherry picked from commit 2492cf2e0703dbe6fcfe26e199505b7b43a65f50)\nConflicts: neutron/services/portforwarding/drivers/ovn/driver.py\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82671b55024ddff8b9593ca9a76898c67468fcea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82671b55024ddff8b9593ca9a76898c67468fcea"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
