)]}'
{"id":"openstack%2Fneutron~839714","triplet_id":"openstack%2Fneutron~master~I4502e9e45421457a1eeb0263eaef55fcd79d6d33","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-04-28 10:28:31.000000000","reason":"yatin replied on the change"}},"hashtags":[],"change_id":"I4502e9e45421457a1eeb0263eaef55fcd79d6d33","subject":"[DNM] Test multinode ovn grenade skip stop_ovn","status":"ABANDONED","created":"2022-04-28 10:22:53.000000000","updated":"2022-05-31 07:41:56.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"af89a9a8c9a22eb9247889a3f12ef401e3c058f5","_number":839714,"virtual_id_number":839714,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-04-28 11:21:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-28 10:23:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9e4c83339c115e14d444f02e1c992d456eaaef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-28 10:22:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a065522030b7d06f715e13eaaf8974da5219dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 10:23:29.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/devstack for 837303,1","accounts_in_message":[],"_revision_number":1},{"id":"ff9e59a2c7ba642ee1b07532f97f2afe1bd23289","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 10:23:36.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/devstack for 837303,1","accounts_in_message":[],"_revision_number":1},{"id":"78429dc3602f0b7998c312f28fbde2c30be1e7df","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-28 10:28:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebdff166a8ace79d9525c188d55331e71c741f95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 11:21:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-ovn-grenade-multinode https://zuul.opendev.org/t/openstack/build/a050a2a06d954187b3a5742f83b8c029 : SUCCESS in 50m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e4157f9bf3825723f6ca97ae6e6204a7030d4fb4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-05-31 07:41:44.000000000","message":"Topic ovn-grenade-fixes removed","accounts_in_message":[],"_revision_number":1},{"id":"af89a9a8c9a22eb9247889a3f12ef401e3c058f5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-05-31 07:41:56.000000000","message":"Abandoned\n\nNot needed, moved testing to other patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8580287102a30ea6ce13048d2ec0ea81da4e8830","revisions":{"8580287102a30ea6ce13048d2ec0ea81da4e8830":{"kind":"REWORK","_number":1,"created":"2022-04-28 10:22:53.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/14/839714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/14/839714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/839714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/839714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/14/839714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/14/839714/1"}}},"commit":{"parents":[{"commit":"720a1c3de9a557478ed93d591f07276ae8df21f3","subject":"Merge \"ovn migration: Turn validations off by default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/720a1c3de9a557478ed93d591f07276ae8df21f3"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-04-01 11:17:26.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-04-28 10:15:54.000000000","tz":330},"subject":"[DNM] Test multinode ovn grenade skip stop_ovn","message":"[DNM] Test multinode ovn grenade skip stop_ovn\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/836191\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/837303\nChange-Id: I4502e9e45421457a1eeb0263eaef55fcd79d6d33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8580287102a30ea6ce13048d2ec0ea81da4e8830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8580287102a30ea6ce13048d2ec0ea81da4e8830"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
