)]}'
{"id":"openstack%2Fneutron~839740","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ic8f6472fe708d12f8fc129729465367083e1095a","project":"openstack/neutron","branch":"stable/xena","topic":"fix-docs-use-uc-stable/xena","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-04-29 10:18:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic8f6472fe708d12f8fc129729465367083e1095a","subject":"Docs build: Add requirements.txt to deps","status":"MERGED","created":"2022-04-29 06:57:43.000000000","updated":"2022-04-29 10:20:00.000000000","submitted":"2022-04-29 10:18:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839740-fix-docs-use-uc-stable/xena","cherry_pick_of_change":839860,"cherry_pick_of_patch_set":1,"meta_rev_id":"957e34a55eda0d03d366df656d14ee48e68eaf39","_number":839740,"virtual_id_number":839740,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 10:18:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-04-29 08:40:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-04-29 08:35:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-04-29 08:40:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-29 08:01:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-04-29 08:25:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-29 08:35:45.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-04-29 08:40:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-04-29 08:44:26.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"}],"messages":[{"id":"1353012a8a113e575f83e78b0a8bbec55e89503b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-04-29 06:57:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"f2756f7fbfafb4fcf8f1f3eb3a67f0aceb69e490","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 08:01:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/08a9e97374c44b649b58b13b307c6e91 : SUCCESS in 48m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6c6181974d145abbe50245fdb2bba02 : TIMED_OUT in 1h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18a1f56aa933ca507272bbe68c16a880f96052a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 08:25:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79b7ca36065b4851afadae3a59db6607 : SUCCESS in 39m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e607fae3bb4048ac9f2a10aa70502ffb : SUCCESS in 16m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15e2d55b007040978d7f5ff19df46f0b : SUCCESS in 31m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56e2a0347f5c45b5b1f07d464f1de667 : SUCCESS in 21m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/831c378bd10b4ae3afb02ec55760e3fc : SUCCESS in 21m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d08f8c8404410a9d1dccdee8543acb : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f32986483eed48b88e8a46146cec8187 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/18cec878fd1f45feab31b251da4f61fb : SUCCESS in 41m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/943b4e03598f409f817d86db3baf78a7 : SUCCESS in 1h 22m 42s","accounts_in_message":[],"_revision_number":1},{"id":"1818612b5333a36a36e37768082639ce143271ed","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-29 08:35:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"140a5f073a02f67cd42b6561a90c71466d45fe0c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-04-29 08:40:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d271282049fbbffd279335deb6a1e3a0bb615e5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 08:40:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43f1c0b2aa79b7c9cad617f53130d6a3ea18b87b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-04-29 08:44:26.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/839740/1 : FAILURE in 1h 45m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4a3ca58d8712f473c80260051d9647d27933de52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 10:18:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceca328e0ddc4f3fbf98cd9a66b122f3 : SUCCESS in 15m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f197fb841e34c0e804245d03b2f8805 : SUCCESS in 25m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e8c5b0c7be7430d9424bab82c1653c4 : SUCCESS in 24m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3763b61f04f04203bfe406bddfdb6de4 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43ad0f9352f340e4b1cb4d0110fb7468 : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7569db46088d4818ab00b551c1ad3a04 : SUCCESS in 39m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/41c5baaaf57b4e64933ae4d234133da1 : SUCCESS in 1h 11m 24s","accounts_in_message":[],"_revision_number":1},{"id":"012e21965ad54434de472f39e9fee525a5453a4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 10:18:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"957e34a55eda0d03d366df656d14ee48e68eaf39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 10:20:00.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf36d1bc80a43ab957688635a7d6e63 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09d4a7a58aab462abe559a7151220423 : FAILURE in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15c078cad5f009ee21e9f78797965a9de1cab880","revisions":{"15c078cad5f009ee21e9f78797965a9de1cab880":{"kind":"REWORK","_number":1,"created":"2022-04-29 06:57:43.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/40/839740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/40/839740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/839740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/839740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/40/839740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/40/839740/1"}}},"commit":{"parents":[{"commit":"753f86d0758c42a93c1ea62708b0bd868fe83ea9","subject":"Merge \"[DHCP] Break reference chain to any Exception object when resync\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/753f86d0758c42a93c1ea62708b0bd868fe83ea9"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-04-28 14:47:54.000000000","tz":330},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-04-29 06:57:43.000000000","tz":0},"subject":"Docs build: Add requirements.txt to deps","message":"Docs build: Add requirements.txt to deps\n\nSince we use usedevelop\u003dTrue, as a result neutron\ninstallation happens at \u0027develop-inst\u0027 phase and it installs\nrequirements from requirements.txt without considering\ntox constraints, This can lead to issue where broken\nrequirements are installed in docs job and as a result\nit will fail.\n\nIn order to get constraints be considered for this case, add\nrequirements.txt to deps for docs env.\n\nAlso squashed below commit:-\n\nDocs build: Add NOTE to not remove requirements.txt\n\nTo avoid unintentionally remove requirements.txt\nfrom deps list from docs env, add a NOTE.\n\nChange-Id: I6ec150a8f31bf841f1c8829719a0f08df531e53a\n(cherry picked from commit 2a8187124b104c32eea5d7730b187897b504f5b9)\n\nChange-Id: Ic8f6472fe708d12f8fc129729465367083e1095a\n(cherry picked from commit f15baf16dd18969efb94719798f5d4a731b0c4fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15c078cad5f009ee21e9f78797965a9de1cab880"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15c078cad5f009ee21e9f78797965a9de1cab880"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
