)]}'
{"id":"openstack%2Fneutron~840087","triplet_id":"openstack%2Fneutron~master~If17736114adedf44b502fd7c91423a7822d7170b","project":"openstack/neutron","branch":"master","topic":"drop-l-c","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-05-02 23:58:47.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-05-02 19:04:09.000000000","reason":"Lajos Katona replied on the change"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-02 23:58:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If17736114adedf44b502fd7c91423a7822d7170b","subject":"Drop lower-constraints.txt and its testing","status":"MERGED","created":"2022-05-01 01:25:45.000000000","updated":"2022-05-03 00:01:05.000000000","submitted":"2022-05-02 23:58:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840087-drop-l-c","meta_rev_id":"cc3773b294b1ec1100256d56feba34b564ef1bdc","_number":840087,"virtual_id_number":840087,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-02 23:58:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2022-05-02 22:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-05-02 19:04:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2022-05-02 22:10:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-01 02:13:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-01 03:07:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-01 03:25:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-05-02 08:10:26.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2022-05-02 12:17:00.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2022-05-02 19:04:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-05-02 22:10:09.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"f9c368cb6ae724ed931ccb278f9aedab2c4c3d8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-01 01:25:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b898ba859988fc2d14a5aab30540815b9aaf097","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 02:13:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca37f9b36e894731addf545c21e981bb : SUCCESS in 44m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4aaaa30134f24a009792d0112b4a5dab : SUCCESS in 45m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c5a5c26ff94a5b28287a4751b089c2701e21a0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 03:07:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/39a8867b3e544604aa66af169213a12d : FAILURE in 41m 42s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/27e2dc36e940429a90ef144aa9f41d84 : SUCCESS in 57m 14s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/331238d00a22499390a8fc2932b651fe : SUCCESS in 1h 04m 12s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/795ba5bda0c94dedb1d863b582b4832f : SUCCESS in 51m 24s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/9e63847659ff4c1e8730fde7193b4fbe : SUCCESS in 57m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4ab40997a14c46ccbcb6927b01a1b7bb : SUCCESS in 27m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/532a191e5c9a41f2a07dfe7c4ce57de6 : SUCCESS in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a6daecbd4cb4989b59f18c77d9c0a80 : SUCCESS in 11m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d7df25888504a9fba52732afebbbd7b : SUCCESS in 24m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac775ba0e8d8457bbbd1da7ec9d20ace : SUCCESS in 20m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4fd3d7198d7a4833ab210e8ee173a7ba : SUCCESS in 28m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03da476e9a1e44a6b7ecf589fe61293e : SUCCESS in 10m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c52985e2ff294548b9fb13405a2af6fb : SUCCESS in 3m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07b7b6826f8a4c688ca2fe8db9b9f60a : SUCCESS in 9m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fbefce1c8a84b9ba27f41f6d1431292 : FAILURE in 3m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d2b4304479a246d8916693418f385405 : SUCCESS in 57m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6f5cad3472d14660891d9ea32baf2778 : SUCCESS in 1h 35m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4fb0d3a2802744fb9e27b9210badec35 : SUCCESS in 1h 16m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/79c330105c934b0ca95d6c93fc2c7e2b : SUCCESS in 1h 00m 49s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/27aaecff32fd413d970da512532ec3b6 : SUCCESS in 1h 08m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/43b3075b2f2c4233b50f7ed7311ac258 : SUCCESS in 1h 22m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/025b2ec049b948d698c79586e6a95818 : SUCCESS in 1h 17m 13s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48927ebed05141dfa731961baaf371b7 : SUCCESS in 1h 05m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/040369894f054b1ca41c23cdfef0ef78 : SUCCESS in 1h 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"370dbb0c4c57b1e3edd35425dcbc504c745d0fbe","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-01 03:25:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840087/1 : SUCCESS in 1h 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"97718b73fb6f376bcb5637129beedd56c39cbe68","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-02 08:10:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9426411c7f0bdb6f44705827c1c50bcc23bac82c","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-02 08:52:30.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b8a7d072a8f88194b23eadd9a895227537327ee0","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-05-02 12:17:00.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"98508db6bd088e0a4f9fdc26816825e24a2846ea","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-02 18:00:46.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1c12afb3c2b2dde7d298d348314fb7b85ca8c6af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-02 18:02:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c314624ec55d4260e07d3ee566388f8bbf7d0df4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 18:46:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d063057a2af94d46a3001e0478cfcb95 : SUCCESS in 42m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2009d4d62e8148bfbbdc8eb6bd8e74b9 : SUCCESS in 41m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"06a96f26a37447db353f84885f88138329ecb002","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-05-02 19:04:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a78b51f179b313b5543a0e63dd7a666807169a4d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-02 19:37:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840087/2 : SUCCESS in 1h 34m 54s","accounts_in_message":[],"_revision_number":2},{"id":"798202b3ed4077737b037d66f584560f064488c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 19:46:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/da1b3441e0aa451ea00a30120b021db5 : SUCCESS in 55m 24s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/32ea960a8b8e403f967f476eec8ebbc4 : SUCCESS in 1h 04m 07s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/447e103619ee4cb5b1f36f349bdd8336 : SUCCESS in 1h 07m 06s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/869067ae4f5645a3a83c231d86a924b7 : SUCCESS in 1h 00m 48s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/c6a119f826c1457ea93fe9724f71361f : SUCCESS in 55m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7b27c93b26e14dee85f274d16515f168 : SUCCESS in 30m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33356043d6034134b02487558ca55753 : SUCCESS in 38m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3788e7ca5794ddf96914feb928e038d : SUCCESS in 14m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7112b667b3e41c0bbf1c08b47ac1136 : SUCCESS in 31m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dea3b33c1e0742fa987eef9225e219bf : SUCCESS in 28m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/207f40c05955427cab883a61ad9c3f7e : SUCCESS in 30m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9abdbdfc61834ce4a9a71571757f7b61 : SUCCESS in 10m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/56534221bd0d4480b7769862f9090116 : SUCCESS in 3m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40ba094e7eda439b8e456522fe842569 : SUCCESS in 10m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/47250c0ff210469baf8410bd8986f8bf : SUCCESS in 59m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0205dc6fe8644da8b037ce607199876 : SUCCESS in 1h 31m 39s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0fa3f610d525407aac97d1cbf9bfef01 : SUCCESS in 1h 13m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4e5a55b225c4839b3c6fd91bbebde4e : SUCCESS in 1h 07m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aac1345893444e438a8483301475c07b : SUCCESS in 1h 05m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7671226aa3d94739bf1b3729b6decab9 : SUCCESS in 1h 26m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ecbf65124c8f415a95da543c951c04b0 : SUCCESS in 1h 35m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6a863ecd49f4a1284963e8f6110f4fc : SUCCESS in 52m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d189644e8cd4585b31435a7ad1ee449 : SUCCESS in 59m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58014c9851d2e6c21b50f564973b81c547fcc39e","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-05-02 22:10:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9026d9d9160e2cd21730d19ec7bafab6d7b85989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 22:11:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cbb4dca3f9601fd5c0156a7a328c8d413e5167eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 23:58:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c8a220b8d1cc4e4db3effa34dae67969 : SUCCESS in 54m 03s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/0570053d62a441dc9d5857a22bab019d : SUCCESS in 1h 12m 42s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/699f6c1ffacd4990881f2110ff405bf8 : SUCCESS in 50m 57s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/0a9b5eb5a34048c5bd7febb6c146530e : SUCCESS in 57m 51s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e07b0202ebf0460e85ae82cd1e4c1047 : SUCCESS in 1h 00m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d9cc1bb065c4a3fb03233fad8adfb6e : SUCCESS in 13m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72482c00fed741e48abbc311b58a2208 : SUCCESS in 20m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94c0a5ef643043fd8ff7ada35430be5e : SUCCESS in 25m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aafa9c920664582a4193e365ad35c41 : SUCCESS in 12m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c421f4a14e84b2a8b287a45a2fcdba7 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/939961c782d14145924b59b8ba3a8572 : SUCCESS in 6m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf06b247942d4377b96c6c76275a5a31 : SUCCESS in 55m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6803765975a94a388a5e23005c71004c : SUCCESS in 1h 37m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3caedfe75fb54691853cb6cd486de4e0 : SUCCESS in 1h 18m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b14c6a8284b542a5b4e96f501eaf38ef : SUCCESS in 55m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fa3db2c0331143b3897bdbca7761ec37 : SUCCESS in 1h 17m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a492e1eb053e4fe3a6fcd7a5b4325f4d : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":2},{"id":"b81e3e09c0379bd8895d2e168a7bff8210c00631","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 23:58:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cc3773b294b1ec1100256d56feba34b564ef1bdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 00:01:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c78016963aa34867ab9db6e93079eb1b : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ade3b01ab154a28b836381a58b27594 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bc8e512b4a7b68f2e5c31efd320d8b347d06bba0","revisions":{"7df31cb36b2371a33ab3960103deefaa932d3263":{"kind":"REWORK","_number":1,"created":"2022-05-01 01:25:45.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/840087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/840087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/840087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/840087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/840087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/840087/1"}}},"commit":{"parents":[{"commit":"42ce0ea42cc345ab0305e344a1f46a8893772e22","subject":"Merge \"[OVN][Migration] More robust checks on stop-agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/42ce0ea42cc345ab0305e344a1f46a8893772e22"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 00:50:16.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-01 00:50:16.000000000","tz":-300},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: If17736114adedf44b502fd7c91423a7822d7170b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7df31cb36b2371a33ab3960103deefaa932d3263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7df31cb36b2371a33ab3960103deefaa932d3263"}]},"branch":"refs/heads/master"},"bc8e512b4a7b68f2e5c31efd320d8b347d06bba0":{"kind":"REWORK","_number":2,"created":"2022-05-02 18:02:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/840087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/87/840087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/840087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/840087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/87/840087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/87/840087/2"}}},"commit":{"parents":[{"commit":"4f0fc0f74e6e9e237f1dd22f18e1bf14c88ce0e3","subject":"Merge \"doc: update peering network BGP guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4f0fc0f74e6e9e237f1dd22f18e1bf14c88ce0e3"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-02 18:01:42.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-05-02 18:01:42.000000000","tz":-300},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: If17736114adedf44b502fd7c91423a7822d7170b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc8e512b4a7b68f2e5c31efd320d8b347d06bba0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc8e512b4a7b68f2e5c31efd320d8b347d06bba0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
