)]}'
{"id":"openstack%2Fneutron~840415","triplet_id":"openstack%2Fneutron~master~Ifd61e379c3cef3589803c96a276da9827051f660","project":"openstack/neutron","branch":"master","topic":"segments","attention_set":{},"removed_from_attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2022-09-02 09:49:36.000000000","reason":"Change was submitted"},"29759":{"account":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"last_update":"2022-09-02 09:49:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifd61e379c3cef3589803c96a276da9827051f660","subject":"ovs: use a local vlan per network/segmentation","status":"MERGED","created":"2022-05-04 07:07:57.000000000","updated":"2022-09-02 09:51:00.000000000","submitted":"2022-09-02 09:49:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840415-segments","meta_rev_id":"98184202d8a9ce20f286d6f919a839ca4336892e","_number":840415,"virtual_id_number":840415,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-09-01 17:31:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-02 09:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-01 16:21:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-09-02 07:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-09-01 16:21:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-01 16:21:23.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":1,"date":"2022-09-01 16:21:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2022-09-02 07:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-04 08:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 08:57:23.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-05-04 12:35:10.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-20 07:38:05.000000000","updated_by":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"reviewer":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"state":"REVIEWER"},{"updated":"2022-07-25 08:56:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-08-10 21:36:16.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-08-30 11:17:11.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-08-31 10:08:59.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-08-31 10:51:59.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"4844074680681b14c9a2d89352967498ca440442","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-04 07:07:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abf1759c0a5830eb60e77e28db70b7b32b9601a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 08:30:51.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/5e1c6b31e5674e46a0b1d742319af2b8 : SUCCESS in 43m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/182ccc9f777a4fd083317395b07510f8 : SUCCESS in 58m 09s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f4050f4d862f4381aac9a814fee905a8 : SUCCESS in 46m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/992dac750807470aa05b593044950935 : SUCCESS in 37m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f34b26824d4245c3b117c4865c1a6667 : FAILURE in 12m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9cd603cd1cb41fa873cd153d35907b2 : SUCCESS in 20m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5538995310c74f06803dce18512bf2da : SUCCESS in 27m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ab576ad2dc648fcb661e7117dac4cae : SUCCESS in 34m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85cedb3bc9734d7cae635c78c13343fa : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a64b2ecbd0574701ba0703bb89da0627 : SUCCESS in 1h 04m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e044bd4b16ec459aa0efdca4b79beb32 : FAILURE in 1h 11m 17s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/527635d66a9d44bdb1a401b0ae84e005 : SUCCESS in 1h 00m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9daf57724c5c4f6faf8798af3700ac3d : SUCCESS in 1h 00m 52s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d73b5e7890f44c4c93f547c10a932c30 : SUCCESS in 53m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5a4eb05553d54e04a61deadf5490756e : SUCCESS in 1h 07m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fee4e620d7b747058388b957fee8920d : FAILURE in 1h 08m 33s","accounts_in_message":[],"_revision_number":1},{"id":"a05e80b0e6395a464855c2035870160fc7d702f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 08:31:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d6f53939f7b4e0da617630c7ef9c25d : TIMED_OUT in 1h 21m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/841520910fd84b99ad16ef9a98fbf5cc : TIMED_OUT in 1h 21m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30b4d475b828a459dbb4ed824aeeaa354d4fbf2a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-04 08:57:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840415/1 : FAILURE in 1h 46m 43s","accounts_in_message":[],"_revision_number":1},{"id":"78ef71be240e57df0b66d815674862db169f2780","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-04 12:35:10.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/15/840415/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c14f95357a6d9a0e0e9c4b1f0328b4d5810ea0cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-04 14:34:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"36b026f14a143ef51b25e2e76f8490733de8853f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 16:00:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/68189c06f4454d5f8ea29acfe8b25ce7 : SUCCESS in 44m 36s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/b751b7888e224b128faad6452263f180 : SUCCESS in 1h 11m 50s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fdbcd24679144bffae85cf0315f28a50 : SUCCESS in 50m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1100efd9cdc44a2d93d9d3d838e69260 : SUCCESS in 50m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f6d6082593b4b76acb0d0dcf1b4679e : SUCCESS in 11m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/add0a3da2248489d98bf160e9b45e5ea : SUCCESS in 26m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72e78dd2812545889067ea8f21ca3e63 : SUCCESS in 22m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/243ae54d855e4d0aa918fc5165bde400 : SUCCESS in 34m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e041ca51ba940f7a340469e4148ba92 : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8262678fb3ea44b4b401838bca8b2f83 : SUCCESS in 1h 01m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3cbfe716e3f140d18bf4c12b347ec035 : FAILURE in 1h 14m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1fc333c385804cb085b3b0ec3fe0cce1 : SUCCESS in 1h 02m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/af1ecddbeb5449edbf7e3207c153e9be : SUCCESS in 1h 05m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2579968984f345839202f5a1db2043ae : FAILURE in 1h 06m 50s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e391a86ac94c497485a6bb211569433e : SUCCESS in 1h 16m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9863cb8a69454cb7a84e7762424e82d7 : FAILURE in 1h 19m 44s","accounts_in_message":[],"_revision_number":2},{"id":"c37b63eec831a77913441161d81d1be5abc88743","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 16:28:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b55c58c505454d5d9c0f161de45f2a8d : SUCCESS in 1h 08m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b200164cf0c94a23a5bef4ab3f851b65 : SUCCESS in 55m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb6113c564c613f0195be5a709bb91e8fb29250a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-04 16:34:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/2 : SUCCESS in 1h 57m 08s","accounts_in_message":[],"_revision_number":2},{"id":"579f2a71f4e89ea24f6d9b2d42f66c032e89ec15","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-05 05:46:35.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/15/840415/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"783f723d803682170bf292cf6c499d3f7e838086","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-11 17:26:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"5853ea540edd009ed9c569d14403b0388d3b2b5b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-11 18:23:23.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/15/840415/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"91f33eb20e117cef09a2d355e69db3d09516cd42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 18:39:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3996a7bfd3e9492593c2bbe078299ebe : SUCCESS in 1h 09m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fcb0bef18b0f4ce094854927f4be5345 : SUCCESS in 1h 09m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"15cdf4a72e99fc518ba1e21a98e09f697c65350b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 19:07:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/df2bddfdd8ec40f1a31356787c28c51c : SUCCESS in 44m 29s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/25bfc4a58a1a4206b0577d48b156cf40 : SUCCESS in 56m 21s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fe551f3293bd49f4b784f6dd263db41c : SUCCESS in 59m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbdd8d59dc3b4455ae7e57e666130900 : SUCCESS in 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b1c8eb9746940a39fdb289b70c7be72 : TIMED_OUT in 31m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69b8352865f1425b90f96b16fe87db6a : SUCCESS in 22m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/873bd6b465784f07b8b8faec625dc3ea : SUCCESS in 29m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e1ec9f7ee91485caaa0a8a8f058deaa : SUCCESS in 21m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4985d95c3a5e486280fcf9cb60742954 : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6b8d7257b0f0403db3756373dec4b656 : FAILURE in 55m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b20a7832b18b4e6bb02482e6605a4026 : FAILURE in 1h 34m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1d655b7fe33343b591cf024fbca2a8da : SUCCESS in 1h 05m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/098f8a5806224aca9203734dddea221f : SUCCESS in 47m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/022accbfb9c1457e939dc0f254cc8919 : FAILURE in 1h 01m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/63f862bcd01d4f1f859f32ac1dd21ddf : SUCCESS in 1h 10m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6cb37bc4f01a491a82ce970d492f72c6 : FAILURE in 59m 50s","accounts_in_message":[],"_revision_number":3},{"id":"3e0d80c84a1a8fb6340b4e901717503f88931c81","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-11 19:41:27.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/3 : SUCCESS in 2h 12m 21s","accounts_in_message":[],"_revision_number":3},{"id":"2e275e4a5cb545be017da731b782f0ee4c520b2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-16 07:11:36.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"66d2e41d9ed9791abb4153bcd449e569b8a22d91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 08:35:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63403a48430d4b388833e2ec1fac1a0d : FAILURE in 1h 19m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9566fdfe76584ce69d670f3f8e3ae8b4 : FAILURE in 1h 20m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a7de79a86353b98c63cecb87d8b58aa22aa0da6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 08:46:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/74e7f3c78e9944569faf2f0f2d898a3e : SUCCESS in 50m 31s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/aac1db94179045688a2f1391bee75e4f : SUCCESS in 1h 18m 27s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/18bb390aab3845d6aa85e63b2500a46f : SUCCESS in 1h 08m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d9cd2bbd68d410482d974f2f92bb8dd : SUCCESS in 41m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c51d244ae14416b8bcb782c0a7282c : SUCCESS in 12m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e95e89f6a0474e31ad8254290715e9a4 : SUCCESS in 23m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81b063ff2acb4818b5ffa889dda531b1 : SUCCESS in 30m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d59de10c2572412b9de3afc931737bc2 : SUCCESS in 23m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93ad2426708442b59788b5c608ef6bc5 : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/27d3e7c7d9404aa8a7b5f0d2b6de05ce : SUCCESS in 53m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/117f690438c94e1a99a612a2c1c0ef14 : FAILURE in 1h 32m 23s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/12d1f0b55b61483d88a185fb11079254 : SUCCESS in 1h 06m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0b83b5446f0d4c238abffc02bb5b713e : SUCCESS in 57m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/27f2303eabb641ff98febf10b3798070 : FAILURE in 1h 02m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/72e5b0fc71da43bc90764ff352656493 : SUCCESS in 1h 21m 35s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/02ea088a04b94d94acb931833d4417e0 : SUCCESS in 1h 25m 02s","accounts_in_message":[],"_revision_number":4},{"id":"12190aed3bf84e065237693786fe2b83e885abab","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-16 09:11:25.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840415/4 : FAILURE in 1h 56m 34s","accounts_in_message":[],"_revision_number":4},{"id":"1f982ee351448eabca16d144cbd4b4e33322efa2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-16 09:53:56.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/15/840415/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"a5d47d4c544881b54a65dea35798e6648ce84239","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-23 10:21:21.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"4a9ada9bee0c11e08ff31d2922aa578d44d56b59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 11:37:23.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f22e7e01c2b7420c83ebd1ada0b6bb07 : FAILURE in 1h 12m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee5e762583d94a9c93d419679dbc8822 : SUCCESS in 1h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0b039057107769e958346a81753fa2e4c1fb1364","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-23 12:12:07.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/5 : SUCCESS in 1h 48m 11s","accounts_in_message":[],"_revision_number":5},{"id":"4321ed70b60ced973f0c5d5ee85a638d77cd0328","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 12:12:12.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/0cb57579d6e5421780fbf97ea8c08191 : SUCCESS in 44m 51s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/cf52cffe748945ddacc43e3818640bfc : SUCCESS in 59m 16s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4140ce65fc5c4f0b98ce1d62c9c89ad8 : SUCCESS in 54m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43f167d009d14215a3af9f4ae7025003 : SUCCESS in 31m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a5bce07429d4cdab1bb1f08391a6dc4 : FAILURE in 8m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c986bcb1148c4746969127e707e66582 : SUCCESS in 21m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be969ec6f337487d909c7cb2f9162d0f : SUCCESS in 29m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd51d0641acb43fe98a49f59e4dbad11 : SUCCESS in 22m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6045df1a2d764afa8677a307107b9d53 : SUCCESS in 10m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/04c792703c4e41618676dc3a1e0b6e7c : SUCCESS in 57m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e3a62d23dfa54b4888d7fd6fbb721ba8 : FAILURE in 1h 33m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4cb5b263715f4c728d0c4940b358e854 : SUCCESS in 1h 18m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f91b2b13b3e499098737f8f61a9b5d8 : SUCCESS in 1h 01m 18s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c074e4a59267485b9659892fa0575b55 : FAILURE in 1h 12m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0d696adc404f41e1b53621d7b43b60ed : SUCCESS in 1h 43m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a08f0b1dc45e458d92ea87f4a57b5f47 : FAILURE in 1h 05m 39s\n\nWarning:\n  Comments left for invalid file neutron/agent/linux/dhcp.py","accounts_in_message":[],"_revision_number":5},{"id":"62c54b98dfe58c476bf97f49dbf8e0f0a871933c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-23 12:26:23.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/15/840415/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"253400e99de97c0192772513dca62ea730a1e300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-23 12:29:52.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"aee67852af8893dae2798619c27564ddd7f53087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 14:07:31.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/bc0e15f4a04b47d997fdc05576f61b01 : SUCCESS in 54m 34s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/9755cbcc264246c8b0b4dc28f18f843e : SUCCESS in 1h 06m 52s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aacee474752c4ebc936e6634cae63e8b : SUCCESS in 53m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f709866b95184f5c8c6c8f6513e185b7 : SUCCESS in 41m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f56293adb5544ab81b73a8f81ef7008 : SUCCESS in 11m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1051c211e2894f14a1b225f30fd6d4a8 : SUCCESS in 27m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c5843c9339e4e7684322f65775320ee : SUCCESS in 24m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af50bb4b420f4cf4acafcb7f16543d39 : SUCCESS in 30m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe7e1008717d4f5789fbc3f020a81ee4 : SUCCESS in 11m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c15100fdde9d41bdb6a07dec57ea6760 : SUCCESS in 52m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/87503a6b95934dd09a7a33aa409381fc : FAILURE in 1h 15m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ba97ab7bb2074035a90397c6f0f15a27 : SUCCESS in 1h 11m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ad5d53e5d6846ef94454129cc7f298c : SUCCESS in 52m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f9a8462ed37a4f979691f392064073ca : FAILURE in 47m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3c688f9006a7473392c3b7ef26515159 : SUCCESS in 1h 16m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6d2d7aec6eb6419a8b8bcaeee4e93588 : FAILURE in 1h 33m 56s","accounts_in_message":[],"_revision_number":6},{"id":"f0534b72c45d2d87038dd7af75199f52f95a6d52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 14:20:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22dd8eceb3a7406bb755285886bb7386 : SUCCESS in 1h 04m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55528ad37234495c840c481c272cd046 : FAILURE in 1h 16m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7e1166b9a2d436019687682b6bfbb3aa438c25b7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-23 14:40:43.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840415/6 : FAILURE in 2h 07m 50s","accounts_in_message":[],"_revision_number":6},{"id":"f524f6e798826286d6a8e54fe1143248708fc47a","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-24 02:37:14.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/15/840415/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"43b19a8723334fd542fe5c8aaffbac8f499ff156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-06-17 07:10:14.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"9c5c5f44156b9df7d46248ba88ee072a3a3e077a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 08:44:34.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd0859b30f8a42a48c1de216b4b697cd : SUCCESS in 46m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee7518253316488899a5392d7720cf76 : SUCCESS in 54m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dd46608bedb319b8563a2c8e394c1816631aa453","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 08:46:35.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fe18305adf954aabae2bb9c457595455 : SUCCESS in 1h 14m 02s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/30ff78c6ed6444168acf32872b1bf5f6 : SUCCESS in 1h 03m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/998e224466094dd2bec5a30a4e415f27 : SUCCESS in 36m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91bc52acb5b34cd885012382b4a5970a : SUCCESS in 12m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38ecbe3e81124054960f8ee754f43dc2 : SUCCESS in 27m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d014d916d90e4bea8df372f90a190d7a : SUCCESS in 28m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6d7e21414e54ff19d31cbe2b4afa791 : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c519b1ea848240719277b1f46a6c0db5 : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d7c87fdf84d8466eb842ac6b2636436b : FAILURE in 52m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c23f6bf3e2bb44bb8dcbdd9b31e90bf4 : FAILURE in 1h 30m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d518f81edbfc4e2094ec89373c2150b1 : SUCCESS in 1h 19m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5750a2016b3640fbae034724681e9fb7 : SUCCESS in 54m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f0c0042be3bc4ca1a1907b5c5a41a5db : FAILURE in 50m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/51d885d36f724ec7bf435f2c79fae5c4 : SUCCESS in 1h 23m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a2a70e08b3574c949c6ca8a9ee6113f8 : FAILURE in 1h 12m 45s","accounts_in_message":[],"_revision_number":7},{"id":"6992307d358516393e51fdd127ac28317bfc7988","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-17 09:05:24.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/15/840415/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"842130ca523a43abdbf42161a09e249ece3ce084","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-17 09:30:12.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/7 : SUCCESS in 2h 17m 24s","accounts_in_message":[],"_revision_number":7},{"id":"5bb965e1379244fef5b3e530dc361a070459f31e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-06-27 07:15:45.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f76f987b0640d495470cf6f7e7d2b613a077ff6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 08:46:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8796566d3f04a0991ae1102592c0582 : SUCCESS in 1h 15m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/369782dc3c174325be8d2bda8a22d969 : SUCCESS in 1h 14m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7cc180ce5faa5de463be743fd27406d05f5e7c16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 09:16:13.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/97b00f7ddae94807bd26e5442d6fd998 : SUCCESS in 1h 31m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/760c7b23d3db409c8cae69475debc02d : SUCCESS in 1h 26m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adffd4148eb74039a949d54324eddb9d : SUCCESS in 36m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cdd40fc267b4906b86d5d769895c9b1 : SUCCESS in 14m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0b5f419e7764869a32bb022b71988b6 : SUCCESS in 24m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ee30070404546a79724879901f80e3a : SUCCESS in 32m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b131e2087d224c329c7e8ba67b2411a0 : SUCCESS in 20m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/645b9db110e04aba8877fbf3bb773e6e : SUCCESS in 11m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/afe280c7ea724b00a6eb478b55a66f8b : SUCCESS in 1h 00m 53s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d7210190ed04b15b3e2f12d1b059243 : FAILURE in 1h 57m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2eee8ccc61ba4425b05316cfbcc7b2d2 : FAILURE in 33m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fce6805f6ff24013aca45e932a771c04 : SUCCESS in 1h 01m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d57fa1a49fef48b4a52cfca1b36e6a95 : SUCCESS in 1h 12m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/34571318de9443a08fc599ed8376a590 : SUCCESS in 1h 29m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e4082bbbd0b84387b091dca91e1f6616 : FAILURE in 1h 26m 01s","accounts_in_message":[],"_revision_number":8},{"id":"ba1e7cb15a94c52c40ea8457a18508dd2d835de3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-27 09:21:14.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/8 : SUCCESS in 2h 02m 26s","accounts_in_message":[],"_revision_number":8},{"id":"adaf64ab57b3b98a2466a09cff4ae2b18df32257","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-27 10:41:09.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/15/840415/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"e2b038a71b1e30c97d7273140fbc8c36f5222274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-06-30 17:32:53.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"501ac6abac880f59d81c910760071949a2ba26bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-06-30 17:43:20.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent, and commit message as Patch Set 9.","accounts_in_message":[],"_revision_number":10},{"id":"6c380abc46df6231bf4fbe73fab1464995c1125b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 18:35:40.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9408332f466a462e90e6bad91af029c8 : SUCCESS in 46m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f87f5b20a03c4d318779b0fa5b8438fa : SUCCESS in 49m 16s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"acc720424f16590f7340cb0e2ec9cfd59d396d3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 19:24:42.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/037d66dcd832440ea96ba18e7ee1fd03 : SUCCESS in 1h 00m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/48492b26c6df47b48c374a2ec7b01ce6 : SUCCESS in 56m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d9616981dde48229b7e5b11f865356f : SUCCESS in 34m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3396640a330a467eabd2e7aa6a024b2b : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5345a8b7c8c42a490d3dea4f3d4a4ec : SUCCESS in 32m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e199f20101c842cc82524aa1db5c66a9 : SUCCESS in 26m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91da87bf01ac4dc9b0577ec3b275b3a0 : SUCCESS in 28m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adffc5b8f7d649339c914a0caf97f7cc : SUCCESS in 9m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d23a8a159b44eb292de005e2048e401 : SUCCESS in 1h 02m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa50c47875894562b0d66a9a8feae703 : FAILURE in 1h 32m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ad69d3452b824ce7a8c2322d8ccfe80e : FAILURE in 25m 43s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b50f25f075dd4e09bceddcf179169fa7 : SUCCESS in 57m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/146b953bb2334b47bb0b573e7ec7f19a : FAILURE in 1h 13m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8dac3f207aff49579b1f79cca4c78d69 : SUCCESS in 1h 07m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e8b70a5fb53d45b78d47a17885caf387 : FAILURE in 1h 20m 52s","accounts_in_message":[],"_revision_number":10},{"id":"9df2f71248fb4195cc064efc78eb10dcb16945b1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-30 20:22:31.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/10 : SUCCESS in 2h 21m 51s","accounts_in_message":[],"_revision_number":10},{"id":"1a8a76827281b3c2903262dcb3995bd3155715ed","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-30 20:41:14.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/15/840415/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"95e569ffbf433272ad5cf9d3e4ad6de891dc1634","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-01 06:12:56.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"332bb74952034263011543ddae48a274533515a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-04 16:01:24.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"f5fc9ae0db4930a9c2e4e238dc0b20ea25a25f6b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-04 16:16:35.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840415/11 : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":11},{"id":"27b7b1bcfd621bcc91da152d544b5c20b51262de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 17:04:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/badff7808f9e47fe87d2205d2399a29e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bcca03ae06c45fe9f0ed6ea7e0d08d2 : SUCCESS in 46m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/384b47ae179d46268eb1adcfe74c409c : SUCCESS in 59m 31s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"05b6318835bf9f65a2225f1ba6b6d0e5acc9fa2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 17:54:56.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9bb3f1570ec040c3bb92847cec4d0383\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b93acabefc774814abb2573cabe035b4 : SUCCESS in 1h 01m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/fa2cddd3f61c4f6b86e7b056ef9918c0 : SUCCESS in 1h 03m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/024a12fd7fb743129145680fb1e9bd66 : SUCCESS in 38m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f14fc5c200e848a3907bd305da66ce7f : SUCCESS in 9m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae6eac7cc2164b8b834d0321c0c86edb : SUCCESS in 25m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9165c565628842f193f90f9edcee0336 : SUCCESS in 21m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93aa1033d9104318994b9bd74c4670b5 : SUCCESS in 27m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b4232cbadeb42c89389ffee317ce61f : SUCCESS in 10m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06492ca522fb481291ab0169f1423f3f : SUCCESS in 57m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dc87b78a9af749bdb98f6a50d8eb3e3e : FAILURE in 1h 12m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/efaafd6ae9624e35a49d89ae5c16c7d7 : SUCCESS in 1h 16m 04s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4ef8c33de8cd4a978994812699a161c0 : SUCCESS in 46m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c82667745d6d461ba1d9e79b3c99b8ff : FAILURE in 56m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7a9286b4a500462c86da79b6367764fa : SUCCESS in 1h 31m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/472211d981d643e7971de4c5c699f4f6 : FAILURE in 1h 45m 15s","accounts_in_message":[],"_revision_number":11},{"id":"b252561e65e99ed56796b1ed12499a21542f2132","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-04 18:32:44.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"df1ad3b04de1dc3f02c97337ee48838bcdc013a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-07 05:59:44.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"7c42924c706b02e57273eee002efbe0d364f0ac9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 06:54:52.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72793998e662432ba6b0d50873f46840\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1fd5791d0c24cd1a66cec78deb69ef7 : SUCCESS in 49m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06832bd5134b441297fbfdeb31c55e66 : SUCCESS in 52m 02s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"be41d6b0193f346be47120360e4aaf66606683fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 07:30:19.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fabbaf78e2b45b19cfbc190b3bda762\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e80db76f050f4bfb92cb26fee0a9fdcd : SUCCESS in 1h 11m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/2c672aff6fb240a1855872782934b126 : SUCCESS in 56m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5df3a63748d24e12bb204b199fbe0469 : SUCCESS in 31m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6825b26a43f34be7baead22b0403ce62 : SUCCESS in 10m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5134fcc78fe4f6c8086c649f434dfbb : SUCCESS in 21m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e8dac32e31e43418b6e0df3fdecafe0 : SUCCESS in 24m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d09ffce88034568a67b8a5e470f6e81 : SUCCESS in 29m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/232c8c802a484eeca66218512807d891 : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5348a3180f8e468d95a4dac9923e5711 : SUCCESS in 1h 06m 06s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/24208b090fb24d9ea1bec4cd9fd043c8 : FAILURE in 1h 23m 27s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/142b2bebbf2f45b7a91b2a90a1472c63 : SUCCESS in 1h 14m 48s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b6d24c579d949b39cee42d0501fd898 : SUCCESS in 51m 58s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a802ce4d8de34c2582e87c17b5e22bb8 : FAILURE in 55m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9a5bd0b699b048db8ac5bf8ddb783202 : SUCCESS in 1h 28m 08s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0b38c103c7674de8bd4b69f67ddfddc8 : FAILURE in 1h 02m 08s","accounts_in_message":[],"_revision_number":12},{"id":"ba4ef7ac007854c21c443f3e4c5f00a3a206d115","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-07 07:47:57.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"50b6f5637e7d4a7203bcc039d239864f7cac276a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-07 07:52:14.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/12 : SUCCESS in 1h 51m 11s","accounts_in_message":[],"_revision_number":12},{"id":"e6cd51aa881b15db41bf33d2d3ba0b244fd3fb62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-11 07:30:25.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"b80cba7a98a315818271f6f51ef8a355e44b14ed","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-11 08:18:31.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"3e470a4cff69e8ac049f3362b59e205ce2e42c28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 08:57:35.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a02b4c6f21284f7a911ced6f17a3d3b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4282f2883a1f49aba2939c780d93412a : FAILURE in 1h 15m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea027c78fa6b4fbfb1205d5ba4b8ab6a : FAILURE in 1h 16m 42s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"5404f0d0a63e93408161e4b800fd56405e9b6af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:15:31.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/920981ba56a94e0f9b151fbcce2545b9\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a19add19ebd34d2e9d78a098de348a03 : SUCCESS in 1h 40m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b5cbaa7a355d4cc2b7c18d7a87045c6f : SUCCESS in 1h 18m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc08e08a280c42e6baf5d1a63c3a7f20 : SUCCESS in 40m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d85bd4523d47149365482d022e24a0 : SUCCESS in 13m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/637e012304104074984d79c6d2b27176 : SUCCESS in 29m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7f864e6071c43ceb483977cd317e92c : SUCCESS in 26m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b2907f9328143108a69e07c88af0432 : SUCCESS in 25m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b94454c1af945a4a922f9f913cebdbb : SUCCESS in 10m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/81cfa9cf7d68441bae361d64e2241cb9 : SUCCESS in 58m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f8644eb2ff3a41e2914a256be9655318 : SUCCESS in 1h 33m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c55739135a7246e98e37c5e23b33cf4e : SUCCESS in 1h 12m 32s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/deaf6fbd4da14979bc7444243e9425e4 : SUCCESS in 1h 01m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8c4cdebeaee1464e81d35f967a98dcd3 : FAILURE in 1h 07m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/04fc6c3596a5403e8d561d13ef3b28e5 : SUCCESS in 1h 20m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f0e6295f7dc645a09639d7ebd8cb7bc8 : FAILURE in 1h 13m 40s","accounts_in_message":[],"_revision_number":13},{"id":"92f0957c43b4ae5e0eeb62ab6280000ebfc0e37f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-11 09:17:53.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/13 : SUCCESS in 1h 46m 10s","accounts_in_message":[],"_revision_number":13},{"id":"46f5b776a13de323977e5ddce6630ba37ec8a248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-11 12:48:45.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"fbd2f5f6f03f4dac7ee6e264901c199e342e7b10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 13:48:37.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7d83b354c9348b09c2e0711038feed1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80e75de2a397442a80db02528b80055d : SUCCESS in 43m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30c2e523655346939467619a93c51db3 : SUCCESS in 44m 50s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"c24930468729c36a7aa6a265c84094d8ec21a34a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 14:30:09.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ddc34c574b604ac3977c3cececab516f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/cd4cbfbd9f8e41df934d4a7362d3563a : SUCCESS in 1h 37m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/82d27c7f4716425e94f1981780ec48c2 : SUCCESS in 1h 33m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6808703466f34cf194b940dec13c89ee : SUCCESS in 31m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d9e6835317e4571a8f22810685c995f : SUCCESS in 11m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82431ba1a6a84f638844df726270f699 : SUCCESS in 24m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cf47eb38c3c47788f44a58f7a84477b : SUCCESS in 29m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bf4dcd9011043e9aa97bc94ced2c6f8 : SUCCESS in 24m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fde1ea05fc411b8d0cc804606a446e : SUCCESS in 8m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/732f17c57f7c4a12a30987eff0edfc21 : SUCCESS in 56m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/17e521d6aafc48078652dc99f77a5e7f : SUCCESS in 1h 28m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/33f1122b6ef84f19b3ff05e4636df563 : SUCCESS in 1h 12m 50s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc99796266a74fedae4a4f1a49e4bc8a : SUCCESS in 51m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/56ed7ba221a54dae874fd0dafcd5db07 : FAILURE in 52m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/79545162d3864f7fb914347fcc972278 : SUCCESS in 1h 13m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/dc71a257fc2445d3b0330da48f3c486d : FAILURE in 1h 07m 57s","accounts_in_message":[],"_revision_number":14},{"id":"b5574d367e614ec190531b8993639a5e72664ecc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-11 14:55:04.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/14 : SUCCESS in 2h 05m 00s","accounts_in_message":[],"_revision_number":14},{"id":"3d12122c8204ef67380155acf101738eeb818ae0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-11 18:05:02.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"3d94e1073a50c1cbef9b14536e6df5e93af6ba38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-15 07:37:39.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"4dbc8ec4b773e2b4928995c67c63b31304547817","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 08:21:18.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ee1ca239b5a47909e8f2054a89c6e6d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7eec6f567ae4ce0a72e781084c00008 : SUCCESS in 40m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5e05420c76d4566a4e6c99104f3c79b : SUCCESS in 41m 15s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3a43e591f130ed263f99a3408fb402c9898a6ec2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-15 08:34:18.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"f4f1ae94dc7eefb8493293b305634862b609495d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-15 09:23:19.000000000","message":"Patch Set 15:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/15 : SUCCESS in 1h 44m 22s","accounts_in_message":[],"_revision_number":15},{"id":"0edbd6b2304f778c9e37518b175f6f018ba2bbb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 09:32:10.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45aa458b37034828b1baa3dc0d4c6b0f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4b17501f49594494a00de977896a1eeb : SUCCESS in 1h 38m 19s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/01d409936aa3405cb6a95ec67feb3c5a : SUCCESS in 1h 44m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87ecf55971474fd3b1d239cf3aa22a77 : SUCCESS in 34m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd6182ea14ab49768b50150034bc9806 : SUCCESS in 14m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d369dd79c77e451a96b0804cfef60efd : SUCCESS in 25m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42db1ec693df46ff86f43777ffd1f0e9 : SUCCESS in 25m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9ad9088ecbe4cedb43aa2a22edb6ad9 : SUCCESS in 29m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5622cd3c90a94b29b27d76edfeb9237d : SUCCESS in 10m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/10089914ea384493873c86120d717334 : FAILURE in 55m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/52e9f2fa00ab478c88751cf332fe9f45 : SUCCESS in 1h 41m 49s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a65a45a7149b4d0d8954f4a6f033b2c8 : SUCCESS in 1h 17m 32s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f810477216747bfafc6833f7d1eda33 : SUCCESS in 1h 10m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6936530a742142dbaf70cb6c294f4fe3 : FAILURE in 1h 05m 49s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2564662d3a74401d83f72f37cf7363ed : SUCCESS in 1h 47m 14s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/311869d47b174e718afe90a2d39415dc : FAILURE in 1h 21m 19s","accounts_in_message":[],"_revision_number":15},{"id":"57c411455ae7cc365e2b72488fd3cc2fd2424084","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-18 09:45:33.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"b3a9b12f64a23e7dd49bad5854f73cae27f68cec","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-18 10:43:21.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"8a236cb5c7a5ac368eceae91b902de729860b187","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 11:10:58.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d46e370121a48afb3f12c9f71085de5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3653cf2b3ea42d1a8813fea89d6af50 : SUCCESS in 46m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/382363ecdf074cd49feaede62e4f09f8 : FAILURE in 1h 14m 56s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"5a8a741a518591756c57bce45ccf1be3932ca47f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 12:01:58.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/139c1f5ddac24b1c8d1445733001ba03\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dcfdabc0c2384192b686175c694a5a37 : SUCCESS in 2h 08m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/294a80763e8448a193e7bff9578a763a : SUCCESS in 1h 33m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5c21bcfb1dd48ba9ff7fd2ad49bcb3b : SUCCESS in 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/525e17f7a507478ca1cf5d68f5f898b1 : SUCCESS in 13m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4d9410c172a4986977f84941b03b433 : SUCCESS in 24m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29ed299bde0e42b0b55af15e7a508d22 : SUCCESS in 25m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afd42714ee664f84925e0b59de37572b : SUCCESS in 28m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f7553ee43d843c2b3694b4db426c46e : SUCCESS in 11m 56s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/502656f35aa3475eac3881a92dcf770e : FAILURE in 1h 15m 07s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4361ae2aa30743b585ea533bd54f6641 : SUCCESS in 1h 57m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5a1164ce4708417d98d67db79ebf3de9 : SUCCESS in 1h 26m 51s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d714ba8232fc42568a7285860d9d270a : SUCCESS in 1h 13m 13s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/67426370dba841ed9ade80ec95d2b0c1 : FAILURE in 1h 14m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ff08c6bd471d4887943fa5744b26f413 : SUCCESS in 1h 22m 28s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/89117ab419be454aa2c6fb8591ea522d : FAILURE in 1h 26m 26s","accounts_in_message":[],"_revision_number":16},{"id":"01c7da832008e29f5571045d705c37a9e8c67c0b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-18 12:12:09.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/16 : SUCCESS in 2h 24m 22s","accounts_in_message":[],"_revision_number":16},{"id":"ca9dbbd8b43a66de0aa45178f209d6a11195a46c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-18 14:14:48.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"eda331d54e0a43531b2699bdb77a195c8acd2dcd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-18 14:23:47.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840415/17 : FAILURE in 5m 55s","accounts_in_message":[],"_revision_number":17},{"id":"a391a7330c5ec1b6e918f72ebeecce49f58b524b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 14:26:54.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cf940a86430490c8fde72f9f58cdf85\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac8b2e1fd0e84b36985cf20704eb09bd : FAILURE in 9m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62da33e919d2428b8dadbff77f911f65 : FAILURE in 9m 09s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"70b48c1b061a8424273fac78c02559eba1db974b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 16:36:37.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8c6fbe22a814fba892cd2ac9f991776\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9231b62a94164388bec4d0044fb50914 : FAILURE in 25m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f886dc59d82b44149003edc1e2fabadd : FAILURE in 39m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/668a93d093aa4ad08a8dbc4c45b08b4a : FAILURE in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0135178ab4b14affa36946c7ce826771 : FAILURE in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f829a14ac6a495395c6c352f2837d80 : FAILURE in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ddc86fa070242cc8d8541394be21f03 : FAILURE in 7m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce2b3c20b54044ceb5d85ea7370bcedd : FAILURE in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da00425c0b445a6b27fa64f0636a16a : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5e301eccdce04bfd9562f03fc2121aa1 : FAILURE in 24m 30s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/52dcb19035c24c929975af5e5f31a1da : TIMED_OUT in 2h 19m 25s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a5e1287874fb477094aaf2fe7e4236f8 : FAILURE in 21m 56s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1004776e32864b2d8c0a48bf0d6a1ac2 : FAILURE in 1h 04m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fa2c73fa87c24602be9eb4507c5996db : FAILURE in 56m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/294e9bede60045e1ae9c99070a3f7066 : FAILURE in 26m 12s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c9eb1e7d58024e37b820dbe6bf8898bf : FAILURE in 36m 57s","accounts_in_message":[],"_revision_number":17},{"id":"e4c01e1eef56dea2dc71715ca049bbc17f87c287","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-18 23:00:57.000000000","message":"Patch Set 17:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"c67919b42f46afc437c3e631ba345f0bb7d80821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-19 20:00:41.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"47e30087f041a4cb71ee77cbaa9733d3bd3a3e3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 20:54:07.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc5abdc4081a440ca5222a2d2f8f8222\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5030314cd0344888b19deb8c8e2b90d0 : FAILURE in 9m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f241693f3fd48da8f9f4911165860c7 : FAILURE in 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"2fb90c5ee439a15967247c701fb39aef9ddcdcd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 22:27:03.000000000","message":"Patch Set 18: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df398e7bf59943b5b645794e804f6bb8\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2a9251748ab14a9bb88c630c96f22f8f : FAILURE in 33m 30s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/346694893dc04778978e446e9439609b : FAILURE in 38m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc928521633548418db4de82d839c599 : FAILURE in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9dc9a14b550429595db40fc86fa262b : FAILURE in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93c100ce5ac84e9e91577c4a1b253129 : FAILURE in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4617cc32de34989a3d76b54c7ef995b : FAILURE in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/491c07d5db13469eb8840becf8436e7d : FAILURE in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ce9886fba4f4ce7b02d40a2b1b7cb96 : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aed639b41e6e474994c64131c78b9c8a : FAILURE in 23m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6686dde3e3e248a28d4f7609fbbec4fc : TIMED_OUT in 2h 17m 53s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/92886c9b43644437ae4df59a04bbe58c : FAILURE in 31m 43s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2bc50703d1d94e97b81c56c262396e89 : FAILURE in 59m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/17637bbdfdf64ae8858f32fb7dc84099 : FAILURE in 1h 07m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bc800a88eafc4dfe8f675ec63a054a1f : FAILURE in 28m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/52247a2f35604f51b7f5eab07d0fd52d : FAILURE in 31m 54s","accounts_in_message":[],"_revision_number":18},{"id":"e334819226b54fde4050c1b3d107127ccc437f1b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-19 22:39:53.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/18 : SUCCESS in 2h 33m 57s","accounts_in_message":[],"_revision_number":18},{"id":"e6afeebae3746f07139e75647f4eeb89dadaf7b3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-19 22:48:57.000000000","message":"Patch Set 18:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"b2956e5dcb62366a451e4f4356d0f13d56427843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-20 05:30:26.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"29bb8678860cbd7bb25e1e9ab80ac9708c1c13da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 07:16:10.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6423f7c665ae4cc690051e1160a9c0b0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3accf28a7a824e3fade61cddf53957b2 : SUCCESS in 46m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/597091589ee34bd08cf4017944b129c4 : SUCCESS in 1h 02m 05s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"91a67aa6294613e6ace9af7716859dc13aa1288f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 07:24:25.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/18a652e0daa54070ba7f159cef2290d2\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/da8a7f28154b4e1190e03ffcfdb97217 : SUCCESS in 1h 35m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/290016b8cd174bd085488d7b52fbc524 : SUCCESS in 1h 32m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee87caa8760343b4a625cc32f031de59 : SUCCESS in 37m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c77627a48e2240e58f8d0a2842575c3b : SUCCESS in 14m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e23009888694c78aabd02e83e397170 : SUCCESS in 26m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68545c22115648dca471957483bc53df : SUCCESS in 28m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e66bf4843b94d578565b5e935fbf5a6 : SUCCESS in 24m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cc45fc6e16c40d390461c6907b8ea3a : SUCCESS in 10m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fe919a543774d28a2e55bc521443ae5 : SUCCESS in 1h 02m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/92318e888437466094297e81e574bfa8 : SUCCESS in 1h 46m 58s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a319136e250c4c418685c62faff996b3 : SUCCESS in 1h 07m 53s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e4c81c1e07c74ecb902a11ee78feb74a : SUCCESS in 57m 03s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e0bd6017cf06438b8ca929b91649ad25 : FAILURE in 1h 08m 40s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0c717ac5133a4699913ba4a24984f9bc : SUCCESS in 1h 25m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/80441829ee1443f895be92022b7a385c : FAILURE in 1h 13m 16s","accounts_in_message":[],"_revision_number":19},{"id":"3a45e6ee897ef729c9aec50f64fa1f681f54568c","author":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"date":"2022-07-20 07:38:05.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"8215f078aa06221ab4997458c8816e34f668f14c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-20 08:04:59.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"df613c8506ca3c362708368d1a117e2cd6540f4e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-20 08:17:09.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/19 : SUCCESS in 2h 07m 07s","accounts_in_message":[],"_revision_number":19},{"id":"84ffdf0960d3951da1e9e8ec1c3ebd60228a4b57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-20 20:31:17.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"ddebce79087eb62404f22d708c9ef8892c52fe47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 21:29:42.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfdfb3a4a6bd432f8ac7f41435bb4b58\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e528bf5883764debaaf35fe3be3ec8ab : FAILURE in 42m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c89428c4a54466da13d373ba25eb8a5 : FAILURE in 50m 29s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"5d0e3022a4a3c7b5efb05d002c3e5dce9835ae72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 22:04:56.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2da3907f54ff4376a2148cd4796bba31\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/82f7231db61e494ea1aab9bcd0dd3bf0 : SUCCESS in 1h 29m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/9d4de25eedd047ddbb22fc66006dcfc2 : SUCCESS in 1h 17m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02a129299f9047d8aabed5ca55cfdc3e : FAILURE in 33m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d5f825c6f8f425fa8de1a66a5592d5c : FAILURE in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7eb372967db941e8b39cfc2acfab9251 : FAILURE in 22m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f6bc2a987b04b01840446908f319048 : FAILURE in 26m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f3e5f6a35314a49911a8e9ba43b5eb9 : FAILURE in 27m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37870fd4eb6e4f7c92f80930ce3646e6 : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d5bdb7286c784923b699e7e0b63e48ef : SUCCESS in 1h 00m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d8a6a4a712d4e519dfad6e7cd83e72b : SUCCESS in 1h 30m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d9601807f9814fd28a06eb13bf7f60de : SUCCESS in 1h 11m 10s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/947ba38a4e38434b903ac9a6522a9d39 : SUCCESS in 1h 03m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/98771fe71807495fbe31bb260fe82853 : FAILURE in 1h 04m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/496867b1b0ef48fd967091c2b2393c59 : SUCCESS in 1h 21m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ae1747db25e0462a8e6e61c4ffae77ed : FAILURE in 1h 15m 07s\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/l2pop/rpc_manager/l2population_rpc.py","accounts_in_message":[],"_revision_number":20},{"id":"16e12f49bbeb3d99b03190bc731956f5e158ae97","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-20 22:20:28.000000000","message":"Patch Set 20:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840415/20 : FAILURE in 1h 47m 15s","accounts_in_message":[],"_revision_number":20},{"id":"dc30c9275fd6f35a9d47269de2e01ce54da677da","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-20 22:23:22.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"65f16ed00326bcaccc7d9a36ff6247130bbf7bde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-21 06:22:55.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"7e905e3250690c4b8f69ee49ec3ce416ef8c1679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-21 07:14:48.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"8f904adf1beb0426beb9780e395094c21b83f8cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 08:44:11.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3de214de4304faf85b94544207194e5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/769b9ad9ac024e11b312e6ac22ce47f9 : FAILURE in 49m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fad40e5e872a49d4a8361b2cd089ede5 : FAILURE in 1h 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"f91b63cf18b64d784723717e5cc5327f387e9670","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 09:04:45.000000000","message":"Patch Set 22: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d48eedbb5a764e67a853d4935d024358\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/1f5e07ce1eb448088316afd2cd1f6187 : SUCCESS in 1h 46m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08c2b82ced904d4b9cc1f277792487b8 : FAILURE in 1h 41m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b19576e3b0b540f29c757257b30712a4 : FAILURE in 34m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75dfe1c5bd954419baf4ea40679eccec : FAILURE in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b26cf68f4e784f18a8e0a7f54b36969f : FAILURE in 27m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/828b1380d4ac41109c20a0d1cebb38a6 : FAILURE in 22m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb48b4815d524d5fa2ada28780a0f8a0 : FAILURE in 24m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a6163e3ac28449aaa87e87bc4302ba6 : SUCCESS in 10m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c063c774e8134a288f5996dfb76c030a : SUCCESS in 1h 13m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/056c25e6d43245c080f5bb89d0be5f9b : SUCCESS in 1h 34m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2cc9482315e94503b16147ca7715a763 : SUCCESS in 1h 06m 09s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b18d89c5bb34be597941f7c2b21ce10 : SUCCESS in 58m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/53e329fa5039421385330595f95ad7b7 : SUCCESS in 58m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/72501efbc3d6404a83b0a94b482cde1c : SUCCESS in 1h 18m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6e3a79d7775f43d282563f543defdf60 : SUCCESS in 1h 19m 57s","accounts_in_message":[],"_revision_number":22},{"id":"f9823a775372164d5d077ddeb6fdbe64c16638b2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-21 09:20:36.000000000","message":"Patch Set 22:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840415/22 : FAILURE in 2h 03m 07s","accounts_in_message":[],"_revision_number":22},{"id":"a5a8140a87f488598430f513c3a0d30ba66a81e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-21 10:16:49.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"dab7d06372893e4eaa27a8a487e637077eacafa6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 11:09:06.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/572920602ff34aa78ac1e4199b0e1a27\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12abfa257b24412e9cf12b09bf768a86 : SUCCESS in 49m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2be6bf7834d4a84bd82c20f9c06a0f4 : SUCCESS in 42m 33s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"449eb36a3f58479ba00db3be2e82767b90bf6e9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 11:55:30.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5987339e7a54c1fbfe0c28a9180aa13\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a14af19c64c94057abd0e9ca285301d5 : SUCCESS in 1h 21m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c5b5e2fef6124fdca2da1d3b9e842437 : SUCCESS in 1h 08m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1f93eb9f0ab45fc946b6c5272649191 : SUCCESS in 30m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7270a0db334444d3a778d1a5298c27f1 : SUCCESS in 11m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bca2046465d40be8e28ca68216241d1 : SUCCESS in 28m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da1986df4d9348129a676007b4fdbbc4 : SUCCESS in 24m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f62ec224c7e43109b9faab4324bc518 : SUCCESS in 23m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a67474b20b1d49a185e98f301ca16718 : SUCCESS in 11m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/474c30ff8b4d400ea738a7b7edd75e69 : POST_FAILURE in 1h 20m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a5618e799b534eef8e4f2371d4c92229 : SUCCESS in 1h 32m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b13de6c2bd144bad82cd33f15d97f9ee : SUCCESS in 1h 12m 15s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae8a815fb59f4b1c852bfdcd115996f3 : SUCCESS in 1h 03m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/7b59174d71ff424aab9cb7b9b5c2455a : SUCCESS in 56m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0df0fc152f5d412cacbb774452c4f11d : SUCCESS in 1h 23m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb9aec5968874a8bae2a9feb8396b738 : SUCCESS in 1h 33m 40s","accounts_in_message":[],"_revision_number":23},{"id":"63743b6f0e3b26a164b4955415a0e68dbfbd4892","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-21 12:05:48.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/23 : SUCCESS in 1h 44m 23s","accounts_in_message":[],"_revision_number":23},{"id":"e57adc1a2c6b048af0338048cfe66b43651c0f32","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-21 12:52:02.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"ee8529d2e4a40e2454f55525979508b6d5970844","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-21 13:33:12.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"d7a940cba6028c9566e0193ab6f17d9798f7dad6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 14:16:14.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eabdd688ee2249fe9338408c7ecc31a4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/143fa0221206443895a0d275ba2493db : SUCCESS in 41m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82d9c93f96a5453f86f0e5b7dac24315 : SUCCESS in 33m 30s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"916307c21bd0bc49351e946a08c7676d60494b56","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-21 15:13:50.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/23 : SUCCESS in 1h 39m 30s","accounts_in_message":[],"_revision_number":23},{"id":"9f8aeb4ac305470f2c48ed1de2e7417d58d61665","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 15:18:51.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21d3fec0f94547d88e84015e6ea20da3\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/88d77ec57fb7426d8176cba4887a2f26 : SUCCESS in 1h 09m 56s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0625511feb5b49068e0bc9610a08f077 : SUCCESS in 1h 10m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45e2ec4782fb4cf19bec18397b8fdcd4 : SUCCESS in 39m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad48eeb27d144e798d06a117121c3932 : SUCCESS in 11m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a84faf33c6e4b3d95d34644c36d17a6 : SUCCESS in 22m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db046e4d908443ea99d97bb7b1152387 : SUCCESS in 30m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f1ff463df844b04a0648d1b6abfc097 : SUCCESS in 28m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a13a839fd8a04bfcad589f3d7a8d1f1c : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9d264af0c20246baa658d2171532116b : SUCCESS in 1h 00m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e26fd44e3a094662ae9a2b31ee08d3cb : SUCCESS in 1h 38m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/33d4561d8e07456383da77eae7fe0d27 : SUCCESS in 1h 17m 31s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a85ec3f703ad48669a5354fb1de00eea : SUCCESS in 51m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e7961ea081e742cfaa40fca2185ae3a2 : SUCCESS in 1h 04m 05s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/0cf7708e95a440089f0a4c6ab5142c1e : SUCCESS in 1h 09m 02s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8bd6d33b121b40b589ddfdf94a18f6ee : SUCCESS in 1h 29m 51s","accounts_in_message":[],"_revision_number":23},{"id":"40f3e2523cc81d7f5a812a13ce39fdf54b97d41b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-21 21:15:17.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"aa2862f4b808f908016eca3004e8d37b5b78534e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-22 09:51:57.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"7eb4931ac07372c358f5a815ecfbc85cc8647cf7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-25 08:56:18.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"825d80ece36c6e91bc493f7bd4f65267609b1acc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-25 13:00:02.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"d1eb9d50b6b4eefe27b7f7c6541773a12273b3c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 14:01:20.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bffb43a1dfba430393dd825033d2ecfd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8cf34d01bc88487baa016d1e89d8d3d6 : SUCCESS in 43m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2bab68260bdd44f4b05b9a0041f7b8b4 : SUCCESS in 43m 16s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"26a4aa74b0993be5223129c86d510feb5b09edea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 14:36:24.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/516abde887d74999b24a5277559b4f37\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c8a2e18f66534749803d11f1e606398b : SUCCESS in 1h 14m 51s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1df1d321e7bc4c369c5e5a7e11d93443 : SUCCESS in 1h 17m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04d785bc72994ffb9ec837b6fd259ddd : SUCCESS in 37m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a390abebcde47c8a8507630ab7d59d2 : SUCCESS in 12m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b2eea1484724a72b3737fc13bf35228 : SUCCESS in 27m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70d0ed02ad3941769a7a2743b7a0d6fa : SUCCESS in 26m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e84c772ddff44b5e98915b18415e4cad : SUCCESS in 24m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7456d2cccefe49a880a53ccf7d2a7e65 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fffa9c3df47c486b93123890de66123c : SUCCESS in 1h 05m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/219db4a56aea4dcd93d4514f1f7982a2 : SUCCESS in 1h 29m 46s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ff07da2302b645b5afe768cf1b835e97 : SUCCESS in 1h 04m 51s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/174a38edb9104cf59857db82fee14aff : SUCCESS in 1h 03m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8dfbc29cf2ee42b5bd2c7164c74f7016 : SUCCESS in 1h 06m 45s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4d56614a541541ab86a42049289e9be5 : FAILURE in 58m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/329b40763f5745e29140f9b081b36e8c : FAILURE in 1h 23m 36s","accounts_in_message":[],"_revision_number":24},{"id":"75fbb5cc15c9eef58c160ad1feb4ab621a8c85e0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-25 14:44:59.000000000","message":"Patch Set 24:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/15/840415/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"d97a664f93d49a6bd9192b5571f6bddd530f21e6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-25 16:59:26.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/24 : SUCCESS in 2h 16m 25s","accounts_in_message":[],"_revision_number":24},{"id":"e2b8cd45da151db8688e7a566264dcdcdfd71c01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-01 08:33:33.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"170ff31adc4dc0cb1e58f9563e1a3df0d33cf5a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 09:19:11.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/802bafada0ad46f5933e50718df3fd9b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6df1e6c546e64da1a8d30c9a44a3f829 : SUCCESS in 42m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f936d9fb0d84d5e89157abaca4fa97d : SUCCESS in 42m 22s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"6eb53a32953b69b2199a61cb2a97caac3ae93a10","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-01 10:39:57.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/25 : SUCCESS in 2h 03m 36s","accounts_in_message":[],"_revision_number":25},{"id":"52abbf75755ffa9671c0d8a7989375e8ef448a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 11:43:39.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae92c6be3e25431980c4e4f97e3a9445\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/46f89be06eab4274be7314cf09cc044d : SUCCESS in 1h 25m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/dc2a6510f2d64b10966868d2284fde28 : SUCCESS in 1h 06m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5654368449084cd888a32b3be08241d7 : SUCCESS in 31m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/473a65c8135440059f3f03e261da0df4 : SUCCESS in 13m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0029d5f6902c4996894a4174fff982ba : SUCCESS in 24m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce0df6b36b19442e91975a3bbb5f1aa6 : SUCCESS in 24m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1103c11d7d6415681896b5b63fa83f9 : SUCCESS in 24m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c18de493f5af4b6ca7493230ef87bba8 : SUCCESS in 10m 46s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fdffe9caf69b4866b8c2a32864ce9f1b : SUCCESS in 1h 05m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/635ef58e3dd641148c748d6771402b03 : SUCCESS in 1h 42m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/978e5ab7742b4af8b0e963d8f1306dd8 : SUCCESS in 1h 09m 00s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4a6b0cf3dd14401e8290a3f1438d23b1 : TIMED_OUT in 3h 06m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ce0434c6ec1d479a9b0b8ca238925629 : SUCCESS in 1h 03m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ccaaeaa29bf74a50b98cab534bf15e6e : SUCCESS in 1h 17m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d3654f620a514eca95579d2939e56a2e : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":25},{"id":"73cb7cb40e12444b86dee012777faa4ed9b46ece","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-01 12:52:14.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/15/840415/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"f8fc339533f337296d1071c186296c5733d5c266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-01 13:44:40.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"7031d6c1b378eedd4c1371e906395d62eba1c05f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 15:13:43.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41f912f9020f49fb91d61a4bf1c366e6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b33c51a3d0a6444b8451465fa1c0c4ed : SUCCESS in 45m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8be50747f8754d3a9a10338dcd648291 : SUCCESS in 46m 39s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"e0aa3338c705a790f4653b8b7c17af817734a4d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 15:25:46.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb7178cf1cba4ff9a0dfa4fb391ed8f7\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c066ac8a8aa342f38bac2b9307ebed7a : SUCCESS in 1h 20m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c8cacea990484ec095f2f64c12e20b96 : SUCCESS in 1h 12m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaaf6be4b47c4dc98d1d5a0002193cee : SUCCESS in 41m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/457577ae5128448dbb04484d74f65266 : FAILURE in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a586ae6fec4a8992baad9528fec893 : SUCCESS in 32m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a5381da64f743259fcec2e28f221585 : SUCCESS in 22m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a205c1717a6c4cb99820cc9be13f5f79 : SUCCESS in 28m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf67bc33f4b94f74afb94be789ede0ec : SUCCESS in 11m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1ca1d3dba0874db4b0eb6e48b395df18 : POST_FAILURE in 1h 19m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/81e941a7c7ff44889f9d928aa63a1b3a : SUCCESS in 1h 38m 38s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0509b751a65f4326b9c55f12a07a4c50 : SUCCESS in 1h 06m 05s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d05f4ee71b242099055232210bf1c10 : SUCCESS in 1h 01m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d834d8d422cd40079cbe8e9822dd7349 : SUCCESS in 1h 04m 35s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d55681f7de584fcd9c2cea87ad1d6aa0 : SUCCESS in 1h 20m 30s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e908c89be2854da1bb70b45550017345 : SUCCESS in 1h 21m 41s\n\nWarning:\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/drivers/openvswitch/agent/test_vlanmanager.py","accounts_in_message":[],"_revision_number":26},{"id":"e675dbc58a1a0228438243ca62cfff24529d75b2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-01 15:55:26.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/26 : SUCCESS in 2h 08m 17s","accounts_in_message":[],"_revision_number":26},{"id":"55a511c088db35357ef8d480536018414eab4007","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-02 01:38:21.000000000","message":"Patch Set 26: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/15/840415/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"061a75f0a4777bcf0155476cf0a229e30473ba57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-02 09:33:49.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"ccc38df381fcd31c4fc3d6244ca860b27fb7608a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 11:06:19.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a61b8f770f2d404583bf3eaa85b8e1bc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df12003c0e634eca944a946c78961657 : SUCCESS in 45m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42265a4631044e1b98ba1e80f1b3195a : FAILURE in 51m 58s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"83a227e283b87456ab6d51bd56030b9704005532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 11:14:23.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e88a0f02431a4a65b2092d6a83e9a35e\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b7105c37bb3246d4b54047583c4701a9 : SUCCESS in 1h 23m 38s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/da44b458933d470787e526aff86a9241 : SUCCESS in 1h 06m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83dd6f0f10c349a9996ca96dddf5316c : SUCCESS in 40m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a09cc02d28a4dd9a8884961d1c19105 : SUCCESS in 13m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/553dd7bfd2454830a7f0f60b321fd5ba : SUCCESS in 25m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5885179d1c33420cbaf29eb2effcd568 : SUCCESS in 25m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb93040745e1485385c6f9232c74628a : SUCCESS in 26m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170d709c31d14faeb3bfdd8a65ed0106 : SUCCESS in 10m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d1b30d919844c3e9fb53838adbb9a2d : FAILURE in 53m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/28565eb5fb2d477e931398eb87729181 : SUCCESS in 1h 38m 50s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3f39a7d68e5a435f97889067c24d72ff : SUCCESS in 1h 04m 39s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b466f6f84e74adda93912cdbbdfc47e : SUCCESS in 1h 09m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b218f913367843a2afecd76ad6e58922 : SUCCESS in 1h 00m 59s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fd593782c769425abe077a24e3a3c5f6 : SUCCESS in 1h 22m 36s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c6419c597cd94f928423ae5d6277bbae : SUCCESS in 1h 32m 57s","accounts_in_message":[],"_revision_number":27},{"id":"f90312df6ef33b0abcfaf856f0b8b7fd9be184a9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-02 11:49:55.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/27 : SUCCESS in 2h 14m 39s","accounts_in_message":[],"_revision_number":27},{"id":"490dd8175447ac7e9d2fa0a1da59cdff882425e5","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-02 13:36:08.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"91fb45189aa60e2ab4cfba487dad34f4beec4437","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 14:18:26.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c98d4b7a9db48dfbe6c85c4aac22918\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3fbcb6c3304d4143b5699eeaa361fa6e : SUCCESS in 35m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b155b28f11a5422fbcbbc96470314f61 : SUCCESS in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"a42e989a43f965cc6eb54c3913b60adedb25ba95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 15:16:23.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f67cfd7b0b474bfdab4c0759533e8f7b\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f5edbb9ea7394efda842e8e59662c414 : SUCCESS in 1h 35m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5128a74f7907485daa82927a8dcc09fb : SUCCESS in 1h 15m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc644617abed4e47aa8add2223404817 : SUCCESS in 28m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc2b4b09b1654cc490e1d540c354a7fd : SUCCESS in 10m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4510b17895b4f59bdd1b5c6b3061a45 : SUCCESS in 27m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5df44eb9fdd493bab257f903f827463 : SUCCESS in 22m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50f0b0c85a09449b9be8b9000e4d2544 : SUCCESS in 27m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb535e875f764bff9677f45b94ebfed5 : SUCCESS in 9m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c828dc8a68940ce98af74aade42ad6d : SUCCESS in 51m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ce9db162628420da74828471597ecb1 : SUCCESS in 1h 22m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1b524de35427401d83e1b73dd6c02ec4 : SUCCESS in 1h 04m 42s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/37fe333ccc2041ea850d2f97d95a0fa6 : SUCCESS in 52m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cef113278970494ea0c063c1db0301f2 : SUCCESS in 51m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fbdb3b8aec564d03ac4ed896fec904d2 : SUCCESS in 1h 03m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/eae4defc8665470eb90b8448ede127b2 : SUCCESS in 1h 07m 42s","accounts_in_message":[],"_revision_number":27},{"id":"69a668b41dd3dc8d8e731a81cf3585bd41d7bcee","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-02 15:38:13.000000000","message":"Patch Set 27: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/15/840415/27 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":27},{"id":"27554d39a11ae3deb9a9d64f64d294fbfba80250","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-02 15:51:40.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/27 : SUCCESS in 2h 14m 39s","accounts_in_message":[],"_revision_number":27},{"id":"78a560fa540a0dfd477c4ba44d2c0caca0bcb9a1","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-04 13:42:15.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"2151870ace9342b0d0fb50ffce00503182f80a86","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-08-10 21:36:16.000000000","message":"Patch Set 27: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":27},{"id":"1b70e8daa3a55e407d18d7b9ccacedcebc651c9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-29 13:21:26.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"9180600e17a283d816c5977f11e90a9ae946ec14","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-29 14:27:28.000000000","message":"Patch Set 28: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/15/840415/28 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":28},{"id":"dc1de5defbd01ba638439768cd9983f06d2ef374","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 14:43:11.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5d19f20211046ffb7de0fcd4748b76d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ebf053648573490e91c5039e6f29b6ee : SUCCESS in 40m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/34a9401ccc93471b83e3ba492377a761 : SUCCESS in 43m 59s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"dcb46868a577f67e17ba322950412db75c9d27f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:18:29.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a551314e1fe4c25874907ce41fc0293\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e8e27d1c8e3e42f299d9c0ca87f42720 : SUCCESS in 1h 17m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4e6304de75f34f888da15ee36dc8cdf4 : SUCCESS in 1h 09m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5cc58f3359745c9b799f6e72bf1c578 : SUCCESS in 36m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f477b71b2382497688b472baa4b71e3e : SUCCESS in 12m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ba4c9127c0f4e51a5205cbf9420e5ae : SUCCESS in 25m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30213870590548808dc9c6f91ac1c271 : SUCCESS in 24m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7223aa9639e4576aaf5f245128587bb : SUCCESS in 25m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5098d8f5c840f5a5c5bc0bf6035d5c : SUCCESS in 11m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a2169b4b60c4fc7ba4ed00701ff53a1 : POST_FAILURE in 1h 20m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e5372c897024bf29ac202e4ce495f1d : FAILURE in 1h 54m 01s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5920b9de8aab4a68af1376e9999c8353 : SUCCESS in 1h 06m 35s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/08595c202c4645c490698035364a4927 : SUCCESS in 53m 16s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e2dc3981b522430d8ced62ffd990d0eb : SUCCESS in 1h 13m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/12cf12f6c274416b95acd91c868f3621 : SUCCESS in 1h 38m 13s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6fb8fd16e1da41fb8820f59d754931b8 : SUCCESS in 1h 05m 25s","accounts_in_message":[],"_revision_number":28},{"id":"95205443e72b79114f0b282bbe392a188008b60d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-29 21:22:53.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/28 : SUCCESS in 2h 37m 54s","accounts_in_message":[],"_revision_number":28},{"id":"5f55937a8305805e018351d9349360c7d97290d5","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-30 08:08:28.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"6fb13b52a4b6ed7716792c1e50b00acd67de81ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 09:21:52.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbd9ad58ce86486cbc146f29ad54bd03\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25a63686bad64bfab91f25e53314d9d6 : FAILURE in 1h 12m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7515800849f9479a8cc9b4c6d74ff688 : SUCCESS in 1h 08m 43s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"0a80ba1b9b03cfa20bb13a11a181e6a77a24d48d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 10:03:29.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80b6782044bf4b0fab84c73255134bf1\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/eeb0d636c13b468e906865c90360bb94 : SUCCESS in 1h 28m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3ebb6adec0e54296b388eaefdbe92f0b : SUCCESS in 1h 21m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4a334f9da394e2da1ad35834edc5c56 : SUCCESS in 29m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ba1f79d749a490baf87be58e43337ca : SUCCESS in 12m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f239c29d42b4a78a249dfe93b98d37a : SUCCESS in 26m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04423b626fc24a20bc4dece41a9ad668 : SUCCESS in 26m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bee05048cb84af39b0581072e2e009f : SUCCESS in 23m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8decf0ba025e4f26a6874f7d79aee37c : SUCCESS in 10m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/688c51af048e447f8615593faffe0e0d : SUCCESS in 1h 03m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a63803c72e13422d9f271596e3548a39 : SUCCESS in 1h 53m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a0aa9cd5702342d598d056822180b54a : SUCCESS in 1h 04m 43s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e3b379ae16a425784b120be89970743 : SUCCESS in 58m 06s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d13cd2b9c140433a94962feb994a7f82 : SUCCESS in 51m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/635e8736805d4529b4e53c18048119a4 : SUCCESS in 1h 12m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9e6f4b43ca7d489ba062f709f93b75b5 : SUCCESS in 1h 27m 24s","accounts_in_message":[],"_revision_number":28},{"id":"e8ab332223f26de1f4c149a5d2940d87260f1228","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-30 10:04:56.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/28 : SUCCESS in 1h 55m 45s","accounts_in_message":[],"_revision_number":28},{"id":"b2eefd635e4c55a513eeba1c122326bc88641652","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-08-30 11:17:11.000000000","message":"Patch Set 28: Code-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"e82b8e79292238f3a409bbae4820a870106eedaf","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-08-30 17:08:36.000000000","message":"Patch Set 28: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"ee962e1387c4309ff42af128212f7a80cb50b20a","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-08-31 10:08:59.000000000","message":"Patch Set 28: Code-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"2b1be8c6855515b6131b1aea29a135c60d652d2e","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-08-31 10:51:59.000000000","message":"Patch Set 28: Code-Review+2","accounts_in_message":[],"_revision_number":28},{"id":"8c2cb72bc8232f22d752710d03d8c03b18356bd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-09-01 16:21:23.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"989cc4157413ba7dedaa0891f1e1013098a958af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-01 17:31:40.000000000","message":"Patch Set 29: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/15/840415/29 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":29},{"id":"771e94ca610c5ea8bcc85b408577028373f316b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 18:06:04.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e3701a49a6a4774898761a1a8542511\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6e014955497e476dabd8315ea568d6cd : SUCCESS in 1h 40m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/db274a85ca7e496483fccd9fa0b4fabb : SUCCESS in 1h 39m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00c7f1ce914e4546b5735921e634b688 : SUCCESS in 38m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c655d609e6a47a483e6f23f214a8341 : SUCCESS in 13m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecb76a9bce694b59842789cb2bb35ebe : SUCCESS in 32m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c3dfe3406874302a7d2edab15813f32 : SUCCESS in 27m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c94fc18107dd4948bb9e9bcac622cf41 : SUCCESS in 24m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c1bae5065a4e90a8253919bde9e718 : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d420d64f2b0046a7b49e6034dfdd6580 : SUCCESS in 1h 16m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c0d58eeafeb4bf69d5808fddaa2e542 : SUCCESS in 1h 32m 03s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0a5dd0a8ef72486eb157b70dc84b3313 : SUCCESS in 1h 27m 14s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a8abe9fe665467281187b30330c8eea : SUCCESS in 1h 16m 20s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1b37560bd18b47edaab2f5ca9e47e813 : SUCCESS in 57m 16s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6ea0d34e2f474eba9afaa224e63ac7cd : SUCCESS in 1h 34m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1c5730cc0c2d4471859360598db6c138 : SUCCESS in 1h 27m 24s","accounts_in_message":[],"_revision_number":29},{"id":"3fc86a5a99b614224e10ad9591f60fd879b074cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 18:08:41.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6763eeb98df480da54137b5ce425acc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf6ec6533fbd4b7e9ce44cb6e26f4c83 : FAILURE in 1h 07m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b71909b49a224cb6ba735a9bfee48872 : SUCCESS in 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"219a00cabb20779da3807f7b61cf2ae97bfa1605","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-01 18:30:16.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840415/29 : SUCCESS in 2h 07m 25s","accounts_in_message":[],"_revision_number":29},{"id":"f7146593545778f6c05bba0f400f2332fd6e41b1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-09-02 07:39:16.000000000","message":"Patch Set 29: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":29},{"id":"3fdea7542ae87206e4feb792af7b795cbf4ba475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 07:39:54.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"b1c4060ffba9ebd60235fd13955c889c887afdac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 09:49:36.000000000","message":"Patch Set 29: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7362569935046898d4e46eb87a4907d\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/411a9050cb7a465cb9dec385f17d2422 : SUCCESS in 1h 37m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/654a7c6e71d04e42ad37301d93ece819 : SUCCESS in 1h 22m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7881780eb7ff4d2ab584b6e3baff00ce : SUCCESS in 11m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ee93c9d906449b2bad49a5957fa6102 : SUCCESS in 23m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/354fedd64d674f35887c98cd06e23576 : SUCCESS in 28m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/149621677f9b440cbdb9e2a301b06727 : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fa5955c367d04ffebe491d5fd50130c8 : SUCCESS in 59m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0c6b581e5e8245ce991181354bdc7e61 : SUCCESS in 2h 08m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/50b5127a76924c55b4d647dc6e035d70 : SUCCESS in 1h 31m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f02e199da6c845f496ea3c06ce6e50cf : SUCCESS in 53m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9915069adc8c4c05af3b79ea8cb72eb7 : SUCCESS in 1h 14m 06s","accounts_in_message":[],"_revision_number":29},{"id":"392cb8a2a21856fafd024cf55932785fd0a37181","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 09:49:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":29},{"id":"98184202d8a9ce20f286d6f919a839ca4336892e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 09:51:00.000000000","message":"Patch Set 29:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afafbc534bc847f0a60b528cf2475a98\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de43259c7ba4e4aad74cf878b49cc55 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"7a1e253851dfb9b46cf7734f5a58a5a859b7c984","revisions":{"4d1e275ac104e4245bbd81619dbaf6230305cd12":{"kind":"REWORK","_number":1,"created":"2022-05-04 07:07:57.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/1"}}},"commit":{"parents":[{"commit":"67e259e7dae253a25c314c9f83a218fdd132a6ae","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/67e259e7dae253a25c314c9f83a218fdd132a6ae"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-04 07:07:18.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d1e275ac104e4245bbd81619dbaf6230305cd12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d1e275ac104e4245bbd81619dbaf6230305cd12"}]},"branch":"refs/heads/master"},"8a8a0d418d4ca5174aaae90c9d0cb92251406e25":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-04 14:34:28.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/2"}}},"commit":{"parents":[{"commit":"cbf0817feaf8d6f785f334f20aba64d06ae810a1","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cbf0817feaf8d6f785f334f20aba64d06ae810a1"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-04 14:34:15.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a8a0d418d4ca5174aaae90c9d0cb92251406e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a8a0d418d4ca5174aaae90c9d0cb92251406e25"}]},"branch":"refs/heads/master"},"2717d6aad9aabd94ef28b26712c35a255336f35c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-11 17:26:44.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/3"}}},"commit":{"parents":[{"commit":"741afce85b89bfbf3b158c57a48225ae5e2886df","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/741afce85b89bfbf3b158c57a48225ae5e2886df"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-11 17:26:32.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2717d6aad9aabd94ef28b26712c35a255336f35c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2717d6aad9aabd94ef28b26712c35a255336f35c"}]},"branch":"refs/heads/master"},"15e3a802584dc774762e82075d1ca1d9e22b41c7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-05-16 07:11:36.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/4"}}},"commit":{"parents":[{"commit":"c05f3cb1d4b4846f20c742f0d471a058706c2f42","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c05f3cb1d4b4846f20c742f0d471a058706c2f42"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-16 07:11:22.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15e3a802584dc774762e82075d1ca1d9e22b41c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15e3a802584dc774762e82075d1ca1d9e22b41c7"}]},"branch":"refs/heads/master"},"61fda281be841bb96dcd04a026dfe8d4f3eb9095":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-05-23 10:21:21.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/5"}}},"commit":{"parents":[{"commit":"abc01d95f49683b273b974cf8c2ced5871e0ee69","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/abc01d95f49683b273b974cf8c2ced5871e0ee69"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-23 10:08:19.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61fda281be841bb96dcd04a026dfe8d4f3eb9095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/61fda281be841bb96dcd04a026dfe8d4f3eb9095"}]},"branch":"refs/heads/master"},"203937586231eef5cd23bd2a858162e63b4392bb":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-05-23 12:29:52.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/6"}}},"commit":{"parents":[{"commit":"c0f385a9dc1f0db9a200a9d94690c0d5a57c9230","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0f385a9dc1f0db9a200a9d94690c0d5a57c9230"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-23 12:28:56.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/203937586231eef5cd23bd2a858162e63b4392bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/203937586231eef5cd23bd2a858162e63b4392bb"}]},"branch":"refs/heads/master"},"96c70b8f540a17dc4cf6d8badf35d5f36dd12a00":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-06-17 07:10:14.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/7"}}},"commit":{"parents":[{"commit":"1a90a2b1daa7705832c354012494f955115b4d6f","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1a90a2b1daa7705832c354012494f955115b4d6f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-06-17 06:15:18.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96c70b8f540a17dc4cf6d8badf35d5f36dd12a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96c70b8f540a17dc4cf6d8badf35d5f36dd12a00"}]},"branch":"refs/heads/master"},"3e543d2e3da4db6076b413ab2d4b5c046f77b98e":{"kind":"REWORK","_number":8,"created":"2022-06-27 07:15:45.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/8"}}},"commit":{"parents":[{"commit":"0d52279ee1dc3e0f745488a5516299fb0a2ef3d0","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0d52279ee1dc3e0f745488a5516299fb0a2ef3d0"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-06-27 07:15:21.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e543d2e3da4db6076b413ab2d4b5c046f77b98e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3e543d2e3da4db6076b413ab2d4b5c046f77b98e"}]},"branch":"refs/heads/master"},"c58cad3aeb6711cd61357787f4dc2823c28d40e8":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-06-30 17:32:53.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/9"}}},"commit":{"parents":[{"commit":"d1ef1f68bbd15f37b452174a4e890f600d415166","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d1ef1f68bbd15f37b452174a4e890f600d415166"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-06-30 17:32:40.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c58cad3aeb6711cd61357787f4dc2823c28d40e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c58cad3aeb6711cd61357787f4dc2823c28d40e8"}]},"branch":"refs/heads/master"},"ea9b95c4e62cc692cca33138a56f12127c66b921":{"kind":"NO_CHANGE","_number":10,"created":"2022-06-30 17:43:20.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/10"}}},"commit":{"parents":[{"commit":"ae0ca42cb79b410afa406e71b28d7e073843756a","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ae0ca42cb79b410afa406e71b28d7e073843756a"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-06-30 17:43:08.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea9b95c4e62cc692cca33138a56f12127c66b921"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ea9b95c4e62cc692cca33138a56f12127c66b921"}]},"branch":"refs/heads/master"},"0e742b6563f8a2b748eaf78f7030d494bc53a596":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-07-04 16:01:24.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/11"}}},"commit":{"parents":[{"commit":"a6e78edcdeec4be511b3ab36da4cdc843fee1576","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a6e78edcdeec4be511b3ab36da4cdc843fee1576"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-04 16:01:08.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e742b6563f8a2b748eaf78f7030d494bc53a596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e742b6563f8a2b748eaf78f7030d494bc53a596"}]},"branch":"refs/heads/master"},"f35f24f9c5b424073fc78ae23a4c798876385845":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-07-07 05:59:44.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/12"}}},"commit":{"parents":[{"commit":"4d1614b8c7902b61f4b5daa3d980ee8dc1392f41","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4d1614b8c7902b61f4b5daa3d980ee8dc1392f41"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-07 05:59:30.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f35f24f9c5b424073fc78ae23a4c798876385845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f35f24f9c5b424073fc78ae23a4c798876385845"}]},"branch":"refs/heads/master"},"c9b7612cbcb8610b48ae38e41450c6b44394fb20":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2022-07-11 07:30:25.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/13"}}},"commit":{"parents":[{"commit":"f73356a2a58b879a9af45a0b410d3768b3ff6dd3","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f73356a2a58b879a9af45a0b410d3768b3ff6dd3"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-10 20:53:58.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9b7612cbcb8610b48ae38e41450c6b44394fb20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c9b7612cbcb8610b48ae38e41450c6b44394fb20"}]},"branch":"refs/heads/master"},"880c7d4cf2d4d2a9cf41abea0e5e2cf58ed5adca":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2022-07-11 12:48:45.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/14"}}},"commit":{"parents":[{"commit":"bc68eec0cb05b43d537d9df95bf14ed9617a365d","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bc68eec0cb05b43d537d9df95bf14ed9617a365d"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-11 12:48:19.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/880c7d4cf2d4d2a9cf41abea0e5e2cf58ed5adca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/880c7d4cf2d4d2a9cf41abea0e5e2cf58ed5adca"}]},"branch":"refs/heads/master"},"82236853bb1ff9b0fd584441b5c1323d252baca5":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-07-15 07:37:39.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/15"}}},"commit":{"parents":[{"commit":"c3c6a2013c029cff991272e590bcf880f7e1de03","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3c6a2013c029cff991272e590bcf880f7e1de03"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-15 07:37:26.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82236853bb1ff9b0fd584441b5c1323d252baca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/82236853bb1ff9b0fd584441b5c1323d252baca5"}]},"branch":"refs/heads/master"},"ee6b71343e10c075d22fc56c24f5263850bb408e":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2022-07-18 09:45:33.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/16"}}},"commit":{"parents":[{"commit":"8c283b41c8f376ea0aab1fce0f8095909580b398","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c283b41c8f376ea0aab1fce0f8095909580b398"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-18 09:45:06.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee6b71343e10c075d22fc56c24f5263850bb408e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ee6b71343e10c075d22fc56c24f5263850bb408e"}]},"branch":"refs/heads/master"},"6f5c4c8308c7acf08462335a89e8bca1f0168291":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-07-18 14:14:48.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/17"}}},"commit":{"parents":[{"commit":"68a30d29d1e5d689481eacd3571239a8d99eae99","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/68a30d29d1e5d689481eacd3571239a8d99eae99"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-18 14:14:12.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f5c4c8308c7acf08462335a89e8bca1f0168291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6f5c4c8308c7acf08462335a89e8bca1f0168291"}]},"branch":"refs/heads/master"},"77f7910cc4f0be7dbcebb46ed37bfa9354243472":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-07-19 20:00:41.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/18"}}},"commit":{"parents":[{"commit":"f4d02615dfe5f7b9dbf09484b6aa424e1f316130","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f4d02615dfe5f7b9dbf09484b6aa424e1f316130"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-19 20:00:29.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77f7910cc4f0be7dbcebb46ed37bfa9354243472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77f7910cc4f0be7dbcebb46ed37bfa9354243472"}]},"branch":"refs/heads/master"},"e8953b62cf5dec92689cc1023e8d9aefd82c3d31":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-07-20 05:30:26.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/19"}}},"commit":{"parents":[{"commit":"fd978fec19a8a3eb486c69bb2ada18d48fa11c3d","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fd978fec19a8a3eb486c69bb2ada18d48fa11c3d"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-20 05:30:13.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8953b62cf5dec92689cc1023e8d9aefd82c3d31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e8953b62cf5dec92689cc1023e8d9aefd82c3d31"}]},"branch":"refs/heads/master"},"6bec504f83131a83acfa3131d3fe0822bc90bfd0":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-07-20 20:31:17.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/20"}}},"commit":{"parents":[{"commit":"a44ea999d46ba601ef17b802c8b3d3e97648c3b0","subject":"do not merge","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a44ea999d46ba601ef17b802c8b3d3e97648c3b0"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-20 20:30:59.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6bec504f83131a83acfa3131d3fe0822bc90bfd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6bec504f83131a83acfa3131d3fe0822bc90bfd0"}]},"branch":"refs/heads/master"},"80d0b86fbc35f101b703a86e9fdba27fa64f67de":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-07-21 06:22:55.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/21"}}},"commit":{"parents":[{"commit":"749b143354d74986231181d5966dda50905bd2ba","subject":"do not merge","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/749b143354d74986231181d5966dda50905bd2ba"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-21 06:22:12.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80d0b86fbc35f101b703a86e9fdba27fa64f67de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80d0b86fbc35f101b703a86e9fdba27fa64f67de"}]},"branch":"refs/heads/master"},"65cc679c2b028127fafec8dba5c96a242c280dea":{"kind":"REWORK","_number":22,"created":"2022-07-21 07:14:48.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/22"}}},"commit":{"parents":[{"commit":"21b7d3431c1101c24116532ea7a2e09aa1a6c938","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21b7d3431c1101c24116532ea7a2e09aa1a6c938"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-21 07:14:36.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65cc679c2b028127fafec8dba5c96a242c280dea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/65cc679c2b028127fafec8dba5c96a242c280dea"}]},"branch":"refs/heads/master"},"d2e98d922a5b1446055a8c4c8071f7f311c6b348":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2022-07-21 10:16:49.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/23"}}},"commit":{"parents":[{"commit":"116da010e6019d37ff9af7f6149178946f7f5927","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/116da010e6019d37ff9af7f6149178946f7f5927"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-21 10:16:36.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d2e98d922a5b1446055a8c4c8071f7f311c6b348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d2e98d922a5b1446055a8c4c8071f7f311c6b348"}]},"branch":"refs/heads/master"},"1e71dfa7485f20193ede30c9a91c748bb80d3bfe":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2022-07-25 13:00:02.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/24"}}},"commit":{"parents":[{"commit":"146d273edf94df67bea8c53108c6c5c77366aa84","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/146d273edf94df67bea8c53108c6c5c77366aa84"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-25 12:58:53.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e71dfa7485f20193ede30c9a91c748bb80d3bfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1e71dfa7485f20193ede30c9a91c748bb80d3bfe"}]},"branch":"refs/heads/master"},"539f0db4e70be359955c77fcc6bc1cc8feb3c6d1":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2022-08-01 08:33:33.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/25"}}},"commit":{"parents":[{"commit":"c0185775aecfd696139364e2619cf7835907f2d4","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c0185775aecfd696139364e2619cf7835907f2d4"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-08-01 08:33:22.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/539f0db4e70be359955c77fcc6bc1cc8feb3c6d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/539f0db4e70be359955c77fcc6bc1cc8feb3c6d1"}]},"branch":"refs/heads/master"},"dc6f01113af562a4ed841e1a72141fdc58bcde6e":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2022-08-01 13:44:40.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/26"}}},"commit":{"parents":[{"commit":"06e2d89cb9b5e29b41471e19c7ba1825ddcb3a44","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/06e2d89cb9b5e29b41471e19c7ba1825ddcb3a44"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-08-01 13:37:51.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc6f01113af562a4ed841e1a72141fdc58bcde6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/dc6f01113af562a4ed841e1a72141fdc58bcde6e"}]},"branch":"refs/heads/master"},"4963ee100b361e69cc9f9fb3c5308c71adb9da28":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2022-08-02 09:33:49.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/27"}}},"commit":{"parents":[{"commit":"ca3aae34244e9d017abd5776effc20a91e30c345","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ca3aae34244e9d017abd5776effc20a91e30c345"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-08-02 09:31:57.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4963ee100b361e69cc9f9fb3c5308c71adb9da28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4963ee100b361e69cc9f9fb3c5308c71adb9da28"}]},"branch":"refs/heads/master"},"2a2fe025f024dd7091e67cd5b7fa3f61556dad80":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2022-08-29 13:21:26.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/28"}}},"commit":{"parents":[{"commit":"b74d019ed094d658d1a81dedee731941bd674030","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b74d019ed094d658d1a81dedee731941bd674030"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-08-29 09:09:54.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a2fe025f024dd7091e67cd5b7fa3f61556dad80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2a2fe025f024dd7091e67cd5b7fa3f61556dad80"}]},"branch":"refs/heads/master"},"7a1e253851dfb9b46cf7734f5a58a5a859b7c984":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2022-09-01 16:21:23.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/15/840415/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/15/840415/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/15/840415/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/15/840415/29"}}},"commit":{"parents":[{"commit":"6ec0bc70a7907ab2d83d1dfe0b177ddf17c79d61","subject":"ovs: make vlanmanager to handle more vlan mapping per network","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6ec0bc70a7907ab2d83d1dfe0b177ddf17c79d61"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-27 12:20:54.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-01 12:48:54.000000000","tz":120},"subject":"ovs: use a local vlan per network/segmentation","message":"ovs: use a local vlan per network/segmentation\n\nThis is using changes introduced before to support for a network more\nthan one vlan.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ifd61e379c3cef3589803c96a276da9827051f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a1e253851dfb9b46cf7734f5a58a5a859b7c984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7a1e253851dfb9b46cf7734f5a58a5a859b7c984"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
