)]}'
{"id":"openstack%2Fneutron~840418","triplet_id":"openstack%2Fneutron~master~I56b22820d29b2d57faf28a2f9b685ab0b2c924b4","project":"openstack/neutron","branch":"master","topic":"segments","attention_set":{},"removed_from_attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2022-09-05 10:11:26.000000000","reason":"Change was submitted"},"4694":{"account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"last_update":"2022-09-05 04:01:20.000000000","reason":"\u003cGERRIT_ACCOUNT_4694\u003e replied on the change","reason_account":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-09-05 10:11:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I56b22820d29b2d57faf28a2f9b685ab0b2c924b4","subject":"ml2/mech: bind port on segment that subnet belongs","status":"MERGED","created":"2022-05-04 07:07:57.000000000","updated":"2022-09-05 10:13:09.000000000","submitted":"2022-09-05 10:11:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"840418-segments","meta_rev_id":"6bd6017b51c5f9bfbf48d7f24c9a126e3e66f729","_number":840418,"virtual_id_number":840418,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-09-02 10:44:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-05 10:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-05 04:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-09-02 10:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-09-02 10:09:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2022-09-02 14:29:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-05 04:01:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-04 08:32:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-04 08:39:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 09:08:29.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-05-04 13:22:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-04 18:24:24.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-25 08:52:57.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-07-29 07:22:59.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-08-10 20:44:43.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-09-02 14:29:14.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-09-02 16:10:34.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"CC"}],"messages":[{"id":"c606cc1b67e007ff1bfa302105a25bb16c38b891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-04 07:07:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a70d4a355df987a273e790f9697a3eda719f0f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 08:32:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc106c8b8263433d8832fdb3241a5aec : TIMED_OUT in 1h 21m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8de5bd2428674498b15420469b7746df : TIMED_OUT in 1h 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55aa541071ff07c8c9f4508036ae2bf3119fa38f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 08:39:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/c9947381ae5e46f88e5d1195fbac5127 : SUCCESS in 53m 56s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/a2929e880f8b4531b44fe60e0a40fef5 : SUCCESS in 57m 06s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/5263b0c6c1d246109f4b69cb9325aeb3 : SUCCESS in 59m 25s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/312abcdd3e924282955ef251dffb5a66 : SUCCESS in 48m 20s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/058afe3805794de1bba10f4ee9bb0923 : SUCCESS in 1h 07m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76f1fc85a2074e6b8d3840c5f70cc28b : SUCCESS in 43m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da72608e9a644e69a29274918738c475 : FAILURE in 13m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c04e6b535b846e695ee8ecb072381d0 : SUCCESS in 25m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0e914ac54af4d108cd0fee734c43f06 : SUCCESS in 25m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f112e3f3c7284e1cb0d8fb16bdc257a1 : SUCCESS in 27m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5013fc1059044db595daaa92229de159 : SUCCESS in 9m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2afe980907664733b93325ff25cc495c : SUCCESS in 1h 00m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d89012fe58c435aad3d8b30313e7183 : FAILURE in 1h 18m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/eae0e8580b5641bc9e4caf155727181d : SUCCESS in 1h 23m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f553d4b3ed4d42dab01e6455deba3c39 : SUCCESS in 1h 01m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cf4bc2f482274097ab8166c61c9d7e1e : FAILURE in 1h 03m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d6e053a6fa994a489336ef983627f53f : SUCCESS in 1h 15m 21s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0138b96d26614dc6a096712059ceb35d : FAILURE in 1h 21m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1e768c36a49490db625a44bee8259b4 : SUCCESS in 58m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/425c564566ce4fd6a80b57b4260404b9 : SUCCESS in 1h 15m 44s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py","accounts_in_message":[],"_revision_number":1},{"id":"08b83ebc4d2d0134cdb66635786dbd1afb7c8069","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-04 09:08:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/1 : SUCCESS in 1h 52m 06s","accounts_in_message":[],"_revision_number":1},{"id":"dca9d0c51528f7ce715b83f78f53d2a80f09c660","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-04 13:22:23.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/18/840418/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"1a0ac6dd32ac362da7113f66bf777a1b26342aa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-04 14:34:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"26b04e98075d70cf831359289dc2c3cdfcdcecbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 16:12:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/42fcaf7eb8bb4388a8d66ba3d3658c4a : SUCCESS in 41m 29s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/993e51dee3f44d17bbde5c316529c398 : SUCCESS in 58m 25s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/90d563ecd8b44a8c842b1c426345c3ec : SUCCESS in 1h 03m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f87e5fddfb58416994cbb6580d283872 : SUCCESS in 49m 16s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/ad72fe85b995498bbbe029e473f5997d : SUCCESS in 1h 06m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98b14ac697474bf3953aaff02aec7a5d : SUCCESS in 30m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18446281b9fb4277b3dc18541e0cd006 : SUCCESS in 12m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e64d7340d004424b530e89179280e49 : SUCCESS in 21m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ff998a465914764b8c045b04c50564e : SUCCESS in 22m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb028a7284794555b83abb709ebe51b0 : SUCCESS in 25m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c721da56c7a4291828ac3e19f8be422 : SUCCESS in 13m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d408d068ddc6400393c48ccead98737d : SUCCESS in 52m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3475ab99665428098ac4e6e318f2483 : FAILURE in 1h 20m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8a529959d9b64a93a1235c09628d4cfa : SUCCESS in 1h 16m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c56c72eabbdc48a1ba12850bdc54331a : SUCCESS in 53m 56s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/855755a013954bb8ad2f217e8bf663b3 : FAILURE in 56m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5c46e446780f49ecab2dd9f13885c120 : SUCCESS in 1h 31m 29s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f9f4319e775f4e15a787f3f350dcf93d : FAILURE in 1h 01m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b461d04a1f74b139435822958adb34d : SUCCESS in 1h 15m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9bf8229495e44a828159993a0b250476 : SUCCESS in 1h 05m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d5fe6942fcc3de730f312ab771cc7a4a8b76218","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-04 16:37:45.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840418/2 : FAILURE in 1h 54m 43s","accounts_in_message":[],"_revision_number":2},{"id":"c421c8d51960828d3bab3f979d3c81efb2c06cd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 17:36:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83f7d80e9db146718331813660ca68f9 : SUCCESS in 59m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c2f7d21fbfc343119a12db329661fb00 : SUCCESS in 1h 14m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3f2940d372d10720fa70ef6ffeec2ba08b13898","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-05-04 18:24:24.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/18/840418/2/openstack-sriov-neutron-master-direct : FAILURE\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/18/840418/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"46069f1b4572018228f5bff974b44dc3b1e1b131","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-05 08:16:36.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/18/840418/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5ffe179d58ecb35f9418580e3fd300f2db604df6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-11 17:26:44.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"470f9449b747b572ffdf5849ed20e5abc3840d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 18:52:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/9fa1e97933414fd0b1f40391685014b3 : SUCCESS in 36m 34s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/bb7af286fa604760a453f7152ce68b07 : SUCCESS in 1h 04m 08s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/c8024aeaf40445a6892e40dd26aaad57 : SUCCESS in 1h 05m 38s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c0f5e22304484cc29d63ac4c5faa6aba : SUCCESS in 49m 31s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/6088e2e8b8a44f63b433f2e2454597ca : SUCCESS in 54m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4a50110b0fa41e8b65664876ede182f : SUCCESS in 30m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aead22da186a4574a4572e4501c1d837 : SUCCESS in 28m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f87eae6202ff4f85aa0d6ff36cd3d1ac : SUCCESS in 24m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34ea4759b0a24c69ba35d103c430d567 : SUCCESS in 25m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7cf7af90e06649c5a4e7bf9434d8604d : SUCCESS in 21m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4990e2bd7f2d431cab25e02dc851df47 : SUCCESS in 11m 09s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/292b37f4263649658afb4293790cb24f : SUCCESS in 58m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f035c591bcc54d73bb6301364ddf9f76 : FAILURE in 1h 19m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/617ce14fdb7640918e77a2498a5e574b : SUCCESS in 1h 12m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/473b243331b949ccb3751a3db3b422a2 : SUCCESS in 54m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b96f2c73ba8c4f80936c2d16498744e9 : FAILURE in 56m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fd80b98ec16344a38381bfc81fb955f4 : SUCCESS in 1h 11m 16s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e676978093254f02bf97b08bb309f9bb : FAILURE in 1h 12m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1762ac2f71c49b1ac76d93f4ffb8432 : SUCCESS in 57m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f5f3be6f50d24949a6ab785acb001cef : SUCCESS in 1h 19m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"69fc42c79639aff9e6845415e9be01bb512b9efe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 18:55:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c151353169c14bc8936bf6e33aff1000 : FAILURE in 1h 16m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3557f3c6d2ce454892f454d2a0ccad24 : FAILURE in 1h 21m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2a81454b8e6239eaadd9ae0b326e0d7e3795c0d3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-11 19:22:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/3 : SUCCESS in 1h 48m 28s","accounts_in_message":[],"_revision_number":3},{"id":"80085edd18b56669ec3314dafab713349bd2d88d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-11 20:53:56.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/18/840418/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"da817b290c683a33b29b9f886e865689ebf2e74d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-16 07:11:36.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ed44912fa103b2595bff5ca910bba26c5469b245","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 08:37:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1653aa7a6e54279abba194c8e36cef4 : TIMED_OUT in 1h 21m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f78e6db63b234a00a72e8101bebef42c : FAILURE in 1h 21m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1ff64d121d65fc0fee100438dd027b057d9ac617","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-16 08:55:56.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/4 : SUCCESS in 1h 36m 22s","accounts_in_message":[],"_revision_number":4},{"id":"cee7c2f4b618dc7f0b2bd589e30bc2010c4d7dd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-16 09:09:31.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/93aefb1974614ceaa99f931c05f9db95 : SUCCESS in 40m 54s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/4004cca7f38848f7bebd54200c39dcfc : SUCCESS in 1h 14m 54s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/1b3b587c13064125a90e810876d313ef : SUCCESS in 1h 25m 14s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/db4e317310d94577862f1dae5c46527f : SUCCESS in 1h 09m 33s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/d50e223ee9164406a3dca630dbd46ecf : SUCCESS in 1h 12m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a13e73013164b33abca377e34f7f7f0 : SUCCESS in 38m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6fbc534d03a4c7c83b615806c1f8c49 : SUCCESS in 14m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b269a7de283d43e09b5c00860bb386cb : SUCCESS in 23m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95f493859f4347c0a79b86e3165e5c5e : SUCCESS in 26m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96c95300a0ca4461ab1b2ac561484bc7 : SUCCESS in 27m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9918d5b33f7947d2b084c4c91c6f742c : SUCCESS in 7m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fdb7c815b24647d9bca3cab687fb21da : SUCCESS in 52m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c9209aaf9e041eea999f8f14f5fe110 : FAILURE in 1h 22m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e3b96075767a451797ac88a907cbbc54 : SUCCESS in 1h 14m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f639c64b26c4de59f76905d6de21bb5 : SUCCESS in 46m 00s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/40c67a90803c41f1861cd56d8d5deda1 : FAILURE in 58m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/18393008f5c6434b9388983b9bafdbe9 : SUCCESS in 1h 12m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/55b1d6375be74716bd791fb5af39f76d : FAILURE in 1h 08m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac770c40c5ed4cef9b17d4c047bb1df9 : SUCCESS in 1h 06m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2ba80edac0af497a8723717428a31fca : SUCCESS in 1h 18m 13s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"68b25a1a7f05c023bdb1b71f82dfff1833a3cd08","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-16 12:26:43.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/18/840418/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"473fbc322e6d8a958025c9714218f9bbc8067162","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-23 10:21:21.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c52b89c26fc21ebbd5c3e20f2f1cd26d7367a1aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 11:41:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ac73cd83865b49bab04d1ae759b43baa : SUCCESS in 44m 45s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/7de59e4800fe42ae9f8f34c321448d3e : SUCCESS in 1h 15m 27s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/168beac2d2b54db091fc2e1f52220e4a : SUCCESS in 1h 14m 49s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f508f61d32364629a34845e4fcc0fe9b : SUCCESS in 1h 02m 54s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/2c0a2336e0e84e499a68c26d5f795f23 : SUCCESS in 1h 03m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/089412a95aa6470197c873012b857a43 : SUCCESS in 35m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5385d49777a846b4bbc0e7aa4e90798b : FAILURE in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6293e2ccea67405dbba57aba734e4011 : SUCCESS in 31m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68280236bcdf4b9cb369aeb1e29e98fc : SUCCESS in 24m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b8a876f30f84d6ebd86ccdf707c168b : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e2cffdfa124876bed47f4a3d828fbf : SUCCESS in 11m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/00e11f495f5147ada815354ab90ef893 : SUCCESS in 52m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/727c229b1d614ce3b5dc59f7ba90ac16 : FAILURE in 1h 13m 54s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/fc83e4155a614b6fb510556be35b1c9b : SUCCESS in 1h 05m 34s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa3b85a141e84308a4f88f1a072024b7 : POST_FAILURE in 54m 42s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/01571f060700486dbd9fb9282417fe70 : SUCCESS in 52m 54s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c8af9be5a7c74569bdea88008acaeb9c : SUCCESS in 1h 14m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0b06c199343a4a8ea3fbb979d776d46b : FAILURE in 1h 02m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6cc4ccee8bd45539de82444283061b1 : SUCCESS in 54m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f74d603b9dd24d09bfba4b22dd8d04a0 : SUCCESS in 1h 07m 12s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/linux/dhcp.py","accounts_in_message":[],"_revision_number":5},{"id":"19da761147dc18c3f07d90ab9564256130bad0e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 11:59:30.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fac32780300f45f8bf5f7eb8b0067063 : SUCCESS in 1h 02m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/93cd6cfaeb154372b83401f1dccb20ff : SUCCESS in 1h 06m 59s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4506339064acf54c5c05571701f7ec0a595fc021","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-23 12:22:32.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/5 : SUCCESS in 1h 53m 39s","accounts_in_message":[],"_revision_number":5},{"id":"4519b747cebe9ff48715650763aad40944612ec9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-05-23 12:29:52.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"b6bf19db10ffbfd1f9209243f753b5e791cfc683","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 14:14:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/ef1e1f638668444895f29b46612a567e : SUCCESS in 46m 49s\n- neutron-tempest-plugin-scenario-linuxbridge https://zuul.opendev.org/t/openstack/build/f771c1a52dfc43bf9d64dd1834c07c66 : SUCCESS in 1h 06m 30s\n- neutron-tempest-plugin-scenario-openvswitch https://zuul.opendev.org/t/openstack/build/49e9a1cbee914539af5f37ee8064d2bf : SUCCESS in 1h 04m 11s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/74bc521286044785a99b39a56fd5df6f : SUCCESS in 50m 22s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/07f2404b53c241faad95b8da3b1bce87 : SUCCESS in 55m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/967542c9db7e4f3a95aa892130514a6c : SUCCESS in 39m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31fc6ff885744d39b913da0658e16ba9 : SUCCESS in 12m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2061d422ee6f48c4892e1b1d6347b4ac : SUCCESS in 25m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a18b6f719a114e58a4aa16225b0057ca : SUCCESS in 29m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90e0ecc683884b54a62cadd389e90055 : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/794b6531b17b43e782c3b3ec64a328ed : SUCCESS in 13m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/27ec617ef0194d9ba5503b5e7c9befa7 : SUCCESS in 59m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d66d895b34c46c5b6f4f6cd0ec01d7c : FAILURE in 1h 40m 40s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3e66c24d3580466493827ca78a7cb92c : SUCCESS in 1h 02m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/46f6b5257aff42b8b75688f9ed2b0486 : SUCCESS in 44m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/58261c77fffb4ec5b8581622db3b85cc : FAILURE in 1h 07m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2426db3027204a1ca139f24aa0331a53 : FAILURE in 59m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/546e112502544ba9a2f189de1de6f2e3 : FAILURE in 1h 13m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d371c38399e4f7fb60eff0b56e6c3b3 : SUCCESS in 52m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b6572fceb4244feebd53ee7abbde018a : SUCCESS in 1h 04m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"42572e3df85845661b120694d4a4e95ed48d650f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-23 14:36:45.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/6 : SUCCESS in 1h 57m 53s","accounts_in_message":[],"_revision_number":6},{"id":"ac1860a62c33762895b3023857d424eef118a807","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 15:11:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/01c008d1a4464f86bd994cc5d63798de : FAILURE in 1h 20m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc928d3564a44d49a387bd709a78fba4 : SUCCESS in 52m 39s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"7628a108560d1b328ca82a1791278f27987826e3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-23 15:14:27.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/18/840418/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"5a6f780f0fad9b22d7a6d5bf27d133f7c9e99033","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-24 04:30:14.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/18/840418/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"a07aa2473cf72e801fcd226a13a88780a7c0a522","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-06-17 07:10:14.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"a5bd2adc9305b6672116b6aa2034f298a15eccc2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 08:41:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f3b4518bad341898d2eae011da2d221 : SUCCESS in 1h 00m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/253d074051f5471f8caf14f2453493a6 : SUCCESS in 1h 04m 02s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c6222fa61d66d395815eec70f84c9ade75c8ff78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 08:57:34.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f6fe876a80e8472a8ceb17d9d487ea29 : SUCCESS in 1h 29m 23s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f522ef244fef40c8b1dff931fe3f4eb7 : SUCCESS in 1h 25m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/26a67d548cdc446aa386a2c2d0774f6c : SUCCESS in 1h 43m 37s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b728e991671b4710902243c1a223cee5 : SUCCESS in 1h 26m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17017d6f66104c3d9c7848407f488475 : SUCCESS in 37m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05615dba1364433ca6fbbbfcd6c52ced : SUCCESS in 14m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94168fd79a3d475cb733730a41fb4845 : SUCCESS in 26m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/208a3db114704634ba6558bccfc5c17d : SUCCESS in 25m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b67512fc907844bd932ea9e63cdafdc3 : SUCCESS in 25m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3b8a1a2507a4a1b874cfc78253fb0f9 : SUCCESS in 11m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d3d26d8b6704429ea9676c21326ddf17 : FAILURE in 48m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2054fe244ed4050b439e057e0a3ba87 : FAILURE in 1h 30m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/587ecaf21eb74473be2ee3d1964acd56 : SUCCESS in 1h 02m 38s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/da36914b3a2a4c11a8129788d79cf5c5 : SUCCESS in 59m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d705554ed37d430f9e173ce2e1cc1877 : FAILURE in 56m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6868a889fc6f47ffbfafa23d60fe2460 : SUCCESS in 1h 28m 42s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/8a17db16d0f84091875da8bded469948 : FAILURE in 1h 04m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff782ce73a114225a37847c5e3718a79 : SUCCESS in 1h 10m 37s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f7e83a419d6c42b18ff8fce2cc68e020 : SUCCESS in 1h 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e78959f3ff5628b6d877ed246c1bae871532b0e0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-17 09:00:32.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/7 : SUCCESS in 1h 41m 35s","accounts_in_message":[],"_revision_number":7},{"id":"18fa55398b348964adf9af6b5b0482ac6f3b69a5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-17 12:35:36.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/18/840418/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"2ac94c84c3a5a5cf4ac09a97ee178f3db729a45d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-06-27 07:15:45.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"da23a51662f63d50ab25de572c82b64f9b3d8b24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 08:34:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0941ea970aac4d99a27988821babc1f3 : SUCCESS in 44m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d233d5a0ea443ea8c702fb56a606d00 : SUCCESS in 51m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"6bd1eedbff662875663241f6843b4bc25bdac546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 09:00:59.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/506cdda293fd460ab205dc7600813fe9 : SUCCESS in 1h 40m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8aff5e1b5e494b54a1d222c6a63a7ad1 : SUCCESS in 1h 25m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4643caaf712f4452a557d59f6f6fe7ee : SUCCESS in 1h 20m 09s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/afd1f5071be7446e87288804aaa45677 : SUCCESS in 1h 13m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a16f7fe0be942c8bf39eec2443c75d8 : SUCCESS in 36m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f210f8ed6534c8399a0a11524e0982d : SUCCESS in 13m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82086f9796b94c23bef60c40d46e64db : SUCCESS in 26m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb77419714b0478db349c8332b938c96 : SUCCESS in 25m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/816c2a24a66b43a4a595bc2e9bded4a6 : SUCCESS in 24m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a400140b935b416c9affe1953686b6db : SUCCESS in 9m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/67505b5f68c64aa5b56daad29f7d25c5 : SUCCESS in 57m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/be50be4a45ba4e5b982d47651df552eb : FAILURE in 1h 14m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c82014c5ee39433b8b07736f7a502ef2 : FAILURE in 28m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5a7cb36014944cdb3edd203de566de4 : SUCCESS in 1h 04m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8a0b16b6ab3c47549ad1c3c0b5301a56 : SUCCESS in 1h 06m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/00ebb0b72d994c1bacc88b636f09c5e1 : SUCCESS in 1h 16m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7693ba9cbccb4e2c8363d7c587449db2 : FAILURE in 1h 37m 08s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ae83bcb23c64df58f5ef5fb2a53ca29 : SUCCESS in 1h 06m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d388b85b1fa143aa98678ad8aad96065 : FAILURE in 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"974ba76e95854db50ff36d50d824c059ebf6b1e9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-27 09:29:51.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/8 : SUCCESS in 2h 06m 56s","accounts_in_message":[],"_revision_number":8},{"id":"e47e60feb10caf8a0c6247e32274848119d7e3a5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-27 13:14:24.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/18/840418/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"26fa69aaa96e6062d9f0e40438dabe3a362614ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-06-30 17:32:53.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"9700c11d0645dc681c5528192d38a6614d3910d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-06-30 17:43:20.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent, and commit message as Patch Set 9.","accounts_in_message":[],"_revision_number":10},{"id":"f026c85a30fbc2b0a7842bf07d40a8e42ae3f947","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 18:33:13.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/704d26eb38bc4c8095aad2b4471b3775 : SUCCESS in 44m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20ab60f571e14b57835701d8ff875489 : SUCCESS in 37m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c94880a2c82a3439d81c47e5de003b560025e4cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 19:22:58.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/44665f0bed6c4fda80252ed2dbcf9591 : SUCCESS in 1h 32m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/52b57eecd5a74a7b80a256069ccdc4e9 : SUCCESS in 1h 27m 33s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6bcc043ceb8d4fddb7cae7913ba0fed0 : SUCCESS in 56m 16s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/db6ecb577fc94ea8aefd6b5c23439360 : SUCCESS in 1h 21m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4840596270d42df8cf36cd0fbddf7b9 : SUCCESS in 44m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ea9426f794d4333b0719160bcf1abe8 : SUCCESS in 13m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32f72e0ad7d4462495fb94c5c4445f6a : SUCCESS in 25m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/652b570378ae436595d7482d2d625c75 : SUCCESS in 24m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2aef3bf8a404f76b3fe6ebc464e7b43 : SUCCESS in 25m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d3c140ced54ed1b06aaa1a33b209b3 : SUCCESS in 11m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e9c9fc521ff24277b0ff0de62fa88809 : SUCCESS in 1h 00m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d683cca6f5954c0b9bea9dfda5fa37b5 : FAILURE in 1h 18m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7f512680285f44a49b6228a8b32616ce : FAILURE in 31m 21s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d2ae8a43c1364d26a7b32a8d3eff7dd1 : SUCCESS in 1h 02m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/38acafeca1c74087864fa7d77303e966 : FAILURE in 1h 06m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ad4b7fad3c5f46809d67e7fdef27d7e5 : SUCCESS in 1h 25m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e4f26f44bbe843049af9c0ffb7810769 : FAILURE in 1h 20m 34s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/002fd9e5bb94437c8b333853a34030c7 : SUCCESS in 1h 00m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/920aeecc7d614110b4197cdf44b84cee : FAILURE in 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d339591c0bd3e784adf0686967aa3ec00a3ff904","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-30 21:25:09.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/10 : SUCCESS in 2h 26m 00s","accounts_in_message":[],"_revision_number":10},{"id":"70b6628a7137100216f039134ab780e913feba63","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-30 23:19:32.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/18/840418/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"4e9aa09dad05ea9aec71775b69ed3a0ebb2caf88","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-01 08:55:58.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"87e12cae3eb3ba04b7d1f731566f123c892011a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-04 16:01:24.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"ca1b425a7e25b880a85ac9fc3082fcd49f9974e6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-04 16:19:06.000000000","message":"Patch Set 11:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840418/11 : FAILURE in 13m 53s","accounts_in_message":[],"_revision_number":11},{"id":"cc97a1d2975b0c63f1fbd9ad3ca79c5307b8c066","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 17:00:34.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/721b5514ad3e41e68962768184e3905b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/55a776593fb443269d7bf071c7c7d845 : SUCCESS in 51m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a9b27d64fd54133ba253c64554d1b84 : SUCCESS in 51m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0ae372750db761953df52a594285a5697b4bc21d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 17:36:05.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/894ea55fceae4a9aa8d1581e1771237e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/082eff488caa42ed8af752245b908974 : SUCCESS in 1h 03m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/364507fa540d4e4bb161c24e7b10cb06 : SUCCESS in 1h 06m 03s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/30499449c1694538b82964d2e6e5d748 : SUCCESS in 1h 00m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c802efa3a1224d8b8382e1e8e4fa1e09 : SUCCESS in 1h 13m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19f3729ffac043d687d7ff7d45ebf3ed : SUCCESS in 32m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00385d853c824aaab5f99954804c851d : SUCCESS in 11m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eac70bb591c04a8a94a59fd4ac994de5 : SUCCESS in 25m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64a099dc1c9741bb8af36cdc92066c21 : SUCCESS in 26m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/171f6e5994704276980af657d3bad47c : SUCCESS in 20m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a434607b943048f6856165d07ef31ef2 : SUCCESS in 10m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/195afc85dce046efb2e129b5cf01bd42 : SUCCESS in 50m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/420a188ab241426cb3fe534cf282a934 : FAILURE in 1h 26m 10s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2ab8b76ddaec4961a3b4ef26c5afc84e : SUCCESS in 1h 02m 45s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0ba09b8199447c3ac03d87cd6ba5be3 : SUCCESS in 52m 32s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/66a2b403ef8447cda84203b983eeaf43 : FAILURE in 1h 07m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/726e730d525d45dd969d1c25eab6b70a : SUCCESS in 1h 07m 46s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1f5d5fa3e91447a999ca5ef828fa03bb : FAILURE in 1h 17m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4874ed83725439b862be5fba9ffe410 : SUCCESS in 56m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d621e498a4344ff6844574d9b71808b3 : SUCCESS in 1h 25m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a3e0fadf9b9cab21a2a94afc89aa522301b58e3c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-04 20:13:52.000000000","message":"Patch Set 11: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/11 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":11},{"id":"f7b248a814aff701547f5923f641b56be6f98380","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-07 05:59:44.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"1359471516ff09c23e675486fc21b5f88bcd5eea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 07:21:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fba02e8019d4fd6a953700bc2947dab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/989fbb1865a2457ea593b80854a9ce98 : FAILURE in 1h 09m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89e9dc4cab334028a7ebd0ea04f1623f : SUCCESS in 40m 50s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9387a128cbf65ec2de93eb887a00d1b1be35d97b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-07 07:59:27.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/12 : SUCCESS in 1h 55m 54s","accounts_in_message":[],"_revision_number":12},{"id":"2897c88983a1569b44f6b1b21d6f078e5a302f9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 08:01:43.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/746e7d75bc394ef0bdca8966dd62dc10\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8751ddcefe70469c96bc682cfd2de52b : SUCCESS in 1h 53m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5144ce4184ea47db841c04a8cb41e9dd : SUCCESS in 1h 22m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/89773b2035024773adf6b4cb8bdea07e : SUCCESS in 1h 10m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2510ca5e38fe4c09a2a96a405e59ee12 : SUCCESS in 1h 09m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8d6652c52cb4fafb78c24648239d290 : SUCCESS in 35m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3547171ea024bec8b14a628aca303c5 : SUCCESS in 11m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93914a10cd7f47fb8b43ed31c47e4764 : SUCCESS in 24m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c2559053327491385acc497edbf1e86 : SUCCESS in 20m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe44ffb958ba46b1b9902d6bdd0b51da : SUCCESS in 20m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e6e2e5d71240a48cfa28d34a5bb08c : SUCCESS in 11m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e133beacb7740a3a0812469d426496d : SUCCESS in 51m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/67432dc0022349afb8dd11c3e3fa0b45 : FAILURE in 1h 32m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f0c0da2bf7bb416a8d6c54007c65d8fc : SUCCESS in 1h 00m 33s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd997a8e13414450a1ce9e06fc099354 : SUCCESS in 51m 14s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/83c6155124d44a02a2933f7ed26f8e6b : SUCCESS in 1h 03m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/34715def218e415fbe598c018e83e963 : SUCCESS in 1h 05m 23s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a8fb2181610a4958a8945162aa996974 : FAILURE in 1h 08m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7159cfc1369471891a1048c2ec823de : SUCCESS in 55m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1fd0407de70f4ad4bbf28d436ffbf6cc : SUCCESS in 1h 09m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9ac011241c38a98837220e175153292c145a575f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-07 10:28:57.000000000","message":"Patch Set 12: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/12 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":12},{"id":"da480fbb21a78526a33b819e2142ab27432f0834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-11 07:30:25.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"038ca510d170540f3ac313ed5e99e92ccf920304","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 08:57:23.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e115038310be40e0a0e06ff257cd0315\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5244a098a9246ddb510719ee8c2deac : FAILURE in 1h 15m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6ac8858f9324e2ab7c4ba6f49c5652e : SUCCESS in 47m 24s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"cb4c05856f1013c88472ec4fe41ce40f1d6e73e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 09:12:53.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/622b5eee83784cc9800c73eb7a508bd8\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/acdb8e9949fb433db882f53bd1fc7f24 : SUCCESS in 1h 10m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a2fc622e931b4e1f83ed97404ff73e37 : SUCCESS in 1h 23m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/37a5f0e28850449ea8bddfdd0bae3eda : SUCCESS in 1h 21m 22s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/775b835411a549508f4dcb4cbbd4ef09 : SUCCESS in 1h 23m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4648138c1c840a3a1a278e74840bc49 : SUCCESS in 36m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0bb6255d0ad4f3492ce836c579fe191 : SUCCESS in 14m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0f65727d02d4aabad7fec142b926b73 : SUCCESS in 23m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04ac05e597ee4596bd2c9d9b57ea2292 : SUCCESS in 27m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d697aebc5a1348859393f16592feabe5 : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea2496fbf834c6da519f2623943fd62 : SUCCESS in 9m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ea3e61a07a6043e58e8e6a48777ac37f : SUCCESS in 58m 00s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a4222cdec1494980b59632c0f6496020 : SUCCESS in 1h 36m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/171d93c17be74e43a50bb545846b0cd2 : SUCCESS in 1h 21m 44s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd715139371d475e9f02bf0540d8afed : SUCCESS in 1h 03m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f7741503d73445cd93fb93ff4f1e7e6d : FAILURE in 1h 02m 02s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/60f25947a830441e8b0049233aff987b : SUCCESS in 1h 21m 04s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/fbdbcbfe057a46bbaba9974242d2dbc3 : SUCCESS in 1h 24m 17s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25e9ce2107054ea0bbbbb7fbc8030b21 : SUCCESS in 1h 01m 29s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ca42861e2b784dd480249b4523e03e12 : SUCCESS in 1h 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"6d6948c648fd4138fdfbb8e8462de022c303f20f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-11 09:35:30.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/13 : SUCCESS in 2h 00m 52s","accounts_in_message":[],"_revision_number":13},{"id":"9ac3a4bedf336ed9885ea2dbf161104c38aff302","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-11 12:29:14.000000000","message":"Patch Set 13: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/13 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":13},{"id":"10b9bd1be6239c49efaefed17c74f8910f43d2f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-11 12:48:45.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"d4d4145cfe393aba599e0eb50838e6960a359ad7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 13:44:19.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02cbba4180dc47c791e46517a261c371\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/596d26aa671c40aaa82df08406527178 : SUCCESS in 42m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da0e48e90fc049fe923421e0227559ad : SUCCESS in 44m 38s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"97e9b8a3c9bc80c9030076284d950da16c32f3e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 14:39:49.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90f3efb8d5684a078d68a42872baf3ad\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/474602767a004b2c82ae76b9e711e598 : SUCCESS in 1h 17m 11s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9c4dd65572db445e970bddf80e321791 : SUCCESS in 1h 39m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e1daef1e61b24e789d1dda3851b5f8e8 : SUCCESS in 1h 17m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6dcd7c20ae5149648c1853f16d7d077a : SUCCESS in 1h 13m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ceafbb14639b423f852d001ea1255956 : SUCCESS in 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c921ea1721f342c098f781d9dddb1aef : SUCCESS in 11m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffdd3cc3bc03409591382ab29aaf61ae : SUCCESS in 21m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc78e4bd7d2f436091b4697e174413e7 : SUCCESS in 21m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bade7e83d17e4a798c162d9efdaeb73a : SUCCESS in 21m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a629285dec343a7a23775729757a2aa : SUCCESS in 11m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e13a90111d441189afaa1c446da74e3 : SUCCESS in 1h 11m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/673f715e92c64bb6b01bb1b9126a3568 : SUCCESS in 1h 40m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0f9fc6dba83b43bbaae7bf0238c33b81 : SUCCESS in 1h 23m 15s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/350cdfa5cfc4431b87feb61020406282 : SUCCESS in 1h 09m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c9a6b906527148b88c159aa5b6944b35 : FAILURE in 1h 08m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fd1b03f552fb4848affd67a1c0355b1f : SUCCESS in 1h 16m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/054926e43d6f424fb642bc0319586152 : FAILURE in 1h 47m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99c61e35363746c28b42839a25f331d1 : SUCCESS in 1h 18m 27s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8cd83d84589a453eaf4b023d7566a42f : SUCCESS in 1h 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"f2cdb8b8174fca03d1eb30308d11d590cff137c3","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-11 15:25:26.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/14 : SUCCESS in 2h 32m 34s","accounts_in_message":[],"_revision_number":14},{"id":"9ac49b6ea0561ebc82e0c46981e2b63924fdcbfc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-11 21:42:37.000000000","message":"Patch Set 14: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/14 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":14},{"id":"c59014133b80faf241c5d1d29644a6dba0a256b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-15 07:37:39.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"ade82e21e23a8f57aebf34a7abbc45e799bef59d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 08:27:52.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72ddeab4724d4126825c178f0b064d20\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd8e30ccc0144418a8392760ccc63f0b : SUCCESS in 36m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18f7b8e8ef424abcbfb3a2835a3da660 : SUCCESS in 42m 21s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"660f45d934661ea2ee49767c1210701bba2ba5c4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-15 09:25:43.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840418/15 : FAILURE in 1h 44m 19s","accounts_in_message":[],"_revision_number":15},{"id":"68be224599d782960d60cbd08152ee2f317e458e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 09:34:37.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0e87e9f251346d9b6a29f226b2e97d4\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f5aee8b7342d4e1395ece0822ff492d4 : SUCCESS in 1h 18m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/117b8b4a67b14774bc53931d3743a5b4 : SUCCESS in 1h 08m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d39259fc58da49a2a29b5c04705d9a7e : SUCCESS in 1h 47m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/43c077bbeaa5460089cc4ce2e2f2ad0a : SUCCESS in 1h 48m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8ddbd1518274393ac83d350a9f200fa : SUCCESS in 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9350a507d42d49118c01bfd77b0fe5e5 : SUCCESS in 13m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/508bd4f08ff74f0d9e0d6cc1a061edbf : SUCCESS in 27m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/929151bb7ff2432cb8038ca27cf4058c : SUCCESS in 25m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0497b69995c4806a3e09f578becfe4b : SUCCESS in 25m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aecd9f4db05845df9e1ccbf9c857f676 : SUCCESS in 14m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/698740ead7a34b298e93ac3cd8981310 : FAILURE in 1h 02m 47s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/32fd64afaaca45a1a475cb19ac927109 : SUCCESS in 1h 36m 45s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5b451fa82c454a68865ade489c6f7ea7 : SUCCESS in 1h 26m 07s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c982b00e676f4484864244e2a4ed6b2c : SUCCESS in 1h 04m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/cc96c5ef02aa491fb71eb11b74ef51af : FAILURE in 1h 21m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/407a4e593d7744f09291b33d8a5a5e6b : SUCCESS in 1h 25m 44s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/aed0eb647fc340ecb2b38cf02b864583 : FAILURE in 1h 24m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8383ffe3b89e4c6996302dffec39e018 : SUCCESS in 1h 06m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d60eb29e29d946859be46d8a04d33e7c : SUCCESS in 1h 18m 55s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3098edfe0042c0dbb37f3731f4d27b9945d05b38","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-15 12:55:11.000000000","message":"Patch Set 15: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/15 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":15},{"id":"6ebe95f59954f4964a452e7537c954f32344911b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-18 09:45:33.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"8a3839e68cbe916704355754ae02e12a0d80466d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 11:13:16.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/318f9b14abec4be79a770c57d990cfa5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4807b7a666b04ab9820b02806c5ff173 : FAILURE in 1h 14m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3ffc90b2d3ad4e1bb249387a4517441d : SUCCESS in 39m 56s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"743e32dc144852fc03390977abebb548c8f60687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 11:47:01.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db19e6566ec14bdd9d6de95b2540619f\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/cec54c845fc847a38b94795cecfcea6d : SUCCESS in 1h 44m 57s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/0afdeb8ee760471d88d89cea2d2c97f9 : SUCCESS in 1h 51m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b4a934603c5b4d529135947fec5f0531 : SUCCESS in 1h 12m 50s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/240a4710e5c748a0ae08aa39c386f73b : SUCCESS in 1h 39m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1634f47ba9f497a99242882d9dee251 : SUCCESS in 44m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf500d05157244b99e2eab09e90ab21a : SUCCESS in 10m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c18a7376af954d95964ed4a593013d15 : SUCCESS in 32m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee93b21ab70a4fc69f4b5049236eb8aa : SUCCESS in 22m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de015788f5764679b4572d98671d521a : SUCCESS in 28m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eda15565a2947ce9ed8a39d8ef24fcb : SUCCESS in 8m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4446ca2e3a0348b3b8495555a2a5677b : FAILURE in 51m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/cf181f22179746c58a81b8344e63393c : SUCCESS in 1h 38m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1523d3f3dbde40dead1f0b4b78e2f329 : SUCCESS in 1h 21m 37s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b73649ac06ee4668b37ddf4cc66a8330 : SUCCESS in 50m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5ababec55f824507b7b87106200c28ed : FAILURE in 57m 27s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f7d937129bfc44c09e210a82d52e8bd1 : SUCCESS in 1h 09m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/12f0f5a9926641a8bdc51a77a5df09cd : FAILURE in 1h 12m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da9f1199528942a8b45d90c8b62b36a9 : SUCCESS in 1h 10m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a0870929e540474db38fd7de5152b4d2 : SUCCESS in 1h 15m 44s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f3afdca06abe984aa8a7b044f1b01aaa0cc66694","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-18 11:53:31.000000000","message":"Patch Set 16:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/16 : SUCCESS in 1h 59m 55s","accounts_in_message":[],"_revision_number":16},{"id":"5c6c701aa6228ec14274bc37fc82ad421af12d33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-18 14:14:48.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"7fed86b8802e38104eda5807ad7c10d96b70e50e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-18 14:31:29.000000000","message":"Patch Set 16: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/16 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":16},{"id":"52ff2eb67588c2aa1e4d20abcfdc59acd37e9c8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 15:07:58.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76ec4d177dca4b428edaec0b3fb45b17\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb123367b87d406395c9a9494a9569c0 : FAILURE in 31m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf1e9bd9927d486c9020ff21a2c3c310 : FAILURE in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"90ceae6eb051d975a71a0456fe86c206aa6c4c62","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-18 15:13:35.000000000","message":"Patch Set 17:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840418/17 : FAILURE in 50m 20s","accounts_in_message":[],"_revision_number":17},{"id":"d1805542e70a258a8e88ba163a1b1d1be2fbffa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 16:40:41.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e1f4b2546b54d0290c72b5864fe3ecc\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/bf0f475d9932426d879473cf6c923ae7 : SUCCESS in 1h 13m 12s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/73c3e55fd18448d8bc33b5fd583d6d83 : FAILURE in 26m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ebf0705db4984fa5b9919a9a1ef7a1e8 : FAILURE in 30m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f12fc37698db4270bca4c0bd3e4282a4 : SUCCESS in 1h 19m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80292d8649744dd9a8df443868d84dc0 : FAILURE in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ace904895e488eac2fa942ec4382d7 : FAILURE in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/adfa2ca14fb94f12b92451cd0971b0f6 : FAILURE in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a81e17de0ca4834a936e4a4d8d9c7c6 : FAILURE in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7db4ecea91124fae8041ccee52225819 : FAILURE in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd6720cb3b84a60964cb730bc1fdb3a : SUCCESS in 11m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f7eb11dd2f4c406da497208e7f622273 : FAILURE in 15m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d430635618349ee8d0e0f981bf12068 : TIMED_OUT in 2h 18m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/83ba5b28cc8e4f52a386512849fa3879 : FAILURE in 22m 18s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9d9acd68532443ca97403582671e4d1 : FAILURE in 46m 23s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2ec8882c2c6444cbbbdd2ab2834fcc1e : FAILURE in 45m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f5c7df27bc314541970e5f21f4a32efb : FAILURE in 28m 17s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a1f65f01f94f4930b6243658e09e8b98 : FAILURE in 24m 54s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb0eaa5526844db894671174c8c7f922 : SUCCESS in 54m 23s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/363f36b7ab7844339b32f10a7877a32f : SUCCESS in 1h 03m 34s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py","accounts_in_message":[],"_revision_number":17},{"id":"8cdc9d324cd25e9ab225a2dec677c70eb4da0065","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-19 00:59:16.000000000","message":"Patch Set 17:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/17 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":17},{"id":"bf389179b76199f51c91dc4a854a33a282876842","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-19 20:00:41.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"960999878cc0c4f9f4d8281433d3f54f13cdde64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 20:40:31.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ee0a32daf334cdca6fb975f835cbeeb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9f7775a0e3e448f93f6ca58e8716778 : FAILURE in 9m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b0162720cb249ce967891f335344eba : FAILURE in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"b762473cde6afac9c4ee5499b260e26a98e5cf35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 22:22:56.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/618edb366a394ded927c63ccc6965ebb\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/ba72e23f1c1b469d87bdfebc81928ff4 : SUCCESS in 1h 54m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ef700b1ae69d4f7d9e829e9ad80c0405 : FAILURE in 27m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ab4c271f743f40d0a053409cf4679b52 : FAILURE in 39m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/e3c020a2703e4c60ac73e20db69003ae : SUCCESS in 1h 27m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bba8766a8d094264aa289c580c160472 : FAILURE in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32960dc37d3c41308716a89e1c5c2099 : FAILURE in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc79e71941624282b70f6139b8de5cb9 : FAILURE in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55106669738a4c50b51b5e322c485063 : FAILURE in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/110ebcacecc945fe843256d8cfea974e : FAILURE in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aee83f655834602b3cb1170507a3fec : SUCCESS in 11m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b2c122dbc6c40668947063e82afb001 : FAILURE in 24m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6025118b0b149a8a5e79d2495ddcda4 : TIMED_OUT in 2h 19m 19s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1e58a2514aea47d5b87cff6e682b170a : FAILURE in 23m 39s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/03b59342f9f94df0a759553485ff1c0e : FAILURE in 1h 05m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6a758b9fb3a8422fa284b8dc9172a571 : FAILURE in 46m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/255d2fadb8424d5c9b24831203580c3d : FAILURE in 31m 25s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/58ba3d815e6c4871aaccce72bed54dcb : FAILURE in 29m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8b41b87927e4a9397e5eab002f9f1e7 : SUCCESS in 1h 14m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f4a0b77b3dae475caed37916dd834bea : SUCCESS in 1h 01m 38s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py","accounts_in_message":[],"_revision_number":18},{"id":"75d085a7c13da5bac89dfc642c60ffe4765cde77","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-19 22:25:46.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/18 : SUCCESS in 2h 12m 35s","accounts_in_message":[],"_revision_number":18},{"id":"592c0e24d3ca939496b23d7142888d9adedae078","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-20 01:01:44.000000000","message":"Patch Set 18:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/18 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":18},{"id":"ed7f0d72d13c5e46224c5cca4a9532cc8639f77c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-20 05:30:26.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"8da7103fac4f59e4aa47b5b503f207199b6abd60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 07:34:34.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc2530ffef6c41eba69b219ac45fee9c\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5f66d7e7d52444f69af77d6dbdb03642 : SUCCESS in 1h 15m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6fd49d3e5d6e41c89738cb21e674fc1f : SUCCESS in 1h 40m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/792874ad75ca4ff3a10f10cd4fd5ef85 : SUCCESS in 1h 02m 04s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7bf547e5409b42a8918a04333619045f : SUCCESS in 1h 29m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a715289aa42642e4bc0e9db7060bfb43 : SUCCESS in 43m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39da4e3ed8e4759a5025122b23ddcd5 : SUCCESS in 14m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ab3d725f7c34eedafd6f5257592a574 : SUCCESS in 25m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/639069400ede489ea16e68fda8385580 : SUCCESS in 24m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fcdeda9c9ba48d3b2f607f7387ba4ac : SUCCESS in 21m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e419b99b991d4ca6a0a5d8d1a35faefd : SUCCESS in 10m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e52b5a7d157143eea3e6ae82c9dfcd39 : SUCCESS in 1h 07m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8a5be3fb895f4e20a43047d0016ce4c8 : SUCCESS in 1h 49m 34s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b2102c42b4314946823d45cc7888522e : SUCCESS in 1h 24m 50s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/51a2ad0f02394ce6aeb1a35144f06131 : SUCCESS in 1h 08m 45s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9dcb46b410aa411483c63f5f97985634 : FAILURE in 1h 06m 41s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/4758a633f5ce4884b54109437a86a989 : SUCCESS in 1h 30m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/afec787ea63b47a0bec28b09dcc9ca54 : FAILURE in 1h 24m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1cbadad2ab3436c81d0fdc8593829db : SUCCESS in 1h 05m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4873404fce294ecf864076e75c3f80e4 : SUCCESS in 1h 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"12beef230c2cf86497208f88f45064b21b48ef2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 07:36:36.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93357f808e494430ac77c25e7801d2f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44d91d024c7a4f84b8e13b03941b9ab2 : SUCCESS in 42m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fc22254a51a4ef59d6b185ddc98cc4e : SUCCESS in 44m 45s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"6b7b17693927b0c7b6c2a96dd5a86013c4837d2b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-20 08:32:23.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/19 : SUCCESS in 2h 15m 03s","accounts_in_message":[],"_revision_number":19},{"id":"3d48dfb43de46cb6c753ac10c86aaa72df5e4bd2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-20 11:20:01.000000000","message":"Patch Set 19: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/19 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":19},{"id":"3633c2a353057efa86001556a90d3997b566d76d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-20 20:31:17.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"9d248219909a90865a69ff6287f8761ac95baa05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 21:47:05.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ec9618636464f7e8b1f269f4c7bff63\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29a1f4b726054c078a04275f99373051 : FAILURE in 1h 10m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44b8c2b844eb462fb16c85ad37760b3e : FAILURE in 1h 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"7f037389dfe4f3f7c7ade5f38f050271e729e3fd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-20 22:26:02.000000000","message":"Patch Set 20:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/20 : SUCCESS in 1h 46m 34s","accounts_in_message":[],"_revision_number":20},{"id":"145be7c0cdb33ed1b41148cc49f5d8108cefcbbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 22:44:15.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcadbd97da02451eb7c3da148f4e062f\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c4cf5b3497bf4d9c851b6c7b49b142c7 : SUCCESS in 1h 19m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/3799b2a806d6488489473b3169df76bc : SUCCESS in 1h 22m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/a03b37fd82184621ba251462431e8118 : SUCCESS in 1h 02m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/72fb139d96ce4183bc4f0eddc800fee1 : SUCCESS in 1h 12m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dd76ff2557d41108b39ea78ea7860dd : FAILURE in 31m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df5cf02b96ae40279705899cb5d93a08 : FAILURE in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6c4ed3343944fa3abf4dcc86acb8035 : FAILURE in 28m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84157471f0c84108982299b1f4d16295 : FAILURE in 26m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22edd3e21e294b8b940d9e9f307adb0d : FAILURE in 22m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ed08bed200494aac521557d95bc527 : SUCCESS in 9m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/178ed38fe47046c5945490b6ce4f8c97 : SUCCESS in 1h 24m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0ffa950bd30544818136892ceeeeb855 : SUCCESS in 1h 40m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/267ae9db946e4e5ca31ab0ee11a66105 : SUCCESS in 1h 31m 06s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/82a6ee80dba64fd3a57f3d310e42efae : SUCCESS in 1h 22m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a33d8549478646ff886d93d064dcef4d : FAILURE in 1h 17m 11s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fac798cde6e843ac89dad519b145f961 : SUCCESS in 1h 19m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/71bc393a7bd54015bdaf4947d5c79ffd : POST_FAILURE in 2h 04m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89c18f9144f444e5a8b194001a6061cd : SUCCESS in 1h 40m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3171642ea9b44106b93cf87171c06956 : SUCCESS in 1h 18m 08s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/l2pop/rpc_manager/l2population_rpc.py","accounts_in_message":[],"_revision_number":20},{"id":"340caf88320259928df77c9b985163e4e54dc70f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-21 00:21:12.000000000","message":"Patch Set 20: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/20 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":20},{"id":"f364410bafa23fac45dd45d957659c36d17fc969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-21 06:22:55.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"fc02a7c99080b16d40c577336c09c99373604673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-21 07:14:48.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"76b0a85d2c95d02cb355120bb87643785bef909b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 08:47:18.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7779bd15e88f4400addaf949842e0712\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/704ac735daf948edbf9deeedda0cc82d : FAILURE in 1h 18m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8237570f204a40c19c0e3dd651d91693 : FAILURE in 1h 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"62b11382350f2da21213a368c6bedf6823a7edfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 09:17:46.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e758dd5728b416a90df29c6bed2dd5e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9617e4cc71a148839b712d94709dc267 : SUCCESS in 1h 54m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/188fbfd3f47f4153aa5dd8ed2fb51793 : SUCCESS in 1h 53m 45s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b327b4831f2745edb41d6c75a88234ba : SUCCESS in 1h 57m 36s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/503c06caf2b14de88a54c0cd36a12f7d : SUCCESS in 1h 28m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55bc8c07a33740cc8d6a8e1c6f6c2cde : FAILURE in 28m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeb89a11de734476a82b765158a30736 : FAILURE in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c7ca7ba26284adfbdb36e6461a32ab1 : FAILURE in 25m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49351bdf49184c759172bddff91958e3 : FAILURE in 28m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79922d8b23bd43e68f88b0f6123e3b3b : FAILURE in 32m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0099ec4ee598468bb35b8185646ed5c4 : SUCCESS in 12m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d7a369fd7f44e89bb39fa7851933dcb : SUCCESS in 1h 18m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/068e5fa79c1c48b39f6d2dfa5247f024 : SUCCESS in 1h 56m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/d45dfc6d3ed0404e955188e18542b890 : SUCCESS in 1h 28m 26s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ccaf29465ac74c6d8a623c37d576aa7a : SUCCESS in 1h 14m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/71d47a5dd3454b46af621981d5c76618 : SUCCESS in 1h 08m 10s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cb720878fc59468cb09538396eab45a1 : SUCCESS in 1h 36m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/329d6736f30742a7a90110cc35c99ad9 : SUCCESS in 1h 04m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ec4b4caebd845ab8d55e0756f47862c : SUCCESS in 1h 12m 02s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b955d1fb96394911b5364b75dc587974 : SUCCESS in 59m 00s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py","accounts_in_message":[],"_revision_number":22},{"id":"3f7d0d5ffe82208ebecaff3bbd8a50c538e9c4f8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-21 09:31:12.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/22 : SUCCESS in 2h 07m 13s","accounts_in_message":[],"_revision_number":22},{"id":"7ce90b15f7f6f0e6761dadb626e28b9593464d3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-21 10:16:49.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"113ef6443064591e0d5a293eef7df8cbeba309dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 11:38:48.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aeb6f7cc36b642b78561b5ab01ea8e40\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96e765d72c8d48719bcb95fa8e905da7 : SUCCESS in 1h 09m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a275a45701984838a984c5f5044b8870 : SUCCESS in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"c89de91afcfb06ade8b82603cd248109411104b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 12:02:24.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9d029a41b4b424c8c6bc1f686f88f9b\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/678ab844a8a1492c8ad969599cb992b8 : SUCCESS in 1h 15m 53s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bfd72cc6112545dbbf8d2871c3aee01b : SUCCESS in 1h 16m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/257f19fc5d0d4c2f89196045cdc9208d : SUCCESS in 1h 04m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/8fc43e6ded434e8fac79c0895d8951bb : SUCCESS in 1h 09m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2474b9abb51a4943a631c484bcea13d7 : SUCCESS in 30m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5db4c1e3deb4c4aaf6fcc473cc0da8f : SUCCESS in 13m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b9f76a644414f6eb650e530d77d11f7 : SUCCESS in 23m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e44bcff7d6ab4f679f9d8a3bb637dbb7 : SUCCESS in 25m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82469f82643a40e79124fe4057490696 : SUCCESS in 26m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8904f6d9a6e84257bc88659678581ee3 : SUCCESS in 12m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/968331760d4a4202bb26caafe5d4d2b5 : SUCCESS in 1h 02m 59s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/91efe0e30e3a4b27aed2dfc4ace8a3c0 : SUCCESS in 1h 37m 41s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ef8d297d2d74417d80bfdfe5f960b48a : SUCCESS in 1h 12m 52s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/517ce3eae7994fc88f30a698ed04a78a : SUCCESS in 1h 02m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c450bcb323d347529789180cc81c1067 : SUCCESS in 1h 10m 52s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/8a1750126d944c5b9ff403d3c66ca155 : SUCCESS in 1h 21m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/de20d804e71e4334864d49a3ab4c3430 : SUCCESS in 1h 21m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fd4509c7bc143df8c7034141e85022f : SUCCESS in 1h 17m 54s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/7624857b8cac4be4bc7713e056bcc945 : SUCCESS in 1h 15m 12s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"a4995b9ced0a060cfb33ff2439604281b2eb3cd0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-21 12:31:43.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/23 : SUCCESS in 2h 03m 06s","accounts_in_message":[],"_revision_number":23},{"id":"7187b4abe9f94ba1e8ef2167208cd6e71c342df9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-21 14:35:23.000000000","message":"Patch Set 21: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/21 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":21},{"id":"c4333232137b8fc1ca9c0bf285390dbcdefccf30","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-22 00:46:10.000000000","message":"Patch Set 22: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/22 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":22},{"id":"8328f80bfe312540873ff691744c4118b1b40322","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-22 12:38:36.000000000","message":"Patch Set 23: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/23 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":23},{"id":"cf0509c3190e04e2513598ef1b232cd8e2a02a6c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-25 08:52:57.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"fa5c8b531016d3731a1dd6dbd47899c81455d3ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-25 13:00:02.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"607e34d67481441f60c978878780e1dd1ebcc79b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 14:58:05.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e18d519670346be866e595bcfe7c444\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/7c46ec8d92d047c1b770f135341b99b8 : SUCCESS in 1h 34m 39s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/6cd4e9d05ce44ec399576f7453664ee6 : SUCCESS in 1h 53m 27s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/1b64727354d940559e6f8e547e2835b5 : SUCCESS in 1h 41m 58s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/137415bcc8cc49b783e3f1a0f9fc5daa : SUCCESS in 1h 42m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c740a484d4734b1ea62c415898560338 : SUCCESS in 35m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d929a1d8f4664f2c8f32057ec3b5aa7f : SUCCESS in 15m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4189574ab9c14a48b744c53dbba99806 : SUCCESS in 28m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5921308bdf5e4a0dbe47f2342a5c7235 : SUCCESS in 30m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a314f58a7fd4496c98e9dde3942a0630 : SUCCESS in 25m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ba4370cb3e445894cf38880f3c5ee6 : SUCCESS in 10m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5d218e1ce52e48b896827b8cec863420 : SUCCESS in 1h 04m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9512b4dbbadd4172b024ec18ca8a73d0 : SUCCESS in 1h 39m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b2e88b963aa343dabec9461596204ee0 : SUCCESS in 1h 18m 51s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/62eaed5eb5804470b341ef44e876abc7 : SUCCESS in 1h 02m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9093ede3de904f02a58243a5410fdddd : SUCCESS in 1h 07m 14s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/44e0de8c460d4ee187e763517ed9e7e6 : SUCCESS in 1h 37m 22s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/39cbcdfb0fb1472b991e53da6032f9de : SUCCESS in 1h 19m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1943b75801b04eafa58c4c147a2422fd : SUCCESS in 1h 05m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b9eb074c19ba435384addc8ff0eda73a : SUCCESS in 1h 14m 16s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"4d667ec27cb5ac42aacc5781920ad2aae4c52818","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 15:11:40.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/345731bc9a1e4bb2b4708ab7d3b44153\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d10a061dcda34be28c56c7d97533a1bf : SUCCESS in 46m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81b55e2170c6428690d03ed7f2becaa6 : TIMED_OUT in 1h 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"336d14ce40377e9f63b2f88d4e8c92f9be0a86c3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-25 16:27:58.000000000","message":"Patch Set 24:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/18/840418/24 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":24},{"id":"4e914a24b8035d36f73dabd38bd4db0975d766c4","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-25 16:46:57.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/24 : SUCCESS in 1h 57m 52s","accounts_in_message":[],"_revision_number":24},{"id":"b9ee688a7ff20bde5901f5a7576f62dd69d95cb3","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-07-29 07:22:59.000000000","message":"Patch Set 24: Code-Review+1","accounts_in_message":[],"_revision_number":24},{"id":"1cab2106d9704e8641d048de77d0e4bfb21f1e22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-01 08:33:33.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"b86428baff32d461467e57df05b31ee644676507","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 10:04:22.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cc4a29bdf0748f69d23eb2618ef8143\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5551b94682cd46198ed2ea9b1c30af27 : FAILURE in 1h 06m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/13efdacfa196461ba7ac9206787ef2a8 : SUCCESS in 48m 00s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"50b26a248902ba97fc0a20e532193a87ddbdf89e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 10:31:46.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91391eecef5141eaabe830b54dee76d1\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b4072722c5894dbb855ad8adf4c1e723 : SUCCESS in 1h 21m 54s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e5b7f361468e410097798463b6483d0b : SUCCESS in 1h 14m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3dad7debff194967a434116021cfae67 : SUCCESS in 1h 07m 10s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/de7cefb5d7ff42b68f7a1436d4f25ec5 : SUCCESS in 1h 20m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b46de0399a924794844e86792e57d518 : SUCCESS in 33m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42e7d134fb104e118c02d115a1a1d724 : SUCCESS in 16m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a909d2bbbf149faae3a324a2f28b359 : SUCCESS in 32m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d2b7feccd004f568fc83388caedc3b1 : SUCCESS in 33m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/448232ac49be4871b1e6be503d96c936 : SUCCESS in 28m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f308225e30c415d99a003e49467cbe9 : SUCCESS in 10m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f7611975d9ea4af18f4b6d378f2acd50 : SUCCESS in 1h 01m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/84253663f5ca415f9d61b58ccfd15a91 : SUCCESS in 1h 40m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b7aec699298b492a865b5befd8d7dd73 : SUCCESS in 59m 54s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4558d23269943e2b942710d15012655 : SUCCESS in 1h 01m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8f6b97af67be438dbb469ef7ab7eda31 : SUCCESS in 53m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c56f79f8399149298a5b64570f7aa2d0 : SUCCESS in 1h 23m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0d782fd105e44b5c820811664f6dc9ce : SUCCESS in 1h 19m 53s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/916c2e5735c146ca87fcc4d163ba25e3 : SUCCESS in 1h 13m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/06d773a1930240b9bf1582e7b85e4404 : SUCCESS in 1h 16m 31s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"7e83484eb3414ac5f60091ff138ed5fd1b55506e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-01 10:42:51.000000000","message":"Patch Set 25:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/25 : SUCCESS in 2h 01m 09s","accounts_in_message":[],"_revision_number":25},{"id":"ee0f213d05e2d3490929af1dca56f8416e57913b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-01 13:44:40.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"8d29814e54082fb913bb90146ed6ac3456a8e666","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 14:29:57.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/131b88c9e3744afb8dfbded96142a8bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e37edbd4ed64403b331d85c5272255f : SUCCESS in 41m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f177399b90374d80b5038fd6e1ebe08d : SUCCESS in 42m 29s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"e7143ae713bb0f7b76e1a07a993247c242a4da03","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-01 14:52:00.000000000","message":"Patch Set 25: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/18/840418/25 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":25},{"id":"1ca2adee8cba72b437fcc26df25948867fbd3d6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 15:34:28.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4988b7fce1748f587d70908a4b6feb1\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/83cba5db0d0c4eba9c200715f2b78268 : SUCCESS in 1h 39m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/df4e854f70d04f41b30a02706cbfb7e9 : SUCCESS in 1h 37m 48s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/b6d5831ae0d44c9387ae2d71628f47fb : SUCCESS in 1h 35m 43s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/56c29133b61e4ff481fad1520d682502 : SUCCESS in 1h 28m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d4cb12ab346450d9cf59aa36771d7fa : SUCCESS in 39m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec7c4116cdf54b7ca7a958664b52c3ca : FAILURE in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4302e58a09b44328677bd984a3fc6e6 : SUCCESS in 26m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d26829ff13e04ec3aad58775e289b3cc : SUCCESS in 30m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36540a725c57467c93f4b1322e80d766 : SUCCESS in 26m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9724888c5a3740b5a70b7f9b0dca3efc : SUCCESS in 10m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7e5c137652d04684a59ada7fcbb36199 : SUCCESS in 1h 05m 41s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0092469790eb45b2acf462351d2096ac : SUCCESS in 1h 46m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a06bee0f17824fad94244021a5614f8d : SUCCESS in 1h 15m 59s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6bccfbe19d604fbab415d7534c636727 : SUCCESS in 1h 01m 07s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e102255970d24f1ba1504b7ef02b3a92 : SUCCESS in 1h 14m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/985d7eb176d24b33a8b5d44dc8a6c8b0 : SUCCESS in 1h 14m 10s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/bd69717e6db64d878c764fd52daffaf5 : SUCCESS in 1h 35m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77946ad77d0d4a32b96b0934f8aaabd8 : SUCCESS in 57m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9ac54d4f8893473bbee4cb21bb7c276c : SUCCESS in 1h 19m 53s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/api/rpc/handlers/dhcp_rpc.py\n  Comments left for invalid file neutron/tests/unit/plugins/ml2/drivers/openvswitch/agent/test_vlanmanager.py","accounts_in_message":[],"_revision_number":26},{"id":"369c21ebcf2365e4b800d37f7d79cc143783a309","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-01 16:08:47.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/26 : SUCCESS in 2h 15m 55s","accounts_in_message":[],"_revision_number":26},{"id":"e814a84e578705f1196df8332f69ddb68363887b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-02 05:01:28.000000000","message":"Patch Set 26: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/18/840418/26 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":26},{"id":"80539da749649638229779939cae7c8f4e27f3df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-02 09:33:49.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"4fcfe7002aee31ad1300668bf4249a641379d34a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 11:08:45.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/343d572ebdb648c582a9278755347665\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4ee4c45150041248d661ccbdf55bb51 : SUCCESS in 1h 05m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a0dcae46f5f46fbb16306b744e89072 : FAILURE in 53m 50s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"6cae1d98febcb52bcb542facee3493044700cc26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 11:19:08.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/643f899c336f4a6683ddfa7018a639c1\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/29de0d7d97cb4bcf8a0f65619458c4a4 : SUCCESS in 1h 15m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/73f6c7a5f9e24326ae82011879a923cc : FAILURE in 1h 22m 18s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/aaa516d223044d149386de35a99badc0 : SUCCESS in 1h 36m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5816dcc0fe094afb864061ea2e4a2487 : SUCCESS in 1h 41m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c34c2dcd88aa4b2194ddc63eb3cabb15 : SUCCESS in 32m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ece665b5de435b8cd0a59456cfee9f : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95a793cea8aa4368ae8a27e23580802f : SUCCESS in 32m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75ad89677fd64955ba323b6d29343f5a : SUCCESS in 26m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/444a1522c092486c8687d278651c8582 : SUCCESS in 32m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4acbf0e04c248b697411238fce66249 : SUCCESS in 9m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/394cbe8e37d04974b9f3a6402f0d5773 : SUCCESS in 1h 05m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dfaab0ceafee4f279053d4bd0f90fab6 : SUCCESS in 1h 31m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6b5048cba9e445ccbc2dd556bc328479 : SUCCESS in 1h 10m 49s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/08e5f7a97e4a4811adf9b3a72b47b2f7 : SUCCESS in 56m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0bf3ccddf4ca4864926de5f2287bf562 : SUCCESS in 1h 03m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d077988218994e9aa9262e752536cdee : SUCCESS in 1h 07m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b766cd5e176740419ec6a5672c4b8fe9 : SUCCESS in 1h 33m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41bc1a6131e64e0193729ec8b747d5b9 : SUCCESS in 1h 01m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/53102e8e18664f999e7a782d7078a71e : SUCCESS in 1h 16m 56s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"99064917164b33e1a7d69f3158b397b1c9f4b373","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-02 11:49:11.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/27 : SUCCESS in 2h 10m 34s","accounts_in_message":[],"_revision_number":27},{"id":"1f21c552cee256c0555b489d28092e94c2747c40","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-02 18:19:10.000000000","message":"Patch Set 27: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/18/840418/27 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":27},{"id":"15eaff96e8f06bd71bbf48d2e3b2f6afc0780569","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-03 06:27:24.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"32c582baadccd8f8b2736051cb76b31ae6398db8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 07:09:32.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1629e2998644b329a84badff2e17e62\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/301c0f017c3b4151a168ace8d0784be4 : SUCCESS in 32m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aea0f8aa1e774c709a1e4c493a03e005 : SUCCESS in 40m 05s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"c0249708e0b230232bfa83db691e94bbefc699b2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-03 08:15:34.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/27 : SUCCESS in 1h 46m 42s","accounts_in_message":[],"_revision_number":27},{"id":"2746bfdac42e230d412a5356895ea7ff20e48a6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 08:27:14.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47508dfec82946cea8ef16872498ca35\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8b3dff18849745aa929c222907164f48 : SUCCESS in 1h 25m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/000d539fe8f149709242d3c35fe3245b : SUCCESS in 1h 26m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/76e65d0992f3491dbc29f4453e3c0641 : SUCCESS in 1h 10m 54s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f95a35d353844bb0a5aed18370a3f1c8 : SUCCESS in 1h 18m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb40952b8aee45e585b8d6313c3d226c : SUCCESS in 36m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ffa1d420d2a4fa596dd441f43c3c6be : SUCCESS in 13m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/788c846c54bc437eba766cde591f1be1 : SUCCESS in 32m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93ef13eaf7614ecd8d57aa8403454e75 : SUCCESS in 32m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8044930ac6064d7cb340f711b96643ce : SUCCESS in 25m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ffa876bfdd4e5fb7cc0e68d02b7975 : SUCCESS in 13m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/683e5dd81e734cbb973f42c5facf634a : POST_FAILURE in 1h 25m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f1de68b1e35454fac8cbe986ca51420 : SUCCESS in 1h 50m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f812f3c35e7641e58e2e06bc6287db16 : SUCCESS in 1h 15m 44s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f0e7a7fe9a074ae39a366199ab8e239a : SUCCESS in 1h 10m 10s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6572891dae2b4618951195102b845852 : SUCCESS in 1h 20m 29s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/71788f9663f445afbb3b23d73b62d28f : SUCCESS in 1h 37m 01s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4d6c7db578914793b19b060d688c396d : SUCCESS in 1h 31m 34s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd17fec96d7146afb63b0c2740a81e13 : SUCCESS in 1h 11m 15s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0537713b817448f5a0eed0b6fe39ff10 : SUCCESS in 1h 26m 52s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"b962c667d0683d6d96bce64582034dd8de37bf0e","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-03 11:40:57.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"bc117a3939591c56b30ae3cba01b652f26910b3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 12:24:10.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a18e2bdaac404ab1ab951e7b18338633\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b42f68da8ad47c18b697c2d4df51ed0 : SUCCESS in 40m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bea2a23bea6f48f389cc03c3f6991f51 : SUCCESS in 35m 09s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"d54aa49a092011adfcd6b592ca03b2af2970f147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 13:17:59.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b920ab4666040f2bb944530bf4120b4\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d0fbd23db08f4a53980f30889c8c4152 : SUCCESS in 1h 34m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bb933d0841d94e69a4929829c7f7377b : FAILURE in 1h 31m 28s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/38f82ced19b647e8b5d1882e762dc07e : SUCCESS in 1h 22m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/1f704b8d96d34c68a6debb43d2d868af : SUCCESS in 1h 25m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8009a41ba4de4819b98b173d4db3247b : SUCCESS in 32m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd6c78a1577490baad75a813c9463af : SUCCESS in 13m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/122be3ce3cd04d5c81c895393e088e5b : SUCCESS in 25m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae5b70703493462ab92d0528eee3948e : SUCCESS in 26m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7af518d72c934bf390634be29329d2f2 : SUCCESS in 29m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d200b27ff2d04fe6804f7ac64f8707d0 : SUCCESS in 10m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/93e637af611d4bb2a0bf366ab0f004c9 : SUCCESS in 1h 06m 26s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a424203ba2847c5801f0026e51ceaad : SUCCESS in 1h 32m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4737b369bdf04468bf22b3494dab9eec : SUCCESS in 1h 08m 40s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cade56d21d16419a87a9c01adffec38e : SUCCESS in 54m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5d1bbd07539d447c967085900229acc7 : SUCCESS in 54m 53s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ddb8fd3d3b0d4fae8a20c2307c5a2588 : SUCCESS in 1h 27m 54s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d54fb250dc9446a29eac4bf26ecf77b4 : SUCCESS in 1h 26m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9aa0bd5010149d6aa4258d0b8e39443 : SUCCESS in 1h 07m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4d256e0be4334ca3bc5a06eae81999d8 : SUCCESS in 1h 21m 51s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"9ca354ba832e628e5b6967a6c30e159b9ed324ee","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-03 13:47:23.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/27 : SUCCESS in 2h 04m 58s","accounts_in_message":[],"_revision_number":27},{"id":"f9af6cccac83548f4533c036d95aace23c808d03","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-03 17:15:12.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"7859bfc4cb188c1e3f7743971a5f807e520f40e4","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-03 17:15:19.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"a8674dd8c522060e17e6926a03bf9f5e5370e279","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 18:00:12.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a73a695ba11345bcab8b0093da884304\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a15abb2333b4a3bb35ac5c2a2e8ff52 : SUCCESS in 39m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8882abc65c2d43bcae0a14eaafb226f8 : SUCCESS in 37m 44s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"b152cb9b5e52a6de8ea37583f7d54a53486c050c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 19:05:47.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2e66a495ef84fb4964904e86dabf29a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/5fa3594b355c4daebe3b33fa7e5039cb : SUCCESS in 1h 28m 04s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/45235f249fa64bb78a85e76150b28804 : SUCCESS in 1h 31m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d7b90ce3dcc54708b8496354a290b107 : SUCCESS in 1h 13m 48s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/76604295c69842fabbde4c12af6fb57b : SUCCESS in 1h 16m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a9d9220617f4e77ba4d624f6917af36 : SUCCESS in 39m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d7e0bffeb864a758a441e1180d1a463 : SUCCESS in 12m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a876d7015c2430abba8b0d7494316c6 : SUCCESS in 29m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70805b8b5dff451f8db4bb2c6a5e86a4 : SUCCESS in 29m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/675eb8ca51124d13ac9e0ae1f04707e4 : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a270a4600d854961b2613a387228ce1d : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8809b7c25f33489c878c67f69ae5a10e : SUCCESS in 1h 10m 33s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b8af37a234084682af1ca4c74fcefccd : SUCCESS in 1h 48m 52s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/3c4d711dcee0448e8739b7973346857d : SUCCESS in 1h 04m 13s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd3dfca1686341f88d85bd305202e8e9 : SUCCESS in 58m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/092a430b4044417787ad7b4767f84d43 : SUCCESS in 57m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/5b499a08c9cd43b2aced2d053a2ea451 : SUCCESS in 1h 26m 39s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9b7313747831439eb360aed45bf07b43 : FAILURE in 1h 11m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d279ccb70e2a4a6788c3360043d6ed12 : SUCCESS in 1h 00m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5e3c4624ba4d4edea72a0a740fb0277f : SUCCESS in 1h 18m 11s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"e2f3e23195282e1cfab7bc6bc3c7f10cbdaeb68f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-03 20:22:25.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/27 : SUCCESS in 2h 21m 08s","accounts_in_message":[],"_revision_number":27},{"id":"33a8869f76fe03f8dfe8b2590369238b425e8fa2","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-04 06:59:11.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"5154f0aea291dc198dec50f27bd09b917a9c84c4","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-04 06:59:18.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"e07696fcedd8fd783c70fa15a80487c7fc4ec14c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 07:41:23.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9588e6e8858459e85b758b7b6dfbd2f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92cb3a0caeae4e8da90dee98403a79bd : SUCCESS in 40m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cf1de7c03cc4e27a86cf3dd43b1aa16 : SUCCESS in 38m 56s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"de0a33ad813317d2b2a997c533c5e127d92e85ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 08:51:07.000000000","message":"Patch Set 27:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92034c70f64d494599e923117e833639\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/34226e2e0c2f440589c372edcae2f4e3 : SUCCESS in 1h 49m 28s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bff2ae3fd8e945b6a0f45cd7123a70f4 : SUCCESS in 1h 49m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4f094554cd494d3ebf03a732ee389b39 : SUCCESS in 1h 41m 57s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2adffc33538d41838bc35ee5ca9f922f : SUCCESS in 1h 19m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3113ce6b4497438d9cf9299585a416b7 : SUCCESS in 39m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bacccc59e6eb4881b477adcedf061bd2 : SUCCESS in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00839f0d8696428f911d36f788d576b2 : SUCCESS in 31m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d39cb977b5b4cf79e3f113012888fff : SUCCESS in 31m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b35d473df23417da477aa6aa49c66e2 : SUCCESS in 27m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0254f54e6a46c3953779e2a2114828 : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/654ec235c1ae44e3a19cc9c00e651aed : FAILURE in 1h 00m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/522d24aa98fa47be98281f553f0579ab : SUCCESS in 1h 38m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/9b88ce9565474fcb9648f95ab1f7b524 : SUCCESS in 1h 18m 02s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bac4b8009552408f99f1e71d53bc2b8b : SUCCESS in 59m 29s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/ba529ad7b5d1452aa8f83cd7d8e262ae : SUCCESS in 1h 08m 13s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/078111896f55430baaf479bca39ff38a : SUCCESS in 1h 27m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ecd3ae6f5f6b444abb92372c8323cbc7 : SUCCESS in 1h 35m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9cd2586ff5342ffadca67ae253d1c8c : SUCCESS in 1h 08m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/848abd4849cd45fcbd789f9d3be1b464 : SUCCESS in 1h 23m 08s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"e79873b14ead124c37aacf9f85f5eb03e7c7ad6e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-04 08:59:31.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/27 : SUCCESS in 1h 58m 09s","accounts_in_message":[],"_revision_number":27},{"id":"e0f8ed3074009a2a80e73361937e94eaa68dc638","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-04 13:40:42.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"45b08192d7ae7e69588f83a0b139f5074f3ddcb4","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-04 13:40:48.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"72efa5ef0f768fbef1c5951b8f7d1e20859bd36f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 15:18:20.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b656fd7b90d342fc865c1d41bdd3be89\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89d7b61cd05f4200ad72fdd729c4e4f4 : TIMED_OUT in 1h 21m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/180a8c5076084bb089c364f08e5c335d : TIMED_OUT in 1h 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"2028d8bac9ab7e13390c55a715e08429ec1c10dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 15:33:38.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab7168c4f69348e7b14c18a618d454e6\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6f28dd71cf08431f96d1d036c0ef2114 : SUCCESS in 1h 33m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d42c8ee2e58e4f56991f1425eb7c6212 : SUCCESS in 1h 15m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7fe5e887163c4c03aed84ae384c425c3 : SUCCESS in 1h 05m 03s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/322eb1305c6449dda1bb0635bdc9efaa : SUCCESS in 1h 12m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96d67501616d4f93a0cc91d659228435 : SUCCESS in 37m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fe63fbc04c9418bb561dd11c46d9afa : SUCCESS in 14m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/314fa383997941d3abed34d56ca8a8f7 : SUCCESS in 28m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f97e6d65e1f4fdc9547ff2575bd1149 : SUCCESS in 28m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8df13796b12649f9adb988477fbff46f : SUCCESS in 28m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78ad87f804e844c99e03ca1a9bdbf45e : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4d9059b05cf946a1bc17a66eb8437d2e : SUCCESS in 1h 10m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c56baeb6032d493daf50cfc62f0cefdc : SUCCESS in 1h 45m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ac01f63cc6724e9094aa418ebd47ea5f : SUCCESS in 1h 10m 27s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/64078673025b4daabcb64e189f61e97d : SUCCESS in 1h 06m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1961909b3e0f4264acb49db97591e6b3 : SUCCESS in 1h 11m 01s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b6b4182a59184362aec89aafbd36d9ae : SUCCESS in 1h 26m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3d03aa9043b348d38be0fc1d838b69ec : SUCCESS in 1h 36m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e209ace7cfd94336b5287f4622f958bd : SUCCESS in 1h 03m 03s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/92ae4f2f73394f54859d61aabb171392 : SUCCESS in 1h 16m 25s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"72fb9c9990899b50a85d0d87a2f9b54fb5cfa89b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-04 17:58:37.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/27 : SUCCESS in 2h 19m 43s","accounts_in_message":[],"_revision_number":27},{"id":"7b0942c66ca5bd4b403f9d76b7b5f44fe787d0bd","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-08-10 20:44:43.000000000","message":"Patch Set 27: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"d5def2bffefe01d845aba03c563a435b8634cca0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-29 13:21:26.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"5c22dc81149cfbf0fbc93e43b4cffe139c06f650","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 14:58:54.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327c7f4d08f846d8929bf34777e0925e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/001d0f147a2d4c7d996aea843dbc2271 : SUCCESS in 42m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cd95d5239c9f4f77ac1fc53329ad31d0 : SUCCESS in 44m 19s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"2235c690dc3b320443396b7b57bff3b0b02ac21c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 15:20:40.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebaa8d8b72614abcace53dba8fadc576\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dca075196e554a0e82b9ee2241cf146b : SUCCESS in 1h 16m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8584c5e4387e4e07b4ef44a13b144c21 : SUCCESS in 1h 23m 01s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f911f5bde86b427f9c181cc6f4c6f264 : SUCCESS in 1h 05m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/48583a9c357c400dafe5874046a30df9 : SUCCESS in 1h 21m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d91ce19002a14eaab25e8bc7973d1281 : SUCCESS in 39m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d7ccf22c32466e889b1cba820d8731 : FAILURE in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ec899923ecb4f88abca075aaa2fe93d : SUCCESS in 29m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a878f27f589545e1a503965112b9772a : SUCCESS in 29m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff28177274364b878a3c754f0eddd45e : SUCCESS in 24m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abe3d74c0f754ba4a4dfe725b593e2cd : SUCCESS in 9m 52s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1017fc7e473c414b8e97d11608d9c2e5 : FAILURE in 1h 02m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd04a373911045dd9f7f9b812286fe06 : SUCCESS in 1h 50m 51s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/98f06fc8318e41d0a29f9aaa301f7d57 : SUCCESS in 1h 11m 03s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a145fd0548cb470a8ffcb1f96d92d927 : SUCCESS in 1h 01m 40s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/77c3517f64724de48783dffb6644d770 : SUCCESS in 1h 13m 32s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a2ad156b22644a5da5fabdcdb54e0457 : SUCCESS in 1h 22m 19s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/db1913c70f504fec80e14a0b759f4b5e : SUCCESS in 1h 19m 58s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20bc0adceed8425da28a0b6438855cdc : SUCCESS in 1h 02m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5bf25202375f4919a7c73795bd633bde : SUCCESS in 1h 14m 02s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/api/rpc/handlers/dhcp_rpc.py","accounts_in_message":[],"_revision_number":28},{"id":"fbe6af59f8b79f8c1197a0776b0d73b78e6c20af","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-29 16:46:07.000000000","message":"Patch Set 28: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/18/840418/28 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":28},{"id":"f2324ff008d050a74dbf22fb48f6b5353379cb2a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-29 19:31:30.000000000","message":"Patch Set 28:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/840418/28 : FAILURE in 42m 38s","accounts_in_message":[],"_revision_number":28},{"id":"f9bf56cbf50f725ed7f298a3861433fb4560eabb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-08-30 08:10:47.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"5a99d1cf45e12ee93f68e01e029f3a152fe30a55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 09:05:00.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/705368e6b8504761a42c6877b6a5b649\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/024f48998e364ff190ae82e3f368bc31 : SUCCESS in 45m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41529bcd778e4c1380e32574750c03f4 : SUCCESS in 51m 32s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"9bd3d257d4b8a1eb09c241732602de3c3e8b734f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-30 09:57:11.000000000","message":"Patch Set 29: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/18/840418/29 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":29},{"id":"e87b0e04cfdeb3f9ed8ee1fb0e4a03182dba41f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 09:58:36.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc5f855ce5e049a9ae6f38f5c884a777\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/87a3b343b6404d429551781b6452ccb3 : SUCCESS in 1h 28m 30s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bb540458f2f348ad87dae5e36bcb9491 : SUCCESS in 1h 17m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6a343edcd94d4a2ca6c0e7fe99fdd8bc : SUCCESS in 1h 11m 28s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9267681ed4654e84abedad83f39ccade : SUCCESS in 1h 26m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06e7abe1ee25487a86ed59ebb2db40da : SUCCESS in 29m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df77feec7a1484884127851031c2dab : SUCCESS in 12m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/078a069c4abe403eb31e5218efe08c95 : SUCCESS in 37m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de6719e6d17247a2ae3c9e5c1da8ab36 : SUCCESS in 26m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ceb917f3e9594583935593bd045451a7 : SUCCESS in 38m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98c2798fe7d44b83a7827e7fe0ae1750 : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/96f898342eb141ffb9d25eb0be6d45b6 : SUCCESS in 55m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/31877c1c5c504f799a08c7ec8cf2d69c : SUCCESS in 1h 41m 30s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/7809dcaa105047a8bb617f97c27458a7 : SUCCESS in 1h 03m 37s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7b8dc5807794e96897a1c404836a165 : SUCCESS in 1h 15m 51s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/18314f36ca7f48f68406b3916c20b7c2 : SUCCESS in 52m 28s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/049654200f2e46399f2e67ce90abe129 : SUCCESS in 1h 20m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0f527a82e83547b695daea6e38d638f7 : SUCCESS in 1h 10m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b949b9b10d14fc1b53df479c5a8334d : SUCCESS in 1h 03m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f2e19a0b198e4da3a0e69cfa2ebe132e : SUCCESS in 1h 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"d1cb256b43734308f3677383550e97db3fc1a49b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-30 10:15:05.000000000","message":"Patch Set 29:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/29 : SUCCESS in 2h 00m 25s","accounts_in_message":[],"_revision_number":29},{"id":"b8a4f886b18b5cbfdd4b7723c30b13d9984cb9e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-09-01 16:21:23.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.","accounts_in_message":[],"_revision_number":30},{"id":"d518d08db221cb1f439d5efa809133eff6702f83","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-09-01 16:29:25.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"5017750496df1eb62eecd9f0f77192a4e0801f11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 18:07:41.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9761dbaeecbb48fea2d28fdd50551bd2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a58cc960f0ac4dd5a9d71a97bbba297b : SUCCESS in 40m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70724722c77d4755aba7068a62c6189a : FAILURE in 1h 08m 45s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"0c0d11d90a4a4efb0f645fe7dfb33917bcfd5315","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-01 18:09:49.000000000","message":"Patch Set 30:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/30 : SUCCESS in 1h 43m 09s","accounts_in_message":[],"_revision_number":30},{"id":"d35fcc9d2a5fb7dccaa679595c7cf23d33b58df9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-01 18:11:39.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66665d3c96fd4851a245390f4525c118\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f49b244f9ae74655af666be4ae054870 : SUCCESS in 1h 33m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/939bc244a57242d98967420ff348f012 : SUCCESS in 1h 22m 47s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4a4d4fbb5f604843a891c287db71e163 : SUCCESS in 1h 11m 41s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/679b4a1547284e11b9f02e6f90eea24a : SUCCESS in 1h 11m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e397769c49d4fe2856d1d60630bab3a : SUCCESS in 32m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5854b744a694db9a77bf595abb2f90e : SUCCESS in 12m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce0cad600c8b46ffb6afa57467e5c00f : SUCCESS in 26m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c1110a973af4f3390e8d6ecc600039a : SUCCESS in 31m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68b67715a64c4f34abeefb801275f752 : SUCCESS in 26m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff78180aa1044bcc91e5f3093904023a : SUCCESS in 11m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d23501a16a9549718449a78abf3e0ef6 : SUCCESS in 58m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e98d5778ddf543d38a5f83e75e271f28 : SUCCESS in 1h 36m 14s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/57b09ddc2eb14bb5b44213019bc01ebc : SUCCESS in 1h 19m 36s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72644b900a2742e5930d42f85d4514ea : SUCCESS in 1h 09m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/0473d306120b4581a549b1918b9125ee : SUCCESS in 1h 15m 23s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7f4f745ac4eb43e39eedbf478aa353f5 : SUCCESS in 1h 42m 32s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1b2dd15f83c14ac38da77cb6b8d7ba50 : SUCCESS in 1h 26m 26s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd762acc4ce4431c86f7d0481386cdc3 : SUCCESS in 1h 07m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ba9c706f19f64dc586311c1cb45f7477 : SUCCESS in 1h 33m 04s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"ee5b31d3e4deb1668bff524c3890e0a0412de0fd","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-01 19:14:34.000000000","message":"Patch Set 30: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/18/840418/30 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":30},{"id":"6bb58a85f301a627ec281f2be4ff2c771f4a8cfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-09-02 10:09:52.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"f01c9464c420ad5d1ca545807f67545c03f4baac","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-09-02 10:44:43.000000000","message":"Patch Set 31: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/September-2022/Arista_Tempest_Testrefs/changes/18/840418/31 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":31},{"id":"30fec09a74a85ef936ff5162990558276d842106","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:30:51.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7e45203aaf04ad7a3f0a1e2f160762f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/252ec33c91364eef85580f30bfd6a18f : SUCCESS in 56m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79bcdcc1833040b6a9b5c443f6daf27a : FAILURE in 55m 16s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"d6f1e4b40131c3a29a1e994a89670b72d083d8db","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-02 11:53:15.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/31 : SUCCESS in 1h 42m 31s","accounts_in_message":[],"_revision_number":31},{"id":"3bd567e863020f1a085c0eb4cc6b698d266bf569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 11:55:54.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0ac8c5b2e1dc4b0f864cb40e39e06924\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/85bf91bec5ac442eb5e338bfdcd54bd1 : SUCCESS in 1h 14m 43s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b41edc7f69f04dbfa5dfa3aad5813ebf : SUCCESS in 1h 29m 44s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/834ebb8242ec419f98f38b2e8889d6bb : SUCCESS in 1h 16m 17s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/01f31b45116c46bdbf8a0e9ae8fdd1e9 : SUCCESS in 1h 30m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dd27689a9094d23ad898b9f3f12ebd5 : SUCCESS in 34m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c2002d498944d5a74c26592945bbf4 : SUCCESS in 12m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/121dd12688b44ae89c7c7677cd3e517b : SUCCESS in 29m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/848e335a1a6a48908d60a6708498ef99 : SUCCESS in 31m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b83abaf2caf44cdb9d2374a2ba9d7fb : FAILURE in 47m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc84f62deb3941f5867448333469c1c4 : SUCCESS in 10m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d77d0bbfb73493a89d8346af7d35ec4 : SUCCESS in 1h 06m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3552d6eefadc4ef993cba20556aa4083 : SUCCESS in 1h 39m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/5e88283a451c450cbc3d7cce6ad8018f : SUCCESS in 1h 23m 40s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c50e45e25157429592069f4874b392d3 : SUCCESS in 1h 04m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/92f9542d08a54f028ce60ef48cba87f8 : SUCCESS in 1h 08m 19s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/973234b718a640649681cd94ffb42070 : FAILURE in 1h 10m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/49d4214af6954409800be7eee0c41867 : SUCCESS in 1h 31m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8977a2f419ab4195967f21cee5ebd6b2 : SUCCESS in 1h 07m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5358cf906038479bb883f16921294751 : SUCCESS in 1h 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"a231211f19fada445b4195269c63fccbfbf073d6","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-09-02 14:16:53.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"4ae5055534ca483db9c93056a703da73c578951a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-09-02 14:29:14.000000000","message":"Patch Set 31: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"3375b85d2117c61a3184bf70541be8fc9f719bbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 15:01:12.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d02cb6984cc4034a1dc523549385afc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47c3e6f5befe4c828b6b6e35cc87f5d8 : SUCCESS in 39m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d047e5f5a924fe59a31bf1968c9d6fa : SUCCESS in 39m 04s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"1e039419c58569d4d87dcd13af94c51f6f61f77e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-02 16:16:09.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5f8e468c1c847d7bb551c59ce0515c0\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/1f6b7efa9db746a787ad1c27465419bc : SUCCESS in 1h 56m 56s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/fa3f2dae34d2496bbdc009c13eb6cb04 : SUCCESS in 1h 44m 04s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6365f2d2cfca4fc1b1d253d71263b697 : SUCCESS in 1h 34m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/95328073482a41118c824a958ebad7cd : SUCCESS in 1h 15m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7ef2d395ded4590a8cda1c3081cddc9 : SUCCESS in 30m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c7b173235644e469318120183617295 : SUCCESS in 14m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1309135254cf41b0b15e2c82a1bcfcfe : SUCCESS in 29m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e77b9e93fc5c481babe2780edd3e5e4c : SUCCESS in 31m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37fa9f3c028645b7a6e972935040e6ed : SUCCESS in 29m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c14032e7a44586a3356c980a618a18 : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f30c4c44b19746a1a8fcf64d0406375f : SUCCESS in 59m 36s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a450fa0f469e4e0da655e5d71d63cb23 : SUCCESS in 1h 41m 28s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/87448e0b5f5b47b5a88839e42ce4c1f7 : SUCCESS in 1h 26m 28s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ea7861da96e45488a0afd551b509166 : SUCCESS in 1h 11m 55s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/3a362dbc7ef446da84fc6146f1b3d028 : SUCCESS in 1h 08m 09s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6e0a5aef073b461aa805e1de57c05f2a : SUCCESS in 1h 25m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0ad9eb6b8ae249a3b0b112c00f5f488c : SUCCESS in 1h 28m 34s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4303b18c8444bcd8f14fbf363eef4af : SUCCESS in 1h 02m 55s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/053e965603684e08b61810c09a6e9bc6 : SUCCESS in 1h 11m 46s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"c5150443c41f446aa114445af2901ad6262a741e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-02 22:15:51.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/31 : SUCCESS in 2h 15m 55s","accounts_in_message":[],"_revision_number":31},{"id":"fd3774d37a9d209e1dc601c1dd1d030e334051c1","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-09-05 04:01:20.000000000","message":"Patch Set 31: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"4da809a0cbd316c0545ff134a6abd8a3bee290a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 04:02:00.000000000","message":"Patch Set 31: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":31},{"id":"afc4fc1b711248212e6a23931d9f57ddd9086f6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 05:47:25.000000000","message":"Patch Set 31: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c74f0f97dad41e38904a75e753c3d38\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dc4a8845c2034940a396a2e2565d3af6 : SUCCESS in 1h 16m 18s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/80a0ea6f38654a28b0757380c0479201 : SUCCESS in 1h 23m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/41262f1d5d0343389893a6a35c85ba04 : SUCCESS in 1h 12m 52s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/183f9c422bf8495095b3b24615e87339 : SUCCESS in 1h 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d2d2c76e35c42529348cfa4036c2296 : SUCCESS in 13m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6decb6d501d140a1ae1cf284ea3327ed : SUCCESS in 26m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/074579fe68e247c69c2954426f3250b4 : SUCCESS in 23m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/954e4026d5224502a08a201f875eac99 : SUCCESS in 8m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/380dc9d0b2f048c29f3b7e8f36992b8c : SUCCESS in 1h 03m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4a4cdfef40ce41afba2fcc44dce17f9e : SUCCESS in 1h 40m 57s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b7b061d73bf147d1943a2ae486279cff : FAILURE in 1h 03m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5670807b6144fdfae7fbd35a459d3f6 : SUCCESS in 1h 01m 15s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/aa2f98d2ebcc4273898a060f27729385 : SUCCESS in 1h 08m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87594e83e18f4cb8b8345e49a53b8366 : SUCCESS in 1h 05m 04s","accounts_in_message":[],"_revision_number":31},{"id":"dbe3dc3cd55198a8c52c5bea830a36913158d4eb","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-09-05 06:20:42.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"f3904ea23d60bd65aa6a403f4ed0b2d88dc11782","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 07:07:40.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fdc3ad0accf415ba620561b249852fb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38373bf69679462a923406aad2c2b7df : SUCCESS in 32m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b6014aa8b3ad45cea939312863c8df61 : SUCCESS in 44m 38s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"f826fdcb554bbbe71b3d33b5daf9a5ca9f03c48d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-09-05 07:55:05.000000000","message":"Patch Set 31:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/840418/31 : SUCCESS in 1h 33m 49s","accounts_in_message":[],"_revision_number":31},{"id":"2ac3b1758e5632bbddc8a67e94ca95ba6eeb1972","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 08:22:59.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f72c1104314eeca30e16db61043fad\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/58a5ff760f844f22a45dff8ab491c27a : SUCCESS in 1h 17m 14s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/bbb6763e76a64012b477911d718e0b86 : SUCCESS in 1h 08m 37s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/babf0b86ac544133963c207eabf244de : SUCCESS in 59m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/71b8a6907ebd4913ae38037e2ce80730 : SUCCESS in 1h 10m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/784ae0bfe71b4426954d5dd6394c9d9b : SUCCESS in 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9faa37ff65214d8f969f105b4867d97f : SUCCESS in 10m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da133534c5d447f98ac2c1a3a8d240a9 : SUCCESS in 29m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d72e3d51901142a48218321da84e6572 : SUCCESS in 30m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b4ff530f7a74358b004842103e6376c : SUCCESS in 29m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce701cc6ad14531b7fac4c1a427f1a4 : SUCCESS in 13m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/650e24a9244b4890970eb85100cc71e8 : SUCCESS in 1h 03m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b5bdf3c65953456299a5f08642d347bc : SUCCESS in 1h 50m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/0f4937d8471146078b215a071a23559a : SUCCESS in 1h 09m 57s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdd996ed29e041f78bdcc3bf369e37a3 : SUCCESS in 1h 11m 37s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a40128033c334b78b69012a446d7915b : SUCCESS in 1h 02m 12s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e9179d9018fd44648e52adc08c32cd29 : SUCCESS in 1h 32m 38s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ab3f39aa24514e93a404149f40b19349 : SUCCESS in 1h 20m 35s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c19e78852c8a48adaffd967ac02184b7 : SUCCESS in 1h 00m 52s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4cbcdf9063a94df1aad6e3f494bfed80 : SUCCESS in 1h 22m 04s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"f94f2f37d10c4df1243a0546b0a433525f663847","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 08:23:34.000000000","message":"Patch Set 31: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":31},{"id":"f445f56cff6f377ffb1c32ddb12bda971583e848","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 10:11:26.000000000","message":"Patch Set 31: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9cb0ca588ac4998adde17285562c918\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/053971cc381144b3be04e6cb07818719 : SUCCESS in 1h 36m 25s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/dfbd3328bc90480a90981e91ac802eba : SUCCESS in 1h 41m 56s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d5ee85fb0cc64814b7ef725f165f0bda : SUCCESS in 1h 43m 43s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7525f65542164c5597888d7ed2fa98ee : SUCCESS in 1h 00m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d15e98e7a4114ba89e9df5a8407abacd : SUCCESS in 12m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e42bc0e58324b3cbd86304b9934d9a5 : SUCCESS in 25m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aaf6d59315c4dc7af70a363c434d173 : SUCCESS in 28m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b392a48016048f2958c38b794dbd99d : SUCCESS in 9m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/937ee72373244ff582793c919a725d04 : SUCCESS in 1h 08m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8aaeeb0275994117a92e88036d0ee66b : SUCCESS in 1h 43m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e205fd4bfdc04d1887e62e3783826291 : SUCCESS in 1h 35m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9918327165e94eb0b6f08b80a4d2816e : SUCCESS in 1h 01m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a3523f5cf8ac407985dd4414547a2e07 : SUCCESS in 1h 14m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cd34d3932b64167b4ca1e91ab6fd006 : SUCCESS in 1h 01m 11s","accounts_in_message":[],"_revision_number":31},{"id":"9d74ff202c670f6eed5c363cf42bfbfb2a2edbfd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 10:11:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":31},{"id":"6bd6017b51c5f9bfbf48d7f24c9a126e3e66f729","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 10:13:09.000000000","message":"Patch Set 31:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85243006517443db91e7129765cd8ca4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71fa599875a3400f868d7100d6d3a286 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":31}],"current_revision_number":31,"current_revision":"6db791ac0fde866a61d8cfbf9679dbd02d327635","revisions":{"6b67ab1df38331ede9c51ccdf85732a7121f3575":{"kind":"REWORK","_number":1,"created":"2022-05-04 07:07:57.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/1"}}},"commit":{"parents":[{"commit":"1806e25a61b54a47c1a271f6227e18b760f6206c","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1806e25a61b54a47c1a271f6227e18b760f6206c"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-04 07:07:30.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b67ab1df38331ede9c51ccdf85732a7121f3575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b67ab1df38331ede9c51ccdf85732a7121f3575"}]},"branch":"refs/heads/master"},"cd22dc8902cf39b03e698610287ecc4d56c785b6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-04 14:34:28.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/2"}}},"commit":{"parents":[{"commit":"1afd337e624380ea4328d98a01510fd0703acf6e","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1afd337e624380ea4328d98a01510fd0703acf6e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-04 14:34:15.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd22dc8902cf39b03e698610287ecc4d56c785b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd22dc8902cf39b03e698610287ecc4d56c785b6"}]},"branch":"refs/heads/master"},"7d697c1acbaa709a5412d10fa2aee494b26f3316":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-05-11 17:26:44.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/3"}}},"commit":{"parents":[{"commit":"9d5c3b6a37ec15ed18071fada4a7a750ecefa6bf","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9d5c3b6a37ec15ed18071fada4a7a750ecefa6bf"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-11 17:26:32.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d697c1acbaa709a5412d10fa2aee494b26f3316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7d697c1acbaa709a5412d10fa2aee494b26f3316"}]},"branch":"refs/heads/master"},"f7fbd47a0a868cbaafcbf0a8ce555cf9bfde2907":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-05-16 07:11:36.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/4"}}},"commit":{"parents":[{"commit":"5dbe808c5714e9c0a844d2d8953940eedc69122e","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5dbe808c5714e9c0a844d2d8953940eedc69122e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-16 07:11:22.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7fbd47a0a868cbaafcbf0a8ce555cf9bfde2907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7fbd47a0a868cbaafcbf0a8ce555cf9bfde2907"}]},"branch":"refs/heads/master"},"595ed87f8de4c25163911adab70ca10fe18d6c36":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-05-23 10:21:21.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/5"}}},"commit":{"parents":[{"commit":"9711dfa8968de59021d0a270d0c0e6e03c708d3f","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9711dfa8968de59021d0a270d0c0e6e03c708d3f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-23 10:08:19.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/595ed87f8de4c25163911adab70ca10fe18d6c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/595ed87f8de4c25163911adab70ca10fe18d6c36"}]},"branch":"refs/heads/master"},"6eec1b4a4bee2efa830d27a825c8e98909cf088c":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-05-23 12:29:52.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/6"}}},"commit":{"parents":[{"commit":"2412545319ba1b02da0cd396fe06d22eb7fd69e3","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2412545319ba1b02da0cd396fe06d22eb7fd69e3"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-05-23 12:28:56.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6eec1b4a4bee2efa830d27a825c8e98909cf088c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6eec1b4a4bee2efa830d27a825c8e98909cf088c"}]},"branch":"refs/heads/master"},"b1d734647bd934d6399df53954f79ed8a945883f":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2022-06-17 07:10:14.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/7"}}},"commit":{"parents":[{"commit":"f84701417258b76f96e629b8c8266e3dfce4b843","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f84701417258b76f96e629b8c8266e3dfce4b843"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-06-17 06:15:18.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1d734647bd934d6399df53954f79ed8a945883f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1d734647bd934d6399df53954f79ed8a945883f"}]},"branch":"refs/heads/master"},"862b1dc2ed47143e93d0b392cae688aa70b6fe00":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2022-06-27 07:15:45.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/8"}}},"commit":{"parents":[{"commit":"11abe5119c43d2deb164a2cba1d4c6d6e0cdca72","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/11abe5119c43d2deb164a2cba1d4c6d6e0cdca72"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-06-27 07:15:23.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/862b1dc2ed47143e93d0b392cae688aa70b6fe00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/862b1dc2ed47143e93d0b392cae688aa70b6fe00"}]},"branch":"refs/heads/master"},"5cbf336a34d1fdd6580b586c05cb87f298a3d6d5":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2022-06-30 17:32:53.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/9"}}},"commit":{"parents":[{"commit":"a740246b5ba0153dd644ac9ac735999ad8a4b668","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a740246b5ba0153dd644ac9ac735999ad8a4b668"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-06-30 17:32:40.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cbf336a34d1fdd6580b586c05cb87f298a3d6d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5cbf336a34d1fdd6580b586c05cb87f298a3d6d5"}]},"branch":"refs/heads/master"},"9c15b346375d475ee7629845b1b80fdc5f771f71":{"kind":"NO_CHANGE","_number":10,"created":"2022-06-30 17:43:20.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/10"}}},"commit":{"parents":[{"commit":"d374f6bb766aa60f3a99e2cbf798d8510a3222c8","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d374f6bb766aa60f3a99e2cbf798d8510a3222c8"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-06-30 17:43:08.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c15b346375d475ee7629845b1b80fdc5f771f71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9c15b346375d475ee7629845b1b80fdc5f771f71"}]},"branch":"refs/heads/master"},"93f3f9eb725e1d3647a2325c59fc6a61c2adf26d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2022-07-04 16:01:24.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/11"}}},"commit":{"parents":[{"commit":"21c6fb9f2e3973f6dfe445761241350d92d1b65a","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/21c6fb9f2e3973f6dfe445761241350d92d1b65a"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-04 16:01:08.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93f3f9eb725e1d3647a2325c59fc6a61c2adf26d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/93f3f9eb725e1d3647a2325c59fc6a61c2adf26d"}]},"branch":"refs/heads/master"},"a05c77608e5606816e5958b17faeede6f43cc78f":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2022-07-07 05:59:44.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/12"}}},"commit":{"parents":[{"commit":"56bad468071f6043940ab69eb65e1eada8f632f8","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/56bad468071f6043940ab69eb65e1eada8f632f8"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-07 05:59:30.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a05c77608e5606816e5958b17faeede6f43cc78f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a05c77608e5606816e5958b17faeede6f43cc78f"}]},"branch":"refs/heads/master"},"1c1cf443c429f12dcfa52800b61eb2bf77b446fb":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2022-07-11 07:30:25.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/13"}}},"commit":{"parents":[{"commit":"fed8d3a0c472fdfc3c1acd4eb4f339b8cc5e3aae","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fed8d3a0c472fdfc3c1acd4eb4f339b8cc5e3aae"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-10 20:53:59.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c1cf443c429f12dcfa52800b61eb2bf77b446fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1c1cf443c429f12dcfa52800b61eb2bf77b446fb"}]},"branch":"refs/heads/master"},"96d7e76b0c5f1261d79aacd2f2bade6ceb56d982":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2022-07-11 12:48:45.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/14"}}},"commit":{"parents":[{"commit":"b6503dbf9803d3059fc784aef02c7be5ccbed22f","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b6503dbf9803d3059fc784aef02c7be5ccbed22f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-11 12:48:19.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96d7e76b0c5f1261d79aacd2f2bade6ceb56d982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/96d7e76b0c5f1261d79aacd2f2bade6ceb56d982"}]},"branch":"refs/heads/master"},"33673a2c9c5ab88d266ca359107e92cf25ff10ea":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-07-15 07:37:39.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/15"}}},"commit":{"parents":[{"commit":"35326daec4592b891a67e4f10049954d17e923e5","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35326daec4592b891a67e4f10049954d17e923e5"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-15 07:37:26.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33673a2c9c5ab88d266ca359107e92cf25ff10ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/33673a2c9c5ab88d266ca359107e92cf25ff10ea"}]},"branch":"refs/heads/master"},"9b425aebf3e99338765e2051348ff7e9de90481e":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2022-07-18 09:45:33.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/16"}}},"commit":{"parents":[{"commit":"cdd99cd021f1a2659450d1d10060491256997744","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cdd99cd021f1a2659450d1d10060491256997744"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-18 09:45:06.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b425aebf3e99338765e2051348ff7e9de90481e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9b425aebf3e99338765e2051348ff7e9de90481e"}]},"branch":"refs/heads/master"},"8b86b668d1f873532359f1480a9517d2476a963a":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-07-18 14:14:48.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/17"}}},"commit":{"parents":[{"commit":"510cafdb7b6f4a751dad7d3a22a65a5bc5e17cc8","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/510cafdb7b6f4a751dad7d3a22a65a5bc5e17cc8"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-18 14:14:13.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b86b668d1f873532359f1480a9517d2476a963a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b86b668d1f873532359f1480a9517d2476a963a"}]},"branch":"refs/heads/master"},"b8548a166186f2d3f71dbb8769d53c9a3972f626":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-07-19 20:00:41.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/18"}}},"commit":{"parents":[{"commit":"c3dbb6ee11ecb9925390baf9bfa3a1b366934d20","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c3dbb6ee11ecb9925390baf9bfa3a1b366934d20"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-19 20:00:29.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8548a166186f2d3f71dbb8769d53c9a3972f626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b8548a166186f2d3f71dbb8769d53c9a3972f626"}]},"branch":"refs/heads/master"},"972ff41c53d633936a4a9fc0fa7a8d1e67ef3b13":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2022-07-20 05:30:26.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/19"}}},"commit":{"parents":[{"commit":"bbd9b72566cd7dae09b5a8811198d5964d97de9c","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bbd9b72566cd7dae09b5a8811198d5964d97de9c"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-20 05:30:13.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/972ff41c53d633936a4a9fc0fa7a8d1e67ef3b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/972ff41c53d633936a4a9fc0fa7a8d1e67ef3b13"}]},"branch":"refs/heads/master"},"3a4aa9b95ddf6737ee5ac47a36d3315718202479":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-07-20 20:31:17.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/20"}}},"commit":{"parents":[{"commit":"80635408752e93912ee75609a00ce884e1d97492","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80635408752e93912ee75609a00ce884e1d97492"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-20 20:30:59.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a4aa9b95ddf6737ee5ac47a36d3315718202479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3a4aa9b95ddf6737ee5ac47a36d3315718202479"}]},"branch":"refs/heads/master"},"0815d7f557b06e3c28741de4111fae7f454561e5":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2022-07-21 06:22:55.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/21"}}},"commit":{"parents":[{"commit":"d71b903690c4cc3dcaeb9928f737cec4d230df17","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d71b903690c4cc3dcaeb9928f737cec4d230df17"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-21 06:22:12.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0815d7f557b06e3c28741de4111fae7f454561e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0815d7f557b06e3c28741de4111fae7f454561e5"}]},"branch":"refs/heads/master"},"72bf4cbf3872f815af48d7666270d6775569d9e0":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2022-07-21 07:14:48.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/22"}}},"commit":{"parents":[{"commit":"cca8f8a338b7e2c7f6e64a64974eb837af2002e0","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cca8f8a338b7e2c7f6e64a64974eb837af2002e0"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-21 07:14:37.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72bf4cbf3872f815af48d7666270d6775569d9e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/72bf4cbf3872f815af48d7666270d6775569d9e0"}]},"branch":"refs/heads/master"},"98f1f1609fb578f22c7512d487a1cf5a5a603b6f":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2022-07-21 10:16:49.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/23"}}},"commit":{"parents":[{"commit":"0fd0831421308fbf6119381aa29410262cd5cb7f","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0fd0831421308fbf6119381aa29410262cd5cb7f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-21 10:16:36.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98f1f1609fb578f22c7512d487a1cf5a5a603b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/98f1f1609fb578f22c7512d487a1cf5a5a603b6f"}]},"branch":"refs/heads/master"},"8b41188381a85b5ee92a4050f16d263b11bb8cbe":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2022-07-25 13:00:02.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/24"}}},"commit":{"parents":[{"commit":"627667b7b344c68c158be5a230d8ab52b54c96e6","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/627667b7b344c68c158be5a230d8ab52b54c96e6"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-25 12:58:53.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b41188381a85b5ee92a4050f16d263b11bb8cbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b41188381a85b5ee92a4050f16d263b11bb8cbe"}]},"branch":"refs/heads/master"},"83e0c01af0267ceac101249ed9a335292e34a123":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2022-08-01 08:33:33.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/25"}}},"commit":{"parents":[{"commit":"bf62840560d1cd1b24ed8decc574c02918b26826","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf62840560d1cd1b24ed8decc574c02918b26826"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-08-01 08:33:22.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83e0c01af0267ceac101249ed9a335292e34a123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/83e0c01af0267ceac101249ed9a335292e34a123"}]},"branch":"refs/heads/master"},"cf88b5908fcc67b97fd8abd0f787ae431a4a394f":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2022-08-01 13:44:40.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/26"}}},"commit":{"parents":[{"commit":"5023c52f39ae1b5772849c4f319c491a62c4180f","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5023c52f39ae1b5772849c4f319c491a62c4180f"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-08-01 13:44:31.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf88b5908fcc67b97fd8abd0f787ae431a4a394f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cf88b5908fcc67b97fd8abd0f787ae431a4a394f"}]},"branch":"refs/heads/master"},"ba72563dee76a5865d1e4cc70070535634521f25":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2022-08-02 09:33:49.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/27"}}},"commit":{"parents":[{"commit":"cb45073038420951fdb58024fad6f6d2701cccff","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cb45073038420951fdb58024fad6f6d2701cccff"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-08-02 09:33:25.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba72563dee76a5865d1e4cc70070535634521f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ba72563dee76a5865d1e4cc70070535634521f25"}]},"branch":"refs/heads/master"},"18ffef8988477ea758c5e8669c744cb5da13a0a2":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2022-08-29 13:21:26.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/28"}}},"commit":{"parents":[{"commit":"04d419a9079a68fc34f1935544b1540e9b4f3dec","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/04d419a9079a68fc34f1935544b1540e9b4f3dec"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-08-29 13:10:20.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18ffef8988477ea758c5e8669c744cb5da13a0a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18ffef8988477ea758c5e8669c744cb5da13a0a2"}]},"branch":"refs/heads/master"},"18a302aad11743c5628878b85c04a3343d21d758":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2022-08-30 08:10:47.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/29"}}},"commit":{"parents":[{"commit":"c2ccd46f054ae8a27eb4293ac77fafd1440b5c5e","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c2ccd46f054ae8a27eb4293ac77fafd1440b5c5e"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-08-30 08:10:28.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18a302aad11743c5628878b85c04a3343d21d758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18a302aad11743c5628878b85c04a3343d21d758"}]},"branch":"refs/heads/master"},"584f82008d6efa0f7ebf4a324fbdb596cc686466":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2022-09-01 16:21:23.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/30"}}},"commit":{"parents":[{"commit":"81853192480c4d011e452cccbe8a15f56a48689d","subject":"utils: add option to generate dhcp device id per segmentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/81853192480c4d011e452cccbe8a15f56a48689d"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-01 12:48:54.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/584f82008d6efa0f7ebf4a324fbdb596cc686466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/584f82008d6efa0f7ebf4a324fbdb596cc686466"}]},"branch":"refs/heads/master"},"6db791ac0fde866a61d8cfbf9679dbd02d327635":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2022-09-02 10:09:52.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/18/840418/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/18/840418/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/18/840418/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/18/840418/31"}}},"commit":{"parents":[{"commit":"12b21e235e6f3b235f576bdc4986e0e4626277ea","subject":"Merge \"utils: add option to generate dhcp device id per segmentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/12b21e235e6f3b235f576bdc4986e0e4626277ea"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-04-28 14:12:24.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-09-02 10:09:20.000000000","tz":120},"subject":"ml2/mech: bind port on segment that subnet belongs","message":"ml2/mech: bind port on segment that subnet belongs\n\nIf subnet is attached to a segment we want to endure that the bind\nwill happen on it.\n\nPartial-Bug: #1956435\nPartial-Bug: #1764738\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: I56b22820d29b2d57faf28a2f9b685ab0b2c924b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6db791ac0fde866a61d8cfbf9679dbd02d327635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6db791ac0fde866a61d8cfbf9679dbd02d327635"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
