)]}'
{"id":"openstack%2Fneutron~842292","triplet_id":"openstack%2Fneutron~master~Ia3831b18463c29f676c253edb64419667b5f2c0b","project":"openstack/neutron","branch":"master","topic":"bug/1982951","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-08-22 10:04:30.000000000","reason":"Change was submitted"},"34118":{"account":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"last_update":"2022-08-22 10:04:30.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-06-15 14:40:51.000000000","reason":"Change was abandoned"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-08-22 08:16:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"9656":{"account":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"last_update":"2022-08-22 10:04:30.000000000","reason":"Change was submitted"},"18904":{"account":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"last_update":"2022-08-22 10:04:30.000000000","reason":"Change was submitted"},"32586":{"account":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"last_update":"2022-08-22 10:04:30.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-08-19 19:12:40.000000000","reason":"\u003cGERRIT_ACCOUNT_5756\u003e replied on the change","reason_account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}},"8655":{"account":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"last_update":"2022-08-22 10:04:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3831b18463c29f676c253edb64419667b5f2c0b","subject":"[OVN][QoS] Add minimum bandwidth rule support to ML2/OVN","status":"MERGED","created":"2022-05-18 09:19:53.000000000","updated":"2022-08-22 10:05:45.000000000","submitted":"2022-08-22 10:04:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":42,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"842292-bug/1982951","meta_rev_id":"9a913ebce48750afc5a0e59d64250b4fc7921ef2","_number":842292,"virtual_id_number":842292,"owner":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-08-19 18:08:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-22 10:04:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-22 08:16:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-19 19:12:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-22 08:16:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"value":0,"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"}],"reviewers":{"REVIEWER":[{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"}],"CC":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 09:25:33.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-05-18 10:02:44.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-18 10:32:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-18 10:47:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-18 11:10:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-05-18 12:24:21.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-05-18 13:55:05.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"CC"},{"updated":"2022-06-15 14:40:51.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2022-07-08 06:43:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-19 00:57:00.000000000","updated_by":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"reviewer":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"state":"REVIEWER"},{"updated":"2022-07-19 14:55:30.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-08-10 14:40:35.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2022-08-16 11:21:07.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":32586,"name":"Elvira García Ruiz","display_name":"Elvira","email":"egarciar@redhat.com","username":"elvira"},"state":"REVIEWER"},{"updated":"2022-08-16 20:17:21.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-08-18 08:21:19.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":34118,"name":"Arnau Verdaguer","email":"averdagu@redhat.com","username":"averdaguer"},"state":"REVIEWER"},{"updated":"2022-08-18 12:31:53.000000000","updated_by":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"reviewer":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"state":"REVIEWER"}],"messages":[{"id":"ed760646620e315eaf53287d14507f3d4d38e0f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"date":"2022-05-18 09:19:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93a117418cc15fbafbcf45c21776f65d35a9f7a7","author":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"date":"2022-05-18 09:23:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"fab355845aafcdd032907e28394fa2c87a6517ab","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-18 10:02:44.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/92/842292/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c4f780c55c1e9d27165bea336976ace0bf037731","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 10:32:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b819901b74d44e5a8b2347afaf10ec24 : FAILURE in 1h 06m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c17a3a83ee04609ab2d460515a0d243 : SUCCESS in 1h 06m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4386534487a253a50e05a887d85c112e263fa3e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 10:47:37.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api https://zuul.opendev.org/t/openstack/build/a446a3f1a5a746f194bcf31b5e7068d9 : SUCCESS in 52m 56s\n- neutron-tempest-plugin-scenario-ovn https://zuul.opendev.org/t/openstack/build/af8e4f1b646e45e6b6ff880e2994ae8d : SUCCESS in 1h 13m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa4b38ba650e4390864a4d51a74bc7ed : SUCCESS in 35m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a0465fbe3a24d6cb01d162c93aa2731 : FAILURE in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3541cd7b0c1346ccbd7475f4d0de6ed2 : SUCCESS in 26m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4288d18fed1a4951a016160745b6e0ee : SUCCESS in 25m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c501be9f75f459f853b0de28bd2ca44 : SUCCESS in 24m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c3bb3e98d4b4b928216f8f956726c8a : SUCCESS in 11m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cae7702ea6d1424099269794c2184d4d : FAILURE in 56m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/804a27b88c354274b06b058911ee2bd0 : SUCCESS in 58m 14s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a84e367c49b94bdeb9e62f8ebc128dc5 : SUCCESS in 1h 20m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e43753d87852cc557bd2427b6cc17e41c7e39ebb","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-18 11:10:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/842292/1 : SUCCESS in 1h 49m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8b35a8e4b34efcec15479a3bd211e8c86026b977","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2022-05-18 13:55:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83e95c26f8462cbc32c9336af18b52a4e22fa0b8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-15 14:40:51.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Zuul jobs the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":1},{"id":"c049207c740e32a24f441403ea6a07cfc8bc2349","tag":"autogenerated:gerrit:restore","author":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"date":"2022-07-08 05:40:00.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"5c42d8643f4042642a7799101782dfce1a95b3ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"date":"2022-07-08 05:46:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"422a2f2bf9fdb2798ada34f816e1d83547533dab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"date":"2022-07-08 05:50:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6d340604b69321bca4cd685e7547f5879511cd8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 06:32:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26c90771df9a46c68452873482c3b006\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13688257fd5d47e9912e7fb9c6d16596 : FAILURE in 33m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc097f19b7bb46668e3dc32355dc26cb : FAILURE in 39m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5c9cb0792db20ae9176e308c82bfc88eb5a4fb66","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-08 06:43:00.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/92/842292/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"d516d6656e770f8d7942156f4a56c33cf9d07350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-08 07:11:10.000000000","message":"Patch Set 3: Verified-1\n\n(11 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96ee96fdc8f34f8ba997da9d0ae2829d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/58fdf7242bc6407c9d821c8e92a4fb3f : SUCCESS in 1h 18m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25f5d23a5b1743069cfafed0c8181e01 : FAILURE in 33m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c950906840ef403dad008b015c862605 : FAILURE in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7469c6e8f2e4a1b80d07dae589f3f19 : FAILURE in 21m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abbd9d3df1d94d13b666bfe7f3a770b9 : FAILURE in 23m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8b026f62fa5417a8e6b8cab17f05691 : FAILURE in 21m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2ddf2a6f5494f79826a13d3c75ce203 : SUCCESS in 9m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7babebfc2efe48a0b7fe403816bb494e : FAILURE in 42m 31s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43c2b862c99f48bca227dedef168d72e : SUCCESS in 57m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a663737584e3429d9a508525d5c5aa3e : SUCCESS in 1h 08m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"77c8cbaeed1e26e6f9108c724dad285ff378461e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-08 07:36:43.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/842292/3 : SUCCESS in 1h 45m 08s","accounts_in_message":[],"_revision_number":3},{"id":"5077bd071d5c7c517406d77e464bfe949460d3d9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-08 07:41:37.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/92/842292/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"530de0e31617fa27fe0abe7699da6a3f32997c65","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2022-07-19 00:57:00.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"98be7c1fd9a2478bf9a3ad621917a50be2621bdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"date":"2022-07-19 11:58:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fce16a9662ced5d17045a62ccb44eaee0c856922","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 12:42:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29cde43349504454a447f569067138a2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/220f470ed7fe48af86071d4f6e434813 : FAILURE in 29m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32a064f466b04d55aa70b3278fbdfc18 : FAILURE in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"09de6ba89f3a80e443ed8dde5b448fdb43f622d6","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-19 13:36:20.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/842292/4 : SUCCESS in 1h 36m 44s","accounts_in_message":[],"_revision_number":4},{"id":"440cb6e38314ee8a9567ab71a20673b54095be3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 13:42:13.000000000","message":"Patch Set 4: Verified-1\n\n(15 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd2f93e6ac6c46ee986311668489204f\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/b609094007aa4093bebc6a1b539e1a2d : SUCCESS in 1h 17m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3447b586b70d41e489c641d2e7d55e74 : SUCCESS in 36m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b22db959fb6e45b5aaefa31e91618291 : FAILURE in 32m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d58f034b6fe4a08b6b3f7cc9804f863 : FAILURE in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5470aaeb8e604832b0fac995c1a5a9a1 : FAILURE in 24m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70fc879070d84eb78dc1fcc8cad8bf5b : FAILURE in 27m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c1c12d0360445828cc7b8f88cc6f575 : FAILURE in 24m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cf7963385fb4081aa619b5dc0ce8293 : SUCCESS in 10m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/01c474c704e0467597aeedfb979f1535 : FAILURE in 1h 00m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/76fc1b2495484ced9ab3a0cd221fcdd9 : SUCCESS in 1h 40m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7217e38f47e04ef48ffcc1e779209750 : SUCCESS in 1h 08m 16s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fb939af9e9234dd38a0531c5368d13d3 : SUCCESS in 1h 16m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"08b6b5f58e7cd42715e68c383da2e981a536bf3d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-19 14:55:30.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f5be169669f6cd05190def0a9078814aa0ec5880","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-19 17:41:15.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/92/842292/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"059bc56d21bcc332ffdd414fe9f3f675f83ac4d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-27 14:01:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7c0eec7d101fa57e31516a1a25d3c0add117e9e0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-27 14:01:35.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f65fc013e7066fc22dadfb948077aa0d2f8e7ad6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 14:40:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0d19d19b30641ce8c18b07b6d37e36d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/346846af711049cea926103276c707bd : SUCCESS in 33m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/531fe849185a4ca8bc9781d3268fd274 : SUCCESS in 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"13b4c5ef835ef2c84f30de6a274a76601830dd45","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-27 14:46:49.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/92/842292/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"e460ade1403637d31636f3f25007625bd78685b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 15:43:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2dd4183ff8f349ddacb4c24e2d6daeea\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/400e7ebaecb947fab0829a77a1422cc1 : FAILURE in 22m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/28520614bf944909a7fc129383617c53 : FAILURE in 24m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56ea58b0cc9f451bb5fef82b1858ef8f : SUCCESS in 38m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb65cc0df8284bbeb8c64ff4689982b3 : SUCCESS in 14m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cd6b262eb5d420e91c0e4f0ab8f735a : SUCCESS in 28m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d64165d6cbf4e7aa5297da32b5e6247 : SUCCESS in 30m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7838e5c64274d9ca8e7e6c302ba11b1 : SUCCESS in 30m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84df595e0cdf4eedb1ae5a8b2b7d519e : SUCCESS in 11m 48s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/67ce578051464324b27189116f103406 : FAILURE in 55m 13s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1c6df3d0c73f483daaa933e64ce7fd97 : SUCCESS in 1h 37m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/724c32e0234943e89d228fa5d324e779 : FAILURE in 31m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ae7431e2a2b54c88a93dd6af0842c08f : FAILURE in 37m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5744b75def6636e67e13e89d83a87da59435c9fe","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-27 15:46:07.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/842292/5 : SUCCESS in 1h 44m 12s","accounts_in_message":[],"_revision_number":5},{"id":"ea46eec5ab011b524a8a550b1e774dbd4afd1ac2","author":{"_account_id":9656,"name":"Ihar Hrachyshka","email":"ihrachys@redhat.com","username":"ihrachys","status":"Red Hat Networking Systems Engineer"},"date":"2022-07-28 00:11:18.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"460ee1a5531ae2eacb7368a5c186263d57eacec1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-29 16:29:58.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"83866df1482963a17bd6a6d7fc47e1ef7c12c0a2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-29 16:30:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"abc1a37fa01130d82c85f083e285fd7363da3aa1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 17:13:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e48150c58954b98a8f148c09d675077\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b241da633f54d069941dd7922685a53 : FAILURE in 41m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11ff2459a1a24674b1e534752b828901 : FAILURE in 40m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e9c85ee2c3a60669b69565a3da959f2363b58496","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-29 17:23:11.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/92/842292/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"3f148d832dfc88e3f43bdc76f6658c02186ec9ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 18:12:18.000000000","message":"Patch Set 6: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5b28db2cd834b488c3dcc862f81ab20\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/639b5e73abe34571a740e1ae2b526476 : FAILURE in 1h 40m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89a1b1efe6d64e8eb2aeaff9e6536e6e : SUCCESS in 35m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/739476168381484d8ffe75844018e12d : FAILURE in 29m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f558687523bd4aaf805ede90f39e0391 : FAILURE in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ccd550d2da0467db14b8dd338c6b4fe : FAILURE in 24m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/125a89b7a4104ed197a13c3490eade81 : FAILURE in 23m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/44e8a19ffac84e9da73402020dcb7c43 : FAILURE in 24m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a35ea8d51d4bde8b6ef0e2c1ed63e9 : SUCCESS in 10m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ba55ae3c48ce45a095a6d13e85549d3a : FAILURE in 55m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4ea80aa91a8c42c4be6008640c589b41 : SUCCESS in 1h 34m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6e4aee21e3d46f7ae54f801836e46e8 : FAILURE in 1h 14m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fad4fd8d0bd74efead821deb0eac7d56 : SUCCESS in 1h 15m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9c2be8b708806e5efd20aaa24042bc0d1bb0d983","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-29 18:18:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/842292/6 : SUCCESS in 1h 48m 09s","accounts_in_message":[],"_revision_number":6},{"id":"b52596eca15a5d839894124acd5ee5cbd4c83dc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-04 09:51:16.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0f4ba7e4b25a9eaaf936f6a41d63a22e8c336a88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:34:35.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23d77758bf054285b82bf122a52741ee\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1adc5c0b12e4227a4ca8e15935c2876 : SUCCESS in 38m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23e6d2bf6afa4dff9e182db9dd3704b7 : SUCCESS in 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a801bb461549f1b6fbaa463a373c8cd2d42a856e","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-04 10:40:50.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9c5762a548c733b8355dd250a8a181c69bfd730c","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-04 10:46:11.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/92/842292/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"88cc6591df8c20945ac8eefabf8ca617ffef76e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 11:30:21.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7006f8d118b240a2a7b6d690c2b45def\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2733d88611264f08bc8c186739c885c7 : SUCCESS in 1h 07m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6f08680976c47e881967ec1244f0afb : SUCCESS in 36m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f78b17495d0c49c686aaee5a17d4c905 : SUCCESS in 39m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd4bb47eaac641239b6da47939f5a54b : FAILURE in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3401270dd5ca49ab878c1ad47ba5d1cf : SUCCESS in 22m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8581e0807c884dc4b297fab45608700b : SUCCESS in 23m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62befb5f6d9c4ec6afc222981ef59bec : SUCCESS in 23m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08448e1a18094e8a871d9c35931edf98 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91ebd5cc87674ce0adc7fc92762693c0 : SUCCESS in 9m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b002b2d3a34c40ff8ad0877ac95837cd : SUCCESS in 57m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5fe23bdd6584ad28a1f2ad0dfe22bf0 : SUCCESS in 1h 36m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa37fd8fc9c8478ca9ac147b116878cd : SUCCESS in 56m 41s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5503e369c33a4105886851020a19d519 : SUCCESS in 1h 08m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9846b2fea67cad5e83012bb285b5dc2ac703c05a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-04 11:58:13.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/842292/7 : SUCCESS in 2h 06m 16s","accounts_in_message":[],"_revision_number":7},{"id":"e85a35dbb442d26c101b92ce4fb1d0aa2ddaebde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-09 11:15:56.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"26b6ce2928d6917489dde074821fecafed725857","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:02:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccc4313c56d548b79a38b0d4b84b45b7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5450775273324536b5a01a79c11786ed : SUCCESS in 41m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55c4155d4d134fb499f94bb0e1637d7e : SUCCESS in 41m 27s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"704afb8bd160391aea8c085750e0328cf058e730","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-09 12:45:21.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/92/842292/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"df19ec139e766118e2aba6a9571d3a6b81571675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 13:08:15.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df8f89d66ce94252b0e75b9779ec3963\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/bd0c15aed5e143f291feab342c7062fc : SUCCESS in 1h 39m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5caa3041a72b49c99fbe0a4b4ae262b3 : SUCCESS in 40m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b16f7ac951b45faa87dd0176347056e : SUCCESS in 36m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79ecc4482025410babd6822f8c1a65f7 : SUCCESS in 15m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69f4f08c38bf4fb8925a7f9ea4bf2a9a : SUCCESS in 26m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6415ccfffe694d718b72df21970b6132 : SUCCESS in 26m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75264b4d6a704969996484baeef1ba06 : SUCCESS in 27m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51b15f38969440609c239da459498ee2 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23aa3debca0641968dfa930a02d5839b : SUCCESS in 12m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cbebb2487abf47a19495b78aeac1eaa0 : SUCCESS in 1h 16m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/83f73090440d4baf8bb7d236a1cbe9c3 : SUCCESS in 1h 45m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e41e14447744fd3b6ff12b85ad0c57a : SUCCESS in 1h 04m 07s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/243d25145bf0477e8cd72933d4f97e7e : SUCCESS in 1h 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a2ff5b0ae74f8d43046edf06af8fc8b8d3002b27","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-09 13:19:11.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/842292/8 : SUCCESS in 2h 02m 33s","accounts_in_message":[],"_revision_number":8},{"id":"04ab6c9b76755940da09577ba84ec102b02853de","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-08-10 14:40:35.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8667c83b42b99e059cf9f4896a2708bde0324549","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-10 15:59:23.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"f188c571d64ca47baaf893c248cb80b9fe9ffdba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-10 16:43:08.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4cc7842847ab083667523068b3e39e1469dd22f7","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-10 17:38:49.000000000","message":"Patch Set 9: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/92/842292/9 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":9},{"id":"25b2d6dee7793329eb6b20bd424ce17f6bf1a849","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 17:48:38.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d495292910e84c15885868ac4cf9087d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b77920a14dc4d0ab4f7a34d8b97109b : SUCCESS in 44m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f77740dbe134785bf1a5f595d593b69 : SUCCESS in 43m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"af361821b9370e9bc58c688b942dd912bf01f0cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 18:20:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb1c76fd1ed64d738bbf2d803ca95e1d\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/0876567c5c2b42358f747cd8f814a760 : SUCCESS in 1h 22m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd58bec004b04756a6a4108ede570e66 : SUCCESS in 43m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df078f0ce0de4b0e9ff8bc254ad16e5c : SUCCESS in 32m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc21eec19a414a379954b074fac4590b : SUCCESS in 12m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10c524abd7e24168a4dee21a6e7b54e7 : SUCCESS in 27m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/145b8f7e5b344f789013fbe8fbc016b3 : SUCCESS in 26m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2299b4dd79134dc7aeafd65044ba5b1e : SUCCESS in 26m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59cee4042d854517840e641c36456cd6 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6fdd577766b450c84246f0e89c1e08d : SUCCESS in 8m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e9e4e1b8c90403b9e6d5587541b0625 : SUCCESS in 1h 02m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f0066a530d74f469e05c7318b381d45 : SUCCESS in 1h 35m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aee0cc5bbfa344c3bdf07c016f96ad7e : SUCCESS in 1h 07m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/4dc65061f5bd4764ae6abacc97787ecf : SUCCESS in 1h 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ab298eb3416a1621e35d5d4b339a16d4bbe51c36","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-10 23:47:37.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/842292/9 : SUCCESS in 1h 48m 50s","accounts_in_message":[],"_revision_number":9},{"id":"6efeba910669942d746d13aff09a2d082100b98e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-08-16 20:17:21.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"23d5aa5302f43d624cb79b2003877635ae3b89e0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-17 11:43:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b0a1e1dee43d98a0e44293c332f57aefcf2876c6","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-08-18 12:31:53.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"8dbf6c47bdda50ba75ba4e9997d86c5b0bc18403","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-19 10:08:38.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b8ef5a5a0d4d68632e8cee53e7dac8f68d7efa32","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-08-19 15:57:00.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9b3cee5a5be47cd3c9ad6e3444dd98be38eb2868","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-19 16:53:32.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f99b0397c8cd1692af502a5cdd28f3e51cee1e5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-19 17:13:28.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"e99d9b5a499cd7e28dcbd6bf00e0ce0f6c18894b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 17:55:43.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c6763bd912740ad82eae2430ad77fa6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d3bd0f4ec0441889567467fe2b29a81 : SUCCESS in 40m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd5d14463313496f9deb873a6c24c2bf : SUCCESS in 30m 51s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"12572cfe4b484f53bec0975c42509c52a07a8090","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-19 18:08:24.000000000","message":"Patch Set 10: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/92/842292/10 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":10},{"id":"eacbe894f68556f4fd653616e82e2998896bc92f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 19:11:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07fceac1c2354d1d916b7beab7a976ca\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/2460ae124c0346e682867698c51b0490 : SUCCESS in 1h 40m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d005af549f56470d83a2bff83ba6d830 : SUCCESS in 35m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4967e18cccd14ea0a9f917cad37a9602 : SUCCESS in 36m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9350eb423af74e54b8774eb80006ba18 : SUCCESS in 12m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/575535b8b38d4fdebcf92a5569798a8d : SUCCESS in 29m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/315a4abd0c2342c894b8e55730ccdf56 : SUCCESS in 28m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd8176d396d44764a0b7b42ee11efe20 : SUCCESS in 25m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a84f52cd6b04c86974a578b83679e5f : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af5dd8c19f5441d3bf5d7dfa6d21e5da : SUCCESS in 8m 06s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d18b584e66de4bb7b3804d053cd789f9 : SUCCESS in 1h 02m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/855f22bd9f774c8482f21c40cd76cd5a : SUCCESS in 1h 55m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f72c789cd2aa41a896990da693f7ce39 : SUCCESS in 1h 04m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8c422ce388e2473eb885df2fe7e39173 : SUCCESS in 1h 10m 22s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0d0f2ac900744e04b001564c777bb58201d633a2","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-08-19 19:12:40.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"b9e7c91312bac13f918e9c9c38da1d425cb5979d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-19 19:35:53.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/842292/10 : SUCCESS in 2h 21m 44s","accounts_in_message":[],"_revision_number":10},{"id":"b793e5d92622095a1ecae4be90561f97c623f23f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-08-22 08:16:14.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3ba1e8893de99d7a3381f5b713b275270a6991a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 08:16:52.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"0b3e99530e421440b56292c165ea7d52e8e347f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 10:04:30.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/776522b8579445f5b0a8649a52f82bde\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/25d0b015e61e41c3b25cc15fdfdd209a : SUCCESS in 1h 03m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6ee1efeeabd4cbe81a59b1ef9a9b2b7 : SUCCESS in 11m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21aba2d69b424dbfa9f30fd857090a78 : SUCCESS in 22m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41bee659bf0c4803bcb81ef2606cc6ae : SUCCESS in 29m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f22c2d53fed4b769bf8491f88e2ed2c : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90abb0f2c18c4e83a529219796318fd1 : SUCCESS in 9m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9ff43aa9808144e69052f8e973a132c5 : SUCCESS in 1h 03m 12s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a1024d6a8ba4469a71c6b25bdcacf9a : SUCCESS in 1h 46m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46ce3705fe114de6a3ca2e20e9d4a3c1 : SUCCESS in 54m 36s","accounts_in_message":[],"_revision_number":10},{"id":"ba618d014f15da67571037ea0bc781dd4e072e8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 10:04:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"9a913ebce48750afc5a0e59d64250b4fc7921ef2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 10:05:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e316679b3e24bbfb7e0f3e9d79c35bc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41f922287644967a07dc10a30ad4a8f : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6834f0b14e1c4a2d895835d6dcbebb5e : SUCCESS in 44s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"846737dac4acb0fd47fb20058663513c4e054ee6","revisions":{"bfce8dcfe501fa27e8c2e7cd31b7cea8bfc9fc84":{"kind":"REWORK","_number":1,"created":"2022-05-18 09:19:53.000000000","uploader":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"ref":"refs/changes/92/842292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/842292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/842292/1"}}},"commit":{"parents":[{"commit":"b12a6a07473c777e015cb9f387a93d3e51ff5659","subject":"Merge \"[L3HA] Don\u0027t update HA router\u0027s ports if router isn\u0027t active on agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b12a6a07473c777e015cb9f387a93d3e51ff5659"}]}],"author":{"name":"Karthik S","email":"ksundara@redhat.com","date":"2022-05-18 09:14:23.000000000","tz":0},"committer":{"name":"Karthik S","email":"ksundara@redhat.com","date":"2022-05-18 09:19:46.000000000","tz":0},"subject":"WIP: Add min bandwidth support to ml2-ovn","message":"WIP: Add min bandwidth support to ml2-ovn\n\nChange-Id: Ia3831b18463c29f676c253edb64419667b5f2c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bfce8dcfe501fa27e8c2e7cd31b7cea8bfc9fc84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bfce8dcfe501fa27e8c2e7cd31b7cea8bfc9fc84"}]},"branch":"refs/heads/master"},"35c2fa9d002416eee81cc0980f8f5f7d59ffdc85":{"kind":"REWORK","_number":2,"created":"2022-07-08 05:46:41.000000000","uploader":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"ref":"refs/changes/92/842292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/842292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/842292/2"}}},"commit":{"parents":[{"commit":"8ab5ee1d174388c9ae31f083ec656f6f95adcc14","subject":"Fix remaining typos in comments and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ab5ee1d174388c9ae31f083ec656f6f95adcc14"}]}],"author":{"name":"Karthik S","email":"ksundara@redhat.com","date":"2022-07-08 05:44:38.000000000","tz":0},"committer":{"name":"Karthik S","email":"ksundara@redhat.com","date":"2022-07-08 05:46:36.000000000","tz":0},"subject":"WIP: Add min bandwidth support to ml2-ovn","message":"WIP: Add min bandwidth support to ml2-ovn\n\nChange-Id: Ia3831b18463c29f676c253edb64419667b5f2c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35c2fa9d002416eee81cc0980f8f5f7d59ffdc85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/35c2fa9d002416eee81cc0980f8f5f7d59ffdc85"}]},"branch":"refs/heads/master"},"ce5763d4827e630d9c85b8ae114d627b3e073f37":{"kind":"REWORK","_number":3,"created":"2022-07-08 05:50:58.000000000","uploader":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"ref":"refs/changes/92/842292/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/842292/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/842292/3"}}},"commit":{"parents":[{"commit":"8ab5ee1d174388c9ae31f083ec656f6f95adcc14","subject":"Fix remaining typos in comments and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ab5ee1d174388c9ae31f083ec656f6f95adcc14"}]}],"author":{"name":"Karthik S","email":"ksundara@redhat.com","date":"2022-07-08 05:44:38.000000000","tz":0},"committer":{"name":"Karthik S","email":"ksundara@redhat.com","date":"2022-07-08 05:50:27.000000000","tz":0},"subject":"WIP: Add min bandwidth support to ml2-ovn","message":"WIP: Add min bandwidth support to ml2-ovn\n\nChange-Id: Ia3831b18463c29f676c253edb64419667b5f2c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce5763d4827e630d9c85b8ae114d627b3e073f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ce5763d4827e630d9c85b8ae114d627b3e073f37"}]},"branch":"refs/heads/master"},"f7255710dce76365d9bc1f7c714ff1c1dfc92ae9":{"kind":"REWORK","_number":4,"created":"2022-07-19 11:58:48.000000000","uploader":{"_account_id":18904,"name":"Karthik S","email":"ksundara@redhat.com","username":"karthiks"},"ref":"refs/changes/92/842292/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/842292/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/842292/4"}}},"commit":{"parents":[{"commit":"8ab5ee1d174388c9ae31f083ec656f6f95adcc14","subject":"Fix remaining typos in comments and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ab5ee1d174388c9ae31f083ec656f6f95adcc14"}]}],"author":{"name":"Karthik S","email":"ksundara@redhat.com","date":"2022-07-08 05:44:38.000000000","tz":0},"committer":{"name":"Karthik S","email":"ksundara@redhat.com","date":"2022-07-19 11:58:06.000000000","tz":0},"subject":"WIP: Add min bandwidth support to ml2-ovn","message":"WIP: Add min bandwidth support to ml2-ovn\n\nChange-Id: Ia3831b18463c29f676c253edb64419667b5f2c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7255710dce76365d9bc1f7c714ff1c1dfc92ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7255710dce76365d9bc1f7c714ff1c1dfc92ae9"}]},"branch":"refs/heads/master"},"5f3c330440f6dcb9716ab5b42c68f23ebe1c8c4c":{"kind":"REWORK","_number":5,"created":"2022-07-27 14:01:03.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/92/842292/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/842292/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/842292/5"}}},"commit":{"parents":[{"commit":"fb1c84c23a95ed7c044716b77e753234fda8032c","subject":"Merge \"QoS minimum bandwidth documentation: OVN backend\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/fb1c84c23a95ed7c044716b77e753234fda8032c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-21 20:26:09.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-27 13:45:14.000000000","tz":120},"subject":"[OVN][QoS] Add minimum bandwidth rule support to ML2/OVN","message":"[OVN][QoS] Add minimum bandwidth rule support to ML2/OVN\n\nThis patch adds support for QoS egress minimum bandwidth rules in\nML2/OVN. The enforcement is done in the network backend.\n\nSince [1], in v22.06.0, OVN is capable of guarantee a minimal\nbandwidth for a logical switch port. The enforcement of this rule\nis done in the physical bridge interface.\n\n[1]https://github.com/ovn-org/ovn/commit/dbf12e5fe1f7ab2acef4152854c239b999b70188\n\nCloses-Bug: #1982951\n\nChange-Id: Ia3831b18463c29f676c253edb64419667b5f2c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f3c330440f6dcb9716ab5b42c68f23ebe1c8c4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5f3c330440f6dcb9716ab5b42c68f23ebe1c8c4c"}]},"branch":"refs/heads/master"},"c62753e093f16d7a8739b6a313430b8c884011aa":{"kind":"REWORK","_number":6,"created":"2022-07-29 16:29:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/92/842292/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/842292/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/842292/6"}}},"commit":{"parents":[{"commit":"4c7e8de57a895d2c0af4f5d368b782c38377d70a","subject":"Merge \"Port update will trigger less notifications to the DHCP agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c7e8de57a895d2c0af4f5d368b782c38377d70a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-21 20:26:09.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-28 10:19:02.000000000","tz":120},"subject":"[DNM] [OVN][QoS] Add minimum bandwidth rule support to ML2/OVN","message":"[DNM] [OVN][QoS] Add minimum bandwidth rule support to ML2/OVN\n\n\u003d\u003d\u003d STILL UNDER DEVELOPMENT \u003d\u003d\u003d\n\nThis patch adds support for QoS egress minimum bandwidth rules in\nML2/OVN. The enforcement is done in the network backend.\n\nSince [1], in v22.06.0, OVN is capable of guarantee a minimal\nbandwidth for a logical switch port. The enforcement of this rule\nis done in the physical bridge interface.\n\n[1]https://github.com/ovn-org/ovn/commit/dbf12e5fe1f7ab2acef4152854c239b999b70188\n\nCloses-Bug: #1982951\n\nChange-Id: Ia3831b18463c29f676c253edb64419667b5f2c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c62753e093f16d7a8739b6a313430b8c884011aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c62753e093f16d7a8739b6a313430b8c884011aa"}]},"branch":"refs/heads/master"},"b87f9880364e85e252501eeb20b01af294179c6b":{"kind":"REWORK","_number":7,"created":"2022-08-04 09:51:16.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/92/842292/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/842292/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/842292/7"}}},"commit":{"parents":[{"commit":"4c7e8de57a895d2c0af4f5d368b782c38377d70a","subject":"Merge \"Port update will trigger less notifications to the DHCP agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c7e8de57a895d2c0af4f5d368b782c38377d70a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-21 20:26:09.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-29 22:21:04.000000000","tz":120},"subject":"[DNM] [OVN][QoS] Add minimum bandwidth rule support to ML2/OVN","message":"[DNM] [OVN][QoS] Add minimum bandwidth rule support to ML2/OVN\n\n\u003d\u003d\u003d STILL UNDER DEVELOPMENT \u003d\u003d\u003d\n\nThis patch adds support for QoS egress minimum bandwidth rules in\nML2/OVN. The enforcement is done in the network backend.\n\nSince [1], in v22.06.0, OVN is capable of guarantee a minimal\nbandwidth for a logical switch port. The enforcement of this rule\nis done in the physical bridge interface.\n\n[1]https://github.com/ovn-org/ovn/commit/dbf12e5fe1f7ab2acef4152854c239b999b70188\n\nCloses-Bug: #1982951\n\nChange-Id: Ia3831b18463c29f676c253edb64419667b5f2c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b87f9880364e85e252501eeb20b01af294179c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b87f9880364e85e252501eeb20b01af294179c6b"}]},"branch":"refs/heads/master"},"1071de73b31599933cf89fd53eb85c474d9d8dc7":{"kind":"REWORK","_number":8,"created":"2022-08-09 11:15:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/92/842292/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/842292/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/842292/8"}}},"commit":{"parents":[{"commit":"4c7e8de57a895d2c0af4f5d368b782c38377d70a","subject":"Merge \"Port update will trigger less notifications to the DHCP agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c7e8de57a895d2c0af4f5d368b782c38377d70a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-21 20:26:09.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-09 11:15:32.000000000","tz":120},"subject":"[OVN][QoS] Add minimum bandwidth rule support to ML2/OVN","message":"[OVN][QoS] Add minimum bandwidth rule support to ML2/OVN\n\nThis patch adds support for QoS egress minimum bandwidth rules in\nML2/OVN. The enforcement is done in the network backend.\n\nSince [1], in v22.06.0, OVN is capable of guarantee a minimal\nbandwidth for a logical switch port. The enforcement of this rule\nis done in the physical bridge interface.\n\n[1]https://github.com/ovn-org/ovn/commit/dbf12e5fe1f7ab2acef4152854c239b999b70188\n\nCloses-Bug: #1982951\n\nChange-Id: Ia3831b18463c29f676c253edb64419667b5f2c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1071de73b31599933cf89fd53eb85c474d9d8dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1071de73b31599933cf89fd53eb85c474d9d8dc7"}]},"branch":"refs/heads/master"},"cc136f73150f0f5a3e9e9c17d8c1125025a97adb":{"kind":"REWORK","_number":9,"created":"2022-08-10 16:43:08.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/92/842292/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/842292/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/842292/9"}}},"commit":{"parents":[{"commit":"4c7e8de57a895d2c0af4f5d368b782c38377d70a","subject":"Merge \"Port update will trigger less notifications to the DHCP agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c7e8de57a895d2c0af4f5d368b782c38377d70a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-21 20:26:09.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-10 01:28:47.000000000","tz":120},"subject":"[OVN][QoS] Add minimum bandwidth rule support to ML2/OVN","message":"[OVN][QoS] Add minimum bandwidth rule support to ML2/OVN\n\nThis patch adds support for QoS egress minimum bandwidth rules in\nML2/OVN. The enforcement is done in the network backend.\n\nSince [1], in v22.06.0, OVN is capable of guarantee a minimal\nbandwidth for a logical switch port. The enforcement of this rule\nis done in the physical bridge interface.\n\n[1]https://github.com/ovn-org/ovn/commit/dbf12e5fe1f7ab2acef4152854c239b999b70188\n\nCloses-Bug: #1982951\n\nChange-Id: Ia3831b18463c29f676c253edb64419667b5f2c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc136f73150f0f5a3e9e9c17d8c1125025a97adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cc136f73150f0f5a3e9e9c17d8c1125025a97adb"}]},"branch":"refs/heads/master"},"846737dac4acb0fd47fb20058663513c4e054ee6":{"kind":"REWORK","_number":10,"created":"2022-08-19 17:13:28.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/92/842292/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/92/842292/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/92/842292/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/92/842292/10"}}},"commit":{"parents":[{"commit":"4c7e8de57a895d2c0af4f5d368b782c38377d70a","subject":"Merge \"Port update will trigger less notifications to the DHCP agents\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4c7e8de57a895d2c0af4f5d368b782c38377d70a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-21 20:26:09.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-11 22:58:17.000000000","tz":120},"subject":"[OVN][QoS] Add minimum bandwidth rule support to ML2/OVN","message":"[OVN][QoS] Add minimum bandwidth rule support to ML2/OVN\n\nThis patch adds support for QoS egress minimum bandwidth rules in\nML2/OVN. The enforcement is done in the network backend.\n\nSince [1], in v22.06.0, OVN is capable of guarantee a minimal\nbandwidth for a logical switch port. The enforcement of this rule\nis done in the physical bridge interface.\n\n[1]https://github.com/ovn-org/ovn/commit/dbf12e5fe1f7ab2acef4152854c239b999b70188\n\nCloses-Bug: #1982951\n\nChange-Id: Ia3831b18463c29f676c253edb64419667b5f2c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/846737dac4acb0fd47fb20058663513c4e054ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/846737dac4acb0fd47fb20058663513c4e054ee6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
