)]}'
{"id":"openstack%2Fneutron~842297","triplet_id":"openstack%2Fneutron~stable%2Fussuri~Id1907481077d230e4461906a1a2a1447abac330a","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1973276","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-02 09:06:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id1907481077d230e4461906a1a2a1447abac330a","subject":"[OVN] Allow VIP ports with a defined \"device_owner\"","status":"MERGED","created":"2022-05-18 09:35:35.000000000","updated":"2022-06-02 09:07:49.000000000","submitted":"2022-06-02 09:06:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842297-bug/1973276","meta_rev_id":"be21be62cbb46532a0e6efaa81371699da6a6c68","_number":842297,"virtual_id_number":842297,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 09:06:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-05-30 09:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-05-30 09:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-05-30 09:28:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-18 12:54:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 09:14:50.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-05-30 09:28:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-06-02 05:35:54.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"4204b06e4a30d926435a5c8dea582dad496a7233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-18 09:35:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d9dd92c488ad959dedb4c6e0eea331e11e54cd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 12:54:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/7c2c36250b274287b9ed18c60fe0f8ea : SUCCESS in 57m 40s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/c4c75e914ad544c6a00385f175da3990 : SUCCESS in 1h 59m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a51fdd7630f4244a176f31a1ba1fd46 : FAILURE in 32m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99caf44b1d745d480687d5bac9c9725 : SUCCESS in 12m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d678487f6314839b05e8ad225b59d8d : FAILURE in 29m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc92cb2470da40989c5bfbe5cbb3388a : FAILURE in 29m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bd4475f5b0d43be91eb86f9ad7dac57 : FAILURE in 30m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65994208bcf74dceb52be7dbf8f8877d : SUCCESS in 10m 33s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/9c6fb8a18c88492a9a3e86d7611366fd : SUCCESS in 58m 18s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/94da27e05ac148a9898ef3c051b2fca6 : SUCCESS in 1h 31m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dcd98bafa8e442b485c204bab7c625ca : SUCCESS in 1h 30m 35s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/1d1a942ae49a4d63975e2be52a4d0942 : SUCCESS in 1h 35m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e05b74a49114802a39b16a73342d697 : SUCCESS in 1h 15m 14s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/06d0bacf4da347d7acd287650c085b48 : SUCCESS in 1h 07m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/eaefeba961334de0bed66b3046bb9ac1 : SUCCESS in 1h 08m 48s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2c9174182e548cd8975c8014c79e0d1 : SUCCESS in 1h 10m 07s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/e59d21edaac1495fb27811e675b4897b : SUCCESS in 1h 07m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/64e11c6c153f434b82f944c12ad1768c : SUCCESS in 1h 35m 26s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/f133cbf501924919ae1d265904f32524 : SUCCESS in 1h 25m 35s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/77c3e3d56901450e864bd7becda34176 : SUCCESS in 1h 08m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8d949dacec1048be8ec9e8aad27fad01 : FAILURE in 46m 26s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c139de7f166744b0aa40171bec968916 : SUCCESS in 1h 24m 56s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b1a36a483fe4758a42eb009842d7a9c : SUCCESS in 51m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/3eec368d7a1145a6a0131f8fcb51d0be : SUCCESS in 58m 01s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/221d203f6a2a4e568eadab24e320390e : SUCCESS in 30m 56s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/619978c41bb447bebea7b045d5573bed : SUCCESS in 1h 09m 11s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/22a2b9fa941e4bb39b8b44dd797d75db : SUCCESS in 44m 40s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/693afec74a2f4fc8979fcc2995795b61 : SUCCESS in 3h 16m 23s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d1d7003fb0434d77bdde5dbbaf6cd0d0 : SUCCESS in 1h 06m 09s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b8e184219ea0451f80a717bc988d83a1 : SUCCESS in 1h 35m 03s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6d910aec2d734f9cb13ce70305d485f6 : SUCCESS in 1h 40m 26s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/6dc8b1a5815f441d884193f6be4b4dad : FAILURE in 1h 17m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e370bd734afa0b3e2ff50d6e1aafb4360245b895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-18 16:15:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"07c61497bd73ea090d26c871630b62a09c838bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-18 19:52:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/a838de679d4d425e9f3c80945137cb8b : SUCCESS in 43m 24s\n- neutron-tempest-plugin-scenario-ovn-ussuri https://zuul.opendev.org/t/openstack/build/8ac15ebf609c423491141786a88c4548 : SUCCESS in 2h 02m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5459b754dbff4b8baa2ab1f083fb03f1 : SUCCESS in 27m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0ca0405bcb3415b971049a4137ee6ff : SUCCESS in 12m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ac4dc9825984ad3888f1fca0c4f0240 : SUCCESS in 21m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f55975f40f94e18a17e974fb5e4ac2d : SUCCESS in 24m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a551674d3b044c0a41caf3fce2ad12a : SUCCESS in 24m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59bb0087cb7f459997c017987574059c : SUCCESS in 10m 34s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c554bcc1032f415bacbfc32db6f6dd79 : SUCCESS in 47m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/c2354b15d4d444e79b91b2e151b83ae3 : SUCCESS in 1h 15m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5721efea024c4553995552d8f62d94d0 : SUCCESS in 1h 20m 23s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/eaa6d8793d514fb3b760c45e9ae120dc : SUCCESS in 1h 17m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1401bb738f384e8d80b0b5a9df025c87 : SUCCESS in 1h 04m 27s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4b3a928f1f084fe98338e6f548755433 : SUCCESS in 1h 06m 14s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a7b18e0effb74cd5abcbae4dfd7d1aca : SUCCESS in 1h 00m 35s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c2a8e67c0014992a3a9de41b6fab8d0 : SUCCESS in 1h 08m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b35fc14e13d341b0abb80b0a7f0b63c6 : SUCCESS in 1h 01m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cc0ccaa7feda406191eaabca4741e472 : FAILURE in 1h 18m 18s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3845f6142de64567adf65cfc6a167832 : FAILURE in 1h 03m 43s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/db33668787884633a457293040a65d7c : SUCCESS in 1h 02m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e1fcf84fd2744008aab1ec4286d73427 : FAILURE in 55m 03s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1246f308cfdd4d21b5e51c55c387ab03 : SUCCESS in 1h 23m 19s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dedba8dd83744fa6ba17172e4a332441 : SUCCESS in 1h 00m 02s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/93aad6af4a6e4909990ff2b707a58a89 : SUCCESS in 1h 02m 41s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/0c7923c17cfc4494854727190580bd85 : SUCCESS in 35m 15s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/3fc5aea5d5494129bed8b707e4cb1ef9 : SUCCESS in 1h 14m 40s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9122e8481281485d9ed431339ccac575 : SUCCESS in 47m 35s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/aa3869ff78794b1896b35a6277d5f3e5 : SUCCESS in 3h 29m 19s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/13da62a385a54712a4c8d2d0e67728db : SUCCESS in 1h 15m 02s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c7d03fd685c741c3a9f80e8df5bb2008 : SUCCESS in 1h 30m 05s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0e5cdd5ff7044b66a1c62d7d3f167dfe : SUCCESS in 2h 00m 24s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/78aa229bfe604c1c9e8e67efdf50ff9d : FAILURE in 1h 24m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"95a1a0adfbb82271c851b82c6e17cef460060353","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-05-30 09:14:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a516d5064524c8228af9eb5397f7d6284836b4bd","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-05-30 09:28:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4b9664a890ae42fb743e8d2cd445452913f74081","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 09:29:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b31ef2a5dd2c3e830cd3f59e7b39189e9548aab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 13:07:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/beb86c0e8e124e7eb53a03e240eab5e9 : FAILURE in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e32f9a238a41c7bc26df9a090696ab : SUCCESS in 12m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68d1d51243ec43d4876811561277100a : SUCCESS in 25m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/667946f61b7749bbbfc74c74c0c3c5ca : SUCCESS in 22m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef40a45551d40a582c73a995b525545 : SUCCESS in 10m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/65c7d8e6102b4473bb2c309f2daaf774 : SUCCESS in 42m 34s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f1724d2cc4e94e36a630dcb513e7bc7c : SUCCESS in 1h 21m 44s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4e4f1f51702048d0972267521ae6a261 : FAILURE in 22m 18s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dc021717ed8e454d9d5c363a5251569d : FAILURE in 22m 24s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6cd92cd3096c458b97bd311e039d57a6 : FAILURE in 11m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ee03bc751f8c4408829395b943b908be : FAILURE in 55m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4ae61d1e213f412b98dfb7ab6db7e94b : FAILURE in 56m 51s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dbc0753e158f4415950e7f3be159540e : FAILURE in 16m 07s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf1d3ab854bd4de1b6a2070c7c8f3068 : FAILURE in 14m 59s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/d5d458f5c36a4f4db34c0d462142d5a0 : FAILURE in 11m 00s","accounts_in_message":[],"_revision_number":2},{"id":"24923e0e45b5969ac93d9b567162c14e9991f025","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-31 15:51:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3444db95b1fcc1d89a6257b0deca433ee55a616f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 15:52:41.000000000","message":"Patch Set 2: Verified-1\n\nUnable to freeze job graph: Job neutron-centos-8-tripleo-standalone depends on tripleo-ci-centos-8-content-provider which was not run.","accounts_in_message":[],"_revision_number":2},{"id":"9763602eab25e54b6ec1c7b67c341d7db5e69ff5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-01 06:39:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4840247d79e037fbcb1ee0418960c20eb6fdc498","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 06:39:50.000000000","message":"Patch Set 2:\n\nUnable to freeze job graph: Job neutron-centos-8-tripleo-standalone depends on tripleo-ci-centos-8-content-provider which was not run.","accounts_in_message":[],"_revision_number":2},{"id":"8c73fea2760cc4aa3ed0cb5c5a355bc6199ab631","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-02 05:35:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3041612329eb550ad95b62d3c4a6f1503112adc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 07:24:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/6144264736aa4f62a6fa9f681443b35b : SUCCESS in 57m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffcf70a4f33748b0a5ad7a696fd56101 : SUCCESS in 37m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/103b4da6e9c04f65a36735bd09069a44 : SUCCESS in 10m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c827de9c320740df8dfba75f46c7dd9f : SUCCESS in 24m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c9601ed67144ef69960eee26daf4bae : SUCCESS in 23m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85c2d49978bd4715bb71909735f5cacf : SUCCESS in 22m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e7b68ac98048ff971a340763b7c3e5 : SUCCESS in 9m 56s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/c23825ec21704160bd26a5cd636e871a : SUCCESS in 48m 43s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ee5cb8c1a2334d45ac54ae540bb7bc8e : SUCCESS in 1h 29m 16s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/76adaf3c5e5542f1bc5c3637ac0365fa : SUCCESS in 1h 25m 48s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fc2663e2de2d4179bf621f10f54e2fce : SUCCESS in 1h 14m 38s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2964bf0186d4b199fb38b087d985225 : SUCCESS in 1h 18m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/6a35373f575f432bbec2b37b9a09c2cb : SUCCESS in 56m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d0470925e1d04fb1a0b698730f94c22a : SUCCESS in 1h 02m 45s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/919e32eb82d14d1596e263ab15fd9bf2 : SUCCESS in 57m 43s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/c78784af95904af2b9621492c2fdfd8c : SUCCESS in 1h 10m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b4a7daf35bfc4fcc978b5446aba8c8a3 : SUCCESS in 1h 27m 48s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a07943e83494433a952d7338a408be24 : SUCCESS in 1h 08m 33s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/df5f68028be24de991aa0479ed2697ab : SUCCESS in 1h 14m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3bd57a7686a04a47ae8462a50ffe13fa : FAILURE in 1h 06m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0838252729e24a54a5661954105e1a07 : SUCCESS in 1h 40m 27s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5910b22df5c4dcd84bf301e68710583 : SUCCESS in 51m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/74b89d9092bb472184ee1d659916ad6c : SUCCESS in 1h 20m 04s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c99798332aa840789065071ef3ae1c68 : SUCCESS in 32m 06s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/a74c62b7986246f7ae707f5f579864ee : FAILURE in 30m 17s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/900128069e2b4d5fb758fd967ae08264 : SUCCESS in 56m 22s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b3f4ceed4aef447da01097b62bc63fee : SUCCESS in 1h 25m 40s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/ba3e87f2bfa642e1865919f7b0e94ab0 : FAILURE in 1h 30m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"222728dbb775915d7ae85bbd4e2c166ec18c08cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 07:25:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"404342eb80571552aa41fe417b408021fb7380a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 09:06:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-ussuri https://zuul.opendev.org/t/openstack/build/24576080043a49b5bfa8e2796a95fb88 : SUCCESS in 57m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/781e2e84391c4cb8b8397e797e36a170 : SUCCESS in 14m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b70848cbbec04379982b5cd2164951c7 : SUCCESS in 25m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d93d7eb2050241c492943574749d05e1 : SUCCESS in 31m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78354834e59b43a6855ea37f6798fbdc : SUCCESS in 10m 31s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/348441b8936949c29d48dc9e5057079c : SUCCESS in 59m 24s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b6bd14e8b25f4336add0aadb00001e80 : SUCCESS in 1h 21m 32s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/09abd3b70c544d0ab6516212b7a5e66b : SUCCESS in 1h 16m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/89fa92e5e760462e81bc207e8d7f8297 : SUCCESS in 1h 13m 32s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/42021afaa3794ed5b9a98c33aa3b173f : SUCCESS in 1h 21m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/185f28ae3f2f4973bd571f6596e47859 : SUCCESS in 1h 13m 43s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/1fdfdff22e904f878ac5fb11f762422c : SUCCESS in 1h 15m 05s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ef271a83bcc6487ca990488bb1268976 : SUCCESS in 1h 35m 16s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dca38f2b59a422f801d0205b52c72ce : SUCCESS in 1h 03m 32s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/dc36ebfd9060497bbd57f6864bb63777 : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":2},{"id":"80547427aa763a94173500fd1bc0026a77d92bf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 09:06:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"be21be62cbb46532a0e6efaa81371699da6a6c68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 09:07:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ad112b75b7042d18eade65945e1cdb8 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"be933757faa3e26f689b0bb9a11216ab2dc8082d","revisions":{"48c34dcf4bfdedfeefc055220e22c46f29f7e5ad":{"kind":"REWORK","_number":1,"created":"2022-05-18 09:35:35.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/97/842297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/842297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/842297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/842297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/842297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/842297/1"}}},"commit":{"parents":[{"commit":"24c4bf0867d15ac7c1c15ac0f185397f09c381c5","subject":"Merge \"Fix setting table monitoring conditions\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24c4bf0867d15ac7c1c15ac0f185397f09c381c5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-12 09:26:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-13 05:41:35.000000000","tz":0},"subject":"[OVN] Allow VIP ports with a defined \"device_owner\"","message":"[OVN] Allow VIP ports with a defined \"device_owner\"\n\nNow OVN VIP port can have a \"device_owner\" defined.\n\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n\nCloses-Bug: #1973276\nChange-Id: Id1907481077d230e4461906a1a2a1447abac330a\n(cherry picked from commit 4c37497e7c98180fb6b1118021acd8e96b7028d0)\n(cherry picked from commit becc90e8ce0ee815fe85a24239eeea5afdacbddd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48c34dcf4bfdedfeefc055220e22c46f29f7e5ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48c34dcf4bfdedfeefc055220e22c46f29f7e5ad"}]},"branch":"refs/heads/stable/ussuri"},"be933757faa3e26f689b0bb9a11216ab2dc8082d":{"kind":"REWORK","_number":2,"created":"2022-05-18 16:15:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/97/842297/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/97/842297/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/842297/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/842297/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/97/842297/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/97/842297/2"}}},"commit":{"parents":[{"commit":"24c4bf0867d15ac7c1c15ac0f185397f09c381c5","subject":"Merge \"Fix setting table monitoring conditions\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/24c4bf0867d15ac7c1c15ac0f185397f09c381c5"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-12 09:26:36.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-13 12:54:44.000000000","tz":0},"subject":"[OVN] Allow VIP ports with a defined \"device_owner\"","message":"[OVN] Allow VIP ports with a defined \"device_owner\"\n\nNow OVN VIP port can have a \"device_owner\" defined.\n\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n    neutron/tests/unit/plugins/ml2/drivers/ovn/mech_driver/test_mech_driver.py\n\nCloses-Bug: #1973276\nChange-Id: Id1907481077d230e4461906a1a2a1447abac330a\n(cherry picked from commit 4c37497e7c98180fb6b1118021acd8e96b7028d0)\n(cherry picked from commit 9e2425111b6f7413098bbb38f6013d18b3804303)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be933757faa3e26f689b0bb9a11216ab2dc8082d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/be933757faa3e26f689b0bb9a11216ab2dc8082d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
