)]}'
{"id":"openstack%2Fneutron~843389","triplet_id":"openstack%2Fneutron~master~I41679cd7e39b0f7d64f99f509605ac9bc760ac5d","project":"openstack/neutron","branch":"master","topic":"bug/1966035","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-03 14:24:28.000000000","reason":"Rodolfo Alonso replied on the change"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-06-03 15:51:33.000000000","reason":"Change was submitted"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-06-03 15:51:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I41679cd7e39b0f7d64f99f509605ac9bc760ac5d","subject":"Add some logging to test_good_address_allocation","status":"MERGED","created":"2022-05-26 07:09:54.000000000","updated":"2022-06-03 15:53:24.000000000","submitted":"2022-06-03 15:51:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843389-bug/1966035","meta_rev_id":"b4f2a46835e2beb8ea8237e44d7a212322c128fe","_number":843389,"virtual_id_number":843389,"owner":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-05-30 07:25:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-03 15:51:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-06-03 14:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-30 09:36:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-05-31 12:15:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-05-31 14:47:52.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-06-03 14:24:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-26 07:57:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-26 08:11:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-26 11:39:07.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-05-30 09:36:04.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-05-31 12:15:51.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-05-31 14:47:52.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2022-06-03 13:49:29.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-06-03 13:49:29.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"6844033d49f00a83ff8af924218d21828ced8846","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-26 07:09:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b46af3fcff3a945db09151f03926adcaf2c3771e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 07:57:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06fa24a609f40e5a953dd2e4abd4ac0 : FAILURE in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c244879eacc4857a1492955d661cf24 : SUCCESS in 10m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ae96e36e70b4fe4806e9e0dacb62e91 : FAILURE in 44m 55s","accounts_in_message":[],"_revision_number":1},{"id":"115c20391d107046806681b3926a44884a7e9558","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-26 08:11:23.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/89/843389/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"39c2b4ecbe930724c3f012f93de5545ae0bfb44b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-26 08:45:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3b613821f165760bc2b5b536b49ea4ba0aa51e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 09:46:52.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09357e9b955d43189713e65702079563 : FAILURE in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42d9d7e4b9394d1395d45494f14122d7 : SUCCESS in 9m 13s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f188425242f3439baab74f1bbb497748 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":2},{"id":"304773bad06e46fbe03875535d571b794abdf8f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-26 09:49:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"690b1e4292d5fb16154b0b3884642dbe9ef482fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 10:47:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89f56cd22755410f9fa64e4c7d65a3da : SUCCESS in 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6573d6532a74a2c9ceb33c5fbc58ce8 : SUCCESS in 11m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b1bec06f23b4db68afec687c0f1c9cf : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":3},{"id":"b04141450cf572d268deaf5b7d91f4f0bd94e2dc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-26 10:59:23.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/89/843389/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"69eacbdd84b7279d2c3cbbf3f7703dbe16fb0713","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-26 11:19:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8272e6bcc2d3649a643aec218606bcd0caa7d539","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-26 11:39:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/843389/3 : SUCCESS in 1h 49m 26s","accounts_in_message":[],"_revision_number":3},{"id":"ae960361727db8b197113b5ca097144bc9241f93","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-26 11:54:54.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/89/843389/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"8987a0c0adbabcc9eda5fc22a41659157d0c293e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 12:08:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e66f2968d7ed40c7b110761c754c92aa : SUCCESS in 10m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3f30253bcd54f8d8eef64f27d38d924 : SUCCESS in 7m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9333f4abb7a9463e9a63e8f746b31f27 : SUCCESS in 46m 43s","accounts_in_message":[],"_revision_number":3},{"id":"f305b7bf7065331fb1f4621fd9f8ec09efa6e58f","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-26 12:31:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5978061b482fee2614fa6eaba7fc501315bc1fb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 13:27:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6c1139e37c04c7184fbb36b96575c59 : SUCCESS in 12m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c001ad4399fd44d48e8742bdd6abad3a : SUCCESS in 10m 40s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/232bbf622b974337bb1399a959a091a9 : SUCCESS in 52m 43s","accounts_in_message":[],"_revision_number":3},{"id":"282a1fe3e0f226df0bb00f74c2bc67cc2d2b4648","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-26 13:34:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4b63af2518e3a00521183a65a3d43569b7da1813","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-26 13:46:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4ec22569ba1c315f2def8573fd24112727331a60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 14:53:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac01adc3df8e4fa3af38ee0b8bd94d5d : SUCCESS in 13m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca62cd7aec3146c88da504dc5a6e36b6 : SUCCESS in 10m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c42875f1bf04f26b48b64d5bb634101 : FAILURE in 1h 05m 50s","accounts_in_message":[],"_revision_number":4},{"id":"164ba4a86fb97e5df3d266bd62e10ff666ce10f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-26 15:11:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"df1432eb2e3f4507cf6fd243b205069ba42e3edc","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-26 15:14:10.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/89/843389/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"af3de0ef8909ce23e81541498e4ff4e770797f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 16:08:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f81e4be3c84364acad293a70c1a3ea : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf47e77669c749b8af238a140414f805 : SUCCESS in 10m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/d8c48308c42346d49abb44442ba852dc : FAILURE in 54m 49s","accounts_in_message":[],"_revision_number":5},{"id":"0bff653a7640a26c9fc83e5c32ded4aa9ea57b90","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-26 17:31:06.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/843389/5 : SUCCESS in 2h 18m 32s","accounts_in_message":[],"_revision_number":5},{"id":"f0b122523b0dddfeed1356743858532a690f14b5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-26 18:48:10.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/89/843389/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"4a9c772d9e0632d73e920d75dfbd8649f9a91ed4","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-27 06:17:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"943ddbb3a5683fb928af279b752048b1d4e636c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 07:20:51.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bdaa19fc5c849cb899451e8e59945a5 : SUCCESS in 12m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feaaeb1742c44dbb880b459a3cc6ce55 : SUCCESS in 11m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/06c94ccb681b48f08b089a5d89c0644b : FAILURE in 58m 29s","accounts_in_message":[],"_revision_number":5},{"id":"7203eb48a329ab0c0b5d97ed4224c9f63c5f61db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-27 07:40:05.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5b43d11de86b2eb053fb8241f55ad659760c1b5e","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-27 08:27:05.000000000","message":"Patch Set 6: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/89/843389/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"fec30a5f8b80b73ff4a9c8b37f9b6d227482b7cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 08:36:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e165c3f077d7402d832b6abeb07be9ea : SUCCESS in 11m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fec2e901f044c088c9c75a595e93f3e : SUCCESS in 10m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5edfc2afedd445af899a5b6afe13fea7 : FAILURE in 52m 06s","accounts_in_message":[],"_revision_number":6},{"id":"de7101cc980f9740bb257feb03c96eeeec973ce0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-27 09:26:50.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/843389/6 : SUCCESS in 1h 45m 54s","accounts_in_message":[],"_revision_number":6},{"id":"7a16398b4825b7c5c2cec53a591bb36c7e1c3365","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-27 09:28:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c9e362c682300e35c33ea1466d24899c4fbcc084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 10:15:48.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60b1f67746da4318b24ccc812b7bbac1 : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f15b91228b0a40488ba55ba6884ae5b4 : SUCCESS in 9m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f2836275f414499283105dcc7317d773 : FAILURE in 45m 21s","accounts_in_message":[],"_revision_number":6},{"id":"a7122b31aa03bb913bf92f7cf5443272c3e9ec87","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-27 11:32:47.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/843389/6 : SUCCESS in 1h 53m 56s","accounts_in_message":[],"_revision_number":6},{"id":"f766643f940bca4ce2fcf155ac40c66104e38251","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-27 15:31:17.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"ce3ca006cddf1459a700d2847c7e5ef5058da6ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-05-30 06:52:41.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"df1586fb59e9724b572a6f7ccd440bc1240f7bd2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-30 07:25:36.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/89/843389/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"8c52c37824ceb235f226f47b7645c58b717c4cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:58:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a66e1c0c71e7458f9dc92d2e71bd8e67 : SUCCESS in 13m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4256ccd3b026474283e13ef1f28382f5 : SUCCESS in 9m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7964fbcb08dd446f8ba7550517ca80f5 : SUCCESS in 1h 00m 47s","accounts_in_message":[],"_revision_number":7},{"id":"e81e30ce66cd5d91a374f4ad2021f768012f9ee1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-30 08:49:31.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/843389/7 : SUCCESS in 1h 56m 14s","accounts_in_message":[],"_revision_number":7},{"id":"b2c5492426dc7852479d3f53edc91feb660b477a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-05-30 09:36:04.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"50e55425b6438db7bd24edc409cf76e8c6910e3b","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-05-31 12:15:51.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"909d0dbfd0cee9d550693c4b52452799485497bd","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-05-31 14:47:52.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"38c903cc8c72091c1846d82bbe76617d871578f6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-03 14:24:28.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"56c402e1738d06ccd57c7a78209230a3cb6ec225","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 14:25:04.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"8c7d2fd432ea765389b264c6f51c1e6d59ff9cde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 15:51:33.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d2622c162f4a4ba5bde38864292df7 : SUCCESS in 12m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2214aef353f42c888a346000271ef4d : SUCCESS in 13m 55s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/811e6f204191437694692b1c391d5f29 : SUCCESS in 1h 05m 28s","accounts_in_message":[],"_revision_number":7},{"id":"a7f6104f63ade02f6ed223a623052e86af595f18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 15:51:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"b4f2a46835e2beb8ea8237e44d7a212322c128fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 15:53:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/360f3fb60048453b8b6da8292d7ef397 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"763d8af1a3c17ab123f4f4d05f5c15a7ed169283","revisions":{"b1fc1fef0e3b0d47a92b80e977e72d1045797f59":{"kind":"REWORK","_number":1,"created":"2022-05-26 07:09:54.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/89/843389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/843389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/843389/1"}}},"commit":{"parents":[{"commit":"bdea1425840e1cc2399801409330056143c96de5","subject":"Merge \"Start using security-groups-shared-filtering from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdea1425840e1cc2399801409330056143c96de5"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 07:06:33.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 07:09:13.000000000","tz":240},"subject":"DNM: Debug test_good_address_allocation","message":"DNM: Debug test_good_address_allocation\n\nCannot reproduce locally, the bug is probably happening only\non loaded test environments. Trying to reproduce with some\nlogging added.\n\nRelated-Bug: #1966035\nChange-Id: I41679cd7e39b0f7d64f99f509605ac9bc760ac5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1fc1fef0e3b0d47a92b80e977e72d1045797f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b1fc1fef0e3b0d47a92b80e977e72d1045797f59"}]},"branch":"refs/heads/master"},"79e98906448709f3213f3f9c497e495103a6689b":{"kind":"REWORK","_number":2,"created":"2022-05-26 08:45:05.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/89/843389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/843389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/843389/2"}}},"commit":{"parents":[{"commit":"bdea1425840e1cc2399801409330056143c96de5","subject":"Merge \"Start using security-groups-shared-filtering from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdea1425840e1cc2399801409330056143c96de5"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 07:06:33.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 08:44:53.000000000","tz":240},"subject":"DNM: Debug test_good_address_allocation","message":"DNM: Debug test_good_address_allocation\n\nCannot reproduce locally, the bug is probably happening only\non loaded test environments. Trying to reproduce with some\nlogging added.\n\nRelated-Bug: #1966035\nChange-Id: I41679cd7e39b0f7d64f99f509605ac9bc760ac5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79e98906448709f3213f3f9c497e495103a6689b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/79e98906448709f3213f3f9c497e495103a6689b"}]},"branch":"refs/heads/master"},"305bf714812f88a142ebe1d338fa00cfe5d7f596":{"kind":"REWORK","_number":3,"created":"2022-05-26 09:49:04.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/89/843389/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/843389/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/843389/3"}}},"commit":{"parents":[{"commit":"bdea1425840e1cc2399801409330056143c96de5","subject":"Merge \"Start using security-groups-shared-filtering from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdea1425840e1cc2399801409330056143c96de5"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 07:06:33.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 09:48:52.000000000","tz":240},"subject":"DNM: Debug test_good_address_allocation","message":"DNM: Debug test_good_address_allocation\n\nCannot reproduce locally, the bug is probably happening only\non loaded test environments. Trying to reproduce with some\nlogging added.\n\nRelated-Bug: #1966035\nChange-Id: I41679cd7e39b0f7d64f99f509605ac9bc760ac5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/305bf714812f88a142ebe1d338fa00cfe5d7f596"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/305bf714812f88a142ebe1d338fa00cfe5d7f596"}]},"branch":"refs/heads/master"},"b9b1d4923e76cedfad98967903d3d444a6cb9d20":{"kind":"REWORK","_number":4,"created":"2022-05-26 13:46:24.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/89/843389/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/843389/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/843389/4"}}},"commit":{"parents":[{"commit":"bdea1425840e1cc2399801409330056143c96de5","subject":"Merge \"Start using security-groups-shared-filtering from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdea1425840e1cc2399801409330056143c96de5"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 07:06:33.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 13:41:28.000000000","tz":240},"subject":"Functional test: wait for dhclient in test_good_address_allocation","message":"Functional test: wait for dhclient in test_good_address_allocation\n\nForce test to wait for dhclient before checking IP address assigned\non port. Not sure it will completely fix the bug but it seems quite\nlogical.\nThis patch also adds some logging to make it easier to debug if still\nhappens.\n\nRelated-Bug: #1966035\nChange-Id: I41679cd7e39b0f7d64f99f509605ac9bc760ac5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9b1d4923e76cedfad98967903d3d444a6cb9d20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b9b1d4923e76cedfad98967903d3d444a6cb9d20"}]},"branch":"refs/heads/master"},"217dde0ca680fa73a716a7b201ce7b0931b26462":{"kind":"REWORK","_number":5,"created":"2022-05-26 15:11:54.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/89/843389/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/843389/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/843389/5"}}},"commit":{"parents":[{"commit":"bdea1425840e1cc2399801409330056143c96de5","subject":"Merge \"Start using security-groups-shared-filtering from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdea1425840e1cc2399801409330056143c96de5"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 07:06:33.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 15:11:38.000000000","tz":240},"subject":"Functional test: wait for dhclient in test_good_address_allocation","message":"Functional test: wait for dhclient in test_good_address_allocation\n\nForce test to wait for dhclient before checking IP address assigned\non port. Not sure it will completely fix the bug but it seems quite\nlogical.\nThis patch also adds some logging to make it easier to debug if still\nhappens.\n\nRelated-Bug: #1966035\nChange-Id: I41679cd7e39b0f7d64f99f509605ac9bc760ac5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/217dde0ca680fa73a716a7b201ce7b0931b26462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/217dde0ca680fa73a716a7b201ce7b0931b26462"}]},"branch":"refs/heads/master"},"8cd006066312322e49019583b34d978a89e8444a":{"kind":"REWORK","_number":6,"created":"2022-05-27 07:40:05.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/89/843389/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/843389/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/843389/6"}}},"commit":{"parents":[{"commit":"bdea1425840e1cc2399801409330056143c96de5","subject":"Merge \"Start using security-groups-shared-filtering from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdea1425840e1cc2399801409330056143c96de5"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 07:06:33.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-27 07:39:53.000000000","tz":240},"subject":"Functional test: wait for dhclient in test_good_address_allocation","message":"Functional test: wait for dhclient in test_good_address_allocation\n\nForce test to wait for dhclient before checking IP address assigned\non port. Not sure it will completely fix the bug but it seems quite\nlogical.\nThis patch also adds some logging to make it easier to debug if still\nhappens.\n\nRelated-Bug: #1966035\nChange-Id: I41679cd7e39b0f7d64f99f509605ac9bc760ac5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cd006066312322e49019583b34d978a89e8444a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8cd006066312322e49019583b34d978a89e8444a"}]},"branch":"refs/heads/master"},"763d8af1a3c17ab123f4f4d05f5c15a7ed169283":{"kind":"REWORK","_number":7,"created":"2022-05-30 06:52:41.000000000","uploader":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"ref":"refs/changes/89/843389/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/843389/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/843389/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/843389/7"}}},"commit":{"parents":[{"commit":"bdea1425840e1cc2399801409330056143c96de5","subject":"Merge \"Start using security-groups-shared-filtering from neutron-lib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bdea1425840e1cc2399801409330056143c96de5"}]}],"author":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-26 07:06:33.000000000","tz":240},"committer":{"name":"Oleg Bondarev","email":"obondarev@mirantis.com","date":"2022-05-30 06:51:31.000000000","tz":240},"subject":"Add some logging to test_good_address_allocation","message":"Add some logging to test_good_address_allocation\n\nLet\u0027s see how many times the test asks for IP addr list\nduring 10 sec timeout pediod. Probably sporadic failures\nare caused by waiting for GIL for too long.\n\nRelated-Bug: #1966035\nChange-Id: I41679cd7e39b0f7d64f99f509605ac9bc760ac5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/763d8af1a3c17ab123f4f4d05f5c15a7ed169283"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/763d8af1a3c17ab123f4f4d05f5c15a7ed169283"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
