)]}'
{"id":"openstack%2Fneutron~843601","triplet_id":"openstack%2Fneutron~master~I3a91f442a1fd72f9027f10f2b1b6572cee3f8360","project":"openstack/neutron","branch":"master","topic":"bug/1975743","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-02 11:09:10.000000000","reason":"Rodolfo Alonso replied on the change"},"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2022-06-10 18:55:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3a91f442a1fd72f9027f10f2b1b6572cee3f8360","subject":"[OVN] Make binding profile validation more robust","status":"MERGED","created":"2022-05-27 10:30:47.000000000","updated":"2022-06-10 18:55:00.000000000","submitted":"2022-06-10 18:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"843601-bug/1975743","meta_rev_id":"c22fa86a9fd90ecb7f02a29b32b36237205e39f9","_number":843601,"virtual_id_number":843601,"owner":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-06-02 11:11:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-10 18:55:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-06-02 11:09:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-10 14:44:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-10 14:44:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":34357,"name":"Itai Levy","email":"itailev@nvidia.com","username":"itailev"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-27 10:35:33.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"state":"CC"},{"updated":"2022-05-27 10:35:33.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":12171,"name":"Moshe Levi","email":"moshele@nvidia.com","username":"moshele"},"state":"CC"},{"updated":"2022-05-27 10:35:33.000000000","updated_by":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"reviewer":{"_account_id":34357,"name":"Itai Levy","email":"itailev@nvidia.com","username":"itailev"},"state":"CC"},{"updated":"2022-05-27 11:14:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-27 11:24:00.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-27 11:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-27 12:31:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-06-02 06:16:17.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-06-02 10:07:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"0dcc4a417e3cc51ba8e163e12b84e3913931d31e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-05-27 10:30:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1e2a84c2b090d06fb282ed537ce57e13ec9aeed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 11:14:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cb31081e75b47b68cea0e4aef9c5c70 : FAILURE in 38m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aaf8e7897fae4607a531bd448b6267b7 : FAILURE in 41m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa2e2e0f3ec7d450500749465c4ed54fe1668644","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-27 11:24:00.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/01/843601/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"60520ac41636d18f7a0d818a57c06ea81d4d97ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 11:57:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d0eb410a9ce04228aaa7c0ca4f20f50f : SUCCESS in 1h 07m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5c5edb3425c432d93d99d23b38fbd8d : SUCCESS in 39m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e11aa6dcb094b008c802369abe5fc73 : FAILURE in 28m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec19ad116554332b80ec17de82e229e : SUCCESS in 9m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2ba6bd2a3bd45f5b4c9e372e7c1dce8 : FAILURE in 23m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f7ea166fabe47f7997fe6c972733d47 : FAILURE in 23m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ced472e604144913a21da232af71d7bc : FAILURE in 22m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c7310e8ba084141bad1b4e8868006f5 : SUCCESS in 9m 03s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c12a65b2a9c041f5aced009ad57c4e2c : SUCCESS in 55m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/d034898523fb472ba3bd2978a191cb8a : SUCCESS in 1h 24m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84c93a2058ef419c9f1532bd0c6e1990 : SUCCESS in 56m 08s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/264b21dfb5004beeb68a77794e6f2b10 : SUCCESS in 1h 08m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37821dcb21c27563f6bd0bcbc64f0d2182055030","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-27 12:31:36.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/843601/1 : FAILURE in 2h 00m 09s","accounts_in_message":[],"_revision_number":1},{"id":"08f7320e39b3c0897dfb1f4e5d68339b07cd4e44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-05-27 12:39:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a8801bd6c8c5c89448d326caa38fb44ac8f1690f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 13:17:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f3245e1bf0548b5adacb003d8a01bd5 : SUCCESS in 33m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eac1f3c50ccf4cc884e4726a5daf1d7f : SUCCESS in 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47b66cc1218125c5e233546a460c3615c9863739","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-05-27 13:30:26.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/May-2022/Arista_Tempest_Testrefs/changes/01/843601/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"8fbac871256175f4fc4f1c38b84b33953a422d35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 14:12:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/31268816defb4761becc14fe7cb0d7a1 : SUCCESS in 1h 20m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5bfc7f48a8864e90b38d3f37dc8cda07 : SUCCESS in 29m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18f7673eccfe410784dd5f5e4a90e649 : SUCCESS in 32m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/516714d0231d4dcb841561f61d954064 : SUCCESS in 12m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89b8a6b43fa742fe99e274feec31e78f : SUCCESS in 28m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a6ecedba7054310977e744450b92414 : SUCCESS in 25m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/59626cec1e0849b79beeb904d8c08ca6 : SUCCESS in 21m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec9e47f390cd4d3ca8d31fc04bdac099 : SUCCESS in 9m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f0818942b92c45cca941efc43ec01c98 : SUCCESS in 55m 08s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/121b087b32fa4fb7858392cb47b3df9d : SUCCESS in 1h 28m 13s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91ac7f4db0864270bfd427dba13d1416 : SUCCESS in 57m 39s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a2c6e5bf491e499bab9beda380577892 : SUCCESS in 1h 14m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ea7265307cb6f22e89924117c0b7a0d29c7b0968","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-27 14:50:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/843601/2 : SUCCESS in 2h 10m 10s","accounts_in_message":[],"_revision_number":2},{"id":"10d4c68fbb06d2f237e5690181013df540c79ae0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-02 06:16:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"70f6b7bb087d96360b9e815746664c3f7d7f4281","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-02 10:07:23.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4d8a7f6aa395d0ab5c0cbc58b35c82e7a000069","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-06-02 10:16:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"de9871b0b0c78b97d3cf9f97f4f98ad45594a8b5","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-06-02 10:16:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"512a241abd6a52854690ec57455d54604ac536f6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-02 11:09:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0410cd40a5530edb1f76de266e7179fd21b5da12","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-02 11:09:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60f08abe993e3ea4377add467152d1ffbea85395","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-02 11:11:32.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/01/843601/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"1efa417117143586c5b60b95c2af818b481c8bc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 11:16:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c158721444724bf3bdd17878bac9a79e : SUCCESS in 51m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d021e8bc6590485a8b2c17a7925922fc : SUCCESS in 53m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5b3f2b31b4b97ec667b3b36626d2921e8197c468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 12:11:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fca1648c005b45d08592e3fe6d0300bc : SUCCESS in 1h 25m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fed1272d641543c6ad7bd9b842a2650e : SUCCESS in 38m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/923ba0d8f7a1432a9c5d77f6187493ef : SUCCESS in 38m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16c6a26ed77c42e1b5992dac11165050 : SUCCESS in 12m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b94b2dab11694b76bd60b3b30854d705 : SUCCESS in 27m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06a4706c0a254121b308d33e32c04a10 : SUCCESS in 25m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2ee1420633842a0bf72a4f7b630525a : SUCCESS in 28m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b56f9a8df6a454eadfc477b04aba808 : SUCCESS in 9m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bf0be2999cf64e80b8d0a150f9f60c71 : SUCCESS in 57m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/90d24aeefb444a66a13fa63e5e31feeb : SUCCESS in 1h 47m 37s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44aca74d1ba4477ea657e7f380dfc432 : SUCCESS in 1h 05m 24s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5a2daa5a3c0a45c49095db23e9ec83ce : SUCCESS in 1h 22m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ce88dda73f44aa59e90a276bb69f7570539e0094","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-02 12:12:07.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/843601/3 : SUCCESS in 1h 55m 05s","accounts_in_message":[],"_revision_number":3},{"id":"9f158f35d3bd15dd70cf200d32c32eb86020d44e","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-10 14:44:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"01180d4b121c1c296079f70089b785969f92ca32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 14:45:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"23cf70fed8b2c5280d22861d4e5f22dbad1fb718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 18:55:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/27c289d2822043f5beb6ab43caf42d8b : SUCCESS in 1h 03m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/942d09844c314eff97b14694a2159628 : SUCCESS in 12m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3b49abcae2648a99729efe5503a2307 : SUCCESS in 22m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7aed123f7bfe4c88af91a16304033848 : SUCCESS in 27m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f227a39df354e0885008448f44668a2 : SUCCESS in 9m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3843d82375c74a3491a6d601594eb8f0 : SUCCESS in 1h 01m 27s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/79a6238fe7cc45ffaa752a98f256bec0 : SUCCESS in 1h 26m 31s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9583e2709a04256a917674d19b255f1 : SUCCESS in 1h 07m 12s","accounts_in_message":[],"_revision_number":3},{"id":"c22fa86a9fd90ecb7f02a29b32b36237205e39f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 18:55:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8a9ffcb0d4c0c44fc226810fd7f08dab9b79fcc0","revisions":{"1dc617c6bde9ede7bccad2763e16e7f535480fa8":{"kind":"REWORK","_number":1,"created":"2022-05-27 10:30:47.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/01/843601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/843601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/843601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/843601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/843601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/843601/1"}}},"commit":{"parents":[{"commit":"2b77f12d2a698d4ca28825a42677b3d073d005eb","subject":"Merge \"Create an index for \"ports.network_id\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b77f12d2a698d4ca28825a42677b3d073d005eb"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-05-27 09:58:04.000000000","tz":120},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-05-27 10:27:31.000000000","tz":120},"subject":"[OVN] Make binding profile validation more robust","message":"[OVN] Make binding profile validation more robust\n\nThe main purpose of Neutron\u0027s validation of the binding profile\nis to make sure expected keys are present and that their values\nare of the expected type.\n\nThe Nova compute component updates the binding profile as part of\ninstance creation.  Depending on the version of the Nova compute\ncomponent and which hardware it interfaces with, the information\nprovided by Nova in the binding profile may differ.\n\nNova also has limited information at its disposal at the point in\ntime it updates the port binding profile, so it would be\nnon-trivial for it to provide information conditionally based on\nthings like VNIC_TYPE and existing binding profile data.\n\nMake the Neutron binding profile validation more robust for both\nupgrade and heterogeneous hardware scenarios by accepting the\npresence of surplus keys in the binding profile.  The data that\nNeutron expects will still be validated and any surplus keys will\nbe pruned before further processing internally in Neutron.\n\nCloses-Bug: #1975743\nChange-Id: I3a91f442a1fd72f9027f10f2b1b6572cee3f8360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1dc617c6bde9ede7bccad2763e16e7f535480fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1dc617c6bde9ede7bccad2763e16e7f535480fa8"}]},"branch":"refs/heads/master"},"173b02bb670769c30d166c467cf644263dc855da":{"kind":"REWORK","_number":2,"created":"2022-05-27 12:39:18.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/01/843601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/843601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/843601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/843601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/843601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/843601/2"}}},"commit":{"parents":[{"commit":"2b77f12d2a698d4ca28825a42677b3d073d005eb","subject":"Merge \"Create an index for \"ports.network_id\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b77f12d2a698d4ca28825a42677b3d073d005eb"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-05-27 09:58:04.000000000","tz":120},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-05-27 12:39:07.000000000","tz":120},"subject":"[OVN] Make binding profile validation more robust","message":"[OVN] Make binding profile validation more robust\n\nThe main purpose of Neutron\u0027s validation of the binding profile\nis to make sure expected keys are present and that their values\nare of the expected type.\n\nThe Nova compute component updates the binding profile as part of\ninstance creation.  Depending on the version of the Nova compute\ncomponent and which hardware it interfaces with, the information\nprovided by Nova in the binding profile may differ.\n\nNova also has limited information at its disposal at the point in\ntime it updates the port binding profile, so it would be\nnon-trivial for it to provide information conditionally based on\nthings like VNIC_TYPE and existing binding profile data.\n\nMake the Neutron binding profile validation more robust for both\nupgrade and heterogeneous hardware scenarios by accepting the\npresence of surplus keys in the binding profile.  The data that\nNeutron expects will still be validated and any surplus keys will\nbe pruned before further processing internally in Neutron.\n\nCloses-Bug: #1975743\nChange-Id: I3a91f442a1fd72f9027f10f2b1b6572cee3f8360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/173b02bb670769c30d166c467cf644263dc855da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/173b02bb670769c30d166c467cf644263dc855da"}]},"branch":"refs/heads/master"},"8a9ffcb0d4c0c44fc226810fd7f08dab9b79fcc0":{"kind":"REWORK","_number":3,"created":"2022-06-02 10:16:24.000000000","uploader":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"ref":"refs/changes/01/843601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/01/843601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/843601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/843601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/01/843601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/01/843601/3"}}},"commit":{"parents":[{"commit":"2b77f12d2a698d4ca28825a42677b3d073d005eb","subject":"Merge \"Create an index for \"ports.network_id\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/2b77f12d2a698d4ca28825a42677b3d073d005eb"}]}],"author":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-05-27 09:58:04.000000000","tz":120},"committer":{"name":"Frode Nordahl","email":"frode.nordahl@canonical.com","date":"2022-06-02 10:15:39.000000000","tz":120},"subject":"[OVN] Make binding profile validation more robust","message":"[OVN] Make binding profile validation more robust\n\nThe main purpose of Neutron\u0027s validation of the binding profile\nis to make sure expected keys are present and that their values\nare of the expected type.\n\nThe Nova compute component updates the binding profile as part of\ninstance creation.  Depending on the version of the Nova compute\ncomponent and which hardware it interfaces with, the information\nprovided by Nova in the binding profile may differ.\n\nNova also has limited information at its disposal at the point in\ntime it updates the port binding profile, so it would be\nnon-trivial for it to provide information conditionally based on\nthings like VNIC_TYPE and existing binding profile data.\n\nMake the Neutron binding profile validation more robust for both\nupgrade and heterogeneous hardware scenarios by accepting the\npresence of surplus keys in the binding profile.  The data that\nNeutron expects will still be validated and any surplus keys will\nbe pruned before further processing internally in Neutron.\n\nCloses-Bug: #1975743\nChange-Id: I3a91f442a1fd72f9027f10f2b1b6572cee3f8360\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a9ffcb0d4c0c44fc226810fd7f08dab9b79fcc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8a9ffcb0d4c0c44fc226810fd7f08dab9b79fcc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
