)]}'
{"id":"openstack%2Fneutron~844016","triplet_id":"openstack%2Fneutron~stable%2Fyoga~Ied5ef53882d4249e0ceaa731239befdc3ba67d03","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/1974183","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-02 18:43:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied5ef53882d4249e0ceaa731239befdc3ba67d03","subject":"[Trunk] Check if parent port belongs to OVN to activate","status":"MERGED","created":"2022-05-31 09:19:00.000000000","updated":"2022-06-02 18:46:02.000000000","submitted":"2022-06-02 18:43:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844016-bug/1974183","meta_rev_id":"db242a6793b6640778b2e944c22b36f29b3e4383","_number":844016,"virtual_id_number":844016,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 18:43:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-06-02 09:29:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-06-02 11:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-06-02 11:58:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-31 10:30:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-31 11:01:21.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-05-31 11:02:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 09:29:36.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-06-02 11:58:33.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"d63b6f6c83446c4262ffb0675b46a49dd9c668db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-05-31 09:19:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e7c14d680f3515840f7dea73f0d2befa5d203c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 10:30:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa90b73ab26e4a94830c7b995cff529d : SUCCESS in 1h 09m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df913dba1fdd4d37b19f10b48d5eb48e : SUCCESS in 1h 08m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3aea98695ab105b71ecfe9df7e94dcfaf665c48","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-05-31 11:01:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844016/1 : SUCCESS in 1h 41m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0bf39a885a223c0fa182d03a1725e5a0fe312f5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 11:02:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/06f5ab4cd1c6490491e343d2cbab0b90 : SUCCESS in 46m 28s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/3ad06ba5ef684367800db88e9d0dc8ca : SUCCESS in 48m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a5908f2ef3a4bc3ab99deaacf8ca622 : SUCCESS in 42m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3754d22881a41648e508ab9b5750240 : SUCCESS in 14m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6eb6abc04cf4a47b70dcae28b2f39ac : SUCCESS in 26m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3bb33d4ff1f499e87897c344e23d454 : SUCCESS in 29m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c00c1e83f25b4b9d9a6281e16c80dcce : SUCCESS in 11m 16s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a3a1d65469504f74b1dadc0e4fe56231 : SUCCESS in 50m 17s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a52c6f11abf421caa52be11d769cfee : SUCCESS in 1h 37m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12687dfd36664e55ae69b900453e2ea6 : SUCCESS in 1h 04m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1a0a2731c85c4311bf459592bcb28645 : SUCCESS in 1h 07m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0821c91d40e59effe43dd20a9f7cbea5fe8e5117","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-02 09:29:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03c6d0e433b1fa84a87dd8dd733d69e34bc976b3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-02 11:58:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"32be341231e8bbdc6803a5234e10a471f872072f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 11:59:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e0a65954d79ff55a8a2672088ebc065212628c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 13:55:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/2a64d263105e4d219d0d7563be7832b3 : SUCCESS in 41m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e88127213d4ffaa0095b268b4130a7 : SUCCESS in 11m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb3d81389ca14fcbb1fb29c8717c534d : SUCCESS in 22m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9cbc40ac5314f4a9d799ae6f83a35a7 : SUCCESS in 20m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e4174929de46fba451d5745964d36f : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/989a4d0651f34ad48a5083cd5ba68a3d : FAILURE in 40m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/30f02bb158a24e8c8e22342719550f04 : SUCCESS in 1h 20m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c126fc8f46df47858ef269eba2ff0390 : SUCCESS in 49m 24s","accounts_in_message":[],"_revision_number":1},{"id":"75cd2e45b8ba7b6da0ca95b0e65f3c33131b0bc9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-02 14:44:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef40cb22ffa3f18acd018915b68b9f64ec595dcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 15:46:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d731761f5dc14b03930a34736366d4db : SUCCESS in 53m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf465f389b664fdaa40c3e8a8ffbae60 : SUCCESS in 45m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b471eb6f03b1735c79583b955a0a59367eb9d5a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 16:21:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/10b1c226f45f442b838039809a20ad58 : SUCCESS in 52m 13s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/c2d0f2d48484402f979f0d0adf659cb4 : SUCCESS in 1h 12m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daeb4ee77bb4494787d572b68639539f : SUCCESS in 41m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbbf880d03a0472ebea7c37fd655b504 : SUCCESS in 13m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5d74f9b00f64452a6f3e1e9372e2b11 : SUCCESS in 26m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0927009c6bd4716a172b6d4b283e394 : SUCCESS in 27m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13e16109eae4a949dbf368a5ca7d55a : SUCCESS in 11m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b9c350f96864ccc9e24eed1382c50c8 : SUCCESS in 51m 31s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ec2f9b4d984441ea3fb03d027f96896 : SUCCESS in 1h 33m 50s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68defb2841644c5485e017e5e87677bb : SUCCESS in 55m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d729f9a8e9c74b1cbf82ec7d5c488698 : SUCCESS in 1h 11m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d877936a1a79eda06e4a7c67e633df6f06750dc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 16:23:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3016bef637c5d4e1f567bacc6de73824555a859","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-02 16:25:07.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/844016/1 : FAILURE in 1h 40m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8662d865d1e7b470f24358101e9f62c19973d427","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:43:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/ca12276068634f70ad99cb2badceac21 : SUCCESS in 1h 01m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7eb723fe6d436ba0ff383f4148c774 : SUCCESS in 14m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de4acb9912a3465eb52d566849c0687b : SUCCESS in 28m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77e15780bee64b5aa400f4aeb42008ba : SUCCESS in 22m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9b645c9e4bc45e082659c4263fe9e0b : SUCCESS in 12m 53s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/303ec214516b4b978f4fcb7e11db197b : SUCCESS in 54m 14s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7b6e01b3cd24418a8c788c13358e6fa5 : SUCCESS in 1h 42m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd564e8a6cf3475984ca1073d9259fb8 : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b6bd354dfa507d662b36179ce83efcfcb420645f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:43:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"db242a6793b6640778b2e944c22b36f29b3e4383","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:46:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/313075ca4ea340b1822e628742184c43 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9326c7732592987299fb34f62f936d4326bb4a3","revisions":{"d9326c7732592987299fb34f62f936d4326bb4a3":{"kind":"REWORK","_number":1,"created":"2022-05-31 09:19:00.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/16/844016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/16/844016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/844016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/844016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/16/844016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/16/844016/1"}}},"commit":{"parents":[{"commit":"c92831b84ece5f272112866beb8f971551567aae","subject":"Merge \"Revert \"Use Port_Binding up column to set Neutron port status\"\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c92831b84ece5f272112866beb8f971551567aae"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-14 09:41:25.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-16 02:24:33.000000000","tz":0},"subject":"[Trunk] Check if parent port belongs to OVN to activate","message":"[Trunk] Check if parent port belongs to OVN to activate\n\nIn Trunk service, the OVN was setting always the trunk to \"ACTIVE\"\nanytime the trunk was created or updated. This patch changes this\nbehaviour: only when the trunk parent port belongs to OVN (that\nmeans, there is a related logical switch port to this Neutron port),\nthe methods will set the trunk to \"ACTIVE\".\n\nCloses-Bug: #1974183\nChange-Id: Ied5ef53882d4249e0ceaa731239befdc3ba67d03\n(cherry picked from commit 68bbd3586bdc313a6b67f2983ba2331a7956856c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9326c7732592987299fb34f62f936d4326bb4a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d9326c7732592987299fb34f62f936d4326bb4a3"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
