)]}'
{"id":"openstack%2Fneutron~844172","triplet_id":"openstack%2Fneutron~master~I3e900d4b028be58af8e716b53704eee6ce0828b3","project":"openstack/neutron","branch":"master","topic":"bug/1976345","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2023-05-11 05:35:19.000000000","reason":"Change was abandoned"},"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2022-06-09 03:14:26.000000000","reason":"LIU Yulong replied on the change"}},"hashtags":[],"change_id":"I3e900d4b028be58af8e716b53704eee6ce0828b3","subject":"Save one time-consuming _routers_to_sync","status":"ABANDONED","created":"2022-06-01 01:09:50.000000000","updated":"2023-05-11 05:35:19.000000000","total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"363655a0ed2425feed3a6a5a5415e63f6229d645","_number":844172,"virtual_id_number":844172,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-06-01 04:54:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-01 04:49:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-06-01 20:15:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 01:47:38.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2022-06-01 01:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-01 02:21:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-06-01 02:22:18.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-01 02:55:35.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-01 04:49:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-01 04:54:40.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eefbb335ee550959c34dfc81e800854f53bc9790","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-06-01 01:09:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a4e577e6c252565ece7f92246dcb71893844f30","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-06-01 01:47:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b32f4e29e5bf62eb56860d41c6f9b04a4460e5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 01:52:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04ea3a6194644ec4a7b6acc98b099b42 : FAILURE in 31m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8bc2382aa86d457e97f45d5ebb6254a1 : FAILURE in 40m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38955d994d611b6a7cc3d96526e39a99546551ae","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-01 02:21:36.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/844172/1 : FAILURE in 1h 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7b3182c15a658a7595f6bd793670efd5765a9a14","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-06-01 02:22:18.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/72/844172/1/openstack-sriov-neutron-master-direct : ABORTED\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/72/844172/1/openstack-sriov-neutron-master-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b55c9a1b560591be360315b1b6c84c9c8bef9ec5","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-06-01 02:39:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a27c2ed8e50f1cde8f8608f6fb623942c0000f05","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-01 02:55:35.000000000","message":"Patch Set 1:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/72/844172/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"f89e2fe3785e20db62c6a4ed4e739b5a546d946d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-06-01 03:10:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"59b2c04f8469fda5e2e28c10262625869ff0b9ec","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-06-01 03:11:32.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/72/844172/1/openstack-sriov-neutron-master-direct : ABORTED\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/72/844172/1/openstack-sriov-neutron-master-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"657986a50720e1da8dc1136d795da547041067af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 04:26:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/398dced31dd245e5bdf9a5bc2d8ba324 : SUCCESS in 49m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/baf5b0b120f2417a8a5886d578f49a91 : SUCCESS in 41m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6dd6072482c80721f7d2258ae869953eec4cb063","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-06-01 04:28:25.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/72/844172/2/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/72/844172/2/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"ff80a0fc813588c2db5bc0537a81d29a12efbceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 04:49:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/3124f79c88bc4cda8f0e633478fbaa8e : SUCCESS in 1h 36m 27s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/456dd91014ab4409be3f66a6e042a421 : SUCCESS in 1h 09m 50s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/17e5c1ce7ac04a16b8eb9a73ecc514ce : SUCCESS in 1h 05m 01s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/920961553ce84a09b0f7769ac5296ecb : SUCCESS in 1h 03m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ecee631014c442ae83f673263ad3ae94 : SUCCESS in 30m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/381abf6ac3da43a6b748fd194866c76c : SUCCESS in 38m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc1acae38d474858b45402efc0f6ae5d : SUCCESS in 13m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c775d476eb34106b6572453a614470a : SUCCESS in 25m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e150ab65bb684fe080c42948b376a9c1 : SUCCESS in 27m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79e5ff1b1fc44a0eace9245b2691bd6d : SUCCESS in 29m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30cd42c56ed541b38ee554fc4caa76d6 : SUCCESS in 10m 26s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1e4ae2d87584dcf9af0d9baf50feaa0 : SUCCESS in 57m 43s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/92a3a37c2f6945ab99c243689462c017 : SUCCESS in 1h 28m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/20eab5d26b3e4614860164ad293b534d : SUCCESS in 1h 16m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1af341012d6349e9bc58c023adc4636d : SUCCESS in 1h 01m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e16a6f3d1dbb42efbd108acc2a680d3b : SUCCESS in 1h 01m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/dc3fef116253448e8c3be0e4f2519dc4 : SUCCESS in 1h 20m 24s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/af99ab35a6304da19e8da57b35d4f9a6 : SUCCESS in 1h 17m 16s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b19daa06759242159df2f82562616d03 : SUCCESS in 57m 42s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/dd18e011918d441397b175f74a0534f1 : SUCCESS in 1h 09m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e71833c6a8fda309d581056d199bc541d775dc99","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-01 04:54:40.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/72/844172/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"0503dda063abb59bb262b4b415d8b08f1628d969","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-01 05:31:46.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844172/2 : SUCCESS in 2h 20m 35s","accounts_in_message":[],"_revision_number":2},{"id":"7951448f48c9a6854b69c088a6f17eeff04c4a94","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-06-01 20:15:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"05a80de43bdb757bb3fe0ec1977211602abb4f49","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-06-02 03:00:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a3de054e44434e637e7a2279929700b5872b7477","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-06-02 14:51:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"762e76e90a958a1db4d82c6402434e95d1490e73","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-06-09 03:14:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"363655a0ed2425feed3a6a5a5415e63f6229d645","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2023-05-11 05:35:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b477f7bfbdbb464a419ecc9797d2c8938b67a00","revisions":{"af153be374e69b328e6a660802bb42925fd1540e":{"kind":"REWORK","_number":1,"created":"2022-06-01 01:09:50.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/72/844172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/844172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/844172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/844172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/844172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/844172/1"}}},"commit":{"parents":[{"commit":"f966ed8c1507706c01a3ac4f6f38315670ed2ba0","subject":"Merge \"[Trunk] Check if parent port belongs to OVN to activate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f966ed8c1507706c01a3ac4f6f38315670ed2ba0"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-06-01 01:07:10.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-06-01 01:09:19.000000000","tz":480},"subject":"[WIP] save one time-consuming _routers_to_sync","message":"[WIP] save one time-consuming _routers_to_sync\n\nCloses-Bug: #1976345\nChange-Id: I3e900d4b028be58af8e716b53704eee6ce0828b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af153be374e69b328e6a660802bb42925fd1540e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/af153be374e69b328e6a660802bb42925fd1540e"}]},"branch":"refs/heads/master"},"5b477f7bfbdbb464a419ecc9797d2c8938b67a00":{"kind":"REWORK","_number":2,"created":"2022-06-01 03:10:34.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/72/844172/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/72/844172/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/844172/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/844172/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/72/844172/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/72/844172/2"}}},"commit":{"parents":[{"commit":"f966ed8c1507706c01a3ac4f6f38315670ed2ba0","subject":"Merge \"[Trunk] Check if parent port belongs to OVN to activate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f966ed8c1507706c01a3ac4f6f38315670ed2ba0"}]}],"author":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-06-01 01:07:10.000000000","tz":480},"committer":{"name":"LIU Yulong","email":"i@liuyulong.me","date":"2022-06-01 03:05:15.000000000","tz":480},"subject":"Save one time-consuming _routers_to_sync","message":"Save one time-consuming _routers_to_sync\n\nUpdate the binding host information to L3 router ports, then\nwe can save one time-consuming _routers_to_sync. And port\ninformation will be updated in _ensure_host_set_on_port\nafter update_port.\n\nAnd L3-agents do not use the port binding informations,\nit should be fine to return L3-agent.\n\nCloses-Bug: #1976345\nChange-Id: I3e900d4b028be58af8e716b53704eee6ce0828b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b477f7bfbdbb464a419ecc9797d2c8938b67a00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5b477f7bfbdbb464a419ecc9797d2c8938b67a00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
