)]}'
{"id":"openstack%2Fneutron~844179","triplet_id":"openstack%2Fneutron~master~Idf522a20f9735829ee568020bfed46345a95e294","project":"openstack/neutron","branch":"master","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-09 11:01:47.000000000","reason":"Rodolfo Alonso replied on the change"},"30380":{"account":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"last_update":"2022-06-10 16:13:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idf522a20f9735829ee568020bfed46345a95e294","subject":"[ovn]neutron agent show real heartbeat_timestamp","status":"MERGED","created":"2022-06-01 03:09:08.000000000","updated":"2022-06-10 16:14:54.000000000","submitted":"2022-06-10 16:13:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844179","meta_rev_id":"e1d19981b5929ead9030d0f54c60ae1a5c601bd4","_number":844179,"virtual_id_number":844179,"owner":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-06-09 11:58:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-10 16:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-06-09 11:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-10 14:36:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-10 14:36:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 04:04:48.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-01 04:13:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-01 04:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-01 04:36:42.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-01 05:18:53.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-06-07 11:33:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-06-09 09:28:31.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-06-10 14:36:30.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"8fa03adfea051f21281184e449a228a20b6f0f7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-01 03:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7165beca31b3f10d1e543815d721b56009515c33","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-01 04:04:48.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/79/844179/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"6f44cf4f9524331767eed0b72abfa64fa0f4dd2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 04:13:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6883ee18e5134766844ecdedc55f5ea5 : SUCCESS in 45m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ffafe9f613d4c399a37c6d6ff9b9907 : SUCCESS in 50m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6be11ed645561ce12bac3412c29e7a9af962b264","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 04:32:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/874b03cefeaa41bba50d37332d44fa93 : SUCCESS in 1h 04m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13779aa5338f457dac7496dd17faaeca : SUCCESS in 37m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b41953fe39ed4fe8badea1b3ec3abf48 : SUCCESS in 13m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cc9be4e0492480f9dc6c1e76faf95d4 : SUCCESS in 32m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e329891587f941c7ac8ac1203531ec35 : SUCCESS in 33m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f93fff07412c47209d3b26c26cb7bf07 : SUCCESS in 27m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08ad50954574f6da30001256feb1b96 : SUCCESS in 10m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/79d2ee378efe4d549d8a6d68e6159c60 : SUCCESS in 58m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd8c292c960b43ad820e30a505515e4a : SUCCESS in 57m 43s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/607ad7b1643440f1b6df99a567329117 : SUCCESS in 1h 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ff506ee2a246bfe308edd3848b6876daba1ce7c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-06-01 04:36:42.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* openstack-sriov-neutron-master-direct http://13.74.249.42/refs/changes/79/844179/1/openstack-sriov-neutron-master-direct : SUCCESS\n\n* openstack-sriov-neutron-master-macvtap http://13.74.249.42/refs/changes/79/844179/1/openstack-sriov-neutron-master-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e03c883379f9b430096078093e0ecdea5a95b091","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-01 05:18:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844179/1 : SUCCESS in 2h 09m 09s","accounts_in_message":[],"_revision_number":1},{"id":"e848f4c39429e3ca4502bd51c7cbc6bf5a0ba524","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-01 05:40:20.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"e6f200118e79df6462195387200256399b84c14f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-04 02:42:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a1ee7a9eb7006d749e10de7de1c57f428c67e283","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-04 02:43:27.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"da25898e02799be81cfce84cc9e269767143557d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 04:24:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab318e8a7caf43888376cfc881b0c07d : SUCCESS in 54m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e3673a7bd504315bf7dc0fb51a6010f : SUCCESS in 49m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dc15ee8b6813563e4d5685bdad93c194f8304728","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-04 04:24:23.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/79/844179/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"b0ed0ba3e39879411f63e0d895b23af55c0a5896","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-04 04:28:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/11323c26c08749cfafad935afdfc54c8 : SUCCESS in 1h 36m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/886e934738fd431694f34d4f6cf9eae5 : SUCCESS in 30m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83633bcf75574e4e9c11415b085641c2 : SUCCESS in 11m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7e4f642e2854ba1b1c9e9ed23e63ec0 : SUCCESS in 26m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6af2cedd730d4acda2b594e8bdb2e203 : SUCCESS in 20m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d3281d9b7214505ad8f26869b530284 : SUCCESS in 21m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d7de937aeb643339c73379f439cf993 : SUCCESS in 9m 28s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ccc7aae576354f3984ed7da8ebc9d8c4 : SUCCESS in 51m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b9064ebb98645de82bb9205edb1357b : SUCCESS in 58m 53s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8747af3cb76f4ca58abad7973264999d : SUCCESS in 1h 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a350b4271904239da2600cce0edaae80024e6f5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-04 04:50:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844179/2 : SUCCESS in 2h 07m 15s","accounts_in_message":[],"_revision_number":2},{"id":"53df3f577c9bc615e28050d126c2e978d409753b","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-07 11:33:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7892f83bca6e1ef37b67ccdba0099b5e01b1cec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-08 06:25:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a869977c9609e1889606c5f3eea5f0a0c5e342d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 07:08:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/85b9b8eef8634147b2d026ce2de8505d : SUCCESS in 40m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35c238ef90b243e6ae532f3664957889 : SUCCESS in 41m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"be1ec0fce71253280d96d8c13be6bff584f877f2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-08 07:18:45.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/79/844179/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"56c386dd84591d771faaabd2debe9bbe5f5b65fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 07:58:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a72681d0be26435aab6873dd5db8c0ff : SUCCESS in 1h 09m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b84ad6fff844dbc93bfdccda8c8b31d : SUCCESS in 37m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c475e9333654ff6861358716b554791 : SUCCESS in 13m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/147951a6c7d04f16bd6d6121186330b6 : SUCCESS in 25m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f0f6d0c5886440f98a42f687072fe17 : SUCCESS in 25m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/065242967bf04305808d79307aebca19 : SUCCESS in 28m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f1c437e72a04cdba867cc1d97cb41a6 : SUCCESS in 9m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2bf25acc88b44592ad98b6188ac84cdf : FAILURE in 54m 56s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8c09e8965bf4979b82c9826c729bd2a : SUCCESS in 1h 00m 50s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/43cdaea9cda54abca39dec1963253d6e : SUCCESS in 1h 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c40c60fe48ead98d361ac6de6183237c46388ab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-08 08:04:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8a5fe00be92c4ed18a98e2a75d0a842aaa1d30a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 08:47:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/665a14c52e97427099bc532e4fc5ee57 : SUCCESS in 38m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cefe1b25f1c74cfc83344c18e6d4cad8 : SUCCESS in 39m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"65b24cf6e836af3c00bca8cf595b49c3f1c46608","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-08 08:54:38.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/79/844179/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"590d8c243b43b4b634271ad207eaeaf8693b1c05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 09:18:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d8f7fba4c497407ab9d47590f9a55c27 : SUCCESS in 56m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60cc32431f4c4349968b771d11c3b118 : SUCCESS in 30m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc05b3ddbc2496ea958dcd439ebab0c : SUCCESS in 11m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac3b7addd77847c2932cec72e3e91ac4 : SUCCESS in 24m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15d87465aea545c487f8dab0a0c72e12 : SUCCESS in 21m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a569bef36bf84ae1a214e6f9446c9b17 : SUCCESS in 27m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/592630b89aa349da975aa33fdef36030 : SUCCESS in 9m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1d24740f46254cc7b48958ffd76b653e : SUCCESS in 48m 49s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/578867ca9a9c478ba9e75ff78aea5cd1 : SUCCESS in 54m 34s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/a96ee465ed6f4351af6af58d12d4e673 : SUCCESS in 1h 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f40f045dbac667b30085d26d8d1fa928a7026421","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-08 09:32:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6f8434648f6aa9b3499d2e2890128c2e31b29440","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 10:14:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f334d5c25bda4be4a1654754927968b6 : SUCCESS in 32m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08c4099e83a143d89682819745c9e340 : SUCCESS in 39m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"58d1846fafe48b3c021920585015ff4e6f8b61ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 11:02:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/65e0be50f24c41d1aee2dc9332bca665 : SUCCESS in 1h 04m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac22fd39e08f4c779c2bec068e0471b9 : SUCCESS in 38m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3fe11eec43408c909701037c84e35d : SUCCESS in 13m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da56a9222cec4dcd8855c2e274474eb2 : SUCCESS in 26m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fb5407c6d6d4c81886c016d0a526766 : SUCCESS in 27m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2f975abaa04437d897d39b7533e06fe : SUCCESS in 26m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03c0857667843b1959afade47f8ca39 : SUCCESS in 9m 45s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d38a07a662146e7a689966ab5d41960 : FAILURE in 54m 32s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/631cc87f33064ff593488ee49eea01b2 : SUCCESS in 58m 11s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/816167c1da004dc09de7eccb82f2242b : SUCCESS in 1h 22m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"01aeb8479a50a121a50dc0f83efd4442c0331397","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-08 11:08:12.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"e7ded636669317ab28b76ba349408459794076e1","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-08 11:20:35.000000000","message":"Patch Set 5: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/79/844179/5 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":5},{"id":"65c47adea187c8c8e7ffae1259926a435647dce1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 11:50:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9e6405f23fcf43d487dcda2ec27fdb44 : SUCCESS in 39m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8f1b317f6e3434b83c3878db722bf65 : SUCCESS in 40m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"947d210ff6c62b1b2b9b73a8a1c8407ce7177ff4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 12:31:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eeb69e7555e24552a316317230d759ee : SUCCESS in 1h 20m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df758099bd2f4d3fa3bae9f9106e8b72 : SUCCESS in 40m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bae683f51d454afab381fcc5b02b7311 : SUCCESS in 13m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dd63d86449e4a89964bea44d7bc11ad : SUCCESS in 24m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3732ef2fa0243769bd1333de4e5e047 : SUCCESS in 28m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87a7341ce8644d4d8e085430ad7d7ecf : SUCCESS in 28m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b65a50eb358480ca128cbf8f499dd8c : SUCCESS in 10m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/902387a827ca435b92e1abbb94df6893 : FAILURE in 49m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d0166f772ee4e53a836a45e2d620053 : SUCCESS in 54m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ff46314908424d4491e3ba03d0fcd37b : SUCCESS in 1h 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d3f3b4989a2b9e1bc156914a33928e12dcc2c174","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-08 13:00:58.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/844179/6 : FAILURE in 1h 52m 05s","accounts_in_message":[],"_revision_number":6},{"id":"93b56a1777793c4bb9602b4e5c65f2262867887d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-08 13:48:19.000000000","message":"Patch Set 6:\n\nArista third party testing FAILED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/79/844179/6 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":6},{"id":"28c1a5c68c7e5a4dea355a40d741178130b0dbab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-08 23:51:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4e354e2ab0523c452c9c626fa4470d9a30cae52a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 00:30:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cbbf1e81ea0e4008906edeedf25e2ea7 : SUCCESS in 36m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/891aa8fab12e4f8b8182ccdf4b45cc09 : SUCCESS in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"15f4cbc1acaa67b6ab32d9081ccfdcbb80dce212","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-09 00:50:20.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/844179/7 : FAILURE in 58m 17s","accounts_in_message":[],"_revision_number":7},{"id":"a1d54f9645aab9f55f75d75e2c3bce8855cc286f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 01:04:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d2935fc610ed4b4a89d9b3aad00741b4 : SUCCESS in 1h 10m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e31e46fb1b2840ac81de3e007a58f8ee : SUCCESS in 35m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1f9b7eac7a44a53a22532e74430ab69 : SUCCESS in 12m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f180b6033c334fb099d8652a628cc4b0 : SUCCESS in 20m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3066055ec798411ca8ee0facfa5f6eff : SUCCESS in 25m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1ba7a2eb3d2d4176830ae03a695b78a6 : SUCCESS in 28m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e6db2e173064fc7ad40229e390036fa : SUCCESS in 7m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c0d3dbbdff0b466d846b7a64b7eaaa82 : SUCCESS in 51m 05s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1edcbfd256d04349b0e0d9371a4b4570 : SUCCESS in 44m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d22c15ed00d440f8988faf6bccabd044 : SUCCESS in 1h 00m 18s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a0de478445e5d948e2f7f714907503b2d2f27238","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-09 01:20:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"441a62c5f690280645d6c68b9a0a396dc7fe9c45","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-09 01:27:21.000000000","message":"Patch Set 7: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/79/844179/7 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":7},{"id":"78dcd0a1be99169cdfdaf79830c4f41d162afad9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-09 09:28:31.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"ed0c85478a746a52d5483fce5ef3dc8a5987a3ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-09 10:35:18.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1769832306d7c89f5138204998fa802a80533b4b","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-06-09 10:36:58.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4796ceb673e872abf36958efb9353982f9f5fa53","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-09 11:01:47.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"dd52218fbbac91177536fe10e4c4083afb634ee8","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-09 11:07:36.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/844179/8 : FAILURE in 31m 26s","accounts_in_message":[],"_revision_number":8},{"id":"80a6e77cdf83eab7989bea704e69007ac3200417","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 11:38:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/374c54b1cc70421eb8746a4562163942 : SUCCESS in 40m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab66b0096e6248e4b1396fb90f12b082 : SUCCESS in 40m 57s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0c88c5a42eb9454ac1c712d2d8f62dc409eedd8d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-09 11:58:10.000000000","message":"Patch Set 8: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/79/844179/8 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":8},{"id":"1d9b073a4b869d43b9ad40d37dc3d54a2434269d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 12:41:40.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/9f50caaeac9443eab12dc88ae7d43756 : SUCCESS in 2h 01m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d37308b50e34d889776c975c22c41f4 : SUCCESS in 53m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c60742346e97415482f91132653a8db7 : SUCCESS in 22m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a55ce7ab6caf4ef4854238e660db3b4f : SUCCESS in 42m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c488e77853d848988b481643cae258cd : SUCCESS in 45m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eeb4bcb45cb04857864ab4f5875b4e6a : SUCCESS in 40m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d12af826def433a936a804a961ff401 : SUCCESS in 10m 22s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b755894618814ca0b00e727d4cbbfc7c : SUCCESS in 1h 17m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adcd5bcc87da46d091507af414c7d9da : SUCCESS in 1h 02m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ea2a4ac4ecba4d779c20ab46e1b92a86 : SUCCESS in 1h 20m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cbeb44e10e1af0cf811d1c4993de8536dc06b6c1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-10 14:36:30.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"b8154cb33e5f8f2c1c1710136c632e7cb751097e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 14:37:09.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"cf85470a61bb3126fdb0e7c15791bf1abf6bc0a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 16:13:02.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/64084d2637474eb8a680566565e31d17 : SUCCESS in 1h 03m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c9e7e33cb040c6828ba130d28ba6ba : SUCCESS in 13m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8a81ce247c84018b3d465de6dcf9d5e : SUCCESS in 33m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35bfd4687f604234a8cc9371e5271fef : SUCCESS in 26m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df7fb43c79f94b59b0959c0ee7c738c4 : SUCCESS in 10m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6ba7418970214fb9bdc488efaa5a27dc : SUCCESS in 1h 01m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b4618237a294359a9887e4828b8b9ac : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":8},{"id":"8b00cc43a3369179bf8740425cfcb97e6567b25d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 16:13:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"e1d19981b5929ead9030d0f54c60ae1a5c601bd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 16:14:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a38371b52c4eb6951ee3423c44ad39 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"411ecc4865198e2bed5e3d0ab498e3a42408c17d","revisions":{"c899c8d9d74e9a80814096a3650e22983e8e20cf":{"kind":"REWORK","_number":1,"created":"2022-06-01 03:09:08.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/79/844179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/844179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/844179/1"}}},"commit":{"parents":[{"commit":"f966ed8c1507706c01a3ac4f6f38315670ed2ba0","subject":"Merge \"[Trunk] Check if parent port belongs to OVN to activate\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f966ed8c1507706c01a3ac4f6f38315670ed2ba0"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-01 03:08:25.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-01 03:08:25.000000000","tz":480},"subject":"[ovn]neutron agent show real heartbeat_timestamp","message":"[ovn]neutron agent show real heartbeat_timestamp\n\nChange-Id: Idf522a20f9735829ee568020bfed46345a95e294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c899c8d9d74e9a80814096a3650e22983e8e20cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c899c8d9d74e9a80814096a3650e22983e8e20cf"}]},"branch":"refs/heads/master"},"cebe30f57cab363ad8ce80bf4c43954684621828":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2022-06-04 02:42:59.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/79/844179/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/844179/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/844179/2"}}},"commit":{"parents":[{"commit":"d253b9fd08b969ebab09a5624df127a4bf1bfb1f","subject":"Merge \"Add some logging to test_good_address_allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d253b9fd08b969ebab09a5624df127a4bf1bfb1f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-01 03:08:25.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-04 02:42:53.000000000","tz":480},"subject":"[ovn]neutron agent show real heartbeat_timestamp","message":"[ovn]neutron agent show real heartbeat_timestamp\n\nagent\u0027s heartbeat_timestamp returns the current time, not\nconducive to agent status troubleshooting.\nthis patch use agent\u0027s updated_at as heartbeat_timestamp.\n\nCloses-bug: #1977629\n\nChange-Id: Idf522a20f9735829ee568020bfed46345a95e294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cebe30f57cab363ad8ce80bf4c43954684621828"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cebe30f57cab363ad8ce80bf4c43954684621828"}]},"branch":"refs/heads/master"},"7c42fc82a7003da98c730e5813a610d6e0b43b1d":{"kind":"REWORK","_number":3,"created":"2022-06-08 06:25:41.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/79/844179/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/844179/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/844179/3"}}},"commit":{"parents":[{"commit":"d253b9fd08b969ebab09a5624df127a4bf1bfb1f","subject":"Merge \"Add some logging to test_good_address_allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d253b9fd08b969ebab09a5624df127a4bf1bfb1f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-01 03:08:25.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-08 06:25:09.000000000","tz":480},"subject":"[ovn]neutron agent show real heartbeat_timestamp","message":"[ovn]neutron agent show real heartbeat_timestamp\n\nagent\u0027s heartbeat_timestamp returns the current time, not\nconducive to agent status troubleshooting.\nthis patch use agent\u0027s updated_at as heartbeat_timestamp.\n\nCloses-bug: #1977629\n\nChange-Id: Idf522a20f9735829ee568020bfed46345a95e294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c42fc82a7003da98c730e5813a610d6e0b43b1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7c42fc82a7003da98c730e5813a610d6e0b43b1d"}]},"branch":"refs/heads/master"},"5d3b338fbfba025f90ae67dcc7e127d2c45ab36e":{"kind":"REWORK","_number":4,"created":"2022-06-08 08:04:01.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/79/844179/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/844179/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/844179/4"}}},"commit":{"parents":[{"commit":"d253b9fd08b969ebab09a5624df127a4bf1bfb1f","subject":"Merge \"Add some logging to test_good_address_allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d253b9fd08b969ebab09a5624df127a4bf1bfb1f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-01 03:08:25.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-08 08:03:34.000000000","tz":480},"subject":"[ovn]neutron agent show real heartbeat_timestamp","message":"[ovn]neutron agent show real heartbeat_timestamp\n\nagent\u0027s heartbeat_timestamp returns the current time, not\nconducive to agent status troubleshooting.\nthis patch use agent\u0027s updated_at as heartbeat_timestamp.\n\nCloses-bug: #1977629\n\nChange-Id: Idf522a20f9735829ee568020bfed46345a95e294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d3b338fbfba025f90ae67dcc7e127d2c45ab36e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5d3b338fbfba025f90ae67dcc7e127d2c45ab36e"}]},"branch":"refs/heads/master"},"1085d44719499f7d7fa1c749c71fd456ee06b892":{"kind":"REWORK","_number":5,"created":"2022-06-08 09:32:55.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/79/844179/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/844179/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/844179/5"}}},"commit":{"parents":[{"commit":"d253b9fd08b969ebab09a5624df127a4bf1bfb1f","subject":"Merge \"Add some logging to test_good_address_allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d253b9fd08b969ebab09a5624df127a4bf1bfb1f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-01 03:08:25.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-08 09:32:11.000000000","tz":480},"subject":"[ovn]neutron agent show real heartbeat_timestamp","message":"[ovn]neutron agent show real heartbeat_timestamp\n\nagent\u0027s heartbeat_timestamp returns the current time, not\nconducive to agent status troubleshooting.\nthis patch use agent\u0027s updated_at as heartbeat_timestamp.\n\nCloses-bug: #1977629\n\nChange-Id: Idf522a20f9735829ee568020bfed46345a95e294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1085d44719499f7d7fa1c749c71fd456ee06b892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1085d44719499f7d7fa1c749c71fd456ee06b892"}]},"branch":"refs/heads/master"},"0a30b4984afd4a653bc3da44ad6785b0b4dd255b":{"kind":"REWORK","_number":6,"created":"2022-06-08 11:08:12.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/79/844179/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/844179/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/844179/6"}}},"commit":{"parents":[{"commit":"d253b9fd08b969ebab09a5624df127a4bf1bfb1f","subject":"Merge \"Add some logging to test_good_address_allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d253b9fd08b969ebab09a5624df127a4bf1bfb1f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-01 03:08:25.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-08 11:07:49.000000000","tz":480},"subject":"[ovn]neutron agent show real heartbeat_timestamp","message":"[ovn]neutron agent show real heartbeat_timestamp\n\nagent\u0027s heartbeat_timestamp returns the current time, not\nconducive to agent status troubleshooting.\nthis patch use agent\u0027s updated_at as heartbeat_timestamp.\n\nCloses-bug: #1977629\n\nChange-Id: Idf522a20f9735829ee568020bfed46345a95e294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a30b4984afd4a653bc3da44ad6785b0b4dd255b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0a30b4984afd4a653bc3da44ad6785b0b4dd255b"}]},"branch":"refs/heads/master"},"3dc68af60a197477cbb10c07c2f0bf380b925d2a":{"kind":"REWORK","_number":7,"created":"2022-06-08 23:51:25.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/79/844179/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/844179/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/844179/7"}}},"commit":{"parents":[{"commit":"d253b9fd08b969ebab09a5624df127a4bf1bfb1f","subject":"Merge \"Add some logging to test_good_address_allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d253b9fd08b969ebab09a5624df127a4bf1bfb1f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-01 03:08:25.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-08 23:49:51.000000000","tz":480},"subject":"[ovn]neutron agent show real heartbeat_timestamp","message":"[ovn]neutron agent show real heartbeat_timestamp\n\nagent\u0027s heartbeat_timestamp returns the current time, not\nconducive to agent status troubleshooting.\nthis patch use agent\u0027s updated_at as heartbeat_timestamp.\n\nCloses-bug: #1977629\n\nChange-Id: Idf522a20f9735829ee568020bfed46345a95e294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3dc68af60a197477cbb10c07c2f0bf380b925d2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3dc68af60a197477cbb10c07c2f0bf380b925d2a"}]},"branch":"refs/heads/master"},"411ecc4865198e2bed5e3d0ab498e3a42408c17d":{"kind":"REWORK","_number":8,"created":"2022-06-09 10:35:18.000000000","uploader":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"ref":"refs/changes/79/844179/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/79/844179/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/79/844179/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/79/844179/8"}}},"commit":{"parents":[{"commit":"d253b9fd08b969ebab09a5624df127a4bf1bfb1f","subject":"Merge \"Add some logging to test_good_address_allocation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d253b9fd08b969ebab09a5624df127a4bf1bfb1f"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-01 03:08:25.000000000","tz":480},"committer":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-06-09 10:34:09.000000000","tz":480},"subject":"[ovn]neutron agent show real heartbeat_timestamp","message":"[ovn]neutron agent show real heartbeat_timestamp\n\nagent\u0027s heartbeat_timestamp returns the current time, not\nconducive to agent status troubleshooting.\nthis patch use agent\u0027s updated_at as heartbeat_timestamp.\n\nCloses-bug: #1977629\n\nChange-Id: Idf522a20f9735829ee568020bfed46345a95e294\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/411ecc4865198e2bed5e3d0ab498e3a42408c17d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/411ecc4865198e2bed5e3d0ab498e3a42408c17d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
