)]}'
{"id":"openstack%2Fneutron~844203","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I364fc7981458374ed25eb8837d1ed3afff046b95","project":"openstack/neutron","branch":"stable/yoga","topic":"bug1967742-stable/yoga","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-06-10 14:40:30.000000000","reason":"Slawek Kaplonski replied on the change"},"5948":{"account":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"last_update":"2022-06-10 18:54:47.000000000","reason":"Change was submitted"},"34271":{"account":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"last_update":"2022-06-10 18:54:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I364fc7981458374ed25eb8837d1ed3afff046b95","subject":"Notify mech drivers with original and modified network.","status":"MERGED","created":"2022-06-01 18:49:05.000000000","updated":"2022-06-10 18:56:24.000000000","submitted":"2022-06-10 18:54:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844203-bug1967742-stable/yoga","cherry_pick_of_change":840146,"cherry_pick_of_patch_set":8,"meta_rev_id":"da6d28662f60b95232394bcfb70aebfadb037b75","_number":844203,"virtual_id_number":844203,"owner":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-10 18:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-06-08 07:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-06-10 14:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-06-10 14:40:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 19:56:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-01 20:26:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-01 20:58:51.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-06-02 14:18:07.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-06-02 14:18:07.000000000","updated_by":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-06-08 07:38:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"9bf0f14953f4ec95ca952cc438b4f44a02c0898f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"date":"2022-06-01 18:49:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9898469e4d9ca2d3d57dad9137e3cf9427506ec6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 19:56:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7f5a44e4abd42188b851aa29cc3a272 : SUCCESS in 1h 04m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09ce3e7532d74d1a842e5b787d3010a4 : SUCCESS in 1h 05m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a05203bc9bff58646ff80f3f883f602af46fa8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 20:26:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/fe72a23c98544a528593b473ff5b0f34 : SUCCESS in 1h 02m 11s\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/8c7966c0a121477c8f110eaa26b04c0c : SUCCESS in 1h 06m 05s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/08f93db350c343a59d2a1e1377734137 : SUCCESS in 53m 01s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/db62a5c647df4d13a90aced9c32e8044 : SUCCESS in 46m 23s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/82079d28b4ad4d03b6805432f46d8ec2 : SUCCESS in 1h 00m 43s\n- neutron-tempest-plugin-designate-scenario-yoga https://zuul.opendev.org/t/openstack/build/8e004dfb477f44cca9a8e2facd1b0d28 : SUCCESS in 28m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45d8f4b3a6f04e9383ccc568c81bb82b : SUCCESS in 42m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/391d8c3cb8fb4b35b2be4901d1c260cf : SUCCESS in 14m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/def17978291047d6998301d12019909b : SUCCESS in 30m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b5b30a5785043c2bb9c38aaf0584258 : SUCCESS in 29m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52f3c64ca0640a4ac170d150f4d4bd1 : SUCCESS in 9m 37s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6e1b4c0d0874b7bb8446a6cbcd9389e : SUCCESS in 55m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3423050cb1964b0eaa82a33b63835267 : SUCCESS in 1h 32m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b030be3c3dd1459cb28860d0137248f9 : SUCCESS in 1h 09m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/403f8728f10f4db38c837ac2132f678c : SUCCESS in 49m 38s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/842d4cdb0d2a45b19359efcde5095084 : SUCCESS in 1h 02m 00s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/29e1b2ce7ebf4bc79586f23c893e8d0f : SUCCESS in 1h 07m 48s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/ce291eb1770744bbac9d5f7c408afc49 : SUCCESS in 1h 08m 42s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3303ba3c2bbc458393de9508e014d64f : SUCCESS in 54m 20s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f8b34236d9cf49b6b72e56fd3b99ba3c : SUCCESS in 1h 22m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7c010cb00e75df5a4728b21a317efca38bee874","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-01 20:58:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844203/1 : SUCCESS in 2h 09m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4a9b602e015877f67937ad7aee8acfecf1d03ae7","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-08 07:38:15.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"491e81878202f059dd17eac1f4eb6b6b2af28523","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-10 14:40:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2dca49959260377f3bfe831b65bffd2321d3794","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 14:41:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64fd014bcf4705ff12d5fb0fa92a49011a8106ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 18:54:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-yoga https://zuul.opendev.org/t/openstack/build/8ede81480b744a69b60e1a3523a25162 : SUCCESS in 43m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b66ead683740ac930f26d98421e714 : SUCCESS in 15m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1f8a56d88c34d809b900382b42a807c : SUCCESS in 35m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f4dac27785f47e9b7be9129572ad852 : SUCCESS in 23m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e104576d2614c69b646368c0fd4122b : SUCCESS in 10m 25s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/aba3ad76bded4e9092d6c0c07e4dcaf7 : SUCCESS in 50m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b53dffd4e7a5416f8eb1011930cfefe8 : SUCCESS in 1h 34m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bd67d01513114e79a4b996c85f2e9fc8 : SUCCESS in 1h 10m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/38da29d5eff247518dc64ab94ebfef5e : SUCCESS in 56m 57s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f1a54479f30d4993951351e45d8b8f32 : SUCCESS in 1h 02m 40s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8138495f47c447a1a5805d5e6df2f2ef : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":1},{"id":"128be1ce338c8c4925a65cb0e8d8c4c0f3bc59da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 18:54:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da6d28662f60b95232394bcfb70aebfadb037b75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 18:56:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4452671381a74b5aaac138f9b592a408 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a8de0cde7003544ecf998a08a293af64f95009c","revisions":{"9a8de0cde7003544ecf998a08a293af64f95009c":{"kind":"REWORK","_number":1,"created":"2022-06-01 18:49:05.000000000","uploader":{"_account_id":34271,"name":"Miro Tomaska","display_name":"Miro Tomaska","email":"mtomaska@redhat.com","username":"mtomaska"},"ref":"refs/changes/03/844203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/03/844203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/844203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/844203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/03/844203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/03/844203/1"}}},"commit":{"parents":[{"commit":"e202bf6a475d1b9bc8974dcad0dfcf80013f6b74","subject":"Merge \"Remove session active check in \"_add_segment_host_mapping_for_segment\"\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e202bf6a475d1b9bc8974dcad0dfcf80013f6b74"}]}],"author":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2022-05-06 18:42:11.000000000","tz":-300},"committer":{"name":"Miro Tomaska","email":"mtomaska@redhat.com","date":"2022-06-01 18:49:05.000000000","tz":0},"subject":"Notify mech drivers with original and modified network.","message":"Notify mech drivers with original and modified network.\n\nInclude original and modified network when notifying mechanism drivers in PRECOMMIT event. PRECOMMIT_CREATE modified network includes new segment, while original does not. Exact opposite for PRECOMMIT_DELETE\n\nCloses-Bug: #1967742\nChange-Id: I364fc7981458374ed25eb8837d1ed3afff046b95\n(cherry picked from commit d1fe14d366e501869addff17e851ac03a599adca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a8de0cde7003544ecf998a08a293af64f95009c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a8de0cde7003544ecf998a08a293af64f95009c"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
