)]}'
{"id":"openstack%2Fneutron~844274","triplet_id":"openstack%2Fneutron~master~I26ebf751b7d187399e8b3d8089e09b9d51ba3625","project":"openstack/neutron","branch":"master","topic":"typos","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-02 18:43:19.000000000","reason":"Change was submitted"},"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2022-06-02 14:54:13.000000000","reason":"Brian Haley replied on the change"},"20733":{"account":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"last_update":"2022-06-02 18:43:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I26ebf751b7d187399e8b3d8089e09b9d51ba3625","subject":"Fix typos in log/error messages","status":"MERGED","created":"2022-06-01 10:19:13.000000000","updated":"2022-06-02 18:44:55.000000000","submitted":"2022-06-02 18:43:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844274-typos","meta_rev_id":"f0da22284fd8615bb76a54bb67dedfe8e26c5c88","_number":844274,"virtual_id_number":844274,"owner":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-06-01 17:15:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-02 18:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-02 14:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":2,"date":"2022-06-02 06:57:44.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-02 14:54:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 10:23:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-06-01 11:01:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-01 11:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-01 11:57:09.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-06-01 12:04:03.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-01 13:58:24.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2022-06-02 06:57:44.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"46a63d973123d51b890f162b43f7f295f9dcaf6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-01 10:19:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1b6f1154dc3a0414362d435543e46a6bdb6bb6f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-01 10:23:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"99abd3b3351ecc692804f533324497686272a80b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 11:01:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7896f01666dd4132b244a7dc0f7ac2e3 : FAILURE in 40m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0b343b8b1ac41938bd37ba898933369 : FAILURE in 33m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a1a921a698845abd2d0845217f936bf8e16ee34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 11:54:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/4a0088a17ac340ee8ec2c111e6994197 : SUCCESS in 1h 28m 24s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/047f8f5dc5c342c1a8b406b55025af6a : SUCCESS in 1h 25m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d266a52f200143e39687b6442d66eb49 : SUCCESS in 1h 07m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/945f04ac7a5a4e228ea07add796881bd : SUCCESS in 1h 25m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/943a788dbf7b4787b8f283b5b1ef1eab : SUCCESS in 30m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ebf1753afe74787b0616d31f30b06ff : FAILURE in 35m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b71e4d77b61d46e680d1e695ab4852b7 : SUCCESS in 13m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c418c030a09e4a3ba237435b0d92d535 : FAILURE in 27m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97502edbb3214f5ca8baee5668958e1e : FAILURE in 28m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1017b6de03624d60ba4f11180b20cab8 : FAILURE in 21m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4dda789c0564473b1cc783aac54818b : SUCCESS in 10m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f6023e9aa6824d8195d228994dc3e79d : SUCCESS in 58m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f10cf6ca34914405bedf05c5f21c21df : SUCCESS in 1h 29m 57s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e673cd72ae1c4a0fa89bb36df1fcf5bc : SUCCESS in 1h 04m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e29682afac44c5f8b58ff5cadfbeda8 : SUCCESS in 57m 21s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/5565c3eb98944403a92a382870568ed9 : SUCCESS in 1h 02m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/028fdf2a70574a8493732de3eb207052 : SUCCESS in 1h 19m 33s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3afa61c838244448b731acb77c4e37e2 : SUCCESS in 1h 16m 25s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/443346db05e4432d8db6aab4e1ac6886 : SUCCESS in 1h 00m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5d0e0d5d1c944ce6a0a77a6fdba36c3d : SUCCESS in 1h 16m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65670729f7c95270762d2e8ec2e177fb34dbf9b0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-01 11:57:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844274/1 : SUCCESS in 1h 37m 19s","accounts_in_message":[],"_revision_number":1},{"id":"751c3696552d8908d97217f54b7c2067e878f35b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-01 12:04:03.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/74/844274/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9d00be05181e6980ca50778fdd4fced6565e750b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-01 12:38:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fc29d7a8881c7ef514d798b7c9d8c27e99362add","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-01 12:55:30.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"7a275988be864a60393e9e85670eba4409f472a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 13:24:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/285350bc8cce4647998f1816e93f5a52 : SUCCESS in 42m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/800a3682c0ae4cd78211820ff680198f : SUCCESS in 42m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b88282e412bb49c76dff4d4be9c7719f86ec5ed0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-01 13:40:16.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/74/844274/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"5bd9144b6705c9e236c3c5b91384bd780df592e2","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-06-01 13:58:24.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6797415d5ce1458757a6edce1451e86660669c6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 14:22:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c290589977f0473cbd162644837f5aaf : SUCCESS in 1h 36m 46s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b12d4f48e95348638a2f897855484fd8 : SUCCESS in 1h 23m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/49e9e8e4a68d411caad97bf1deb684aa : SUCCESS in 1h 24m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5ac8372653ff422c94d0983446e30388 : SUCCESS in 1h 33m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c99bd70f8f6144a78ce28a766331884c : SUCCESS in 30m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/143c0480079e4e72970657a83f7956f3 : SUCCESS in 38m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e28d1d58252e4052bb2bc50184e47ce5 : SUCCESS in 13m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3b420aec9a3498bba25fddca333d0b0 : SUCCESS in 26m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10ef5b04bc574dc4b1c7840b6edcb879 : SUCCESS in 24m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/546a09c0e3de4a548100d5ef73bc5b23 : SUCCESS in 26m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b10824449746f3986a948e2155f034 : SUCCESS in 9m 35s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/40e23d73e10a4118bacaf7e18d4fdbc5 : SUCCESS in 1h 02m 10s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2c303313692747018579bd2ecdab1bb1 : SUCCESS in 1h 36m 13s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/983403ff7b04488abb770854a2b62514 : SUCCESS in 1h 12m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/625297524fc24141a207c730ac2bd207 : SUCCESS in 59m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b9838ad9f6d34d14b0caf1531b05f7be : SUCCESS in 1h 07m 31s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e37c42c09b0a41099bee1ed575f39ea5 : SUCCESS in 1h 23m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/4a34e77ed8e743dc90237efa8c2c85f9 : SUCCESS in 1h 18m 30s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a6b4604a4e4481a84c83ee0589929eb : SUCCESS in 1h 05m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/6cffc67ae82a4ecc8946db8ab9ff6f11 : SUCCESS in 1h 21m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42f326c6466921bcce0f968d7d16da10937b0a18","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-01 14:48:18.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844274/2 : SUCCESS in 2h 09m 20s","accounts_in_message":[],"_revision_number":2},{"id":"dac6872ea0f943fd86cd8e7c2a7d5823b6a134d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-01 15:47:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"99d7a2c6d1b772cf8edbdfa2c69378b2a54f6e41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 16:31:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/812ce61b37a546f08ffa9fc921db85b8 : SUCCESS in 41m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7dfa06803f1a4b01bb738fd0bc2f6faf : SUCCESS in 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dced396dc0ffedfc4eb3f011a7136a636802c4e2","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-01 17:15:00.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/74/844274/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7fe2a9cd8d662a95159f1e1a5de73686b8b15950","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 17:27:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/108f828e11f74c55bb4bc4ed1f64a92b : SUCCESS in 1h 32m 22s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b6014c4366804636bf929139eac0802d : SUCCESS in 1h 10m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/efefbbb79dc849f886b444eb2b96bef6 : SUCCESS in 1h 22m 40s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6975d4901f6a4070bebf279095136ad1 : SUCCESS in 1h 07m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6810696084e442c9b347c1774fea2f0 : SUCCESS in 34m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10998b76d6db49d0a6c7aec46fd04591 : SUCCESS in 40m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5be52a2b7b7c42fd8051c4205c631b12 : SUCCESS in 15m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84d8b938b35d4b4490ea1b5ce2212113 : SUCCESS in 29m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac9434f457fb456db7cd825ae3772a8f : SUCCESS in 23m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/581b21387c25427d8a9508b12236f7f6 : SUCCESS in 23m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c61acbd4759406586746e0d42b74ebe : SUCCESS in 10m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3254602c26104948a17ea2c16dc66fcd : SUCCESS in 1h 05m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d4eeadbca264dd081ed6e7351fc7a08 : SUCCESS in 1h 33m 42s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/c1e51bea3d6644d59ef97f9df10eb8e3 : SUCCESS in 1h 11m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f81dfed6352145fea0e776926ac8851a : SUCCESS in 57m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/195d6d2920fd4f8585b2525590a622ad : SUCCESS in 55m 04s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f4c3ba31bc474aa39bf537012afc292a : SUCCESS in 1h 13m 50s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3d7e8eb7de4440fcb8f323e9f2f34b86 : SUCCESS in 1h 19m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/869aa462983d49569681fc5b979e43fe : SUCCESS in 58m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/d53415b3c7984e3893a0b43d625befed : SUCCESS in 1h 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d4c149ffcf222c0de8d1435d8e52ff184fb7f29","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-01 17:50:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844274/3 : SUCCESS in 2h 01m 57s","accounts_in_message":[],"_revision_number":3},{"id":"bd9a4d4de7e939b84d30e606a8880ee810b1a0eb","author":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"date":"2022-06-02 03:35:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9f93bc4b334dcfb5e89851ed47f971bec29f7817","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-06-02 06:57:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a39d0947b7131dd2a663b662b675f4e3925a027c","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-06-02 14:54:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4a962a2bf70da9ff70a04187e89992b69a4cd4c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 14:56:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ec4caa4d53801d9d2283f8e2f3054503f14ebc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:43:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/428a210a55434fba980d5f9c332bf06b : SUCCESS in 1h 24m 34s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b652995d2d794fb8ae1d4e8d2c8f5a5a : SUCCESS in 1h 40m 13s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/6e39b192d3814f0db4f4418d5db2768b : SUCCESS in 1h 08m 11s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/db333e58117747ca9bc90dee4d4c3776 : SUCCESS in 1h 12m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1eacb90ef4048fda28a9e85e6b04954 : SUCCESS in 13m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0cc5f8914254e379845c9b0abf51243 : SUCCESS in 28m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51b6fa43b48d4937b0adad370e1b5d7f : SUCCESS in 21m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af2e66af9e044c8a8265347bb4f89051 : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c6cc7af92efb494db44eee00f943847b : SUCCESS in 1h 02m 03s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/31605a7b20cc4a05a724b91c21320b9e : SUCCESS in 1h 46m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/da883b81b6f0421bb9ab81519a4335ca : SUCCESS in 1h 35m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9da63b6fd6244c98893273cd2fae299 : SUCCESS in 50m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4e8c4192d4cd4fe1ba505ce061b25ffb : SUCCESS in 58m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/510f533e04e948c9a8acdbdd77a3e819 : SUCCESS in 1h 14m 04s","accounts_in_message":[],"_revision_number":3},{"id":"5e8a3b4ee46d5a23afa0993be6d3300780167990","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:43:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f0da22284fd8615bb76a54bb67dedfe8e26c5c88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 18:44:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc1264fe8b13476b9fef9fcfd1e70c45 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"18b1b614013e5a0f12d5c0ea1e19064f94e5c306","revisions":{"15fc9892848d6f4c4dd2382324161c215ccb49e6":{"kind":"REWORK","_number":1,"created":"2022-06-01 10:19:13.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/74/844274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/844274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/844274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/844274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/844274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/844274/1"}}},"commit":{"parents":[{"commit":"8c1ad2dfadd89fb397fbb57d984d29238db08916","subject":"Merge \"Check the Chassis_Private nb_cfg_timestamp with current value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c1ad2dfadd89fb397fbb57d984d29238db08916"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-05-31 16:17:33.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-01 10:19:09.000000000","tz":330},"subject":"Fix typos in log/error messages","message":"Fix typos in log/error messages\n\nThis change fixes some of the typos in LOG and ERROR\nmessages.\n\nChange-Id: I26ebf751b7d187399e8b3d8089e09b9d51ba3625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15fc9892848d6f4c4dd2382324161c215ccb49e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/15fc9892848d6f4c4dd2382324161c215ccb49e6"}]},"branch":"refs/heads/master"},"48a9e6f00e31304e9692a4482abcb0eca3f3b8e5":{"kind":"REWORK","_number":2,"created":"2022-06-01 12:38:13.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/74/844274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/844274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/844274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/844274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/844274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/844274/2"}}},"commit":{"parents":[{"commit":"8c1ad2dfadd89fb397fbb57d984d29238db08916","subject":"Merge \"Check the Chassis_Private nb_cfg_timestamp with current value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c1ad2dfadd89fb397fbb57d984d29238db08916"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-05-31 16:17:33.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-01 12:32:42.000000000","tz":330},"subject":"Fix typos in log/error messages","message":"Fix typos in log/error messages\n\nThis change fixes some of the typos in LOG and ERROR\nmessages, comments and variable names.\n\nChange-Id: I26ebf751b7d187399e8b3d8089e09b9d51ba3625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48a9e6f00e31304e9692a4482abcb0eca3f3b8e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/48a9e6f00e31304e9692a4482abcb0eca3f3b8e5"}]},"branch":"refs/heads/master"},"18b1b614013e5a0f12d5c0ea1e19064f94e5c306":{"kind":"REWORK","_number":3,"created":"2022-06-01 15:47:43.000000000","uploader":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"ref":"refs/changes/74/844274/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/74/844274/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/844274/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/844274/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/74/844274/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/74/844274/3"}}},"commit":{"parents":[{"commit":"8c1ad2dfadd89fb397fbb57d984d29238db08916","subject":"Merge \"Check the Chassis_Private nb_cfg_timestamp with current value\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8c1ad2dfadd89fb397fbb57d984d29238db08916"}]}],"author":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-05-31 16:17:33.000000000","tz":330},"committer":{"name":"Rajesh Tailor","email":"ratailor@redhat.com","date":"2022-06-01 15:47:29.000000000","tz":330},"subject":"Fix typos in log/error messages","message":"Fix typos in log/error messages\n\nThis change fixes some of the typos in LOG and ERROR\nmessages, comments and variable names.\n\nChange-Id: I26ebf751b7d187399e8b3d8089e09b9d51ba3625\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18b1b614013e5a0f12d5c0ea1e19064f94e5c306"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/18b1b614013e5a0f12d5c0ea1e19064f94e5c306"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
