)]}'
{"id":"openstack%2Fneutron~844552","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Icf92685579409282bad0a80ba42531c93738e0b1","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1976292","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-03 16:53:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icf92685579409282bad0a80ba42531c93738e0b1","subject":"[OVN] Protect the AgentCache iteration by copying the local cache","status":"MERGED","created":"2022-06-03 07:14:27.000000000","updated":"2022-06-04 16:23:16.000000000","submitted":"2022-06-03 16:53:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844552-bug/1976292","meta_rev_id":"28a677c2d28406aea6845f64fae3a148f457187f","_number":844552,"virtual_id_number":844552,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-03 16:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-06-03 11:30:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-06-03 12:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-06-03 12:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-03 08:26:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-03 09:07:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-03 09:11:11.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-06-03 11:30:10.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-06-03 12:07:09.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"697109cceeff3db40d9d29547a6a38d855575344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-03 07:14:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71fcc7e6c3108b631e373a32761bfcb196ba1fb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-03 07:22:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d1624797a567bf379a3617d1280118e7396f814","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 08:26:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3be3fb95ea3e48668939bfc0065b182a : TIMED_OUT in 1h 01m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/033c8510336943f890ca0ba82931c161 : TIMED_OUT in 1h 01m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4a8681ec483aec8de86b98804031fb6f58fe7c21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 09:07:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/fa2eb5ad9dc04b818cb0dc7ac27b69a8 : SUCCESS in 1h 00m 04s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/dc16b7d0569b4cb7b3c001f1ddd9a46e : SUCCESS in 1h 18m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b939b5a81cd449898b7a11f9f99e8c3b : SUCCESS in 35m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d2315ac195047069d01f42029b340b1 : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1002ecffce634f2ea4f24c5b67b5be6c : SUCCESS in 30m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9295b342c3864bdf96ba3d67cc1d6dbe : SUCCESS in 25m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3126c826f17422799218f6ba2e518c2 : SUCCESS in 27m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc8071f1d784bc29843c98135d921bb : SUCCESS in 9m 18s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/62f10c64def74c2ea0ed74f17eaab2e0 : SUCCESS in 52m 40s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/e73aa9d837fd4daf81d3ddc828925a4b : SUCCESS in 1h 08m 31s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/da7ef48ecf1c4d45bdf42ff07b0b1550 : SUCCESS in 36m 46s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/03e354ccde114d83a3c8e42b1e04efb3 : SUCCESS in 1h 37m 50s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/473a80b5927d436fb34c6cd0a39a105d : SUCCESS in 1h 23m 03s","accounts_in_message":[],"_revision_number":2},{"id":"6af459d9d90f8640707f6e016fd4e969c27db9e5","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-03 09:11:11.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844552/2 : SUCCESS in 1h 47m 58s","accounts_in_message":[],"_revision_number":2},{"id":"dfb052cfc3963009f9a50bcced3b73722cb99869","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-03 11:30:10.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ebcdb09b3df3ad15d4c1de99916e9791cf73fdc6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-03 12:07:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"13b36b07e6103dd1797a058036a25d035d32a0ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 12:07:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f508436067fc1e134088b1f51a0d89ddd3fd17d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 13:27:44.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/edfff163eba841c599098b1eeeb95ec0 : SUCCESS in 1h 07m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc8ef6cf5804b63abc5d31384e9b0e4 : SUCCESS in 13m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38e80c2fa3354e439a896cadde8e91bc : SUCCESS in 28m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79a7573b52504297af66111128a8954b : SUCCESS in 33m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3831aad0999346638593da5af6b5125a : SUCCESS in 9m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/84f912f3733d47adbe541916f1c6c622 : FAILURE in 1h 00m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/23a93bb556874c498e29006867f3e8d8 : SUCCESS in 1h 15m 39s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/26d68a9eea384cd7a9cfd0b735511b42 : SUCCESS in 33m 54s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/7356164555fb47a68d344ea4eeb68006 : SUCCESS in 1h 08m 51s","accounts_in_message":[],"_revision_number":2},{"id":"6e2f57e9e65dea3e45b5aa813e8d60f5e40e2f84","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-03 14:02:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d53a740a470407474da352d5a5fdcd242acd1f0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 15:06:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ad006f1e11b4e0bb51c57c0aee0cc5f : TIMED_OUT in 1h 01m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4b12076ced3a49469093dfe58ca7f91a : TIMED_OUT in 1h 01m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"adf67747f5924f2fe59f172bb0f42ba9998004d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 15:34:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/1684f3b06e7848cd99e6c1c364b2be9a : SUCCESS in 48m 01s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/46fee4867bc2496a9807b9f2936bdfda : SUCCESS in 1h 29m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db56fb369c3c404ba00e13a57f17f096 : SUCCESS in 30m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/376bbc2a17144fe1874b00767ab74808 : SUCCESS in 10m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a2679c4a5ef4e8e9b0c3bc4a9725ecf : SUCCESS in 22m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3512c3815bd34b6d8e907dbc7ba34f32 : SUCCESS in 21m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7358a57e772a448f966269d34f392e6a : SUCCESS in 22m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844e7802efdc4de0bcc3d09c057bf4e4 : SUCCESS in 10m 32s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8ea2c71be9cf4d67bbc3d78b1def7434 : SUCCESS in 44m 01s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/ebef07e721134a5a9a00f2821703141d : SUCCESS in 1h 04m 33s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/02c1815ab4654756abcd0b71693660de : SUCCESS in 33m 10s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b8d477b58bc6432984416a0f68f83683 : SUCCESS in 1h 28m 36s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/8385b5c347894c929a4bb5c458be9dc7 : SUCCESS in 1h 06m 13s","accounts_in_message":[],"_revision_number":2},{"id":"02c1c5a875fc4f8c525a5d6f9a862ac8a94c71f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 15:35:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1992f34036c521667e1b753688e37903ed8f2e18","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-03 16:23:47.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844552/2 : SUCCESS in 2h 20m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6b2f61c363746adc52475a1daca41ae5cb8b3ba3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 16:53:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/13b9f5cda9e6431490edd40c6add9c50 : SUCCESS in 51m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a549649e8bd4d47bf9fd6addd8865d1 : SUCCESS in 13m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad202e76b0c040b8a5702b3cdbec5653 : SUCCESS in 28m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2d7c3a10d134c0d8190895df4bd06bb : SUCCESS in 21m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3183a66ac7474ef3a0d9cf8e5526b0a1 : SUCCESS in 9m 05s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/cfbe7868582e4430af146794fff881c3 : SUCCESS in 44m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/34366a16b2e5444eb779bdefabfa7d93 : SUCCESS in 54m 10s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/ad6111979cb444fabdf024bee1508987 : SUCCESS in 28m 26s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4f5258c84d9b457ca359ddc3fcc3d86f : SUCCESS in 1h 06m 32s","accounts_in_message":[],"_revision_number":2},{"id":"1e29aa68f33a1c62ed7e35923818f19061b9aba6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 16:53:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"78806bb6c2c5290bb7e261fc6f32016dc19eefde","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 16:55:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04590d996f574290a7ac9a84e7bc0e4a : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2},{"id":"27dab383c6753bafbf4c3a5f7d4e0184c2c726b9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-04 14:20:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28a677c2d28406aea6845f64fae3a148f457187f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-04 16:23:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844552/2 : SUCCESS in 2h 02m 33s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9182712672f05d1b2849957c7632d703180c5e0f","revisions":{"29b934482884dc1505d259c4f1d2d2f64945b655":{"kind":"REWORK","_number":1,"created":"2022-06-03 07:14:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/844552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/844552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/844552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/844552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/844552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/844552/1"}}},"commit":{"parents":[{"commit":"f6821b5b77cac4ef9b9df3065280bf0abea07c22","subject":"Merge \"[OVN] During the DB migration, allow the DB context to finish\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6821b5b77cac4ef9b9df3065280bf0abea07c22"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-15 22:48:46.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-17 07:24:02.000000000","tz":0},"subject":"[OVN] Protect the AgentCache iteration by copying the local cache","message":"[OVN] Protect the AgentCache iteration by copying the local cache\n\nDuring the execution of ``AgentCache.__iter__``, the content of\n\"self.agents\" can change; the Neutron server can attend to an event\nto add or delete an OVN agent. If that happens, the iteration will\nfail with the error:\n  \"RuntimeError: dictionary changed size during iteration\"\n\nTo avoid this, before iterating the list of agents, the cache\nsingleton creates a copy of the local cache. The aim of making this\ncopy is to avoid using locks between threads.\n\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/agent/neutron_agent.py\n\nCloses-Bug: #1976292\nChange-Id: Icf92685579409282bad0a80ba42531c93738e0b1\n(cherry picked from commit a71fe45d9628f0dc67ccd12d94a51a7b801ec9d0)\n(cherry picked from commit 19f856034ffb9520f48db230d9e3c96078b3929f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29b934482884dc1505d259c4f1d2d2f64945b655"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/29b934482884dc1505d259c4f1d2d2f64945b655"}]},"branch":"refs/heads/stable/wallaby"},"9182712672f05d1b2849957c7632d703180c5e0f":{"kind":"REWORK","_number":2,"created":"2022-06-03 07:22:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/844552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/844552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/844552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/844552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/844552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/844552/2"}}},"commit":{"parents":[{"commit":"f6821b5b77cac4ef9b9df3065280bf0abea07c22","subject":"Merge \"[OVN] During the DB migration, allow the DB context to finish\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f6821b5b77cac4ef9b9df3065280bf0abea07c22"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-15 22:48:46.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-17 07:33:04.000000000","tz":0},"subject":"[OVN] Protect the AgentCache iteration by copying the local cache","message":"[OVN] Protect the AgentCache iteration by copying the local cache\n\nDuring the execution of ``AgentCache.__iter__``, the content of\n\"self.agents\" can change; the Neutron server can attend to an event\nto add or delete an OVN agent. If that happens, the iteration will\nfail with the error:\n  \"RuntimeError: dictionary changed size during iteration\"\n\nTo avoid this, before iterating the list of agents, the cache\nsingleton creates a copy of the local cache. The aim of making this\ncopy is to avoid using locks between threads.\n\nConflicts:\n    neutron/plugins/ml2/drivers/ovn/agent/neutron_agent.py\n\nCloses-Bug: #1976292\nChange-Id: Icf92685579409282bad0a80ba42531c93738e0b1\n(cherry picked from commit a71fe45d9628f0dc67ccd12d94a51a7b801ec9d0)\n(cherry picked from commit eb562b60a3f128fc45c9a90e1a849efbd6a62c57)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9182712672f05d1b2849957c7632d703180c5e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9182712672f05d1b2849957c7632d703180c5e0f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
