)]}'
{"id":"openstack%2Fneutron~844589","triplet_id":"openstack%2Fneutron~master~I63e78fdd030f063a1b943d481a9cbd98850552d7","project":"openstack/neutron","branch":"master","topic":"bug/1976323","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-07-29 17:37:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I63e78fdd030f063a1b943d481a9cbd98850552d7","subject":"Make iptables rules diff to be case insensitive","status":"MERGED","created":"2022-06-03 13:30:06.000000000","updated":"2022-07-29 17:38:40.000000000","submitted":"2022-07-29 17:37:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"844589-bug/1976323","meta_rev_id":"f05af4d8463e481964281cb234ed649bab260533","_number":844589,"virtual_id_number":844589,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-07-28 13:49:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-29 17:37:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-07-29 10:09:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-28 20:49:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2022-07-29 07:09:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-07-29 07:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-07-29 10:09:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-03 13:38:05.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2022-06-03 14:18:25.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-03 14:19:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-03 15:47:51.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-06-03 16:26:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-08 14:59:51.000000000","updated_by":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2022-07-28 20:49:46.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-07-29 07:09:23.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-07-29 07:47:37.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-07-29 10:09:00.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"93de25668f4fe423ec7c813d76480cdf8c195d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-03 13:30:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"534f0240928fed54eef418f258253f53d688ee68","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-03 13:38:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"98a75b356ba5154a2f545982c8a517545c478dd8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-03 13:51:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d02628d0db16ccda592df6d41a5ff7febca5ffd0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-03 14:18:25.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/89/844589/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"c56af8a2304f41845f8955afc02219425ebbe7b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 14:19:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c6347e91d8a4c3e87449233fc870640 : SUCCESS in 45m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dca704af57f345ed9ad8a79aaa8d38f4 : SUCCESS in 46m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7fb11bec70abbb04d9cf0129c005ea7c69fc9a5a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-03 15:47:51.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844589/1 : SUCCESS in 2h 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9d37560db5fedc15a170fdfc0959927f40890fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 16:26:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/8fc19228639d4614bdda15fe54a2d759 : FAILURE in 2h 28m 59s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c31baf7e282f416f835b755b844f2da1 : FAILURE in 1h 46m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1fbf5294e5a42b18a14270d630fc346 : TIMED_OUT in 2h 53m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/64ada75e321b46798ae23deaa2ea333b : SUCCESS in 1h 39m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2fa252159c94b8492494200b49de8c1 : SUCCESS in 39m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60a8d5ac6c574afeb85c9a5c7a010042 : SUCCESS in 13m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7473dcddb20492aaec569f8a125606f : SUCCESS in 27m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4410ec6318540d5ac597e10fc59afec : SUCCESS in 26m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e5015413a20409eb1a1011c49f22517 : SUCCESS in 22m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c514cab93ff4c828b6202491b91cde2 : SUCCESS in 9m 39s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d74c90b801541f08c8ee597259795e4 : FAILURE in 47m 49s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/450ed1589883401e931e74d57faab0af : FAILURE in 1h 38m 02s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ea42d53ca74a484692a6d8af7e5db547 : SUCCESS in 1h 12m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/07d14e377c914e35b99609bb339087a2 : SUCCESS in 59m 05s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/46be5915f2604f91adc44702df075642 : SUCCESS in 1h 00m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ced11656435e415080425f045da1014b : SUCCESS in 1h 27m 58s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6b9551775d7943fe908c68665aa64ec5 : SUCCESS in 1h 21m 08s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac0e9847f12f44d390432249a58168fc : SUCCESS in 1h 03m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/8b6a9e63758c4cf6948d6c4537cec0e0 : SUCCESS in 1h 13m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00e76f43ecacc544e4c1d7e6674458930f05dceb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-07 14:31:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"52a65ef1645fab98761723625d89417824d9ef46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 15:12:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616a4336adab47c0bd5af93dfd2403d1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fd213be6b7b4a18836484bbdafb5ee7 : SUCCESS in 39m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d24812c4443d4e1f9ff17a3de97e21f0 : SUCCESS in 38m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"952f7556afba8bb013e55cc20ae7076aa4b6f6fe","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-07 16:30:28.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844589/2 : SUCCESS in 1h 58m 24s","accounts_in_message":[],"_revision_number":2},{"id":"6873eaaa69e52e02df9b7cc79a5aa1d99b6b7422","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-07 17:05:21.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/89/844589/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3d91ac3b019b9d85d2c5ce5d5f6a152adc19525a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 17:29:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e9030e989214549a6db288295a15f38\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/12deab7a15414180b04a7f2305d820a9 : FAILURE in 2h 28m 55s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/ff6d5597dd164fda959e66f4fe1e7c11 : FAILURE in 1h 35m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5c9671b8d8224f578d86e616c859d1d3 : TIMED_OUT in 2h 56m 59s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/d6c40e92c60044418f49fd665ffe9b2d : SUCCESS in 1h 12m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/818454ee7ae44fb5930611f1e8affc42 : SUCCESS in 36m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb816bee628e4b088367b21e72bbcde5 : SUCCESS in 15m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85cc5c1548744ad79d22aa0ffac7ed63 : SUCCESS in 27m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b58cc92e9ea44814a32efb4b832c71b2 : SUCCESS in 20m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2fb2923bd2b949739721e916a3084ca4 : SUCCESS in 26m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3739a37da8194e828d57dd2b2a3f5a68 : SUCCESS in 10m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/fce5e6cdccd74c71894139691621c8e4 : FAILURE in 49m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/ead7ced024754f75bdf5d3179633ce6e : FAILURE in 1h 13m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/a86d55b86f9f4eda870449b37692060b : SUCCESS in 59m 21s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7dba3af59404ccc8be73a7a56613e6b : SUCCESS in 56m 25s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d6d7f64cef89441d9371064820c95e6b : SUCCESS in 56m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/db990d3989af4cdfbb269ca5df1c964f : SUCCESS in 1h 11m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e18933dcfb9f442297cc9508c13ee4d6 : SUCCESS in 1h 07m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f2cdeea995a467382a6005048ba62fb : SUCCESS in 59m 18s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/3d38bf687b96407480ffcfd872485250 : SUCCESS in 1h 07m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d762cb222778eaec231ccb3cbd99a81fd97685a","author":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"date":"2022-07-08 14:59:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00d25148bceb846db8d5898a850c400b73dad91a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-27 18:49:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"02a1b3e14c0cf8323cfd3ff0d25eec37007f43b3","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-27 19:30:22.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/89/844589/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"7b04bdbc0e4c01bee8e7655f639cb18e2aee5a64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 19:33:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d75ceafde449af8d4bd72b2036e380\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37b4611389284358a53f0d2196baaf72 : SUCCESS in 30m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6c18bd8065584ce899ebcd64e7f3163f : SUCCESS in 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"996f089b11cef9dfb397b251954858573bde336d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 20:28:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3caa303c171f4ee2b3c59e1cb2a64e83\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/82efa7218fcc48dd938f8459c5c7e754 : SUCCESS in 1h 09m 29s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/15a16f2f5442435bbbb69375a08b7671 : SUCCESS in 1h 29m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/3e0ad7fd23d94ab6bf7c0e7f8a889ae0 : SUCCESS in 1h 05m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/7d1731e5158c4e31bf28d2e6cd1e2fae : SUCCESS in 1h 27m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ed837dd7a2b4e3a9fd438068bc29252 : SUCCESS in 38m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d1fc208075499a9377c12d2202de40 : SUCCESS in 11m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6590ca48921e4d50be8ec4765be088f0 : SUCCESS in 26m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21a93874865b447d97bf3265ccc4babf : SUCCESS in 23m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa5b3c9550654eacaa3c7a9bf655b29f : SUCCESS in 20m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ac7c2021514077afc16063c9e02688 : SUCCESS in 9m 20s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c1a361b57db4ce4ab246e503c2d58ad : FAILURE in 53m 32s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c705380779094787a546673ae91ff479 : SUCCESS in 1h 32m 05s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e7073a8256604445bd71dda468553dc1 : SUCCESS in 1h 04m 50s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/77c2c4679e914f78843e448ed6daae48 : SUCCESS in 1h 06m 17s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/b7a877acde004010a9ebabcdd25d9099 : SUCCESS in 59m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d0e0b841b2f44bf5a5a793ffbb22626f : SUCCESS in 1h 26m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/e831854ec45f46de813387aaf99ffdf6 : SUCCESS in 1h 13m 29s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/417b9e69029e4091aa3e20e1cd1b5892 : SUCCESS in 55m 06s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/54d6f4ea229c4fd7959b4175f890fd37 : SUCCESS in 57m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8f673d3101399ab7ed8adb80c2cd0920337ede0b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-27 20:43:06.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844589/3 : SUCCESS in 1h 52m 37s","accounts_in_message":[],"_revision_number":3},{"id":"496df225afe6f166c21a3d57d6f172918f0516ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-28 12:08:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"141ff3df4d185070f75dd9ce682be03099bc3cbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 12:47:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eebd367571f4b468d6c4e6e638b3102\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0fd04dadc1346f39a4f0b5762fa24e7 : POST_FAILURE in 33m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8788d2980f6f49c88504ca51047eb9cc : POST_FAILURE in 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"45a73b7732f536f769324a075594013c6d524344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 13:43:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d6ea44684eb4eacbee3ee049e38967d\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/d58c77e7386744cdb1150875e580e5f1 : SUCCESS in 1h 13m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/a10b66dea0a54c83b856971dd3aa9dc9 : SUCCESS in 1h 28m 20s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/64f74c898ede44d39625afc795fd55f1 : POST_FAILURE in 1h 03m 32s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f3692cf287d9434eae53b8ba91597a26 : SUCCESS in 1h 13m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35f5992140e44befa070d053a6f653f2 : POST_FAILURE in 30m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52ed13f997584006bfc8bd62618eb899 : SUCCESS in 11m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0615137556b346d495ceeb278161e46b : POST_FAILURE in 22m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5025314acf8432f80de3e8143f1e613 : POST_FAILURE in 22m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b7ab881fe444cbaac9dcfe3b5221973 : SUCCESS in 27m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/596faca35ebb4545b9199f36edf2bec9 : SUCCESS in 11m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4c1be92058f94af58aa805555731bbbd : SUCCESS in 54m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1fb5009d8d7947c0b487008c6d9c7883 : SUCCESS in 1h 33m 18s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8e7ab48688f5431785abd14a50d28feb : SUCCESS in 1h 05m 08s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c7f486eb221a45c79cdf36269068fbfb : POST_FAILURE in 48m 28s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2bfcee20433b4ab2a08c77876da20a9d : SUCCESS in 55m 22s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/777ce043587c4042bb06a7216d18f217 : SUCCESS in 1h 24m 34s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/cb25fdbbd47b41bdb836cca5b9c7abcc : SUCCESS in 1h 24m 59s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/87626be22e8a4736a314d136af5fd2cc : SUCCESS in 1h 01m 19s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/63c94382ecfd4dd6b57af6a742027dd6 : SUCCESS in 1h 10m 46s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2157d5143b3e37d203cc1986d2b63ea89c8a71e5","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-28 13:49:42.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/89/844589/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"c9e5bd1c26f14141edce3cac5fca4ec98f27bd9b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-28 14:12:05.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844589/4 : SUCCESS in 2h 02m 50s","accounts_in_message":[],"_revision_number":4},{"id":"9a6c9423e44ccd6ba99b38822e6cc73b6b2fcb2c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-28 15:17:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5da082a398de6b66ba39ddd11db392aa766fe285","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-28 15:18:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f56f969b8c3f02d8875a019be1be1dcf57654da2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 16:05:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1116b57cee24411aa1dbfe813d839f8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/56bd9b946f5b46b48d1bd08dd5e06c56 : SUCCESS in 45m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7892854fbf7441c889262c7605d423af : SUCCESS in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1dbcd47e5000f91fd56dd94b0bda2a9cb66d334a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-28 17:09:38.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844589/4 : SUCCESS in 1h 51m 37s","accounts_in_message":[],"_revision_number":4},{"id":"6e9532757f4926a7260fc6178dabfffada064a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 17:10:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b97014f4e537438898b68744e5a72626\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6593c1484d8a4d7ba4515a901c5df5ed : SUCCESS in 1h 19m 02s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b41cb6bf21254ff0a0b922bc21f3752c : SUCCESS in 1h 29m 36s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4ab1068fe1214c799d914e0304efd237 : SUCCESS in 1h 04m 14s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/68d58ed8d6214c528a5c2b6e25b006ae : SUCCESS in 1h 25m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d30a399f5ff4accbbf67d43323a4c41 : SUCCESS in 38m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44cb0359f5ca457b8f4f161ebdc5ac7b : SUCCESS in 12m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06e07e5eb3ba49c993ad9d823cd5d3e7 : SUCCESS in 24m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ef95133bb81472d9c2148fb38f89d05 : SUCCESS in 24m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7148e528f1434b15a2233b37d96c4263 : SUCCESS in 31m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd453dda54e24c869d245424737d0b92 : SUCCESS in 10m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/97231cf404744734bf0e8a1979f4a226 : SUCCESS in 1h 06m 29s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5c875e26085640e081549f46cee80be4 : SUCCESS in 1h 45m 43s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/356a9374655144bfabad61c76ed4d371 : SUCCESS in 1h 04m 08s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fce43fd7b6284241a34490690e2662f5 : SUCCESS in 54m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eb215c866fdf4fb38f87ac50675c2971 : SUCCESS in 1h 08m 38s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9ca9646789eb44f1978fc49049edf2a6 : SUCCESS in 1h 17m 40s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/12a6bcdef57843b5b551e33aa8f02f75 : SUCCESS in 1h 26m 12s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eac8a1b546c24fbe865406dda3a23914 : SUCCESS in 1h 09m 17s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/1d2bb967551245099835abc052ad68c9 : SUCCESS in 1h 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8c30c598da6d0f6a363baf5b115cf13dbf19e0ac","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-07-28 20:49:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3d345b3d2944f17c40b87a7109cababad7e0f33f","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-07-29 07:09:23.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"794e4b9f6cc17b7320f055faf431ad21dea688fc","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-29 07:47:37.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"456414aedffc3ac11bb50872d0c8d85ca888362a","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-29 10:09:00.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b1f2afc7744b89de1c90358e3ae7f662d8b595a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 10:09:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d162bbc773905ded9686f2c0143864488f321456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 12:30:13.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3d06ea427bc4759a735f9073391b097\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/6dd4d5cad5914ea6ba5cf6df44038833 : SUCCESS in 1h 54m 19s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/8b25e0eaa71b4727b7ebca063e51c0b6 : SUCCESS in 2h 13m 55s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/708f3112fcdf46e5bf46a6b9ed5ca9f3 : SUCCESS in 1h 36m 41s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/08c71b622fb945a7b5b8cefbbb30afec : SUCCESS in 1h 49m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c39c977e91d44b98bdf54b439ae7966 : SUCCESS in 11m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e522a477b22443a8f2bde9949c7bf44 : SUCCESS in 21m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cff56ed6bc14f9b9aa26cc369a3a6af : SUCCESS in 22m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83333f4440cd4a499358877d7d5e88df : SUCCESS in 12m 24s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/b6b1081b689d40dc9924db01c7bad6eb : POST_FAILURE in 1h 21m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/80457b9054264cfc8e14bcc9177d849f : SUCCESS in 1h 54m 17s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/67f4444acf444e7d9492f551e22143a1 : SUCCESS in 1h 29m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d6c4c874aa344259caaed9948187fa3 : SUCCESS in 1h 13m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/637f8fe4a84146c4aa3968ea9d8ccb1e : SUCCESS in 1h 11m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a21ff4e5cfd941a4ac4e1722e42b3e8f : SUCCESS in 1h 14m 46s","accounts_in_message":[],"_revision_number":4},{"id":"209b4aab4fc00f60a0cc534908b6347e303c1146","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-29 13:07:02.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"63dd6dd3269683445e500af44656943a820944a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 13:52:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/669b6d4ef4474dffb73a8d2f01282272\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b83aabbb887e47e0851cf2d6a864a9aa : SUCCESS in 32m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2089eb008c84432aba7b6d72a5ad2693 : SUCCESS in 43m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3f644b6776f2fb037d7b1e7a3cf1a63851cefd11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:49:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c16dfe1345504b85b30e95f4cc67fa86\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/927e721a056744259ba22bf27c6096b2 : SUCCESS in 1h 16m 52s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9a83d99f624b43849692a03e31d574cd : SUCCESS in 1h 23m 25s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/86341b980c554bd3ba11c79e995c77da : SUCCESS in 1h 20m 06s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/86836d9f13844982b37ec60ac2ba44a4 : SUCCESS in 1h 13m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d42e505065544b5ebc9cb8bc41cee423 : SUCCESS in 38m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf6859054d45469eb6362c8611b1d307 : SUCCESS in 14m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/639e84bacfec4853a8989149002a8f6a : SUCCESS in 26m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdfdac5a79094b4ebf9fb7e86145dff7 : SUCCESS in 27m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b29a0defb39645b5ab46f710065827b0 : SUCCESS in 29m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1330c711af82437c81f47206efb3d2e3 : SUCCESS in 9m 41s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/845acdabba27434a9ad0209969900511 : SUCCESS in 1h 05m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6cd28a9c6dd64d39ba1a99a70b28e2ea : SUCCESS in 1h 36m 32s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2e7377ec41d9484c827d212104dfb656 : SUCCESS in 1h 04m 10s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0861f151193843159578238806f703b5 : SUCCESS in 1h 07m 46s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e0e8a853cf8242209aaef7357f9d1986 : SUCCESS in 1h 02m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cc7369be43b54dff9a43464172ca1edd : SUCCESS in 1h 18m 09s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/33ee7cd70b0d440ebf4630b9545c4b11 : SUCCESS in 1h 32m 48s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3849ee2ee9a94902a1bb5497dbc60331 : SUCCESS in 1h 02m 57s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/adaafb38ab7b4217a67ea3948ee9ef34 : SUCCESS in 1h 09m 55s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b1ab236d61a3ac35fb2b3a43b4f19c0ba7344ca1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:50:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e119ad50995798dbcb7294c844484a14aae20b62","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-29 15:19:12.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/844589/4 : SUCCESS in 2h 11m 35s","accounts_in_message":[],"_revision_number":4},{"id":"6d1c62d92e57c343dd37f1ab0f56c0de8a2b7b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 17:37:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed9aca7f84e41bea9a1fe4f6f33a4ef\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/84078e9108ae4046b445b6bd8986c360 : SUCCESS in 1h 48m 05s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4bb237424411411da3bead2f2e4bff45 : SUCCESS in 1h 50m 25s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a246c35b43724ecdba232165f5fcfc20 : SUCCESS in 1h 45m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f1513a52e5fe4f1895fc4aa1949e06ea : SUCCESS in 1h 42m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/998443d404c74380bb53139e9654d8ac : SUCCESS in 12m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/339d93aa7bd5434f85216920ee58b1f8 : SUCCESS in 24m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73808fc67eb442458e44c0634049926b : SUCCESS in 25m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1013efa2694067a4c456fa5efec138 : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/200230ef69a7424992873fac7fa2b6df : SUCCESS in 54m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/dd8602c541ae426cbfea0ee653f3c3dc : SUCCESS in 1h 34m 06s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b914d11afe224af8bd83fbfa7f18bd68 : SUCCESS in 1h 24m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5407a0873d354956a0c3f161317bf492 : SUCCESS in 58m 34s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fa62041113144c9c84411465941eb26f : SUCCESS in 1h 08m 43s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21519e532ee54040acdfe9175bdfca32 : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":4},{"id":"9d2d58e13e8cc312530ad0b2971e7d8158fb551c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 17:37:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f05af4d8463e481964281cb234ed649bab260533","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 17:38:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3baffa48112642b49dad4eb0e01be3b9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0775d8674ec44ca084771ebf3266c6ff : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6c2d2744d5e841f431ae637624dd5954429f9607","revisions":{"f162e34e6de3c450c0125e93a30adc813c16e4df":{"kind":"REWORK","_number":1,"created":"2022-06-03 13:30:06.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/844589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/844589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/844589/1"}}},"commit":{"parents":[{"commit":"3261f3bafb0c87e6137153d8b9d8d1508785d03d","subject":"Merge \"Fix typos in log/error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3261f3bafb0c87e6137153d8b9d8d1508785d03d"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-06-03 13:24:51.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-06-03 13:24:51.000000000","tz":120},"subject":"Make iptables rules diff to be case insensitive","message":"Make iptables rules diff to be case insensitive\n\nMAC addresses in the iptables rules are always added by iptables_manager\nwith uppercase. It was done like that in [1].\nFor some reason, iptables on Centos 9 Stream (1.8.7) returns MAC\naddresses all in lowercase and difflib library treats such new and old\nrules as different ones so iptables manager always tries to remove and\nadd antispoofing rules which have MAC addresses set.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/184355\n\nRelated-Bug: #1976323\nChange-Id: I63e78fdd030f063a1b943d481a9cbd98850552d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f162e34e6de3c450c0125e93a30adc813c16e4df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f162e34e6de3c450c0125e93a30adc813c16e4df"}]},"branch":"refs/heads/master"},"1216c8379cea4dc12f4099da860265e7fc571daf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-07 14:31:21.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/844589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/844589/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/844589/2"}}},"commit":{"parents":[{"commit":"ac0f24735205612922be90e41c911056faf8ac14","subject":"Merge \"update the nova host aggregates links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac0f24735205612922be90e41c911056faf8ac14"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-06-03 13:24:51.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-07 14:31:21.000000000","tz":0},"subject":"Make iptables rules diff to be case insensitive","message":"Make iptables rules diff to be case insensitive\n\nMAC addresses in the iptables rules are always added by iptables_manager\nwith uppercase. It was done like that in [1].\nFor some reason, iptables on Centos 9 Stream (1.8.7) returns MAC\naddresses all in lowercase and difflib library treats such new and old\nrules as different ones so iptables manager always tries to remove and\nadd antispoofing rules which have MAC addresses set.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/184355\n\nRelated-Bug: #1976323\nChange-Id: I63e78fdd030f063a1b943d481a9cbd98850552d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1216c8379cea4dc12f4099da860265e7fc571daf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1216c8379cea4dc12f4099da860265e7fc571daf"}]},"branch":"refs/heads/master","description":"Rebase"},"23dbb08fc7b62cfc7f1a20807eed718848db524c":{"kind":"REWORK","_number":3,"created":"2022-07-27 18:49:43.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/844589/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/844589/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/844589/3"}}},"commit":{"parents":[{"commit":"8ab5ee1d174388c9ae31f083ec656f6f95adcc14","subject":"Fix remaining typos in comments and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ab5ee1d174388c9ae31f083ec656f6f95adcc14"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-27 18:48:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-27 18:48:45.000000000","tz":120},"subject":"Make iptables rules diff to be case insensitive","message":"Make iptables rules diff to be case insensitive\n\nMAC addresses in the iptables rules are always added by iptables_manager\nwith uppercase. It was done like that in [1].\nFor some reason, iptables on Centos 9 Stream (1.8.7) returns MAC\naddresses all in lowercase and difflib library treats such new and old\nrules as different ones so iptables manager always tries to remove and\nadd antispoofing rules which have MAC addresses set.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/184355\n\nRelated-Bug: #1976323\nChange-Id: I63e78fdd030f063a1b943d481a9cbd98850552d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23dbb08fc7b62cfc7f1a20807eed718848db524c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/23dbb08fc7b62cfc7f1a20807eed718848db524c"}]},"branch":"refs/heads/master"},"6c2d2744d5e841f431ae637624dd5954429f9607":{"kind":"REWORK","_number":4,"created":"2022-07-28 12:08:34.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/89/844589/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/89/844589/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/89/844589/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/89/844589/4"}}},"commit":{"parents":[{"commit":"8ab5ee1d174388c9ae31f083ec656f6f95adcc14","subject":"Fix remaining typos in comments and tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8ab5ee1d174388c9ae31f083ec656f6f95adcc14"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-27 18:48:45.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2022-07-28 12:08:16.000000000","tz":120},"subject":"Make iptables rules diff to be case insensitive","message":"Make iptables rules diff to be case insensitive\n\nMAC addresses in the iptables rules are always added by iptables_manager\nwith uppercase. It was done like that in [1].\nFor some reason, iptables on Centos 9 Stream (1.8.7) returns MAC\naddresses all in lowercase and difflib library treats such new and old\nrules as different ones so iptables manager always tries to remove and\nadd antispoofing rules which have MAC addresses set.\n\n[1] https://review.opendev.org/c/openstack/neutron/+/184355\n\nRelated-Bug: #1976323\nChange-Id: I63e78fdd030f063a1b943d481a9cbd98850552d7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c2d2744d5e841f431ae637624dd5954429f9607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6c2d2744d5e841f431ae637624dd5954429f9607"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
