)]}'
{"id":"openstack%2Fneutron~845098","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I46b1fe94b2e358f7f4b2cd4943a74ebaf84f51b8","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/1975674","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-10 10:32:10.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-06-10 08:53:02.000000000","reason":"Bernard Cafarelli replied on the change"}},"hashtags":[],"change_id":"I46b1fe94b2e358f7f4b2cd4943a74ebaf84f51b8","subject":"Defer flow deletion in openvswitch firewall","status":"MERGED","created":"2022-06-08 08:09:26.000000000","updated":"2022-06-10 10:33:48.000000000","submitted":"2022-06-10 10:32:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845098-bug/1975674","meta_rev_id":"418aaa330ef4611a601aa5343f69c07950422480","_number":845098,"virtual_id_number":845098,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-10 10:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-06-10 08:53:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-06-08 11:27:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-06-10 08:53:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":34915,"name":"Henning Eggers","email":"henning@eggers.name","username":"henninge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 08:09:26.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":34915,"name":"Henning Eggers","email":"henning@eggers.name","username":"henninge"},"state":"CC"},{"updated":"2022-06-08 08:53:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-08 09:43:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-08 09:52:53.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-06-08 11:27:49.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-06-10 08:10:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"ead1baadae94d67a66a4deb31ea73c8857047d6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-08 08:09:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4b239b9055ac4a62ea623023af6f7a5ad39c3cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 08:53:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5789edf34784c2996e2f98b9dd491c9 : SUCCESS in 38m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a77384d59ba42aabc30042354f15820 : SUCCESS in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a35b1629b26dd5d667a2f82929b06727d1e36cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 09:43:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/558cc816af6f4ea79a1c6644970463d9 : SUCCESS in 56m 22s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/7f690fd026f74d9fbb40b71a9c79d9fb : SUCCESS in 50m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5061a0dee5f04145b3cd24cd2f657a95 : SUCCESS in 28m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5eb3e36a0e14c37b6438d4bdaa1a484 : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d32895e598874257996c66d3a732840a : SUCCESS in 21m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d46b186704a475b940d68263095a532 : SUCCESS in 21m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0020306a5604f2da20087bf124ea81b : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f13e65d02ca4a608041d1ef00bc8071 : SUCCESS in 56m 51s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/06653d8278f44434b82ba263e3b3f063 : SUCCESS in 1h 31m 59s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/1353f2936c0240dc9f12cc845508a6bd : SUCCESS in 1h 09m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1efae946f8534e62988ce265e6361e6d : SUCCESS in 49m 48s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/48a31e15b1fc4fc995b5f4d9a6f7ef42 : SUCCESS in 52m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3f6c03b3d7c849f78019aeeec1d90c28 : SUCCESS in 1h 12m 18s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a2f29320894d4457b72a6b31863efa3d : SUCCESS in 1h 05m 00s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1828e59dd324413aa1caa44515d5745e : SUCCESS in 50m 33s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/583add956d9d4086926cecd03dee4d4a : SUCCESS in 1h 08m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a575f9f14c77cf060c0559a491e3fb0b9bf67a9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-08 09:52:53.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/845098/1 : SUCCESS in 1h 42m 50s","accounts_in_message":[],"_revision_number":1},{"id":"aaaea47d105290ea4cf6d587aaed9333a00714c5","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-08 11:27:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"72c5b8204aa3aaf8b410bf7f9a4b2c66e5f62a69","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-10 08:53:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f1d8ac5029696f474ba18f88d130ecf9ac8be233","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 08:53:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"856037e9e5dc456f6280e0f40cdbe123a8be6a5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 10:32:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bcb73a68be34ee88a0e16030eb65142 : SUCCESS in 11m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01c12364276346b5a5cdbf9656431b1a : SUCCESS in 30m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a5e4d59adb64969b6d37321aec42b77 : SUCCESS in 27m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7981e2096b944769b0ff6f03d8a2f33a : SUCCESS in 8m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/20b2ab0245bf46e78d97a0268be96e26 : SUCCESS in 47m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a2c2ad90d414a7a8698b189fa9e945d : SUCCESS in 1h 37m 39s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a08341ed141e4605a0ebb71111c98a3e : SUCCESS in 1h 20m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/13c8273194fa4127a87abc00115f52e0 : SUCCESS in 52m 44s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/fb68afe05f2546c58e0e655a18fb4e43 : SUCCESS in 58m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71db75c2d6924306b9850b000847d5d7 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8a2adbbec46fdec69462c91c1bdee51f95ae9f90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 10:32:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"418aaa330ef4611a601aa5343f69c07950422480","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 10:33:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bae74c6eaec34c1cad9e38a3ea88ba88 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0ca155596c3699a018993e065332a33801c6b81f","revisions":{"0ca155596c3699a018993e065332a33801c6b81f":{"kind":"REWORK","_number":1,"created":"2022-06-08 08:09:26.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/98/845098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/98/845098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/845098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/845098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/98/845098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/98/845098/1"}}},"commit":{"parents":[{"commit":"6b8c70902b2e5ce5786306021a2f9045519469eb","subject":"Merge \"[OVN] Protect the AgentCache iteration by copying the local cache\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/6b8c70902b2e5ce5786306021a2f9045519469eb"}]}],"author":{"name":"Henning Eggers","email":"henning@eggers.name","date":"2022-05-25 09:17:43.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-18 16:29:31.000000000","tz":0},"subject":"Defer flow deletion in openvswitch firewall","message":"Defer flow deletion in openvswitch firewall\n\nReduces the deletion time of conjunction flows on hypervisors\nwhere virtual machines reside which are part of a security\ngroup that has remote security groups as target which contain\nthousands of ports.\n\nWithout deferred deletion the agent will call ovs-ofctl several\nhundred times in succession, during this time the agent will\nblock any new vm creation or neutron port modifications on this\nhypervisor.\n\nThis patch has been tested using a single network with a single\nvm with a security group that points to a remote security group\nwith 2000 ports.\n\nDuring testing without the patch, the iteration time for deletion\nwas at around 500 seconds. After adding the patch to the l2 agent\non the test environment the same deletion time went down to\n4 seconds.\n\nCloses-Bug: #1975674\nChange-Id: I46b1fe94b2e358f7f4b2cd4943a74ebaf84f51b8\n(cherry picked from commit e09b128f416a809cd7734aba8ab52220ea01b2e2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ca155596c3699a018993e065332a33801c6b81f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0ca155596c3699a018993e065332a33801c6b81f"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
