)]}'
{"id":"openstack%2Fneutron~845232","triplet_id":"openstack%2Fneutron~stable%2Fyoga~I2c6f09548f39cadfe85e57032091a70a5bc978e5","project":"openstack/neutron","branch":"stable/yoga","topic":"bug/1977752","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-09 20:51:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2c6f09548f39cadfe85e57032091a70a5bc978e5","subject":"Add \"max-rate\" value to minimum bandwidth rules","status":"MERGED","created":"2022-06-09 07:54:51.000000000","updated":"2022-06-09 20:52:38.000000000","submitted":"2022-06-09 20:51:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845232-bug/1977752","meta_rev_id":"e5dba69beacd6bb468245b5eab6a93930abdf9e7","_number":845232,"virtual_id_number":845232,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-09 20:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-06-09 13:18:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-06-09 19:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-06-09 19:08:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-09 08:38:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-09 09:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-09 09:51:49.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-06-09 13:18:22.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-06-09 19:08:35.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"6e7fef09de782c2256b361ed15c43ad4afc18087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-09 07:54:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"694e232e1c4d99f7d305a4ebd20be61cec19a1f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 08:38:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a739fd5a1aad4856a9321e8d5fd7e691 : SUCCESS in 39m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79e5299edd92405baaf4cecf5b37a624 : SUCCESS in 40m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7158ff9acc45214562c7276dc76aef06d67e9dbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 09:45:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-scenario-linuxbridge-yoga https://zuul.opendev.org/t/openstack/build/3566038ff28a4998a28407505615f9f7 : SUCCESS in 55m 38s\n- neutron-tempest-plugin-scenario-openvswitch-yoga https://zuul.opendev.org/t/openstack/build/7016b51736f542999f6d8b5d457b6ce4 : SUCCESS in 58m 24s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-yoga https://zuul.opendev.org/t/openstack/build/0efa8c31454b42b4850467f101e26f37 : SUCCESS in 49m 04s\n- neutron-tempest-plugin-scenario-ovn-yoga https://zuul.opendev.org/t/openstack/build/08e510f928814bd394b0b11a187607a8 : SUCCESS in 48m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5956cd02b29f4092a315efd116f36c01 : SUCCESS in 31m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04243463d147487ba957be3789029cb0 : SUCCESS in 10m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/468b1bd779c04f56b56c7f2a3cb922f8 : SUCCESS in 21m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb39002e21984c31a2054b0b6b7e29e5 : SUCCESS in 22m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d72dbdc5274240a9d7ded660d6de8f : SUCCESS in 10m 02s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2e1ac59ba7bd4b5cafb288e031a3820f : SUCCESS in 46m 23s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/da6e974a14d3466884ea2f51df0d676d : SUCCESS in 1h 49m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/cf3da4ffeffc406bba05b551cf967621 : SUCCESS in 54m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d3ad1758d1884f43818662474815942f : SUCCESS in 46m 41s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/089eb98647a543bba6961b3177cbfc21 : SUCCESS in 51m 47s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/b2c41b5715af4778ab8da3196454a541 : SUCCESS in 1h 10m 20s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/2e96c6e00374491aaa24600b8ecc1627 : SUCCESS in 1h 05m 39s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ed596fa78f54ea888f45d704ad231aa : SUCCESS in 1h 03m 58s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/11a3261fad204d868b679808231c5011 : SUCCESS in 1h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"287705ceb7a1b7e89cc4a49be7f234db60e62404","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-09 09:51:49.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/845232/1 : SUCCESS in 1h 56m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1054c504982c15f65ada4ac9b6fef4da83940e9a","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-09 13:18:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c3c8dca81436eaf3e3493fddefc540898251ac1b","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-06-09 19:08:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"feac2657d071a082276189f9aad9b629a0406ebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 19:09:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fd37744af7fe4c6a2dbcc6b000b390a346dad0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 20:51:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/168336728a2d454ca423b8b745b56f1b : SUCCESS in 14m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/543b12e2a4bc4dea8640dacba889d2ee : SUCCESS in 21m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a1b591403d3451796f79f8280736b9d : SUCCESS in 27m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a1f2d61de96472d8cec0d96b793851a : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/61abb608fea44457a1a02163d854f7e0 : SUCCESS in 58m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4145914defa04d679ff248e4b9ef90d8 : SUCCESS in 1h 40m 25s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/694ab345897f4cd79803b38e4824d15e : SUCCESS in 1h 12m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e5307db624441ed8b27d347a0366246 : SUCCESS in 56m 47s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/51cf487999ae4a7f9bf3fb6e9dff1938 : SUCCESS in 57m 36s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67a8303eebc344cb80412d3c7d379ff1 : SUCCESS in 57m 51s","accounts_in_message":[],"_revision_number":1},{"id":"97e501686b7386238aa527fd5699853abb7b3be5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 20:51:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e5dba69beacd6bb468245b5eab6a93930abdf9e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 20:52:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9cf55a60015474c94f6d9fef90a8958 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5728045b843488554ab40dde890b561e44b206d","revisions":{"c5728045b843488554ab40dde890b561e44b206d":{"kind":"REWORK","_number":1,"created":"2022-06-09 07:54:51.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/32/845232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/32/845232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/845232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/845232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/32/845232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/32/845232/1"}}},"commit":{"parents":[{"commit":"bf024e743ff1beed1f6c21cc7ce5bd9376279dd6","subject":"Make irrelevant files for functional/fullstack more accurate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/bf024e743ff1beed1f6c21cc7ce5bd9376279dd6"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-18 03:43:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-19 02:44:37.000000000","tz":0},"subject":"Add \"max-rate\" value to minimum bandwidth rules","message":"Add \"max-rate\" value to minimum bandwidth rules\n\nOVS translates the QoS and associated queue registers,\nattached to a port, into \"tc\" linux-htb classfull traffic\nshaper, applied on the port.\n\nOVS creates a \"tc\" root class on the port device. On top of\nthis root class, the queues are represented as child \"tc\"\nclasses. In order to define a \"min-rate\" value (\"tc rate\"),\na \"max-rate\" value (\"tc ceil\") must be provided and higher than\n\"min-rate\".\n\nBy default, OVS agent QoS minimum rules do not have a \"max-rate\"\ndefined. Before this patch, any minimum bandwidth rule was\nlimiting the maximum rate to 100Mbit/s, that is the default value\nset in \"tc ceil\". This patch provides the maximum \"max-rate\" value\nfor any minimum bandwidth rule.\n\nCloses-Bug: #1977752\n\nChange-Id: I2c6f09548f39cadfe85e57032091a70a5bc978e5\n(cherry picked from commit a575dbc4a1aa7c44bba7ada4571a85d21120261e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5728045b843488554ab40dde890b561e44b206d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5728045b843488554ab40dde890b561e44b206d"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
