)]}'
{"id":"openstack%2Fneutron~845238","triplet_id":"openstack%2Fneutron~stable%2Fussuri~I2c6f09548f39cadfe85e57032091a70a5bc978e5","project":"openstack/neutron","branch":"stable/ussuri","topic":"bug/1977752","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-10 08:28:09.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-06-10 06:49:42.000000000","reason":"Bernard Cafarelli replied on the change"}},"hashtags":[],"change_id":"I2c6f09548f39cadfe85e57032091a70a5bc978e5","subject":"Add \"max-rate\" value to minimum bandwidth rules","status":"MERGED","created":"2022-06-09 08:28:40.000000000","updated":"2022-06-10 08:29:40.000000000","submitted":"2022-06-10 08:28:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845238-bug/1977752","meta_rev_id":"831a944fbfb716319a51f00c17fdea8b05e21249","_number":845238,"virtual_id_number":845238,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-10 08:28:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-06-10 06:49:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-06-09 22:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2022-06-10 06:49:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-09 10:53:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-09 13:25:18.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-06-09 20:37:54.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"CC"},{"updated":"2022-06-09 22:59:40.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"fe2a2565b95bd59200c2095210fe19da9963e8c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-09 08:28:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26b825ce82ba98f1ff8049ee717acec3e697f529","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 10:53:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58e9783cf16b4f1c8c338bd7454cb11d : SUCCESS in 36m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa9511efb47145439bb448249fe3c834 : SUCCESS in 14m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28cfcc46220641bfb9212ac2ad3e1519 : SUCCESS in 26m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/548de5324fef4e4d9a0cf880d169ef19 : SUCCESS in 29m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38b1fdc68dab4bea843fb29ab186204a : SUCCESS in 27m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed18abdc48ac4a1f952edbf0d485456b : SUCCESS in 10m 22s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/14723c724f104c6db676a498f9cc71c4 : SUCCESS in 57m 39s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/e443867e2c724c75b32fb987f7d2ca26 : FAILURE in 1h 12m 01s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/62c6499d6a644d62b9a82d967b5509a5 : SUCCESS in 1h 14m 32s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/54fe3da8d5f54263bbaae2f2fb14cb87 : SUCCESS in 1h 17m 00s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb0654d384e642bd9d3f82aa13b35a72 : SUCCESS in 1h 04m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/279cbf2c66794c95a3a56d44625f2702 : SUCCESS in 1h 14m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/956ad9e2e96c428ab2a9899b05f9e25f : SUCCESS in 1h 17m 27s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2d1593d555dc4915aab15eb4a0fd15bd : SUCCESS in 50m 20s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/833052ff214346a49036edf53d969ca9 : SUCCESS in 1h 04m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8896f0b4f4a248e7b9765da86be9b721 : FAILURE in 1h 17m 04s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/b1eaa90f68bc4976a77753a768aeaceb : SUCCESS in 1h 21m 52s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f946d25223d44ce9bd1f1ee9babbc397 : SUCCESS in 1h 19m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2e5ec435befe4d31be629f5fe28ca012 : FAILURE in 1h 06m 36s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ebab074505c1453cbeaaaf0259153323 : SUCCESS in 1h 38m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81d9ac4fcae94287b7635811b6805d6f : SUCCESS in 1h 06m 31s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/6fe62310e2244df3a2cf19364f759c8e : SUCCESS in 1h 16m 29s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/94f4980f2be749668d90b22458b0e491 : SUCCESS in 46m 27s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/aacb5e57eada489cafbb846597881fca : FAILURE in 36m 31s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1e7573bf8b334a08b802ca344c4adb92 : SUCCESS in 56m 03s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0a681f15877345d388973d3fc2c5746d : SUCCESS in 1h 49m 06s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/1355a739865e4c4bb01f917e051abd95 : FAILURE in 1h 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"065410b5b4a69c102937e87a0dcbfde3cb958413","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-09 11:03:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01e230280ca61ab40c122fe36e4065b982505b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 12:54:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/892c0e5241d5472e93ac966ce7bac66a : SUCCESS in 34m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20968f9610c94da9997bc15606b623fb : SUCCESS in 13m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/663b6e11ae604e149291ffd07b3ffaf5 : SUCCESS in 29m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/74792cd3ad554d749031dd323c1872a7 : SUCCESS in 30m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fd49a983726407a9eedbff618b2da4f : SUCCESS in 30m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c452f420259544aeab094dd2b3c4e144 : SUCCESS in 10m 15s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/21c1ee26552e4849a74e8f673060efd4 : SUCCESS in 1h 05m 40s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/ddae362371234b2bac512f9366215ee9 : SUCCESS in 1h 29m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a6020675fce34955a37f4bc6125c7f0d : SUCCESS in 1h 34m 05s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/c4e100a8e478465e9a8ae8ddac9e5aaa : SUCCESS in 1h 37m 47s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7814906c992140bb88be757819f07ca6 : SUCCESS in 1h 16m 08s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/286d60735023431ea8e432fd515c3483 : SUCCESS in 1h 17m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d3f7735b70234c4990c73aeac9e0f0c6 : SUCCESS in 1h 17m 58s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/99fcb9d3ed1c4084b59d5d4a0095ebda : SUCCESS in 1h 05m 28s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/a77ed65ac34347ce850ab35e93f335c8 : SUCCESS in 1h 21m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fac2a6037ead4c2b8a24c52887cd5e49 : FAILURE in 1h 22m 11s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7afc171088344a179790037fe88d459a : FAILURE in 1h 19m 43s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0ab593da22c6430dbfab915d6672d150 : SUCCESS in 1h 14m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2b2928f3c2494992b922b1ab35e96fed : FAILURE in 59m 01s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4bd3ab64ecf5465ea586efcc9689e6d2 : SUCCESS in 1h 29m 52s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6588580fa38f4ee19114299924f72637 : SUCCESS in 1h 05m 34s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/161ee823f03740fc94f8de00c830603b : SUCCESS in 1h 26m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/84d5b33c94f84845aa597ea2081993cf : SUCCESS in 39m 47s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/1183b66bf70844f6b8baa880ffe5d6c2 : FAILURE in 30m 20s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/71ceb1b62ba54d6b9d517f0e27e7a31b : SUCCESS in 59m 53s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/0f21fc5e1d144dbcb0b3f6f7fa4b6009 : SUCCESS in 1h 45m 54s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/11d9935dfac4422895f60568d7eda7a4 : FAILURE in 1h 22m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"997a50def924f7252c2edd26bb8acdcfb0ab7041","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-09 13:25:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45c736cdf6c34e4848d4c93a0972761533df7e72","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-09 13:27:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f5418488ebbe7aaa00172c2205afb01f954348ad","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-09 16:38:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1fc17d7175f25535b16ddd0e55c09c03511c8c6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-09 16:39:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9db114b47cd43388d790c28c66b200408916987f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 18:40:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac94681d27f4495cb217ca60c47ce4ca : SUCCESS in 25m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1ab0062f6864c84b37cfaed498f69a5 : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa7fc083866d43f282284b6656c48fd6 : SUCCESS in 27m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21540725d3a149d69f429206236f9821 : SUCCESS in 28m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cda762b655f40719b8018fda623cd6d : SUCCESS in 22m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee2d5d163ba7499aa82b4dba8e2db170 : SUCCESS in 8m 53s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/ab5f104ab8be4c44bdf147f4e5a7fd86 : FAILURE in 48m 02s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/821c055d034d48f1a282121c0df1e9a3 : SUCCESS in 1h 18m 19s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2566180f99314a1ba7bb62a9071500fc : SUCCESS in 1h 17m 28s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/bc3d6ad67aa34a368060be537f0c4cf9 : SUCCESS in 1h 13m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6702fed1f5b449f889bc7695bcb783c : SUCCESS in 59m 23s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/59bd8477b38d49f682d5fca582c67d2e : SUCCESS in 1h 03m 47s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a7ab1453d5194418bce8c0420d421e95 : SUCCESS in 59m 04s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2f4749c335f54b80a7f5e4074c9e5d2b : SUCCESS in 51m 33s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/00986c3a2306407ea7a43393f7b04842 : SUCCESS in 1h 08m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bf80d597ceeb4ff58b1e33a05c97fe9b : SUCCESS in 1h 29m 31s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/c3a2173edcf34cc7baf076278af18dbc : SUCCESS in 1h 17m 27s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4ed98353359e4f5c82bbcc84b163acec : SUCCESS in 1h 02m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/276d4b26746243f0a49e946fe7f2f2d2 : FAILURE in 52m 44s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d2f6d945887544ceb7a4531f8877c36d : SUCCESS in 1h 46m 14s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ce4911e435341998b16d6f90cd45ac5 : SUCCESS in 1h 04m 09s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/500a0c60aa9f4f4484ba0da36f6e35b3 : SUCCESS in 1h 08m 03s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/d83baca308654ac89a667e5d58d3534b : SUCCESS in 32m 34s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/84549aebc46c4f8a90c2675556f6876d : FAILURE in 36m 18s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/49b05f83b698461da1e92187172e3c78 : SUCCESS in 52m 59s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/f2424c2e845f45d98ddd5636938a1748 : SUCCESS in 1h 55m 43s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/f6cb3a33a941470b869005874a51f5e5 : FAILURE in 1h 17m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f0bab990dedd2d4dc7fcba4620670d8d6285c6d4","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-06-09 20:37:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"52c5e776349be01891b34cde79be7ecbe291d257","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-09 22:21:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25cdda0a411646fe9265b83b8179154c : SUCCESS in 35m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c47f0f832449baa66948f14edcaf17 : SUCCESS in 17m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84a8a3dd54f84e27b0d1b4378dd6fc00 : SUCCESS in 21m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b44f39228d04ef1a56a4e74c9368596 : SUCCESS in 26m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c96077f0432142f888d350e0bf8e29e8 : SUCCESS in 23m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee4a2fdfdcd47c089af823448508193 : SUCCESS in 8m 01s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/e80e5b99eea44d12a915392b5e03b543 : SUCCESS in 50m 44s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/21023f8c2b2b47ae93c4e933e9aaf01d : SUCCESS in 1h 22m 45s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f3db8710d2ba440994a4614300b92c42 : SUCCESS in 1h 20m 38s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/42d0e52891bb4ac0b0030961dd136b39 : SUCCESS in 1h 36m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9d185216ce29486b994c73e5b0972c93 : SUCCESS in 1h 02m 45s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/db431549655b49289fa9d83574254971 : SUCCESS in 1h 17m 54s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/dd2845832b3e48cd9127a22eaa14b9c6 : SUCCESS in 1h 35m 01s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/2ea1954c3d55412ca97ac4314ac3d85d : SUCCESS in 58m 52s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/221fd253ddd64acf8f43b578ec52fb12 : SUCCESS in 1h 02m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ee9ef1abe09c4009a6cec1bd06943c09 : SUCCESS in 1h 34m 40s (non-voting)\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9c4187037a2b43f7884f0e6e49c9526e : SUCCESS in 1h 19m 14s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f9c2468de83b4023b1ea104a46026751 : SUCCESS in 1h 08m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/739d40496f2146d483e1254e3f34dec7 : FAILURE in 55m 16s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df4fddfe4c694f1eb25d3d0a93130032 : SUCCESS in 1h 37m 24s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a4236d9446346bbaa69494f343bd24b : SUCCESS in 55m 42s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/309ef0667750447f9aa79f221cf76fce : SUCCESS in 1h 05m 49s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/fc04ee7f609449cf87279691b96bef67 : SUCCESS in 36m 10s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/40083c1415354c638e70a01dc5a854a9 : FAILURE in 39m 29s (non-voting)\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/efef141ee2f848b4b51cbcc5e1317d89 : SUCCESS in 50m 51s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ea0b8ae1637c425db6aed4ae7c14388f : SUCCESS in 1h 33m 39s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/a440aa87eb10497c8480239231396727 : FAILURE in 1h 29m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7d0f85c079be0f32cb85d09ec8303d3725cb7cb","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-06-09 22:59:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"28873b37687bd6957909d2bac8b5cc3fa5156087","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-10 06:49:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cac8d1c643490f46723f16bef42ca208804abe9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 06:50:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78a0df29466f8cc0e13212cdd59cd88620797923","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 08:28:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79e882f73181471eba4628c5cdd161bd : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dcf519b8d664272b23a4e0123ec21f3 : SUCCESS in 24m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/aa20728e1306482594c6cbf664b3b500 : SUCCESS in 26m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a91f2afd3234ff6a8707ae543c5a9f5 : SUCCESS in 8m 46s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/1bb0a92e6be04440ba5d6311ca9db872 : SUCCESS in 58m 17s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b714b3512849442985b32ddc23bf6ca9 : SUCCESS in 1h 22m 26s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/4d6e55c133be45ecb3d8384f3189c377 : SUCCESS in 1h 10m 51s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2a5ba7129791473eb5005d7c21aec128 : SUCCESS in 1h 06m 00s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ee90f1c8857547ac85335b775674fe58 : SUCCESS in 1h 09m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/872c563334ce4bb2be06830dae1af142 : SUCCESS in 1h 12m 06s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/65ebf28af8b8499ea82701421ecfa65f : SUCCESS in 1h 13m 02s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9ccce6e5ada248858cd6cbfc48dca18c : SUCCESS in 1h 32m 13s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad5a888cd1544cf9a99df607f8a97a95 : SUCCESS in 1h 05m 53s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/25a057b3143b4444980751b05f58149c : SUCCESS in 1h 06m 14s","accounts_in_message":[],"_revision_number":2},{"id":"029ee7b1e39e11a2af020d63c50a36ca6b4c1b09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 08:28:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"831a944fbfb716319a51f00c17fdea8b05e21249","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 08:29:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d6b849b17d404d8283f88f656e99c3 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0e7013af7a05976a5940dc3045a2bf56fc05c1aa","revisions":{"cd87a7cbd1c56e502046d223f8c01edd6a5c6856":{"kind":"REWORK","_number":1,"created":"2022-06-09 08:28:40.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/38/845238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/38/845238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/845238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/845238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/845238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/38/845238/1"}}},"commit":{"parents":[{"commit":"f7bc92cf22c6633dcdfdac7f14e4834a47e42eb2","subject":"Merge \"Revert \"Use Port_Binding up column to set Neutron port status\"\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7bc92cf22c6633dcdfdac7f14e4834a47e42eb2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-18 03:43:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-19 03:20:57.000000000","tz":0},"subject":"Add \"max-rate\" value to minimum bandwidth rules","message":"Add \"max-rate\" value to minimum bandwidth rules\n\nOVS translates the QoS and associated queue registers,\nattached to a port, into \"tc\" linux-htb classfull traffic\nshaper, applied on the port.\n\nOVS creates a \"tc\" root class on the port device. On top of\nthis root class, the queues are represented as child \"tc\"\nclasses. In order to define a \"min-rate\" value (\"tc rate\"),\na \"max-rate\" value (\"tc ceil\") must be provided and higher than\n\"min-rate\".\n\nBy default, OVS agent QoS minimum rules do not have a \"max-rate\"\ndefined. Before this patch, any minimum bandwidth rule was\nlimiting the maximum rate to 100Mbit/s, that is the default value\nset in \"tc ceil\". This patch provides the maximum \"max-rate\" value\nfor any minimum bandwidth rule.\n\nCloses-Bug: #1977752\n\nConflicts:\n    neutron/agent/common/ovs_lib.py\n    neutron/tests/functional/agent/common/test_ovs_lib.py\n\nChange-Id: I2c6f09548f39cadfe85e57032091a70a5bc978e5\n(cherry picked from commit a575dbc4a1aa7c44bba7ada4571a85d21120261e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd87a7cbd1c56e502046d223f8c01edd6a5c6856"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cd87a7cbd1c56e502046d223f8c01edd6a5c6856"}]},"branch":"refs/heads/stable/ussuri"},"0e7013af7a05976a5940dc3045a2bf56fc05c1aa":{"kind":"REWORK","_number":2,"created":"2022-06-09 16:39:36.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/38/845238/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/38/845238/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/845238/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/845238/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/38/845238/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/38/845238/2"}}},"commit":{"parents":[{"commit":"f7bc92cf22c6633dcdfdac7f14e4834a47e42eb2","subject":"Merge \"Revert \"Use Port_Binding up column to set Neutron port status\"\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f7bc92cf22c6633dcdfdac7f14e4834a47e42eb2"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-18 03:43:40.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-06-09 16:39:36.000000000","tz":0},"subject":"Add \"max-rate\" value to minimum bandwidth rules","message":"Add \"max-rate\" value to minimum bandwidth rules\n\nOVS translates the QoS and associated queue registers,\nattached to a port, into \"tc\" linux-htb classfull traffic\nshaper, applied on the port.\n\nOVS creates a \"tc\" root class on the port device. On top of\nthis root class, the queues are represented as child \"tc\"\nclasses. In order to define a \"min-rate\" value (\"tc rate\"),\na \"max-rate\" value (\"tc ceil\") must be provided and higher than\n\"min-rate\".\n\nBy default, OVS agent QoS minimum rules do not have a \"max-rate\"\ndefined. Before this patch, any minimum bandwidth rule was\nlimiting the maximum rate to 100Mbit/s, that is the default value\nset in \"tc ceil\". This patch provides the maximum \"max-rate\" value\nfor any minimum bandwidth rule.\n\nCloses-Bug: #1977752\n\nConflicts:\n    neutron/agent/common/ovs_lib.py\n    neutron/tests/functional/agent/common/test_ovs_lib.py\n\nChange-Id: I2c6f09548f39cadfe85e57032091a70a5bc978e5\n(cherry picked from commit a575dbc4a1aa7c44bba7ada4571a85d21120261e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e7013af7a05976a5940dc3045a2bf56fc05c1aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0e7013af7a05976a5940dc3045a2bf56fc05c1aa"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
