)]}'
{"id":"openstack%2Fneutron~845366","triplet_id":"openstack%2Fneutron~master~I0d1bc49927b061a91ff6d4ca3f5b826d529d0770","project":"openstack/neutron","branch":"master","topic":"use_sql_in_clause","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-14 16:49:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0d1bc49927b061a91ff6d4ca3f5b826d529d0770","subject":"Use SQL IN clause instead of OR with exact match comparisons","status":"MERGED","created":"2022-06-10 16:13:41.000000000","updated":"2022-06-14 16:51:35.000000000","submitted":"2022-06-14 16:49:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845366-use_sql_in_clause","meta_rev_id":"2e3c49b3afaa327ef6159170bc97091eb6982f5f","_number":845366,"virtual_id_number":845366,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},{"value":1,"date":"2022-06-13 08:33:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-14 16:49:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"recommended":{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},"all":[{"value":1,"date":"2022-06-14 12:10:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-14 08:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-06-13 19:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-06-14 07:54:54.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-14 08:09:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-10 17:02:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-10 17:55:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-10 19:30:50.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-13 19:46:50.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-06-14 07:54:54.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-06-14 08:09:51.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-06-14 12:10:04.000000000","updated_by":{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},"reviewer":{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},"state":"REVIEWER"},{"updated":"2022-06-14 13:36:16.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"}],"messages":[{"id":"c58711b1ffc00ac27b35d2d08d8f3deb04188e27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-10 16:13:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d8d0bb1c79912bbced1531cfbde24f4837e219d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 17:02:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5e08e0d84bc4421be55d5b955c61990 : SUCCESS in 40m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/070b712c6725466caa44620ffd872a30 : SUCCESS in 46m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adb2506df8ce0486823918f589ac7faff4c0c368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-10 17:55:11.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/ac54cfa285994c489d645e84233ed481 : SUCCESS in 1h 33m 49s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b24bed58d28f46f0b8e6d63116a539f0 : SUCCESS in 1h 26m 53s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/5a70183afcd744259046ccec3a02a7c1 : SUCCESS in 1h 24m 52s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/a3841cb2c2c8447287c3f6b2b80fbc4b : SUCCESS in 1h 26m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4e6b7155b376438d8bec3dd8c7764da7 : SUCCESS in 37m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d64f21f98b0c431ea42da89195fdc5a8 : SUCCESS in 37m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf31c43672f341a3ace16013f7198587 : FAILURE in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9a4b4303b344028b27b262ccb777e73 : SUCCESS in 28m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65a7333191ef4036956500078c6a8fb7 : SUCCESS in 21m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd1bb637405b4f128ed8d53124dfae0e : SUCCESS in 26m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ca565e78b7401baf8c601b11808bab : SUCCESS in 9m 19s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3e1a418d710647eda848e214461d4136 : SUCCESS in 1h 00m 55s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/599a29df17c14c8a941ebb355bd4bee5 : SUCCESS in 1h 36m 16s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/bb7c56636dac4f10a19c9c28fa335420 : SUCCESS in 1h 21m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6904276cff6f45d58a72a6a8d4bd56a6 : SUCCESS in 1h 02m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/76f8506ebba945d381868e270396ea7f : SUCCESS in 1h 03m 24s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6d43eee6007248fd95b839b5fe05e002 : SUCCESS in 1h 13m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/9a8c63b42bfe4805be5c42d40c49d844 : SUCCESS in 1h 15m 46s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/428331c9bbf344dc97837d403db71024 : SUCCESS in 1h 05m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ca6b23508ce54d2a84d1122de11cde0d : SUCCESS in 1h 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4112cac6fbf0037561e5bba5b4b2cd680f3fa8a0","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-10 19:30:50.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/66/845366/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"00bf8506da954bb0c7ea0be30ed24082fe0a5e9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-13 06:35:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0cf3d1b4d0a4a1c97bd1eb2bd63778004e92ea8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 07:17:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2733e5fa589460e9ce3ebe21af85069 : SUCCESS in 40m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f47f9d8b7234525b46b74f5394e0d7c : SUCCESS in 40m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46401a723dcba2fb1c9d0a0893217341a0d0ddaf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-13 08:33:44.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/66/845366/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"352c526c5268cc26edcf70480a7317dafdc25249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 08:43:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/2665fb0212d94dd08ec4c5a24dc19a3a : SUCCESS in 1h 31m 41s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/aa3106c0417e4392b442a7f5f6946453 : FAILURE in 1h 27m 46s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ff2277254d2146e6adc2bf3bff949f08 : SUCCESS in 1h 31m 47s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/62bef7c26158470b971c2acf7db51591 : SUCCESS in 1h 37m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/965826693ca14d168e0786fbfb74802a : SUCCESS in 27m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96f549e1b4a845e79d197c58475a9683 : SUCCESS in 40m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed3b41333b6645dd8cc2fd391d3abafd : SUCCESS in 12m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12e734b328c8401f9321e3417ae5593a : SUCCESS in 22m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6236312739ca4dda87c1bca870634d7e : SUCCESS in 26m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a94115e4bee845029df0ae559e896c7d : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb475f989948410980aa19ce96d45ac2 : SUCCESS in 7m 58s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/61ed01b96bc24215b352539e68964755 : SUCCESS in 1h 08m 25s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/93717b5839e14c7fbd8839b17fd561fa : SUCCESS in 1h 26m 20s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/ac2d34eb19f84289a7a6a1721be6c976 : SUCCESS in 1h 10m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ad3212799da4e5b9c46510a12a4f70e : SUCCESS in 42m 22s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f9c83ea369f2410ab50b4502ea275d49 : FAILURE in 34m 46s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2e32afb8302644e7950bd9916763e556 : SUCCESS in 1h 10m 47s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/a66b7ce3c6c645d8b9910d25df94f7f4 : SUCCESS in 1h 20m 07s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/143cc800d0114ba7a1e9621b978c38ed : SUCCESS in 56m 22s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/cb9a4e073d4946829206e74361ebc44b : SUCCESS in 1h 10m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b83bb2de7e2931ca755e8a9fff5129cdaeaef9e9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-13 16:31:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d88ccff3576a1ae872b14c5cf9ceee3eb0cc3306","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 17:15:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9026f81456754b348577a9dded75e2a5 : SUCCESS in 35m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb2fb4331eca49a1a658eac0c8a933bc : SUCCESS in 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"af82da86ec065a22b3931499dd8cf9bf910edfbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 18:49:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/b7d74b8b9ed24ed4b9592631f897aac7 : SUCCESS in 2h 05m 33s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/036a036121ad4651a8f17804956a1910 : SUCCESS in 1h 25m 14s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/4ef61f6e10ce40b68f3544be6d03b682 : SUCCESS in 58m 02s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/cad84b635d274fb29ef6993229133aa3 : SUCCESS in 1h 05m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d940e54bbd1b40739e7a39d8aa207fb9 : SUCCESS in 28m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99f248dacda8480a9dee647dea6d7140 : SUCCESS in 38m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c30bef66b442bf8cc33b8e4ed49750 : SUCCESS in 15m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec78ff6e3fbd48138c6bda136a5a6596 : SUCCESS in 27m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d8c3ffdd84f4953acb412f1a7656f57 : SUCCESS in 26m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/165e2cbf3a2e4a7c81830093124f3596 : SUCCESS in 20m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc72af1c34704e1a8588e8ddd619defc : SUCCESS in 9m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8c50c4b42ea14416834fd6ee129ac482 : SUCCESS in 49m 50s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/db58fd906d46416b87f6d4d01e85259b : SUCCESS in 1h 32m 35s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8b43c844542f423ba8f8070cde2a393e : SUCCESS in 1h 01m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ef071f2e73b42f4a221c90c29051111 : SUCCESS in 56m 36s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8d34cc0377d24f2f8878578781c0ad2e : SUCCESS in 54m 26s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1ef44c6ee1c04018b8110bf3b55a1343 : SUCCESS in 1h 14m 51s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/3d19cbc2d2ff4c7e85b630d50cf5d89e : SUCCESS in 1h 08m 20s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68081c478e604678bd46da55e6dd679a : SUCCESS in 54m 04s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/c6cc342c2cca4e52ae96405354f79a00 : SUCCESS in 1h 03m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a24f931bddc16cfad08d2246e3a8e9bd45f48af","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-13 19:46:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1eedf30c388ced53bb2edc3eba14f2f1b0cab1af","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-06-14 07:54:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"904fed06b9537ddd4206c5c28380241ead7fb941","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-14 08:09:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6ac6a55906a465685553916fbcd5174d16bc74e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 08:10:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b1632a48ec7f625d47cb7bf158f61a3b534cf093","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 09:58:48.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/9ebf93849d1b41e19ea602ba8cb3fc96 : SUCCESS in 1h 35m 13s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/494d0b2ebf41421fae95c2d6e6108091 : SUCCESS in 1h 43m 21s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/154752b23d054a5c92e83e5557444547 : SUCCESS in 1h 30m 00s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f7422a3c2c2e4e0aae7b71956409b141 : SUCCESS in 1h 37m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a060dcf26f7845b98e633957aa9108ca : SUCCESS in 13m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f201c4890def475596f5df2c095b04d2 : SUCCESS in 29m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/219b8d6896da478baa438081bef4fe4f : SUCCESS in 29m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/622f0ccee9dc4a5090619d3173872401 : SUCCESS in 10m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/69fb3c4bf0994ecc809919593d83c845 : FAILURE in 49m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b42d8e65e0eb4c76a1ffde7519daaf69 : SUCCESS in 1h 36m 30s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c86e2eee25e74d2e984680121ec09f68 : SUCCESS in 1h 27m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5a318820b6ef496e9a23dfb39fd5b431 : SUCCESS in 1h 07m 02s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/a48a0be55135464ba8829e182d469d0d : SUCCESS in 1h 07m 06s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6fb1be5a0584c18a8cee4f8895b340e : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e7d554cec82644d8e6eceff747e63a92103f559f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-14 11:44:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"34898041a03609911c37c15286caadb4282f067e","author":{"_account_id":22865,"name":"Jaganathan Palanisamy","email":"jpalanis@redhat.com","username":"jaganathancse"},"date":"2022-06-14 12:10:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4618b52dbb53b9035631645f02889a765891e9ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 12:30:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7de3d6dbe0724c8385b3b778e5e8818a : SUCCESS in 43m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59fdc8a87afe4e69b291480065a42839 : SUCCESS in 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"457d04e22215782b011c88740f131b2712cb33b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 13:19:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/dd15d5d2b88f47679e3df77e02c5ef04 : SUCCESS in 1h 23m 47s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/b42b20db7566418dae2940ef723b9383 : SUCCESS in 1h 30m 54s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/146754fec2d64a66a3f9149a79e0aa96 : SUCCESS in 1h 24m 08s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/f87084e2f3574353b2434b2d2a06f21c : SUCCESS in 1h 19m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/09c15141163141faad09e3d23a2169a1 : SUCCESS in 36m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca27c5ac6c70475080cc2637b9c8e985 : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4439afdd9f4fd49d5873c25906377a : SUCCESS in 12m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a47d0d26ca294efc952e65b8b6c6e1b9 : SUCCESS in 27m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1803d585f294a35b45b15761ff73678 : SUCCESS in 20m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8e86bd387ac9464e8144b70ddb5a9cbf : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfd1ac2d1bad479ba673a1208792b427 : SUCCESS in 9m 04s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/34f5ed28d1924308a26d5b2a8de47d15 : SUCCESS in 59m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2b891afed959405fa97290ad4213a1a6 : SUCCESS in 1h 28m 15s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/f0dcdf6867104f3a8dac6e3999dc962e : SUCCESS in 1h 16m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/090a28abc56d4ae6812df71a2a34f663 : SUCCESS in 1h 07m 53s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/057fe67882e94ce4944103e54dc693b0 : SUCCESS in 1h 04m 18s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/2907cbffe60545b587d8cd5de77a611b : SUCCESS in 1h 26m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/552aedae198d4fe49dacb17f93b77862 : SUCCESS in 1h 15m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c9fa2f79de245ef9a4e1a501b6d05db : SUCCESS in 1h 08m 21s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/b93d1f4f9b434ca59f3856a147b42f66 : SUCCESS in 1h 13m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9ca789e4c00bb9367ad866285ca0e7c6582cee0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 13:20:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5aaf789de453f7e1acc027cf6452228649c6cf90","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-14 13:36:16.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/845366/2 : SUCCESS in 1h 50m 13s","accounts_in_message":[],"_revision_number":2},{"id":"214130cc7f1a84001a22aa880db366f45cfee0f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 16:49:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/92f5b3c98f434b6bbc2be9b1af0fb36d : SUCCESS in 1h 10m 07s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/5a55ba2f60d54cfda118e9786bac9777 : SUCCESS in 1h 04m 00s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/c32e74ace0ec4a70bb197192e81f8e7e : SUCCESS in 2h 07m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/eb884517ff6e48628981eee9e15286cb : SUCCESS in 1h 50m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5110c5baf2e94a09927502a8e2a779b8 : SUCCESS in 13m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e7e03b5872543219b5370b7be8c0e18 : SUCCESS in 21m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95217273402f4717b8c7cd95ac3d918f : SUCCESS in 25m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2008b22916cb40e3af688ec167a96719 : SUCCESS in 11m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/45f387a8d7e5427ab186a4622445e6c5 : SUCCESS in 1h 02m 11s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/f9ff6680aeab47e1b946e5a9cfc6c2ff : SUCCESS in 2h 04m 21s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a4662d1e9bd64f93a0afea8d1b024cef : SUCCESS in 1h 22m 40s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c6c69027c0f42c6abc33d1a8da36347 : SUCCESS in 57m 33s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/29624cdcbe984f2c98b6fe7031a7e848 : SUCCESS in 1h 02m 27s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/666b5eab250143f7bbf6215e47c9f1a5 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":2},{"id":"2d069a6943fa1614a69342564b94ce9636d874f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 16:49:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2e3c49b3afaa327ef6159170bc97091eb6982f5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 16:51:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb1454c6daf84d74a2b33047e307ff46 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1d135e3c0865d825127947c65632551436e8894","revisions":{"95b45e77d417c3b4c5b03b04fad6197c0284a76b":{"kind":"REWORK","_number":1,"created":"2022-06-10 16:13:41.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/845366/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/845366/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/845366/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/845366/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/845366/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/845366/1"}}},"commit":{"parents":[{"commit":"38cdfb88558887b830514b4ef390fd535c11942c","subject":"Merge \"\"_handle_lb_fip_cmds\" to handle the LB members correctly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38cdfb88558887b830514b4ef390fd535c11942c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-19 20:57:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-19 20:59:09.000000000","tz":0},"subject":"Use SQL IN clause instead of OR with exact match comparisons","message":"Use SQL IN clause instead of OR with exact match comparisons\n\nSQL IN clause is faster than several exact match comparisons\nin an OR clause.\n\nTrivial-Fix\n\nChange-Id: I0d1bc49927b061a91ff6d4ca3f5b826d529d0770\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95b45e77d417c3b4c5b03b04fad6197c0284a76b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/95b45e77d417c3b4c5b03b04fad6197c0284a76b"}]},"branch":"refs/heads/master"},"a1d135e3c0865d825127947c65632551436e8894":{"kind":"REWORK","_number":2,"created":"2022-06-13 06:35:25.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/66/845366/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/66/845366/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/845366/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/845366/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/66/845366/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/66/845366/2"}}},"commit":{"parents":[{"commit":"38cdfb88558887b830514b4ef390fd535c11942c","subject":"Merge \"\"_handle_lb_fip_cmds\" to handle the LB members correctly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/38cdfb88558887b830514b4ef390fd535c11942c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-19 20:57:11.000000000","tz":0},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-06-13 06:35:25.000000000","tz":0},"subject":"Use SQL IN clause instead of OR with exact match comparisons","message":"Use SQL IN clause instead of OR with exact match comparisons\n\nSQL IN clause is faster than several exact match comparisons\nin an OR clause.\n\nTrivial-Fix\n\nChange-Id: I0d1bc49927b061a91ff6d4ca3f5b826d529d0770\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1d135e3c0865d825127947c65632551436e8894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a1d135e3c0865d825127947c65632551436e8894"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
