)]}'
{"id":"openstack%2Fneutron~845552","triplet_id":"openstack%2Fneutron~stable%2Fxena~Ib51764778a666050c42de0dfeb9bf9b185d44bb7","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1973347","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-20 17:04:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib51764778a666050c42de0dfeb9bf9b185d44bb7","subject":"Skip \"PortBindingChassisEvent\" if revision number changes","status":"MERGED","created":"2022-06-13 07:09:49.000000000","updated":"2022-06-20 17:06:24.000000000","submitted":"2022-06-20 17:04:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845552-bug/1973347","meta_rev_id":"014c6c880042539bd9f7666c4dc1ce2ff02289e5","_number":845552,"virtual_id_number":845552,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-20 17:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-06-17 16:25:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-06-20 12:13:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-06-20 12:13:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-13 08:03:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-13 09:05:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-17 16:25:12.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-06-20 12:13:53.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-06-20 16:53:55.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"}],"messages":[{"id":"8d4f065a0a99370a7daa9d114856685dca167326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-13 07:09:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2767e8cc1de96398fbab29ad6fbf0cfd4882f21c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 08:03:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74d37261458943039e79b8d2ca0bafbd : SUCCESS in 51m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab299865c1824f6f9cdcb5c24069e515 : SUCCESS in 49m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9dd3c62627c133326659d0e6a4dbe12c4ba02302","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 09:05:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/a08e6d8ad81b4fa0bc0e8f9940127fe6 : SUCCESS in 54m 38s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/10a2eec9d621434abe1a83fa9c379ddd : SUCCESS in 56m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c8ed017361d4c6e9f3c3ba6c1049a49 : SUCCESS in 35m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4fd01f691d843aab8440b54e36f5fc7 : SUCCESS in 15m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a22899ea487345c5ad8591ad1f432d9b : SUCCESS in 27m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8a92d3f4bcf403e939001de79301b74 : SUCCESS in 25m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f19a67c088e4ede9693468b56efe7a4 : SUCCESS in 24m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6262d070f92400fabad965eadccdae2 : SUCCESS in 9m 14s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/5a8e8ec926c843e49ddbc5c34a693009 : SUCCESS in 38m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db56c03732b344758848e1ec1a2f79d6 : SUCCESS in 57m 32s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/77cc6eb90448439c8eba27ac83803b73 : SUCCESS in 1h 20m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4674bf07c91a3048a6bc3d901c9ede7d4f815832","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-17 16:25:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f8f63e82bbbc3c468aa6769722d78cb99206de99","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-06-20 12:13:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35f1a2c6ebbb19522d61a80b32e22c36af9dca6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 12:14:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24f74016690f15a0d96c99eee73286f8238aaed6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 14:35:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/043052718ca44d96ac2bbfb7135e68c1 : SUCCESS in 48m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46654cfd599a457798ca894ffd60f63e : SUCCESS in 12m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f162ebafc20a4aa6b1bfefba5f42ac6b : SUCCESS in 25m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d5282ea644d4bae9a4f33551a08eebe : SUCCESS in 23m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e3c5c5225264d03a34eeed8423a1d45 : SUCCESS in 11m 07s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1750020f356140c49464a4dc241404af : FAILURE in 48m 24s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a79bb84bcb6c490297829513ba1102f0 : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ebafc9186deb956dd501ed733e10bc3e3dddf9b0","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-20 14:50:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1fdcd5ab5675ff84a395bf5bb954f6d1e342ed9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 16:03:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad49890089e04f87b165c420e9483202 : TIMED_OUT in 1h 01m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c437fc1021947759753d679738517b9 : TIMED_OUT in 1h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7008775a8dafaf64a94ae2661a35f92851a4fc29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 16:07:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/505fbab1f9f44b0695880a39639c22e8 : SUCCESS in 38m 37s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/759f96141ab14ebfb52355996fe51b0d : SUCCESS in 50m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c6b62f003bc4bd0a811a93f0204ea4d : SUCCESS in 46m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9773b41776b9434eb03aa86a44678606 : SUCCESS in 11m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca497a0ee065439f85d0ac6b4d3f102a : SUCCESS in 33m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2c0c1d72f0b411e8b472dfa6fc066c9 : SUCCESS in 24m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c793e3bc20c0464892b70865ec2a816b : SUCCESS in 30m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dae3f09825847f996ca99db34b28920 : SUCCESS in 9m 33s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8647d7ab541a44ada99654ae15758b89 : SUCCESS in 1h 05m 33s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c0c21129d774118a792dffd6d5b9ad2 : SUCCESS in 46m 48s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/5ebafde00a544144a1f63c7a3f89d0bd : SUCCESS in 1h 05m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa56e4a794b63e41b541bbe6a79ac9685c58c30b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 16:08:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e77a82a5526ad9c334bed6d7c9fb59c3e4aade2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-20 16:53:55.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/845552/1 : FAILURE in 2h 02m 52s","accounts_in_message":[],"_revision_number":1},{"id":"91cd001b4e3b20e26e67056ced22a855464fc97d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:04:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/8815436a4b984e699a4c89e71fb93e44 : SUCCESS in 52m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fba1a171fb94457dac496ddffd903446 : SUCCESS in 12m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e19b819b199941ea8b665512383a0a37 : SUCCESS in 20m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b959eac3a40a4a27aee52c22a42d5a39 : SUCCESS in 19m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51b28e3126c141e08a2465cafcbd4803 : SUCCESS in 9m 44s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a7446bb201949fdb78bd7c2d0b84ce3 : SUCCESS in 42m 15s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/068b98d7df474702b160d80379d47ccb : SUCCESS in 51m 43s","accounts_in_message":[],"_revision_number":1},{"id":"0e25960a8a7c3028b120e37e05d1a78f01b52b6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:04:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"014c6c880042539bd9f7666c4dc1ce2ff02289e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:06:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8321ca71cd184defb94a550265c57827 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ace772787e9d00608c30b2dd321469c1ced8318c","revisions":{"ace772787e9d00608c30b2dd321469c1ced8318c":{"kind":"REWORK","_number":1,"created":"2022-06-13 07:09:49.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/52/845552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/52/845552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/845552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/845552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/52/845552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/52/845552/1"}}},"commit":{"parents":[{"commit":"9a23fb369980429d7f1a393142736f004a7d7cbf","subject":"Merge \"Notify mech drivers with original and modified network.\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9a23fb369980429d7f1a393142736f004a7d7cbf"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-13 00:54:27.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-05-19 22:19:30.000000000","tz":0},"subject":"Skip \"PortBindingChassisEvent\" if revision number changes","message":"Skip \"PortBindingChassisEvent\" if revision number changes\n\nSince [1], the \"external_ids\" of the NB Logical_Router_Port register\nare copied into the SB Port_Binding \"external_ids\". When a change\nin a Port_Binding register is received, if only the\n\"external_ids:revision_number\" is changed, we skip any update on the\nrelated Logical_Router_Port.\n\nIf not, that will lead to an infinite loop: Neutron will update\nthe Logical_Router_Port with a new revision number and OVN will\ncopy this new revision number to the SB register, triggering again\nthe update of the NB Logical_Router_Port\n\n[1]https://www.mail-archive.com/ovs-dev@openvswitch.org/msg62836.html\n\nConflicts:\n    neutron/tests/functional/base.py\n\nCloses-Bug: #1973347\nChange-Id: Ib51764778a666050c42de0dfeb9bf9b185d44bb7\n(cherry picked from commit 32e8303b3b21e047abcf365c3999cb7379467b0c)\n(cherry picked from commit b7fb963d91af8247c5624d7f1753d5969658b616)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ace772787e9d00608c30b2dd321469c1ced8318c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ace772787e9d00608c30b2dd321469c1ced8318c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
