)]}'
{"id":"openstack%2Fneutron~845844","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Id174cd944c3b883b749046bc49d00e011eb891fe","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1946764-stable/yoga-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-06-15 19:20:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id174cd944c3b883b749046bc49d00e011eb891fe","subject":"[OVN]Any string type dhcp options should be quoted","status":"MERGED","created":"2022-06-15 13:42:52.000000000","updated":"2022-06-15 19:22:25.000000000","submitted":"2022-06-15 19:20:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845844-bug/1946764-stable/yoga-stable/xena-stable/wallaby","cherry_pick_of_change":845843,"cherry_pick_of_patch_set":1,"meta_rev_id":"4b6a077859564f5fc3bba870c994fd54c1be47f2","_number":845844,"virtual_id_number":845844,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-15 19:20:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-06-15 13:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-06-15 13:44:10.000000000","_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-06-15 16:45:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-15 13:42:52.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":28056,"name":"Liushy","email":"liuxie_11@163.com","username":"liuxie_11"},"state":"CC"},{"updated":"2022-06-15 13:44:10.000000000","updated_by":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"reviewer":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"state":"REVIEWER"},{"updated":"2022-06-15 13:54:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-06-15 14:29:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-15 15:32:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-15 15:37:52.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"}],"messages":[{"id":"01d863947113ff6dcaf4a198a1521b3610a18170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-15 13:42:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"6bfcb449bc8fea2d1356d9fc75b60c3f4ec2e15f","author":{"_account_id":6773,"name":"Lucas Alvares Gomes","email":"lucasagomes@gmail.com","username":"lucasagomes"},"date":"2022-06-15 13:44:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bef3174d97e505ce345a07afadeb2ab4798d0263","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-15 13:54:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"414526153e211d2b921bf733ea5a3d8e7399035f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 14:29:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f36004899ba455c8e4d454d86b860c4 : SUCCESS in 45m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ca6d9791a2446938bd4011a034b094b : SUCCESS in 43m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26420bba82f699eed6812eb9c32bf819532086b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 15:32:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/6e62ef75e98c4eba98e8c0e770d3758a : SUCCESS in 51m 22s\n- neutron-tempest-plugin-scenario-ovn-wallaby https://zuul.opendev.org/t/openstack/build/58891793aab34d5182f78aa57e4c01ea : SUCCESS in 1h 02m 45s\n- neutron-tempest-plugin-designate-scenario-wallaby https://zuul.opendev.org/t/openstack/build/f62f0efdea194006bac8587d908cafad : SUCCESS in 27m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffade106e9a2400b8918bc7564930edc : SUCCESS in 40m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3203d63f3fd94e4088cff3f48b52226b : SUCCESS in 12m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cef7c2951d914094a592d0b23ce6aae9 : SUCCESS in 27m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/429a1790c1f54d929547c3be1c8c3bba : SUCCESS in 27m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/030925c14de34596b6bc7470288c8b7a : SUCCESS in 26m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74fce9a24e2a4bd58f1134116f23428e : SUCCESS in 9m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/9b1c6590ee584a6b878e4b57aab98244 : SUCCESS in 51m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4635f9bf62924a4cafcec62dfa4ed632 : SUCCESS in 1h 14m 30s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/69550e7077f14c44990432cbe9f16bd7 : SUCCESS in 1h 14m 54s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/9fcedc38ce334f178dac7afd8d2e6a36 : SUCCESS in 33m 31s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/aba458503dd742be9d161d0344c2bcc0 : SUCCESS in 1h 41m 34s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/62265e8ed84b43de8d01939e47e9d6f8 : SUCCESS in 1h 22m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e8c9b53f6dbe37ca9720e11562bc1a6994d540a7","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-15 15:37:52.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/845844/1 : SUCCESS in 1h 53m 36s","accounts_in_message":[],"_revision_number":1},{"id":"99811b44f953836c150b044132b6b91faaa28bc2","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-15 16:45:40.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae41cd62504b921ae4f3517670b36929bd6567db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 16:46:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b34bbd6ab8e201c9ac2a6ea756e4a3b32cecea1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 19:20:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- neutron-tempest-plugin-api-wallaby https://zuul.opendev.org/t/openstack/build/bfd68d92d54f4797b4ea04356ac36d31 : SUCCESS in 58m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1ebf292bc5142538077cd08f69dc14a : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd8d577640ad4060b21a4fce19bd8ee4 : SUCCESS in 24m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7577c387aed4bfaa38fb97b1f7dca98 : SUCCESS in 23m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f683d46c7184ad6be1b761706b3200d : SUCCESS in 10m 30s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/2fb33d5821f44abb8954f9aa67b78a1d : SUCCESS in 43m 56s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/413fb895d61e4eaeba1cc6b11bc5d2b7 : SUCCESS in 2h 01m 16s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/18a366739419463b9d373500ac314603 : SUCCESS in 1h 02m 34s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/4e9dafa1048b49caae051d3c2cbef26f : SUCCESS in 30m 40s\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/e425406ce41943fd8635f3deca31f32e : SUCCESS in 1h 24m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ec7d5d351ea37bc4bf46177301ab1950331fb790","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 19:20:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4b6a077859564f5fc3bba870c994fd54c1be47f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 19:22:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae815dfd0257493383a7cd39a13161c3 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1e7d20b01e452efc234b1f169ee356333adf499","revisions":{"f1e7d20b01e452efc234b1f169ee356333adf499":{"kind":"REWORK","_number":1,"created":"2022-06-15 13:42:52.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/44/845844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/44/845844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/845844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/845844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/44/845844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/44/845844/1"}}},"commit":{"parents":[{"commit":"f633a893e70b285feb10be07bdc74ccd05a98aff","subject":"Merge \"migration: Fail task if executed script failed\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f633a893e70b285feb10be07bdc74ccd05a98aff"}]}],"author":{"name":"liushy","email":"liuxie_11@163.com","date":"2022-05-11 03:06:50.000000000","tz":480},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-06-15 13:42:52.000000000","tz":0},"subject":"[OVN]Any string type dhcp options should be quoted","message":"[OVN]Any string type dhcp options should be quoted\n\nAny dhcp option[1] which value is string type in ovn\nshould be added quotes, otherwise it would be regard\nas constant.\n\n[1]https://github.com/ovn-org/ovn/blob/main/lib/ovn-l7.h#L116\n\nCloses-Bug: #1946764\nChange-Id: Id174cd944c3b883b749046bc49d00e011eb891fe\n(cherry picked from commit 18c32e4339cbcb0993d9c162e2715f6fcfa932cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1e7d20b01e452efc234b1f169ee356333adf499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f1e7d20b01e452efc234b1f169ee356333adf499"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
