)]}'
{"id":"openstack%2Fneutron~846545","triplet_id":"openstack%2Fneutron~master~Ib16c83eabc9ecdca52edfbb3129f67da373817b9","project":"openstack/neutron","branch":"master","topic":"dnm","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-06-21 05:22:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib16c83eabc9ecdca52edfbb3129f67da373817b9","subject":"[DNM] Test enable-fips with nslookup_target set","status":"ABANDONED","created":"2022-06-20 06:15:31.000000000","updated":"2022-06-21 05:22:35.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d2eb404282b62823ec1d90e75f986c9df9149b1e","_number":846545,"virtual_id_number":846545,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-20 07:44:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-20 07:44:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-20 08:15:44.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"}],"messages":[{"id":"e39d35655ba31ef61bc1e3a0d4365f0c9becee45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-20 06:15:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"781bb0772a49ee24d13eb6a782d0011d64a15840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 07:44:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- neutron-ovs-tempest-fips https://zuul.opendev.org/t/openstack/build/f1d7ff7818754b71b4e8eb850f9c89eb : FAILURE in 1h 23m 45s\n- neutron-ovn-tempest-ovs-release-fips https://zuul.opendev.org/t/openstack/build/3534421cda794aaeaeba86356d598315 : FAILURE in 1h 04m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3fd0acdccb140d1d619c4b834e0c3fce31c306d9","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-20 08:02:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7184ad57ee7bcb9897818d7b4fbe7767bfc39039","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-20 08:15:44.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/846545/1 : SUCCESS in 1h 59m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d2eb404282b62823ec1d90e75f986c9df9149b1e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-21 05:22:35.000000000","message":"Abandoned\n\nBeing fixed with https://review.opendev.org/c/openstack/neutron/+/846001","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f758334a847c5db5f5775bd76ee3009d845347c","revisions":{"7f758334a847c5db5f5775bd76ee3009d845347c":{"kind":"REWORK","_number":1,"created":"2022-06-20 06:15:31.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/45/846545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/45/846545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/846545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/846545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/45/846545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/45/846545/1"}}},"commit":{"parents":[{"commit":"92b70efca072cdf3b2271e6c7cf7131a136c4508","subject":"Merge \"Remove unneeded contexts in ``_create_port_security_group_binding``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/92b70efca072cdf3b2271e6c7cf7131a136c4508"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-20 06:15:05.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-06-20 06:15:05.000000000","tz":330},"subject":"[DNM] Test enable-fips with nslookup_target set","message":"[DNM] Test enable-fips with nslookup_target set\n\nChange-Id: Ib16c83eabc9ecdca52edfbb3129f67da373817b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f758334a847c5db5f5775bd76ee3009d845347c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7f758334a847c5db5f5775bd76ee3009d845347c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
