)]}'
{"id":"openstack%2Fneutron~847005","triplet_id":"openstack%2Fneutron~master~Id41e7c1a67ea9a8fcda1d5d331ab8192aff7dc48","project":"openstack/neutron","branch":"master","topic":"bug/1905391","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-08-01 12:42:47.000000000","reason":"\u003cGERRIT_ACCOUNT_11975\u003e replied on the change","reason_account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},"31976":{"account":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"last_update":"2022-08-19 14:53:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id41e7c1a67ea9a8fcda1d5d331ab8192aff7dc48","subject":"Add vpnaas to extensions supported by ovn","status":"MERGED","created":"2022-06-21 11:44:35.000000000","updated":"2022-08-19 14:55:18.000000000","submitted":"2022-08-19 14:53:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847005-bug/1905391","meta_rev_id":"657a4e661a914717cea806b56cf4d0bed0da60eb","_number":847005,"virtual_id_number":847005,"owner":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-06-21 12:27:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-19 14:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-10 06:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-08-08 23:17:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-08-01 12:42:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-08-08 23:17:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-21 12:27:09.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-21 12:37:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-21 13:18:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-21 13:41:28.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-07-27 14:24:41.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"CC"},{"updated":"2022-08-01 12:42:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-08-08 23:17:16.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2022-08-10 06:02:14.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-08-19 10:05:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"}],"messages":[{"id":"a005579e3b0a02e2cb6c2ff74cba6b796e03b209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-06-21 11:44:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92a4b2157ba178339ba0957728ee64b9ee0b15f9","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-06-21 12:27:09.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/June-2022/Arista_Tempest_Testrefs/changes/05/847005/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"6917891d430a899df5a75e044544150ba1a3d2a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 12:37:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ff2d84373da4d1ea811f89ead6ae195 : SUCCESS in 43m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01161de134f14b789ca18d0d463191b6 : SUCCESS in 50m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31b7c34bd926c4745d83400ae8fd72c610e75113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 13:18:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4d586d9318d14c0d96d1c5ab63dd2082 : SUCCESS in 1h 01m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf9add07d7ba46aa9204c321a7a63191 : SUCCESS in 26m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17c34aa3ccc04d20bc93c5db834488be : SUCCESS in 35m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec580d1e8b54440bc6a1de78935fb6c : SUCCESS in 12m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51681a38d121452cbafdba3090075758 : SUCCESS in 22m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6682f57ee248482a8883476fe987903d : SUCCESS in 22m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6e19af7a80c4c1cb0983a602d060310 : SUCCESS in 25m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/896003658d9645048d51bc3207ceed8a : SUCCESS in 9m 12s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/c7c8352b3a024579a256e1971be0b0cb : SUCCESS in 1h 02m 39s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6d4733a5a6464bd7aa380c5ad5b374ca : SUCCESS in 1h 31m 45s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84eb0201c8254285a4edb4870a159079 : SUCCESS in 56m 45s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/fd2521fa59a64175b187d7e2c994ef6b : SUCCESS in 1h 05m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edff23b06fc7084379c01413fe395bdc8234a21b","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-06-21 13:41:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/847005/1 : SUCCESS in 1h 56m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5386e683496a655f4acf0113cd86abaf018d80dc","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-27 14:24:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4868fe170f6cbc47df46a6f0ae6e5fd85b11013a","author":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"date":"2022-07-28 11:25:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6524a4bfa452aaedbd7d32c4af532cfa25aa431a","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-08-01 12:42:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"62023f6f9d1baadd3e746a7b22c49d42e6dfe7a9","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-08-08 23:17:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9aa3531e6ef9edb0edb21ec387f8e72c0be56d33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 23:17:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96fc811f5ecbc3c0c902ac18ac26be72ea28e8e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 01:21:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ef8925c4158479a81522a6e562919fa\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4e17b83f97be4623bb4d7e9f95d3934b : FAILURE in 27m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cc44b0fa9bb4c81b94f09777e744b0f : SUCCESS in 13m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d497bbee67c14cf1b7422703e0c85fd9 : SUCCESS in 20m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34e957b66b4d459a89d5f8f6c874faa7 : SUCCESS in 20m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5561d90d45e84309bbff1287d1b66715 : SUCCESS in 8m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/e16a7bbc5b454655a810858777de2e74 : SUCCESS in 57m 54s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/4349c8e534ad4da998e2a83d30ebaa2d : FAILURE in 2h 03m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66c1787486f9498fb7e83608eb462ee7 : SUCCESS in 53m 16s","accounts_in_message":[],"_revision_number":1},{"id":"95ef8a662c4608b8ee64aaa4ee90c364e1f577a5","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-08-10 06:02:14.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efc3023cbf4bbc7c0e7b34f39f204ed2bd9a64eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 06:45:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0ef39ee8f8b4511843cb1c3ed08a0d0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba5db378ba1a428c8fabb446b4a95536 : SUCCESS in 41m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba3e6a21510041bdaa14f2864988f85d : SUCCESS in 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a52b6efe1bbabf17591333ebd9c31bb7c789272b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 07:39:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a15a0717a6d7482b9dc4621094e3b666\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/117fad3056b84ac196d3245af667e5a5 : SUCCESS in 1h 34m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/677915281f6c426e92c2e5992f83135e : SUCCESS in 33m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58bf360c94b842b5b5c9445ef48acde6 : SUCCESS in 30m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b40c1153eeb6479c8963464ca4b7e000 : SUCCESS in 11m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78876ae92bcb4cdba90c032853dd1fb0 : SUCCESS in 23m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c62dcb9ca3e04ac1a04518e37042ab68 : SUCCESS in 25m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9af52d68811a4e228f47f355cb54ffe8 : SUCCESS in 23m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff18c6a9934941759ab6c902eaacccdd : SUCCESS in 10m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8427217744f64145a8404b5b6878430e : SUCCESS in 1h 04m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/5802cf9c2cbd4816a4002d5ce51fc0fb : SUCCESS in 1h 35m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9a7d7c59be94fd68b61ffbdac47cab1 : SUCCESS in 1h 02m 09s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ddc83e6b6e634eef81262b61fc5d3b1a : SUCCESS in 1h 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e08566fa1af8d55d138e434ab706cef31a6e3996","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 07:39:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b03a94353a546e57256428eead3dc8bd100b839","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-10 08:24:01.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/847005/1 : FAILURE in 2h 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d49ad152cc9f5eb1418d44f66ade48b521f1e8b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 09:24:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/765d79a0b82d4e489391183b23b2e8da\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/4980e0f7464f4af0b2ea5d0f23277da9 : SUCCESS in 1h 43m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8a6b2000e2467a8c02aee2d79ccd3f : SUCCESS in 15m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8479db1fd394d66820c89406143041a : SUCCESS in 30m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fde8f0a26574f3caa6cae7a3e4d431f : SUCCESS in 27m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619299b98f9d4dac8db54892fd2dc669 : SUCCESS in 14m 10s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ca26b4a25a0f44bda50d7a9b182b6736 : POST_FAILURE in 1h 21m 57s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/b983d60dfa624e7b9a7a9b6cc77d95eb : SUCCESS in 1h 38m 04s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c799d8f530848079e92d19e7a43a2ac : SUCCESS in 1h 13m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bf10d642ac0fec417c364e0380f099271426a2f2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-19 10:05:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e266b35685c83b6adc900fd3e8052faf2405636d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 10:48:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ed23e2a99d419f8bbb77ab082e1c74\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/119f22cb71004bb6af392c790146bc34 : SUCCESS in 41m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90117181b45343849bd30e551c40e38c : SUCCESS in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"566240cb084cdb357d0cb5d68c6029e96322b4e0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-19 11:43:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/847005/1 : SUCCESS in 1h 37m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f8c865e7f9d0c93712c1b686e6871af08b669c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 12:10:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14ba504952dd45369eddbcc9df4192fe\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/fb7bca05442148e1b1606a04c8ef338f : SUCCESS in 1h 18m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc87a69cf6ef42a1a48ba7deed4c9363 : SUCCESS in 30m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/063dea05921447f78d77c141582700e4 : SUCCESS in 32m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12e400758eb242549639fd7cdeab8d4b : SUCCESS in 16m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec50fbac19c342c09ca7db32e1270ca8 : SUCCESS in 26m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce05c5461c704b82aeac2904fc14cbff : SUCCESS in 23m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2c7a38676a143cd82ed03a0162d482b : SUCCESS in 37m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be0e1e5491384cf2bd35db6c2b1c040a : SUCCESS in 9m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/bdf9088d2d3c4c589b5ad0b9ed1b5756 : SUCCESS in 1h 07m 34s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9f5b6ecfde2d4cacb10e30a2072ea331 : SUCCESS in 2h 03m 10s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e007da5c73eb4b4f8f0eb4f61b3efc9d : SUCCESS in 1h 04m 00s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ab68a072b8974167abda745363706cae : SUCCESS in 1h 07m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ca40841ef9f4d19892f384265ce8cc5c4141d80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 12:10:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01bb1b8a0d87dcfd78a7612cd9d80b01d141d6cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 14:53:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d6479d1919748c49887bc8697822231\n\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5bd73018f7df4d6a90abeea469d4ebb1 : SUCCESS in 1h 11m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cc5450f2155447ba058b9d4b9400bea : SUCCESS in 14m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5411718bf33c4dbda2cf24cdbd4d0027 : SUCCESS in 29m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a11a2620196474494ca430361999ba4 : SUCCESS in 29m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15059d4981814f42b5383d51a77645ac : SUCCESS in 9m 59s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/554e2f0350dd4864b57348aa17e08234 : SUCCESS in 1h 08m 21s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/7d903b1ac0d549cca7dc146750117767 : SUCCESS in 1h 43m 28s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80042aa45dfb4397881c6ffd4ab44a3c : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":1},{"id":"feec743fdf7f5ae3b2b9b67af0b2d01c8b443269","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 14:53:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"657a4e661a914717cea806b56cf4d0bed0da60eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-19 14:55:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b5026db1207444e8a9b26b484173f4e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c955aac1954c0c964aa30d5fdd0cd9 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7ea4909d98679715c059ce4f44c952ca6d985e5","revisions":{"a7ea4909d98679715c059ce4f44c952ca6d985e5":{"kind":"REWORK","_number":1,"created":"2022-06-21 11:44:35.000000000","uploader":{"_account_id":31976,"name":"Bodo Petermann","email":"b.petermann@syseleven.de","username":"bpetermann"},"ref":"refs/changes/05/847005/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/05/847005/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/847005/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/847005/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/05/847005/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/05/847005/1"}}},"commit":{"parents":[{"commit":"e867a9b31da66531ee9f96f146d16cc0f1b7cd3f","subject":"Merge \"[OVS][QoS] Unset the min-bw QoS from the phys bridge interface\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/e867a9b31da66531ee9f96f146d16cc0f1b7cd3f"}]}],"author":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-06-21 11:43:11.000000000","tz":120},"committer":{"name":"Bodo Petermann","email":"b.petermann@syseleven.de","date":"2022-06-21 11:44:33.000000000","tz":120},"subject":"Add vpnaas to extensions supported by ovn","message":"Add vpnaas to extensions supported by ovn\n\nThis addition is required to run a devstack setup with ovn\nthat includes vpnaas from the proposed patch [1]\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #1905391\nChange-Id: Id41e7c1a67ea9a8fcda1d5d331ab8192aff7dc48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7ea4909d98679715c059ce4f44c952ca6d985e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a7ea4909d98679715c059ce4f44c952ca6d985e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
