)]}'
{"id":"openstack%2Fneutron~850039","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~I90a130830a6a76cd68adfb6503bb268a60d7de20","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1938766","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-07-25 09:48:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I90a130830a6a76cd68adfb6503bb268a60d7de20","subject":"functional: Add debug messages around OVN services","status":"MERGED","created":"2022-07-19 09:16:01.000000000","updated":"2022-07-25 09:49:45.000000000","submitted":"2022-07-25 09:48:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850039-bug/1938766","cherry_pick_of_change":803936,"cherry_pick_of_patch_set":1,"meta_rev_id":"d820656e3d47604470a3065cf57df491e48604db","_number":850039,"virtual_id_number":850039,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-25 09:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-07-24 20:14:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-07-25 08:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-07-25 08:32:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-19 09:16:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":8655,"name":"Jakub Libosvar","email":"libosvar@redhat.com","username":"jlibosva"},"state":"CC"},{"updated":"2022-07-19 09:17:23.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-07-19 09:53:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-19 10:24:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-24 20:14:39.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-07-25 08:32:05.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"ed52eb51c536b95b15b2055ee91ac564fa37c1c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-19 09:16:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2adfde39b689a8841e47e1d1350a69262d985437","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-19 09:17:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/850039/1 : FAILURE in 43s","accounts_in_message":[],"_revision_number":1},{"id":"a31345c34b9d80671effc14932f4f4ff4204330d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 09:53:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f2b86d1cc484675b8e4c2cf99e9515e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2af556e698b34b1b9822c635c4907c5c : SUCCESS in 34m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df43fee2968e40a0a34734ae2867aa8c : SUCCESS in 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"116f3a27a4d0dbf6d41e05654236215b78db212e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 10:24:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0afba48d05fe461e9a243d969cea21d6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13466792b89b496d8ccc6501b3d23070 : SUCCESS in 13m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96653e50511d4019b521432bf9c6e769 : SUCCESS in 29m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d2e244378e7495cb9365bb0d434aab0 : SUCCESS in 8m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d251dcae84d4c5e850358677872ecb3 : FAILURE in 1h 00m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8be4d90c336d4c7bb10b636bcad449c16e261a2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-22 16:10:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"80b57e4a8bf9ca4145c492daa9b347ed36c1710d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 17:00:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a24296ec96a402bb117040ddec7f986\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccad9a9f1f064340aec0f3f481ba6783 : SUCCESS in 34m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74c78b05b24148358f490b1b67e57a49 : SUCCESS in 48m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf481d9d08731944d9d5b8db67f1f0586ee0fd50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 17:24:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4d52cbd41304a8e96fe1009c6212fb7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81cc817c9b8a4fbfbc7190d1d9df2fc9 : SUCCESS in 14m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e657f799a714ff59736219c198d1c52 : SUCCESS in 26m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d8dd7ae41747dcb27404ca8514b262 : SUCCESS in 12m 17s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/675d2dd851e64cd7b852e1b50c26fcbd : SUCCESS in 1h 12m 02s","accounts_in_message":[],"_revision_number":2},{"id":"eba881ce9a4e02a8715db10659b06204231796cf","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-22 18:33:24.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/850039/2 : SUCCESS in 2h 21m 58s","accounts_in_message":[],"_revision_number":2},{"id":"074600c2eb08e658c1b5fa695bf4a1bbad049959","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-07-24 20:14:39.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e46f8395d0055f9a2aeeaa331ecb8af19929f4ed","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-25 08:32:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"74061f4d3df64e7e441c87deade2c95cd1985de7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 08:32:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f0f96a171302216511d665e2bdef51234036a179","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 09:48:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ee7f1d288bf4fb8bf524f5c460dad22\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c392bb9ef432409ea549f0fa3f41afc8 : SUCCESS in 12m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff845edcbb514fc8ba5f288eafcf0779 : SUCCESS in 28m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b7757603404d308a652d28edd04cd9 : SUCCESS in 11m 34s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/4525ab3d97344eaa966dd79e5da3d4d7 : SUCCESS in 1h 14m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3cc50f8f5099aead314586789fb2913bba8d47d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 09:48:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d820656e3d47604470a3065cf57df491e48604db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-25 09:49:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8de6d1e8df9d4c73b5b27d96bbf9c6d7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f559911c3aa446d6a1a3c360ade772b8 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b183e12fdc1b79c5140abc1b70562c53860e82d6","revisions":{"09599d72dab75eaaba31cd899ece44d874e6c499":{"kind":"REWORK","_number":1,"created":"2022-07-19 09:16:01.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/850039/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/850039/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/850039/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/850039/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/850039/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/850039/1"}}},"commit":{"parents":[{"commit":"a146f58aefe86a43e01b35eada041f5622ef5bec","subject":"Merge \"[OVN] Sync QoS policies\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a146f58aefe86a43e01b35eada041f5622ef5bec"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-08-09 15:26:52.000000000","tz":120},"committer":{"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","date":"2022-07-19 09:16:01.000000000","tz":0},"subject":"functional: Add debug messages around OVN services","message":"functional: Add debug messages around OVN services\n\nThis patch is to help understand LP bug 1938766.\n\nChange-Id: I90a130830a6a76cd68adfb6503bb268a60d7de20\n(cherry picked from commit c267125e20d78f3d9eaedf743049432cf6c3dc23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09599d72dab75eaaba31cd899ece44d874e6c499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/09599d72dab75eaaba31cd899ece44d874e6c499"}]},"branch":"refs/heads/stable/wallaby"},"b183e12fdc1b79c5140abc1b70562c53860e82d6":{"kind":"REWORK","_number":2,"created":"2022-07-22 16:10:27.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/39/850039/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/39/850039/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/850039/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/850039/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/39/850039/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/39/850039/2"}}},"commit":{"parents":[{"commit":"80d408cd960f6900c07e0661d4b0ed97ed86f06f","subject":"ovn: Wait for northd in functional tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/80d408cd960f6900c07e0661d4b0ed97ed86f06f"}]}],"author":{"name":"Jakub Libosvar","email":"libosvar@redhat.com","date":"2021-08-09 15:26:52.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-21 12:08:49.000000000","tz":120},"subject":"functional: Add debug messages around OVN services","message":"functional: Add debug messages around OVN services\n\nThis patch is to help understand LP bug 1938766.\n\nConflicts:\n    neutron/tests/functional/base.py\n\nChange-Id: I90a130830a6a76cd68adfb6503bb268a60d7de20\n(cherry picked from commit c267125e20d78f3d9eaedf743049432cf6c3dc23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b183e12fdc1b79c5140abc1b70562c53860e82d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b183e12fdc1b79c5140abc1b70562c53860e82d6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
