)]}'
{"id":"openstack%2Fneutron~850147","triplet_id":"openstack%2Fneutron~master~Ic6bd7bfadcba8deb132d8af3e295ec25a8d64b50","project":"openstack/neutron","branch":"master","topic":"segments","attention_set":{},"removed_from_attention_set":{"7730":{"account":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"last_update":"2022-07-21 10:23:51.000000000","reason":"Change was submitted"},"29759":{"account":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"last_update":"2022-07-21 10:23:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6bd7bfadcba8deb132d8af3e295ec25a8d64b50","subject":"ovs: improve log message when ofport is not configured","status":"MERGED","created":"2022-07-18 09:45:33.000000000","updated":"2022-07-21 10:25:44.000000000","submitted":"2022-07-21 10:23:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850147-segments","meta_rev_id":"3f11f4c51bdc56a5b1512d6fddee6e5e34d26c86","_number":850147,"virtual_id_number":850147,"owner":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-07-21 08:04:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-21 10:23:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-07-21 06:22:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-07-21 06:54:34.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-21 06:22:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":-1,"date":"2022-07-21 06:22:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2022-07-21 06:54:34.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-18 11:13:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-18 11:22:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 13:31:01.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-07-18 19:50:45.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-19 08:09:54.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-07-20 08:06:42.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-07-20 08:16:12.000000000","updated_by":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"reviewer":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"state":"REVIEWER"},{"updated":"2022-07-21 06:54:34.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"}],"messages":[{"id":"de7f0541bce12701b81ed03b857d4ddf7d567a13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-18 09:45:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcdab5f02862b9fb09d9bb95b9071354d8abc57e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 11:13:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43ca9fc1a2b34c73a9cb60533ae0f3c4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14bbe51548c547118b3093bf4a052ecf : FAILURE in 1h 13m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31ba8820093246219bc11d96d2b57a9f : SUCCESS in 51m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"969bde19cadc6b2362f9666b09498e5b2d48d719","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 11:22:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b5712eefa6c438499bb9f8a69251015\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/2464b728e8944b8daa37c26ee6cf5617 : POST_FAILURE in 1h 34m 13s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/c633d26ad00e45dfb4168382472aeb8b : SUCCESS in 1h 08m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca8180701547497ea973b1fd71d2fb79 : SUCCESS in 36m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db5085c85419426ea1b16b039f429f46 : SUCCESS in 13m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ac15bef9dc04c52b8759a24391e151d : SUCCESS in 24m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21b93180a74645bd908499e510d96958 : SUCCESS in 27m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ab342a622e248a19beb0d789a697649 : SUCCESS in 23m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a051eb688fbf4e2189c6954f9a3e4dab : SUCCESS in 8m 38s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77ee149545e24817b12aad7dd6882a15 : FAILURE in 53m 44s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9c6f89ffe51c41128d60b2eedb13a6cf : SUCCESS in 1h 28m 44s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/83961c2460f74582bade3817e51bf63d : SUCCESS in 1h 06m 08s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/adb98ef1f5e0428db2bbe56d71311e1f : SUCCESS in 57m 39s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e68aee3a9de14a20a4981e8a3d49dadc : FAILURE in 1h 09m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/cf66d5d3099f4055a754cb080cb8623c : SUCCESS in 1h 17m 00s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/15ff6396a03941aa93fed980b786d754 : FAILURE in 1h 16m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9b7b9b6f97da1efc495181a359a48f527002affa","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-18 11:53:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f43540ccaecbe1d3c363aa560358babf35c170cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 13:07:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5ccc69044964f53a9821e7e955532d8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23da4eea5fa2494195422bd770b32d31 : SUCCESS in 42m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a265b0a843f745478d961d721c2fa67e : SUCCESS in 44m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a23601f5631466a184964a88d7c49c396d87c723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 13:25:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/519b309f95b64c048a4e9a8f9da214bc\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/c929237b704e440a918638d7c009dd06 : SUCCESS in 1h 30m 57s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0e50902fecd64c3a87228419e84f10cb : SUCCESS in 1h 22m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5528f2bd22ae40b3b1c737aaa4267878 : SUCCESS in 29m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11ebd27e50843df8fdc736158d2ab23 : SUCCESS in 11m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f60af9cdd547416cae7a695097b92564 : SUCCESS in 27m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/638693ed76984d0a89e62373d6c9c166 : SUCCESS in 27m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/33bf31d320ce42938ac5132d266cfd9b : SUCCESS in 22m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfddedeff5d94fa68f19bb88333be908 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/7c838b6caf9645c08ccce5b19cb731fd : FAILURE in 52m 04s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/56228f9c314547aea586b3a11f5c1684 : SUCCESS in 1h 27m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/37bbaafd8f8e4aa5aa5244a1e99c6fca : SUCCESS in 1h 16m 52s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd9eeb2ec05a4cb9bccda7a08e2b657d : SUCCESS in 1h 08m 09s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/f1e0d4e62c214ee9a4a9ab899d95f9c9 : FAILURE in 59m 20s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/1511ea71f2224218942ec6495937261f : SUCCESS in 1h 21m 37s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/0775cdcea12844b18deb2a4a34f179e2 : FAILURE in 1h 24m 56s","accounts_in_message":[],"_revision_number":1},{"id":"201a0551bca2710583eda8f412dc2c33218835fd","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-18 13:31:01.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/850147/1 : SUCCESS in 1h 46m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0b357ccab5e8526522954ff154611d59ed5352d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-18 14:14:48.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f70d35e064551301f55ac656feeab7fbaab53b5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 15:26:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a888c82c82ed4882b808092ebb68d3a8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0035346343cb4ee69adeb0e1ea11a1fc : SUCCESS in 58m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00eaaf53e70847ef9e6781a4e10d13f4 : SUCCESS in 39m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"48246eed3b6488de1668f51f27a25138fdb3163c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 15:52:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a3fb63d15b94e84ad01933d1507edd4\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/e8777fad49c9485fa7d46965bf471a22 : SUCCESS in 1h 17m 49s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/29e6d1eaaf3c4533a3b7b2ed8a386414 : SUCCESS in 1h 31m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80f9676976ef42a78e5ee8cb49529fe5 : SUCCESS in 33m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e91b56fb0924f7f997a871eb01cd9f5 : SUCCESS in 12m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21e3a92405e4412985b4429e24cca716 : SUCCESS in 21m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41a577c47bc4454abb051f992770ba73 : SUCCESS in 23m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c5f0d6c354a64a798990d260fbce28c4 : SUCCESS in 21m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af65c855884444e6ad238d43d4f51255 : SUCCESS in 10m 21s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/8e2af52ea910403db6d248a00d86b2ca : FAILURE in 55m 20s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8bb7c176c2394b4783fcc518c434218f : SUCCESS in 1h 35m 31s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/8891c53928654bfe927189354a9a73c5 : SUCCESS in 1h 19m 54s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fada53f3ea494b6f8cc04296e4bce253 : SUCCESS in 1h 01m 31s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/bb7655c6661f45b294d7125f8982db5e : FAILURE in 1h 02m 03s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/3f11718200bd4eb5ba036e7c7fde04b6 : SUCCESS in 1h 20m 06s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/7e3e8886d9954ac78421aaa0c9b8ab34 : FAILURE in 1h 17m 53s","accounts_in_message":[],"_revision_number":2},{"id":"50c104632d9af142e0702e935e1edbbe9d3e8dbc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-18 16:21:01.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/850147/2 : SUCCESS in 2h 05m 14s","accounts_in_message":[],"_revision_number":2},{"id":"5c3fb1670fd56da1b9d8325879509367dc32badf","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-18 19:50:45.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/47/850147/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"63bac8a3e2df7b4144fcc64caf332dca8b8e740d","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-18 21:34:42.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/47/850147/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"3ed7044beedafefff283733405686c9bcc2ea139","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-19 08:09:54.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e1bff7e2d2c0c7fdf70dd0ab786419d6c01f204","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-19 20:00:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"e6883efa771e596646b9e37137d222bcd01c500f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 21:22:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fce8e5f840c4fa488bad965c9d0d4f9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f110e6fa7174435bfd5e35a599982a5 : SUCCESS in 41m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7f003c39ea1d4138b563de4346d69432 : SUCCESS in 39m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"335881a005e2e4ee5da0c99cef1e6cb65466126f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 21:48:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e825e43ccebb4bbd9f735b30593eb116\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/4ac61e99536047b0b06386e0c3ad92e4 : SUCCESS in 1h 30m 08s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/6d6b8e3116554c4c9b5fe6f429756f6f : SUCCESS in 1h 12m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ba184835dfd4ba78fff4d5f447b36f8 : SUCCESS in 36m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5f939c1b3574bc8aeb4867d76651c68 : SUCCESS in 14m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f88d29d4e24413bbffc02bc74dd0ff3 : SUCCESS in 27m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c176eafbd4614008ae57804cda177c6b : SUCCESS in 28m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4409ea7fb5a54090a2e05bcc69c6c2e9 : SUCCESS in 26m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3f88a7fe84495c9326c24f0a0781c9 : SUCCESS in 9m 57s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ee7c4384872b4c059b62f19cf9d32694 : SUCCESS in 1h 08m 22s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1594ddc5da354a6483442979a56db308 : SUCCESS in 1h 41m 55s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/4f69343f5fb240379f9cdb8acf186c84 : SUCCESS in 1h 05m 39s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6e714b7bc614f44904e19a7f43d8a2e : SUCCESS in 1h 10m 12s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c7fc384c290f44d89022ec3029182f7b : SUCCESS in 1h 02m 33s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/be34d0617746427fbf0b01e1bfe9118c : SUCCESS in 1h 31m 26s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/1d13a00ee87549e5a3c37ac8b706b3e0 : SUCCESS in 1h 21m 16s","accounts_in_message":[],"_revision_number":3},{"id":"3c526dd13a20d28b7e34daf86f78d817c8736142","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-19 22:07:01.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/47/850147/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"a703adfb2e5d94a6ea038ec036e3a5b480be8d72","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-19 22:17:34.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/850147/3 : SUCCESS in 2h 15m 22s","accounts_in_message":[],"_revision_number":3},{"id":"b5d2bc9fb3d229773853b58f1e0641311363a52d","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-07-20 08:06:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"be54f65195df7a9623552778749512f7f051d355","author":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"date":"2022-07-20 08:16:12.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dba1fbcac5b652f50ed9af39e6c8931712babdf0","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-20 11:06:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"62579b575c6449956c76f814cb7e417941f4cb14","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-20 11:07:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2cc158f7bb973e881c7d862832a92a3f418a01e1","author":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"date":"2022-07-20 14:34:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb053d8e1a4859df24b06ac1ecf98e4b9e3d5b65","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-21 06:04:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"46214f035f4e37a51c477ccc3729356f6c0602e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-21 06:22:55.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"8cc38395ccbfa7c990b8fed98a99283a85bc6124","author":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"date":"2022-07-21 06:25:16.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c4fa680497ace89edf48a671d90317cfaf3f7070","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-07-21 06:54:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd1a5cf36bcc718b16af468028d5935b58a78cb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 07:36:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3c8b3cdb0404ddfaeb1cd86f105554d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d785908a270495baf482b1dcd816ba1 : SUCCESS in 55m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78987acbb8a44995984eddd7e7b54acc : SUCCESS in 50m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ba2c215063251a97efdf49d0e69901504a599bd6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-07-21 08:04:20.000000000","message":"Patch Set 4: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/July-2022/Arista_Tempest_Testrefs/changes/47/850147/4 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":4},{"id":"f1256ffa1b9a5719121615927f6c5212741716d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 08:14:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bc7897a8ccf4955b8773ee33f31aafd\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/edb6adbf1a384b91b18805461085b74d : SUCCESS in 1h 44m 07s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ea83cff5a73e40e59d05180740a950bf : SUCCESS in 1h 37m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/806bfb14497c46018dcf8245ce52272a : SUCCESS in 32m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8cfdb045a340bd852df2eaa684dd37 : SUCCESS in 11m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b3f91feff684f82a757a822f9f8c222 : SUCCESS in 24m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ec635e7623f40a3a5733f2a9c11d2d9 : SUCCESS in 23m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b846ebb68e6444f0a9e04bfc5716ddcd : SUCCESS in 25m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a72bbd4630c44548ee86442cd0d5376 : SUCCESS in 10m 29s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/77cb6c8111884c68b3127c15a85d8d62 : SUCCESS in 59m 40s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/66fa526e930946c5a029903bda194996 : SUCCESS in 1h 49m 47s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/2d263bbfbd7d40ac90477a2ea9540e6b : SUCCESS in 1h 09m 48s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/858622a001b840ecab78b1a4e4d7dde0 : SUCCESS in 1h 05m 59s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/4461ddf58b024fc09c13c7be2c335af3 : SUCCESS in 1h 07m 07s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ecd9801cfff04b03a1a426aa27898d75 : SUCCESS in 1h 23m 53s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/5daa8df990dd49ef93e5c1b4e1186d0f : SUCCESS in 1h 33m 09s","accounts_in_message":[],"_revision_number":4},{"id":"62a70875ec517ff7fbc2add3f3135788ca7ebb11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 08:15:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5b01e2a874bdf57e2ec2786af00065912ae39d2c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-21 08:33:17.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/850147/4 : FAILURE in 2h 09m 21s","accounts_in_message":[],"_revision_number":4},{"id":"f7a24f53fa0babe031b300fd245330bd442eb2f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 10:23:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/943399673bd04e44864260cd351ba46f\n\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/9af2ba1a9c2b4436b8d76ccecc60b7a5 : SUCCESS in 1h 08m 34s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/92a92971eaaf4a58970a060a4b17d311 : SUCCESS in 1h 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec127b7b14644008a9afa98b441bfcb4 : SUCCESS in 13m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4544cebf387f441abdd18ba867b0c1b9 : SUCCESS in 36m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7c0cf7d8c7b464f891a1ab6a5798c02 : SUCCESS in 24m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fee53c6352c4ae9be36389b0624c3c8 : SUCCESS in 12m 00s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0cfb5a216ead4460b88f9c184a8aef2b : SUCCESS in 1h 17m 09s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/1dfe8ffa66b341d5aaedff3e9d73abb8 : SUCCESS in 2h 06m 44s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/f44a78eeffa743f0aedf661320686cf3 : SUCCESS in 1h 32m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/757a56f1707a41f8972b14ed7817ebe4 : SUCCESS in 1h 10m 14s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d21d97a6705046f3b2ad7e61ae378d13 : SUCCESS in 1h 16m 06s","accounts_in_message":[],"_revision_number":4},{"id":"efb7c680b739aa56f19e554c9a7c73d0aaa4c87c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 10:23:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3f11f4c51bdc56a5b1512d6fddee6e5e34d26c86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 10:25:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57adafa2a9e1430b81b478477d9ac919\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/201203d1f0ed4b5ebbd1e8ef23a2d493 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5848c0dd1c951e916c0b769f711d8aafa6aa72b1","revisions":{"77d27e51b104797e29d10efa4a736c82d468833b":{"kind":"REWORK","_number":1,"created":"2022-07-18 09:45:33.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/47/850147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/850147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/850147/1"}}},"commit":{"parents":[{"commit":"c5807cbcf5aca795d4c2eac3a15136bd3ad96986","subject":"update neutron-linuxbridge-tempest for bug 1980948","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c5807cbcf5aca795d4c2eac3a15136bd3ad96986"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-18 09:43:50.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-18 09:43:52.000000000","tz":120},"subject":"ovs: improve log message when ofport is not configured","message":"ovs: improve log message when ofport is not configured\n\nThis switch the warning to error as we may be in a sitation of no\nconnectivity and this should never happen.\n\nAlso improves the condition for an ofport invalid.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ic6bd7bfadcba8deb132d8af3e295ec25a8d64b50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77d27e51b104797e29d10efa4a736c82d468833b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/77d27e51b104797e29d10efa4a736c82d468833b"}]},"branch":"refs/heads/master"},"aa3bbddb018613a0affc1048641c906582c61784":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-18 14:14:48.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/47/850147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/850147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/850147/2"}}},"commit":{"parents":[{"commit":"c53a951450b7758a3c8df1b5d1895a99f086442a","subject":"Merge \"Doc: make the contributor guide more visible\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/c53a951450b7758a3c8df1b5d1895a99f086442a"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-18 09:43:50.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-18 14:14:12.000000000","tz":120},"subject":"ovs: improve log message when ofport is not configured","message":"ovs: improve log message when ofport is not configured\n\nThis switch the warning to error as we may be in a sitation of no\nconnectivity and this should never happen.\n\nAlso improves the condition for an ofport invalid.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ic6bd7bfadcba8deb132d8af3e295ec25a8d64b50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa3bbddb018613a0affc1048641c906582c61784"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/aa3bbddb018613a0affc1048641c906582c61784"}]},"branch":"refs/heads/master"},"689db7122ada23ffa1be2ac76598a611c5bcd9b3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-07-19 20:00:41.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/47/850147/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/850147/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/850147/3"}}},"commit":{"parents":[{"commit":"ff1c6936b874dfcf675af62bead6dde07f8076a6","subject":"Remove linuxbridge mech driver from the ovs based jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ff1c6936b874dfcf675af62bead6dde07f8076a6"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-18 09:43:50.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-19 20:00:29.000000000","tz":120},"subject":"ovs: improve log message when ofport is not configured","message":"ovs: improve log message when ofport is not configured\n\nThis switch the warning to error as we may be in a sitation of no\nconnectivity and this should never happen.\n\nAlso improves the condition for an ofport invalid.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ic6bd7bfadcba8deb132d8af3e295ec25a8d64b50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/689db7122ada23ffa1be2ac76598a611c5bcd9b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/689db7122ada23ffa1be2ac76598a611c5bcd9b3"}]},"branch":"refs/heads/master"},"5848c0dd1c951e916c0b769f711d8aafa6aa72b1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-07-21 06:22:55.000000000","uploader":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"ref":"refs/changes/47/850147/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/47/850147/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/47/850147/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/47/850147/4"}}},"commit":{"parents":[{"commit":"3ffbf831cf6f2390615597ddbbe34a484e596667","subject":"Merge \"Forbid create ndp proxy on same router with same ip address\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/3ffbf831cf6f2390615597ddbbe34a484e596667"}]}],"author":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-18 09:43:50.000000000","tz":120},"committer":{"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","date":"2022-07-21 06:18:41.000000000","tz":120},"subject":"ovs: improve log message when ofport is not configured","message":"ovs: improve log message when ofport is not configured\n\nThis switch the warning to error as we may be in a sitation of no\nconnectivity and this should never happen.\n\nAlso improves the condition for an ofport invalid.\n\nSigned-off-by: Sahid Orentino Ferdjaoui \u003csahid.ferdjaoui@industrialdiscipline.com\u003e\nChange-Id: Ic6bd7bfadcba8deb132d8af3e295ec25a8d64b50\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5848c0dd1c951e916c0b769f711d8aafa6aa72b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/5848c0dd1c951e916c0b769f711d8aafa6aa72b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
