)]}'
{"id":"openstack%2Fneutron~851064","triplet_id":"openstack%2Fneutron~stable%2Fwallaby~Ic366d67770cc6e19a4412eaf2f1ab53780e98ee8","project":"openstack/neutron","branch":"stable/wallaby","topic":"bug/1982206-stable/yoga-stable/xena-stable/wallaby","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-07-27 19:09:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic366d67770cc6e19a4412eaf2f1ab53780e98ee8","subject":"Test: mock out _check_netfilter_for_bridges in unit tests","status":"MERGED","created":"2022-07-27 09:21:15.000000000","updated":"2022-07-27 19:10:49.000000000","submitted":"2022-07-27 19:09:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851064-bug/1982206-stable/yoga-stable/xena-stable/wallaby","cherry_pick_of_change":851063,"cherry_pick_of_patch_set":1,"meta_rev_id":"7ed1ae3c131b8e3e1e6333d22f69833c3026f31d","_number":851064,"virtual_id_number":851064,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-27 19:09:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-07-27 15:49:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-07-27 13:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-07-27 15:49:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-27 10:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-27 10:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-27 11:34:08.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-07-27 13:46:17.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-07-27 15:49:23.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"}],"messages":[{"id":"b65ad21d83445b46f1ff54e7ef0a9f91a1a71ebb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-27 09:21:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"7037520ba1d621828acb6c0942ef22920fcc11ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 10:10:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2002b4f286e43d787be41ff74232f2d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d82a64a46a4a491f81a7a32cc0f66d44 : SUCCESS in 45m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38a0fed059154fb09949dd3349358de5 : SUCCESS in 34m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"446cc81a2fcb0374de41bf10275ffa627d8db83c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 10:46:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b51fbb11e0f347e6a5a8f5f3372514f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1fa620e57c0471e961da7c01e8de0ab : SUCCESS in 41m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd18c33bb4c248a88f7963c60ef89c59 : SUCCESS in 13m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0f18c37cd964402b10fdc4b46c998d7 : SUCCESS in 33m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41ea5017c9fa4933951e6257b4fb47e2 : SUCCESS in 28m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db7ff340e11d4323a862623812b60b40 : SUCCESS in 28m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9598688e1f754c5e9f975eb9f6d99573 : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10a735ddacb946c28e68512947037df3 : SUCCESS in 8m 27s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/76e84b6c4d3244a9ab0ea6a3f9ab6c1f : SUCCESS in 1h 05m 48s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/6fb7b4937df949f9bc3da96066b46de4 : SUCCESS in 1h 20m 56s","accounts_in_message":[],"_revision_number":1},{"id":"8e4f83c2a703db2c36a559a0f9071868c8d48f96","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-07-27 11:34:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/851064/1 : SUCCESS in 2h 12m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1ee7ae5b9a0288bc0452aeb3ada4a65e1f6c7ae7","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-27 13:46:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f4efc939293bbc8439de25b6181f70f279beb11","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-27 15:49:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19df5421d4d2dffa6a83b4c4527aecc9a9bee4af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 15:50:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d80b3dceb76de0b91d9b4820fd9efff21c577ab5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 19:09:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/458bbec08bdb4192836502ecc4cc48d2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19c3236538e49219b87f796fa361e7b : SUCCESS in 14m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b432bfdb7bb04c03aabcbcaf4f7dad20 : SUCCESS in 31m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/208bf56170f04d5dbeaae56df99f41ae : SUCCESS in 33m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b432e00628f74e3eaaaa5e9cd51800f9 : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/847627442370430dbb9f40190add88dc : SUCCESS in 9m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ae7ea9e3aec546edbe448c0726ef1c77 : SUCCESS in 1h 15m 58s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/484b74198e9f42bba3332e2a1c6622a5 : SUCCESS in 1h 39m 33s","accounts_in_message":[],"_revision_number":1},{"id":"edf7511b13fbf0edc98875e8c96a77d5be9183d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 19:09:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ed1ae3c131b8e3e1e6333d22f69833c3026f31d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-27 19:10:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c9ad9950ca94c67acabc38224483009\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb17f426451454dba7b188b3eb303a7 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b273398b0c47410b95c490d9a477c26a : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"868427c104f9ad7cb996143646d587b0f43079eb","revisions":{"868427c104f9ad7cb996143646d587b0f43079eb":{"kind":"REWORK","_number":1,"created":"2022-07-27 09:21:15.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/64/851064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/64/851064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/851064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/851064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/64/851064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/64/851064/1"}}},"commit":{"parents":[{"commit":"b183e12fdc1b79c5140abc1b70562c53860e82d6","subject":"functional: Add debug messages around OVN services","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b183e12fdc1b79c5140abc1b70562c53860e82d6"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-07-21 14:53:45.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2022-07-27 09:21:15.000000000","tz":0},"subject":"Test: mock out _check_netfilter_for_bridges in unit tests","message":"Test: mock out _check_netfilter_for_bridges in unit tests\n\n_check_netfilter_for_bridges do some calls which are not for unit tests.\nSpecifically the call for sysctl -N net.bridge seems to be broken with\na recent kernel upgrade on bionic (see the irc log for this discussion\nin [1]).\nAdd OS_TEST_TIMEOUT:180 to tox testenv, this can help to debug such\nissues.\nDuring the rootwrap to privsep change some of these mocks were added,\n(see [2]) so backporting (before Wallaby mostly) can be done with\ncaution.\n\n[1]: https://meetings.opendev.org/irclogs/%23openstack-infra/%23openstack-infra.2022-07-20.log.html#t2022-07-20T12:59:32\n[2]: https://review.opendev.org/c/openstack/neutron/+/764015/12/neutron/tests/unit/agent/linux/openvswitch_firewall/test_iptables.py\n\nChange-Id: Ic366d67770cc6e19a4412eaf2f1ab53780e98ee8\nCloses-Bug: #1982206\n(cherry picked from commit 119b82f1b10df50a696e936e80672c7ddb00436f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/868427c104f9ad7cb996143646d587b0f43079eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/868427c104f9ad7cb996143646d587b0f43079eb"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
