)]}'
{"id":"openstack%2Fneutron~851330","triplet_id":"openstack%2Fneutron~stable%2Ftrain~I34aabd97fc51ee6393ab3c6f1c6f462db8efbdb5","project":"openstack/neutron","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-07-28 08:47:17.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-07-28 09:28:22.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-28 18:03:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I34aabd97fc51ee6393ab3c6f1c6f462db8efbdb5","subject":"[Train-only] Make neutron-granade job non-voting","status":"MERGED","created":"2022-07-28 04:09:31.000000000","updated":"2022-07-28 18:05:58.000000000","submitted":"2022-07-28 18:03:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851330","meta_rev_id":"a786d7dab5bc2ba3cbb6a78199a3ab87405515d3","_number":851330,"virtual_id_number":851330,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-28 18:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2022-07-28 08:47:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2022-07-28 09:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-07-28 08:36:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-07-28 08:57:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-07-28 09:28:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-07-28 08:57:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-28 05:05:26.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2022-07-28 07:14:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-28 08:02:15.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2022-07-28 08:36:39.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-07-28 08:47:17.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-07-28 08:57:40.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-07-28 09:28:22.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"803c4971e3af04a9b837c2661c37c2b7e723cffc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-28 04:09:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85b3d3a063a59448060e4b954ba4b4c3aaa9fe0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-28 04:10:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"46a3052b61fbd0924f0719625eb0930d8d3a9541","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-28 05:05:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3448caa198a0b75dc5eed520329b44970913ccd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 07:14:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63c3c07842404b13a5998e4341cc5cd5\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/7007bfb8f9b3490bbee47f6a1f6a5a11 : SUCCESS in 44m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc9ba33d07f046a69579b7cf076a3414 : SUCCESS in 24m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f22b94135c4bdfbb5a11c6d676d2db : SUCCESS in 12m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f416f95864a44fd19c92b67e9ae45436 : SUCCESS in 20m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c32e364f16a3401ba0ae841eb6ba8070 : SUCCESS in 19m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd06de4dad0249259a7857d93a146e15 : SUCCESS in 19m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b326fc0e46b04be6a6a25bb5ece8309b : SUCCESS in 9m 59s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cda5190edde440cfa9dcfeae9c75a3ef : SUCCESS in 45m 19s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/fc1228d99cd8456aacb1616cb1d2c4b4 : SUCCESS in 40m 20s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/b0ff971326d043c5b6cf9375c7279ced : SUCCESS in 1h 21m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/fc6c85bf82d548b4beef6f9886c9c884 : SUCCESS in 1h 23m 12s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5fcd85e51f024d4f9d3dc053700f16d5 : SUCCESS in 1h 17m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2945c4c5c8f74582ae2524d937bd5921 : SUCCESS in 56m 17s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b98618fb820e4790a99c39f022ca3df7 : SUCCESS in 1h 14m 25s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/c110bf4477da40c89b69afaaaeaa031b : SUCCESS in 57m 05s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/8f627f5e7ad045a28760bd7b38962606 : SUCCESS in 58m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/43bca258cf024b9aabeb6937b6502041 : FAILURE in 2h 55m 23s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ff47918b291f4c299ef19e3784efee1b : SUCCESS in 1h 23m 28s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/a49a1a7bbbc04554bd19c2be11c28ee1 : SUCCESS in 2h 24m 55s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/b3f4a91ec9fa4dc7be1a64ccbf7d5d6f : SUCCESS in 1h 04m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff8405070e304e60873db867bce8a15a : SUCCESS in 48m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/83c8aa27f34c476eb04a3e6121f48d9c : SUCCESS in 48m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6659ff67a0234fceb8c1c0bff87a9550 : SUCCESS in 48m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dbba19ca1034ffda7f664d7701e42fb : SUCCESS in 57m 36s","accounts_in_message":[],"_revision_number":2},{"id":"5f1861fe5f4a70c170a25e5c7da87e35ae477840","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-07-28 08:02:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7dd3737f2867f24638d232102f37c04a32698be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-28 08:06:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"39756983c33747b6b54940cd28d90273616bd37d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-28 08:06:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc5a89c17b461c69a6535690f45ec7b1ef319e08","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-07-28 08:36:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e9a48d1982bd1cdf2e270b6e2a48fdb321b95b95","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-28 08:47:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"40eddcf633fe72700b7c51365db5f8026240907f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-28 08:57:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c62c34aba688e0907048cce13f4ef506bc7dd642","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-28 08:57:57.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"21d3819fdbc00c3b14b58a8e766fe02f92dba620","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-07-28 09:28:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7a231e01457cdaf3b7c5afa160f06c0c09aa1945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 11:03:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fcadc01768c4fbaaf8cfb7697772c01\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/11f4266900b84e079445869113e572c8 : SUCCESS in 55m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e7c9ec839b344e1a0a3049549c6f3fc : SUCCESS in 23m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c60653498048a5ad96adff04185183 : SUCCESS in 12m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7af4b0e15c1b427db88f60d66c7e2b72 : SUCCESS in 22m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf162ffd6dcf4990bf3b9bf8eb14dc5f : SUCCESS in 18m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1d1cb8ae59084f5db9bb7e86d84984ee : SUCCESS in 25m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89ff11c8de3449269a9ab9fd618c8a99 : SUCCESS in 12m 21s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/d802410124874afca13c0454d0afc6e6 : SUCCESS in 53m 04s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/dcffdee65e854a7a84de23f8aba5dfd5 : SUCCESS in 55m 59s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/832e5a06a12a438ab5f54013b0a39186 : SUCCESS in 1h 34m 29s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/dd554fc95de448f3bb96cdd2f37e25f0 : SUCCESS in 1h 45m 44s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/401c717a728145598e61e8cb1d21fae3 : SUCCESS in 1h 06m 28s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5ac885b8fc754dc9a204778d36042819 : SUCCESS in 1h 10m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/df9ab1a4f80044ae8e220ab36f1558b8 : SUCCESS in 59m 33s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/eff3485b8bf84f9a967001fae1d663e4 : SUCCESS in 1h 14m 09s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/d3d7a2f227014fa78dfc549067fb4270 : SUCCESS in 1h 19m 05s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a4c5f699745f4807ab1206eb7a6b2268 : FAILURE in 2h 55m 20s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3de112ace7294218afef6b6f4514ce4f : SUCCESS in 1h 41m 52s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/58cf2c2d13e445f89accc65e58c7cf5d : SUCCESS in 1h 45m 31s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/783044200e5446a4af87069c9981e92b : SUCCESS in 1h 09m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a874994b127f46b7872f479ac1b25868 : SUCCESS in 1h 08m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/d16c997bbcab46bd85e9602851037547 : SUCCESS in 1h 11m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f5616ff10470411d8e9e82f083c54804 : SUCCESS in 1h 01m 18s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d7d60d9c98e483c9e31546b2b56e43f : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":3},{"id":"52d926ff292a54c8877fed2e96703f1f1feca1cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 11:04:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a8328435bc90f6eefafd1252859c44de92d0f37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 12:43:40.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24ce857593ad4242918bbb41e2e13427\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/e19a488a46a84324875ee755788c6b95 : SUCCESS in 57m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b48430a1e4d4427ac9eb6bc7008a9e6 : SUCCESS in 12m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7a96988a2b940e79559a397f7079041 : SUCCESS in 21m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96140557071e49beb45cba1557b30543 : SUCCESS in 22m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d08bbad6f88e4a6dafd90807d5d885a3 : SUCCESS in 23m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ccd4bc5bed342df904f7d15ee3a9199 : SUCCESS in 10m 08s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/cd8c2226952142569d9b575aa1c52216 : SUCCESS in 55m 02s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/ba96bacc6b6547179de47e8d640027e6 : POST_FAILURE in 53m 27s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/744ea87e324b456d8e38338bad60b243 : POST_FAILURE in 1h 23m 13s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0dd0ef654e1e49138d9975dfc668a366 : POST_FAILURE in 1h 24m 20s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/201aa3220e3f43da9e021c1ebb3e82bd : POST_FAILURE in 1h 14m 48s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/24a6b8a906db429f8d572564dcfbe21a : POST_FAILURE in 1h 27m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd7509ab5794475bb00dfcb1104bc5be : POST_FAILURE in 1h 04m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/d70da07073924a24ba3ada04e55d8778 : SUCCESS in 1h 08m 58s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3add312391f84bbd97254f4dc79af31f : POST_FAILURE in 1h 35m 17s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/0e40c52fdcfb424389a72bcaf8a52850 : SUCCESS in 1h 20m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91a2ffe337344a179b1e14cf1414fe04 : SUCCESS in 1h 01m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/521aa77bd3d04761a33bce66a2fab96a : SUCCESS in 1h 06m 34s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6dee36f16127442eabe66e290c697074 : POST_FAILURE in 1h 08m 12s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/945cc82722ae4b44bdf26101c16f41aa : POST_FAILURE in 1h 06m 29s","accounts_in_message":[],"_revision_number":3},{"id":"7bcfb23a6a8a4dd5ba7140f0e233bc38325b532d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-07-28 13:22:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3fe981b3c12c6c3fd2d5b9493d59bef4d55b9db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 16:22:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b0cbb65b0bf41fd995f9fee40509098\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/5fbf325ca3eb4ffcbdf83cdbb534821a : SUCCESS in 53m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cefdbfa9cc14c5d9924851241aff0e4 : SUCCESS in 27m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/796454c633024a3282e897b33213ef9f : SUCCESS in 10m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f548d2dc0bb419eb6e7570f74fdabe2 : SUCCESS in 18m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b0ae8bd83264f478c82e6ab8d7502b5 : SUCCESS in 20m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1adf463a55440c0b25d3330deda7a88 : SUCCESS in 21m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48eaaa275f0a4d88be12a344c7ce8725 : SUCCESS in 10m 12s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/a80af52ede0a4fb3ab0e75aae4d2861c : SUCCESS in 48m 16s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/622ae22f35864e5f9e335630d22248f2 : SUCCESS in 47m 06s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/f334ac76475544f882541be26b09b19a : SUCCESS in 1h 21m 50s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/4b23918466ae4ba594357a9161720041 : SUCCESS in 1h 27m 28s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/1d65b1db60c5414fb7d72b9ad463135e : SUCCESS in 1h 14m 05s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/55b2b25a1e094a5388bf22a2d0ccbf08 : SUCCESS in 1h 06m 37s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7fc993a236d84d8bb04a4a9bcf0ecd7d : SUCCESS in 1h 04m 31s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/78520afa35ef405692f4cc34f3d4e317 : SUCCESS in 1h 04m 17s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/567b3189adc54d629e3b900f49eef929 : SUCCESS in 1h 05m 46s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/b62d0cd5cea4433284235b7b5bcf89cd : FAILURE in 2h 54m 39s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9741a27729754d3ea7cf3ed2a6b697c9 : SUCCESS in 1h 35m 09s\n- networking-ovn-tempest-dsvm-ovs-release https://zuul.opendev.org/t/openstack/build/dd47625b85dd4c01b402c03eb39986be : SUCCESS in 1h 56m 18s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/1d21d335832947a8b7c91b112cdc9846 : SUCCESS in 1h 02m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21fe65b05a55407cabd39cbeed8ee449 : SUCCESS in 57m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/5736ba2fb1e0447cb8d0e289e8dbdecc : SUCCESS in 1h 03m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/792cd0bbb97341908e92c4bb4a568e70 : SUCCESS in 59m 39s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1410b9fd8e84768ab3726e9eb64af14 : SUCCESS in 1h 01m 01s","accounts_in_message":[],"_revision_number":3},{"id":"d47dfd876d26ec02bbd62f84157d1eaf90079197","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 16:23:36.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"718123138516fc302c411ec58d207b9fd229afd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 18:03:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b2c061bafaf40d6b3fa4a96cc2fdcef\n\n- neutron-tempest-plugin-api-train https://zuul.opendev.org/t/openstack/build/ca5a96fb34874cf6bccf9209accc189c : SUCCESS in 1h 00m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57fac27f8f544c6c8e56eb89f9bbce90 : SUCCESS in 13m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/42dd5e34cf234f10a02b0b14109698fb : SUCCESS in 23m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d0d53e13fdd484bbecd4b9fecb50989 : SUCCESS in 26m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ba2d764a947f40d7814c92debb323e3c : SUCCESS in 26m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e523a3419f34a66b281ed5521f56d48 : SUCCESS in 10m 02s\n- neutron-functional https://zuul.opendev.org/t/openstack/build/76373e072b484b88a397b02e3d0f5de7 : SUCCESS in 50m 39s\n- neutron-functional-python27 https://zuul.opendev.org/t/openstack/build/e6db14687142486d9b80f63c7b591f99 : SUCCESS in 55m 45s\n- neutron-fullstack https://zuul.opendev.org/t/openstack/build/bdeda178566041a6998e43806a58171f : SUCCESS in 1h 30m 59s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/d90f88eeea3a42ccadef105ca47baa45 : SUCCESS in 1h 22m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/cdd2143310b740c180a0c97f41719278 : SUCCESS in 1h 16m 41s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/f101c3ccf9ea48338cacfe5c3821adfc : SUCCESS in 1h 14m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d14436f926bf4b369cc56dbdd7d1c30f : SUCCESS in 1h 11m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/9ef0ce3abf774d43afabe6425aedb266 : SUCCESS in 1h 18m 03s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cbac61847b684f97aad34af934577a61 : SUCCESS in 1h 39m 35s\n- tempest-integrated-networking https://zuul.opendev.org/t/openstack/build/add800d2530a4ad481ce8cbb069b1dc7 : SUCCESS in 1h 05m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcee0edd21074f27b98193bd224e711d : SUCCESS in 1h 08m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/1f13b7f3713041818a91cb9e47efc00b : SUCCESS in 1h 10m 52s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5404dce8be414a68a85602d1b490279c : SUCCESS in 1h 13m 37s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b19581fc10464dbfae3983878845c152 : SUCCESS in 1h 14m 54s","accounts_in_message":[],"_revision_number":3},{"id":"83db6c01bc281d1b18e710d6c479a8cbf7f1315e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 18:03:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a786d7dab5bc2ba3cbb6a78199a3ab87405515d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 18:05:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f358cdbbc835459f97943f5ea6beb916\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da02b1760914cf3a8affd4b25fa8674 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"01c88ae08c1e5dc3f4b57c7531d367ab7f4094dd","revisions":{"d26c875f388cf33914b602fa76ef793c4480a068":{"kind":"REWORK","_number":1,"created":"2022-07-28 04:09:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/851330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/851330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/851330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/851330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/851330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/851330/1"}}},"commit":{"parents":[{"commit":"7740ed078431d2df8160abb207f5c64fb64cea14","subject":"[OVS][QoS] Unset the min-bw QoS from the phys bridge interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7740ed078431d2df8160abb207f5c64fb64cea14"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-28 04:09:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-28 04:09:09.000000000","tz":540},"subject":"[Train-only]: Make neutron-granade job non-voting","message":"[Train-only]: Make neutron-granade job non-voting\n\nCurrently the neutron-grenade job in stable/train is consistently\nfailing because of the job script being stuck.\n\nThis change temporarily makes the job non-voting to unblock CI.\n\nChange-Id: I34aabd97fc51ee6393ab3c6f1c6f462db8efbdb5\nRelated-Bug: #1982720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d26c875f388cf33914b602fa76ef793c4480a068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/d26c875f388cf33914b602fa76ef793c4480a068"}]},"branch":"refs/heads/stable/train"},"295db41a7f6b84b9021c8b9260d3415a204a5050":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-07-28 04:10:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/851330/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/851330/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/851330/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/851330/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/851330/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/851330/2"}}},"commit":{"parents":[{"commit":"7740ed078431d2df8160abb207f5c64fb64cea14","subject":"[OVS][QoS] Unset the min-bw QoS from the phys bridge interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7740ed078431d2df8160abb207f5c64fb64cea14"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-28 04:09:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-28 04:10:00.000000000","tz":540},"subject":"[Train-only] Make neutron-granade job non-voting","message":"[Train-only] Make neutron-granade job non-voting\n\nCurrently the neutron-grenade job in stable/train is consistently\nfailing because of the job script being stuck.\n\nThis change temporarily makes the job non-voting to unblock CI.\n\nChange-Id: I34aabd97fc51ee6393ab3c6f1c6f462db8efbdb5\nRelated-Bug: #1982720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/295db41a7f6b84b9021c8b9260d3415a204a5050"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/295db41a7f6b84b9021c8b9260d3415a204a5050"}]},"branch":"refs/heads/stable/train"},"01c88ae08c1e5dc3f4b57c7531d367ab7f4094dd":{"kind":"REWORK","_number":3,"created":"2022-07-28 08:06:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/851330/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/30/851330/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/851330/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/851330/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/30/851330/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/30/851330/3"}}},"commit":{"parents":[{"commit":"7740ed078431d2df8160abb207f5c64fb64cea14","subject":"[OVS][QoS] Unset the min-bw QoS from the phys bridge interface","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/7740ed078431d2df8160abb207f5c64fb64cea14"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-28 04:09:09.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-07-28 08:05:50.000000000","tz":540},"subject":"[Train-only] Make neutron-granade job non-voting","message":"[Train-only] Make neutron-granade job non-voting\n\nCurrently the neutron-grenade job in stable/train is consistently\nfailing because of the job script being stuck.\n\nThis change temporarily makes the job non-voting to unblock CI.\n\nChange-Id: I34aabd97fc51ee6393ab3c6f1c6f462db8efbdb5\nRelated-Bug: #1982720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01c88ae08c1e5dc3f4b57c7531d367ab7f4094dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/01c88ae08c1e5dc3f4b57c7531d367ab7f4094dd"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
