)]}'
{"id":"openstack%2Fneutron~852131","triplet_id":"openstack%2Fneutron~stable%2Fxena~I99d2a8078dd7acde72b232a5b1de826d7e1d60f3","project":"openstack/neutron","branch":"stable/xena","topic":"bug/1983530","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-08-16 16:39:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I99d2a8078dd7acde72b232a5b1de826d7e1d60f3","subject":"[ovn]Change LogicalSwitchPortUpdateUpEvent old conditions","status":"MERGED","created":"2022-08-04 10:00:04.000000000","updated":"2022-08-16 16:41:05.000000000","submitted":"2022-08-16 16:39:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852131-bug/1983530","meta_rev_id":"7fdffefa71ee5d1d44a0b8187f7a548e673f52a0","_number":852131,"virtual_id_number":852131,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-16 16:39:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":2,"date":"2022-08-05 07:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2022-08-16 12:07:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2022-08-05 00:27:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2022-08-16 12:07:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-04 10:00:04.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"CC"},{"updated":"2022-08-04 10:39:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-04 11:36:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 12:27:33.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-05 00:27:39.000000000","updated_by":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"reviewer":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"state":"REVIEWER"},{"updated":"2022-08-05 07:37:03.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-08-16 12:07:17.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"dccdbb2cd4d87faf8696aca280932c2c4d9e7655","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-04 10:00:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"277e356431cb1f85ee4537d724cba6fd8e939379","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 10:39:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc60d65204314275b20ec2edc7b39ecc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/916f7148bd884c4cb83434976b8fcdf6 : SUCCESS in 34m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3aa89d31fbe54f5b8625161a234547be : SUCCESS in 36m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"854b111cf2c4635ae9cc58fda3a25914327ecd69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 11:36:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4689a12cde7b408f9c9411b569772041\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/0b2cb14be47745f7a406dbf99737d45e : SUCCESS in 49m 45s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/e22135f180f642fbb14b452ed28d54db : SUCCESS in 1h 03m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d79e4471c1d412eb15b85122ddcb985 : SUCCESS in 30m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d982810b81244bd8c6ce1cf9b1bc976 : SUCCESS in 12m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d320b36a97fc498990a83be40b7c0df4 : SUCCESS in 22m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/825ce6ba9d8047a0a7ec14ceeaa81e81 : SUCCESS in 20m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4234220fdd26473c9288aed7b80fa235 : SUCCESS in 21m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0798c5242d4c4b18b575a337c412813e : SUCCESS in 9m 43s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/012222858fe944b489cfe7402ebc2fdc : SUCCESS in 54m 03s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f258458bc35e4010b2fcbbda2342106d : SUCCESS in 1h 02m 12s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/9fe97185986e4b4883b8971e208aa089 : SUCCESS in 1h 29m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7a215f829e4646e4d0396d2cd3ee32246890372","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-04 12:27:33.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/852131/1 : SUCCESS in 2h 25m 55s","accounts_in_message":[],"_revision_number":1},{"id":"0a66b06a7f9eeef183d0bdd895cc85e017e64387","author":{"_account_id":30380,"name":"ZhouHeng","email":"zhouhenglc@inspur.com","username":"zhouhenglc"},"date":"2022-08-05 00:27:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"45703e42e660dfa30c3371b921caead4fb019034","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-08-05 07:37:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9bed93b4cb242c529ce6b16332ddc1d57caae3c2","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-16 12:07:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a99608cc700705c26301c9d8fd5bb67c49f5acad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 12:07:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70052cc2fc8df80a65ddc6904e4db8cc9891d24c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 13:16:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b755a8e5dd834adf868bb316f232a5ae\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/3f6ab7f8ecbb443d87c03f522af0541e : SUCCESS in 1h 05m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a7b0f026e90428694dd73313541c389 : SUCCESS in 13m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a3351f59e4a45c8a0708384375aacdc : SUCCESS in 26m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8bc28ac80044651bc4b595c6f57248a : SUCCESS in 30m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa1fdcc3d9743feb1b8b15a305b9cde : SUCCESS in 9m 31s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3da5bf2bcb5044e7a32e210c5610aabe : FAILURE in 1h 05m 23s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/957bb81218a544f6ba9c46c74d32c3b9 : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5a510500955ccc9a067a5708f5fe4c33ed4ea5e7","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-16 13:26:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d57d5622fd93016c81c0fb7b0c1d2e66421f6875","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-16 13:47:31.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/852131/1 : FAILURE in 13m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5d8832aa995b8792565e3c5c4fb7d6b2001e05fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 14:07:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec12feac807f4badbc0fbc19e0c9a08a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ff8c09153554f0697572b4afb444953 : SUCCESS in 32m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/20dafda93c4842dc9b88354cffaf46ea : SUCCESS in 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"046bf59707056ebdc8d873a07cbf47a00b1ed998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 15:04:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06d1b19fbb794490bacddee1eb98b1cd\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/d3e681aadc5f48af80af49308cce81a3 : SUCCESS in 58m 09s\n- neutron-tempest-plugin-scenario-ovn-xena https://zuul.opendev.org/t/openstack/build/ab382bb2f4c244f1b1d95408de4ef0b0 : SUCCESS in 1h 28m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/009b730284e8459d9c631fc84f017a3b : SUCCESS in 34m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3486ce6843a44e10928670575ea18cd3 : SUCCESS in 14m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70f28198eb6d4d3cab68f99dea94a323 : SUCCESS in 21m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bedafb83c4a845c0a811dcec37e6e42d : SUCCESS in 22m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/241ac828f1424387800817ba69cce85e : SUCCESS in 25m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6848c890ab884b33ba3f2a22b375d3c3 : SUCCESS in 10m 01s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/dcdd21b0cf0d41ad8d47941bc708a5ed : SUCCESS in 55m 14s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6076bdd2f5f64af9a99abb17ffdfe183 : SUCCESS in 1h 08m 35s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2b74075f0c7f4c6d90c03dee57d1bf39 : SUCCESS in 1h 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0843c62dd10aa3490f732624edeb3f63c3a3dde8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 15:04:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f074a243be682504d7ca29af655f773fc12dc9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:39:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/507c59f1f69b48bf9474e09c22b105d3\n\n- neutron-tempest-plugin-api-xena https://zuul.opendev.org/t/openstack/build/894a1e390a4d4f85b17a025d1f385136 : SUCCESS in 1h 01m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8725145f393449d6b589281f3db38baf : SUCCESS in 14m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a633ef9015734ce88c1d134d47b8cd4d : SUCCESS in 26m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9444556bfb642bbbeba55bee0707b41 : SUCCESS in 24m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5fbae997441461f9a7d1553a32dbe03 : SUCCESS in 10m 49s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/15629f55c8f340909430e529e9bb6cbf : SUCCESS in 47m 21s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7de4dd7944e14bb787983ad1ff6bb2e9 : SUCCESS in 1h 13m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c2702fff7455ebd3207cc097b013ad4372eb0e2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:39:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7fdffefa71ee5d1d44a0b8187f7a548e673f52a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 16:41:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/012bcf604d0c40c9b65b836008796429\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02454926edf4782b081184c1f84d9fb : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cfb0672ab71620680b7f26475021c4857944d986","revisions":{"cfb0672ab71620680b7f26475021c4857944d986":{"kind":"REWORK","_number":1,"created":"2022-08-04 10:00:04.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/31/852131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/31/852131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/852131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/852131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/31/852131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/31/852131/1"}}},"commit":{"parents":[{"commit":"ec831cdf124dff1a15b1a64ff3992a822348bc50","subject":"Add workaround for eventlet.greendns bug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ec831cdf124dff1a15b1a64ff3992a822348bc50"}]}],"author":{"name":"zhouhenglc","email":"zhouhenglc@inspur.com","date":"2022-08-04 00:47:10.000000000","tz":480},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-07-29 22:30:38.000000000","tz":120},"subject":"[ovn]Change LogicalSwitchPortUpdateUpEvent old conditions","message":"[ovn]Change LogicalSwitchPortUpdateUpEvent old conditions\n\nSometimes when creating ports, the status of logical_switch_port\nchanges from [] to [true]. we should be able to deal with this\nsituation and ensure that the port is in the active state in neutron.\n\nCloses-bug: #1983530\n\nChange-Id: I99d2a8078dd7acde72b232a5b1de826d7e1d60f3\n(cherry picked from commit e0039371f876fe7ca52d4c49c997be0c73b80adf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfb0672ab71620680b7f26475021c4857944d986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/cfb0672ab71620680b7f26475021c4857944d986"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
