)]}'
{"id":"openstack%2Fneutron~852649","triplet_id":"openstack%2Fneutron~stable%2Fvictoria~I320ac2306e0f25ff933d8271203e192486062d61","project":"openstack/neutron","branch":"stable/victoria","topic":"bug/1976439","attention_set":{},"removed_from_attention_set":{"9531":{"account":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"last_update":"2022-08-11 20:14:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I320ac2306e0f25ff933d8271203e192486062d61","subject":"Clean up db residual record from dvr port","status":"MERGED","created":"2022-08-10 12:21:28.000000000","updated":"2022-08-11 20:16:03.000000000","submitted":"2022-08-11 20:14:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852649-bug/1976439","cherry_pick_of_change":852648,"cherry_pick_of_patch_set":1,"meta_rev_id":"80429327d684bfcc9d80acd1c20544a8ea26e783","_number":852649,"virtual_id_number":852649,"owner":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-11 20:14:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":2,"date":"2022-08-11 13:23:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-08-11 16:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-08-11 16:49:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32352,"name":"liujinxin","display_name":"scilla","email":"hbbfxyxxgcljx@163.com","username":"liujinxin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-10 12:21:28.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":32352,"name":"liujinxin","display_name":"scilla","email":"hbbfxyxxgcljx@163.com","username":"liujinxin"},"state":"CC"},{"updated":"2022-08-10 15:22:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-11 13:23:47.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-08-11 16:49:14.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"7863a25c86747641f413d48f84482502bd1a3aca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-08-10 12:21:28.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"8b4bbaab2316c8bf7087b3c6ab6508f3e775f0a4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-08-10 12:25:00.000000000","message":"Topic bug/1976439-stable/yoga-stable/xena-stable/wallaby-stable/victoria removed","accounts_in_message":[],"_revision_number":1},{"id":"8996a1123c4fb6642cfb5bff347e92152dec4f22","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-08-10 12:25:02.000000000","message":"Topic set to bug/1976439","accounts_in_message":[],"_revision_number":1},{"id":"6a583f7f0cc1eba3c59b9727a97054d9a510fad1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 15:22:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d6dd0f30ae4620ba460db2692d0b5f\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/411879ed8adb490aa2fb1793856d028c : SUCCESS in 54m 27s\n- neutron-tempest-plugin-scenario-linuxbridge-victoria https://zuul.opendev.org/t/openstack/build/63010be534bc4b55bc0fd0b19397ae39 : SUCCESS in 1h 07m 41s\n- neutron-tempest-plugin-scenario-openvswitch-victoria https://zuul.opendev.org/t/openstack/build/1d93cb57fa704069b92e2c5c08685fb0 : SUCCESS in 1h 04m 05s\n- neutron-tempest-plugin-scenario-openvswitch-iptables_hybrid-victoria https://zuul.opendev.org/t/openstack/build/ea6d22f715de4a92a9cd1e4276bf31af : SUCCESS in 1h 02m 07s\n- neutron-tempest-plugin-scenario-ovn-victoria https://zuul.opendev.org/t/openstack/build/fff5d27b8b1a42e09a01c63ba06cf399 : SUCCESS in 1h 04m 23s\n- neutron-tempest-plugin-designate-scenario-victoria https://zuul.opendev.org/t/openstack/build/209849faf228446fb64eb8251466679d : SUCCESS in 35m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e819491db3b4a45b548c0dd8f1dfc8e : SUCCESS in 33m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73bae0ce421b4b5199b4edab5f1b0d4a : SUCCESS in 13m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/570ba50835ec4370bf2bd18fe63ab1b8 : SUCCESS in 22m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6287a4356dcf44ef829d92a7e23b937f : SUCCESS in 24m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9df830e91f86452c8c837458de12beb9 : SUCCESS in 10m 23s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/44bcec3229534adab7b3d1b744903894 : SUCCESS in 1h 30m 18s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/60ae915db7604f35bf93df10418e412e : SUCCESS in 1h 20m 14s\n- neutron-rally-task https://zuul.opendev.org/t/openstack/build/3600558a394244968eda8cb92fb2867c : SUCCESS in 1h 28m 36s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c675628725164fb986e43c18bea27e66 : SUCCESS in 1h 05m 11s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/e4fd915df8404bb7a070ec0b62b53e3c : SUCCESS in 1h 20m 11s\n- neutron-grenade-ovn https://zuul.opendev.org/t/openstack/build/a1b16e8e4f064a96a9aa93673ae25630 : SUCCESS in 1h 17m 48s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/82c460a3f6144002b2eb410cbdde7470 : SUCCESS in 1h 01m 21s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/89d23fca3d2349abb9f2e0d2413c049e : SUCCESS in 1h 12m 41s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0e8ff83e6f914bd1ae92c702c0344c0d : SUCCESS in 1h 18m 15s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/156966e32a8e4f8dbb0ba47051c0755f : FAILURE in 1h 16m 56s (non-voting)\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/85e3e072bb2843b785d1abf8ba96add3 : SUCCESS in 1h 04m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4b7b8597e0534afb9f8b18ec9e82dbce : SUCCESS in 1h 05m 56s (non-voting)\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fd4911d1aad84344a0f42fea73de7986 : SUCCESS in 1h 39m 28s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7051238021940f5a318368bff08a41f : SUCCESS in 1h 02m 21s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/a20934e7964c404da85dc56d8af246b5 : SUCCESS in 1h 05m 13s\n- neutron-ovn-tempest-ovs-release-ipv6-only https://zuul.opendev.org/t/openstack/build/c833fc75377c4477b116160d0c3388c0 : FAILURE in 43m 43s (non-voting)\n- openstacksdk-functional-devstack-networking https://zuul.opendev.org/t/openstack/build/b6d83bdb94e94908afe06c4400b4f045 : SUCCESS in 1h 30m 18s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e6dfd422158248b2adabebd592243e0c : SUCCESS in 2h 45m 24s (non-voting)\n- neutron-centos-8-tripleo-standalone https://zuul.opendev.org/t/openstack/build/11cc53fbd53c404fb8b09101f7a0c18b : FAILURE in 45m 44s (non-voting)\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/243610613d4b4eeab1910492a926c23a : SUCCESS in 1h 49m 51s (non-voting)\n- neutron-ovn-tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4f000ec290914280bc47bdf020373087 : FAILURE in 55m 15s (non-voting)\n- neutron-ovn-tempest-slow https://zuul.opendev.org/t/openstack/build/4489d205b7024fccb5973a29c5c7b56f : FAILURE in 44m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a031645a78c967771bb82fa6473df2f58c478e8","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-11 13:23:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"430b9da182ce868d056bd95c289b1a03e985df68","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-08-11 16:49:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56e66932fc0c18fcddbb8dc12b08eab4a8839050","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:50:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6ce53929d22858f20341943cb5ce1fde5d2d83d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 20:14:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/561022e7c8cf438fa8f31e0fb48ce181\n\n- neutron-tempest-plugin-api-victoria https://zuul.opendev.org/t/openstack/build/48a21816dfb64dd08568a1cb26530bf1 : SUCCESS in 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69eab8e602694ff285f9b3bfe39ed6ab : SUCCESS in 11m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6ee1526dac54f10afe98b6f0fc4a700 : SUCCESS in 21m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f6d96b0494248c492adf6b2b2c13965 : SUCCESS in 21m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31413ff44079460893754daf9ee011d0 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/ff5790827d52496389e56e8d73ffa4e1 : SUCCESS in 1h 24m 52s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/a389a24da5bd401ab89fe2388b4adf25 : SUCCESS in 1h 40m 48s\n- neutron-tempest-with-uwsgi https://zuul.opendev.org/t/openstack/build/e5fe9a9e2fc1454480706473f36915ed : SUCCESS in 1h 15m 46s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2c7449fa59d243c1a7946dd916c428f0 : SUCCESS in 1h 15m 44s\n- neutron-tempest-iptables_hybrid https://zuul.opendev.org/t/openstack/build/ce5ae4aff1cd4e9c876ead61922a9799 : SUCCESS in 1h 12m 32s\n- neutron-tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/99db4603dd684ba5a3976c495e400dd4 : SUCCESS in 1h 24m 30s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bba9b0287feb469a96f5f13401b826ec : SUCCESS in 52m 26s\n- neutron-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/892c4f52f9964c42835ac9d5b56f9468 : SUCCESS in 1h 15m 06s\n- neutron-tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/276af3e8a0f4449a92aed27834721f25 : SUCCESS in 1h 28m 03s\n- neutron-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc3174ce97f84597b2ff3143bc00d268 : SUCCESS in 50m 10s\n- neutron-ovn-tempest-ovs-release https://zuul.opendev.org/t/openstack/build/717e344599294426858c10e1f0f497e8 : SUCCESS in 1h 20m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e07874e4520252125510d6e51089ec2108021c68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 20:14:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80429327d684bfcc9d80acd1c20544a8ea26e783","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 20:16:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a635976628054a90864130293438eee7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e57cd14b0b04f8eb75bc5aa2f663236 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b97b977feef92317ad9abdefc604f4f76fc8e72","revisions":{"8b97b977feef92317ad9abdefc604f4f76fc8e72":{"kind":"REWORK","_number":1,"created":"2022-08-10 12:21:28.000000000","uploader":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"ref":"refs/changes/49/852649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/49/852649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/852649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/852649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/49/852649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/49/852649/1"}}},"commit":{"parents":[{"commit":"ed0d8661080cd3b960e10d6b2c57f310b23cd0ad","subject":"Merge \"Port update will trigger less notifications to the DHCP agents\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ed0d8661080cd3b960e10d6b2c57f310b23cd0ad"}]}],"author":{"name":"liujinxin","email":"hbbfxyxxgcljx@163.com","date":"2022-06-08 08:10:55.000000000","tz":480},"committer":{"name":"liuyulong","email":"i@liuyulong.me","date":"2022-08-10 12:21:28.000000000","tz":0},"subject":"Clean up db residual record from dvr port","message":"Clean up db residual record from dvr port\n\nDelete the DVR port record on the node when the router is deleted from\nthe node. This patch fixes an issue where the DB actions have no context\nwriter which will actually update nothing duing the following\n`delete_distributed_port_binding_if_stale`.\nAs well as fixing that when the dvr router was deleted from a node,\nonly the ml2_distributed_port_bindings were cleaned up and not the\nml2_port_binding_levels records for the dvr port.\n\nRemove the last VM under one router in one host, the\nml2_distributed_port_bindings and ml2_port_binding_levels will remain\nthe record.  So if VMs under one router had spread on many hosts,\nand router (router ports) still exists, even there is only one VM under it,\nthe binding entries will still be equal to amount of hosts where this\nrouter resided before. The result is a large amount of redundant data\nin the database, even if some nodes are no longer in the cluster,\nthere are still records for that node.\n\nCloses-Bug: #1976439\nChange-Id: I320ac2306e0f25ff933d8271203e192486062d61\n(cherry picked from commit ad3f7a8b7d69c2f5612d04b08f7250a21c4488f9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b97b977feef92317ad9abdefc604f4f76fc8e72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8b97b977feef92317ad9abdefc604f4f76fc8e72"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
