)]}'
{"id":"openstack%2Fneutron~852650","triplet_id":"openstack%2Fneutron~master~I5e885e2b2a09d48197de30d8c6de5d74c10f5987","project":"openstack/neutron","branch":"master","topic":"pylint-c0330-2","attention_set":{},"removed_from_attention_set":{"1131":{"account":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"last_update":"2022-08-17 16:07:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5e885e2b2a09d48197de30d8c6de5d74c10f5987","subject":"Fix some pylint indentation warnings","status":"MERGED","created":"2022-08-09 17:29:51.000000000","updated":"2022-08-17 16:08:52.000000000","submitted":"2022-08-17 16:07:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852650-pylint-c0330-2","meta_rev_id":"36970636a5199da55f3fae6713f5ab5f8f86a8fb","_number":852650,"virtual_id_number":852650,"owner":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-08-09 19:54:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-17 16:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"recommended":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2022-08-16 08:30:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-17 14:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-08-16 04:24:17.000000000","_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-17 14:03:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-09 18:12:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-09 18:21:23.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 19:07:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-09 20:55:36.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-16 04:24:17.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"REVIEWER"},{"updated":"2022-08-16 08:30:56.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2022-08-17 14:03:18.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"}],"messages":[{"id":"19a84f02f59745e3fdb04ecf958770de65081379","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-08-09 17:29:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c7fa8e29e3c71d10d994822109c42cc52bdb2f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 18:12:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58445a9d77be419d94bd13f435e7062f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a0174cee7824fbfa37757c3ad9467d8 : SUCCESS in 37m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b038bf7f8d564ea6b2b1c91f34e646d4 : SUCCESS in 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ce766cfff0742396d54e6465d34057f903cbe78","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-09 18:21:23.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/50/852650/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"e138b1d3b8b372f95ac19cfbb04202269beb48bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 19:07:19.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11d56377c24943a8b0bc1fc6fc2f3cce\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/048ab8dc639e455394a7da88cfb5417c : SUCCESS in 1h 23m 35s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/d43364b2a0bb4065951b4447805cd65b : SUCCESS in 1h 18m 09s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/225bd4dbc50049dfb96b04fc9d2f4999 : SUCCESS in 1h 09m 51s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/72f14868abd54ce5bd0596c42a01b86f : SUCCESS in 1h 13m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a5d65c4580914db9b249bc838874d454 : SUCCESS in 37m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47d9b294e2a244589143f24d33c30902 : SUCCESS in 31m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9f5f4f403a24d56b808be13e4b9672d : FAILURE in 28m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2b1fef904b94540b60365a5aa5c335e : SUCCESS in 23m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccecd0800fe6446bb630390c4cd61a80 : SUCCESS in 23m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3492e78e8a3740c88faa1ae4d42cb7c5 : SUCCESS in 23m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29684bcedcf4c40a60b06a93a5897ae : SUCCESS in 9m 50s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1290f98768de4e3081485ee951d422f8 : SUCCESS in 1h 02m 28s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/af2952f1554941c69fa914d3183886bb : SUCCESS in 1h 35m 36s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/b9ae364d1e4c47b2a813f31c848d7f82 : SUCCESS in 1h 06m 15s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f9a51e2984041bd8babddefc4c79152 : SUCCESS in 57m 11s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/de3966e5ccad44a9835ec3af0ec1d76a : SUCCESS in 59m 34s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/a5f1396a0cae4e1faf367bc8ee549343 : SUCCESS in 1h 17m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/d929c70e30764907967bd4d6cb91d5e1 : SUCCESS in 1h 19m 08s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00e024e3f40d4d2d9a94eaf12f14b260 : SUCCESS in 1h 02m 36s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/2865624653f5482fbfd242c4bf7203d5 : SUCCESS in 1h 11m 39s (non-voting)\n\nWarning:\n  Comments left for invalid file neutron/agent/common/async_process.py\n  Comments left for invalid file neutron/agent/common/ovs_lib.py\n  Comments left for invalid file neutron/agent/common/utils.py\n  Comments left for invalid file neutron/agent/dhcp/agent.py\n  Comments left for invalid file neutron/agent/l2/extensions/dhcp/base.py\n  Comments left for invalid file neutron/agent/l2/extensions/dhcp/ipv4.py\n  Comments left for invalid file neutron/agent/l2/extensions/dhcp/ipv6.py\n  Comments left for invalid file neutron/agent/l2/extensions/fdb_population.py\n  Comments left for invalid file neutron/agent/l2/extensions/local_ip.py\n  Comments left for invalid file neutron/agent/l2/extensions/qos.py\n  Comments left for invalid file neutron/agent/l2/extensions/qos_linux.py\n  Comments left for invalid file neutron/agent/l3/dvr_edge_ha_router.py\n  Comments left for invalid file neutron/agent/l3/dvr_edge_router.py\n  Comments left for invalid file neutron/agent/l3/dvr_fip_ns.py\n  Comments left for invalid file neutron/agent/l3/dvr_local_router.py\n  Comments left for invalid file neutron/agent/l3/extensions/conntrack_helper.py\n  Comments left for invalid file neutron/agent/l3/extensions/ndp_proxy.py\n  Comments left for invalid file neutron/agent/l3/extensions/port_forwarding.py\n  Comments left for invalid file neutron/agent/l3/extensions/qos/base.py\n  Comments left for invalid file neutron/agent/l3/extensions/qos/fip.py\n  Comments left for invalid file neutron/agent/l3/extensions/qos/gateway_ip.py\n  Comments left for invalid file neutron/agent/l3/ha.py\n  Comments left for invalid file neutron/agent/l3/l3_agent_extension_api.py\n  Comments left for invalid file neutron/agent/l3/namespaces.py\n  Comments left for invalid file neutron/agent/l3/router_info.py\n  Comments left for invalid file neutron/agent/linux/daemon.py\n  Comments left for invalid file neutron/agent/linux/dhcp.py\n  Comments left for invalid file neutron/agent/linux/dibbler.py\n  Comments left for invalid file neutron/agent/linux/external_process.py\n  Comments left for invalid file neutron/agent/linux/interface.py\n  Comments left for invalid file neutron/agent/linux/ip_conntrack.py\n  Comments left for invalid file neutron/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/agent/linux/ipset_manager.py\n  Comments left for invalid file neutron/agent/linux/iptables_firewall.py\n  Comments left for invalid file neutron/agent/linux/iptables_manager.py\n  Comments left for invalid file neutron/agent/linux/keepalived.py\n  Comments left for invalid file neutron/agent/linux/openvswitch_firewall/firewall.py\n  Comments left for invalid file neutron/agent/linux/openvswitch_firewall/rules.py\n  Comments left for invalid file neutron/agent/linux/pd.py\n  Comments left for invalid file neutron/agent/linux/utils.py\n  Comments left for invalid file neutron/agent/metadata/agent.py\n  Comments left for invalid file neutron/agent/ovn/metadata/agent.py\n  Comments left for invalid file neutron/agent/ovn/metadata/ovsdb.py\n  Comments left for invalid file neutron/agent/resource_cache.py\n  Comments left for invalid file neutron/agent/rpc.py\n  Comments left for invalid file neutron/agent/windows/ip_lib.py\n  Comments left for invalid file neutron/api/api_common.py\n  Comments left for invalid file neutron/api/extensions.py\n  Comments left for invalid file neutron/api/rpc/agentnotifiers/dhcp_rpc_agent_api.py\n  Comments left for invalid file neutron/api/rpc/callbacks/version_manager.py\n  Comments left for invalid file neutron/api/rpc/handlers/dhcp_rpc.py\n  Comments left for invalid file neutron/api/rpc/handlers/l3_rpc.py\n  Comments left for invalid file neutron/api/rpc/handlers/resources_rpc.py\n  Comments left for invalid file neutron/api/rpc/handlers/securitygroups_rpc.py\n  Comments left for invalid file neutron/api/v2/base.py\n  Comments left for invalid file neutron/cmd/ovn/migration_mtu.py\n  Comments left for invalid file neutron/cmd/sanitize_port_binding_profile_allocation.py\n  Comments left for invalid file neutron/common/cache_utils.py\n  Comments left for invalid file neutron/common/ovn/acl.py\n  Comments left for invalid file neutron/common/ovn/utils.py\n  Comments left for invalid file neutron/common/utils.py\n  Comments left for invalid file neutron/core_extensions/qos.py\n  Comments left for invalid file neutron/db/address_group_db.py\n  Comments left for invalid file neutron/db/address_scope_db.py\n  Comments left for invalid file neutron/db/agents_db.py\n  Comments left for invalid file neutron/db/agentschedulers_db.py\n  Comments left for invalid file neutron/db/allowedaddresspairs_db.py\n  Comments left for invalid file neutron/db/data_plane_status_db.py\n  Comments left for invalid file neutron/db/db_base_plugin_common.py\n  Comments left for invalid file neutron/db/db_base_plugin_v2.py\n  Comments left for invalid file neutron/db/dns_db.py\n  Comments left for invalid file neutron/db/dvr_mac_db.py\n  Comments left for invalid file neutron/db/external_net_db.py\n  Comments left for invalid file neutron/db/extradhcpopt_db.py\n  Comments left for invalid file neutron/db/extraroute_db.py\n  Comments left for invalid file neutron/db/flavors_db.py\n  Comments left for invalid file neutron/db/ipam_backend_mixin.py\n  Comments left for invalid file neutron/db/ipam_pluggable_backend.py\n  Comments left for invalid file neutron/db/l3_agentschedulers_db.py\n  Comments left for invalid file neutron/db/l3_attrs_db.py\n  Comments left for invalid file neutron/db/l3_db.py\n  Comments left for invalid file neutron/db/l3_dvr_db.py\n  Comments left for invalid file neutron/db/l3_dvrscheduler_db.py\n  Comments left for invalid file neutron/db/l3_gateway_ip_qos.py\n  Comments left for invalid file neutron/db/l3_hamode_db.py\n  Comments left for invalid file neutron/db/l3_hascheduler_db.py\n  Comments left for invalid file neutron/db/local_ip_db.py\n  Comments left for invalid file neutron/db/metering/metering_db.py\n  Comments left for invalid file neutron/db/migration/autogen.py\n  Comments left for invalid file neutron/db/models/network_segment_range.py\n  Comments left for invalid file neutron/db/models/subnet_service_type.py\n  Comments left for invalid file neutron/db/network_ip_availability_db.py\n  Comments left for invalid file neutron/db/ovn_revision_numbers_db.py\n  Comments left for invalid file neutron/db/port_device_profile_db.py\n  Comments left for invalid file neutron/db/port_numa_affinity_policy_db.py\n  Comments left for invalid file neutron/db/portsecurity_db.py\n  Comments left for invalid file neutron/db/portsecurity_db_common.py\n  Comments left for invalid file neutron/db/quota/api.py\n  Comments left for invalid file neutron/db/quota/driver.py\n  Comments left for invalid file neutron/db/quota/models.py\n  Comments left for invalid file neutron/db/securitygroups_db.py\n  Comments left for invalid file neutron/db/securitygroups_rpc_base.py\n  Comments left for invalid file neutron/db/servicetype_db.py\n  Comments left for invalid file neutron/db/uplink_status_propagation_db.py\n  Comments left for invalid file neutron/debug/commands.py\n  Comments left for invalid file neutron/debug/debug_agent.py\n  Comments left for invalid file neutron/extensions/agent.py\n  Comments left for invalid file neutron/extensions/dhcpagentscheduler.py\n  Comments left for invalid file neutron/extensions/l3agentscheduler.py\n  Comments left for invalid file neutron/extensions/quotasv2.py\n  Comments left for invalid file neutron/extensions/segment.py\n  Comments left for invalid file neutron/extensions/tagging.py\n  Comments left for invalid file neutron/hacking/checks.py\n  Comments left for invalid file neutron/ipam/driver.py\n  Comments left for invalid file neutron/ipam/drivers/neutrondb_ipam/db_api.py\n  Comments left for invalid file neutron/ipam/drivers/neutrondb_ipam/driver.py\n  Comments left for invalid file neutron/ipam/requests.py\n  Comments left for invalid file neutron/ipam/subnet_alloc.py\n  Comments left for invalid file neutron/ipam/utils.py\n  Comments left for invalid file neutron/manager.py\n  Comments left for invalid file neutron/neutron_plugin_base_v2.py\n  Comments left for invalid file neutron/notifiers/ironic.py\n  Comments left for invalid file neutron/notifiers/nova.py\n  Comments left for invalid file neutron/pecan_wsgi/controllers/extensions.py\n  Comments left for invalid file neutron/pecan_wsgi/controllers/quota.py\n  Comments left for invalid file neutron/pecan_wsgi/controllers/resource.py\n  Comments left for invalid file neutron/pecan_wsgi/controllers/root.py\n  Comments left for invalid file neutron/pecan_wsgi/hooks/ownership_validation.py\n  Comments left for invalid file neutron/pecan_wsgi/hooks/policy_enforcement.py\n  Comments left for invalid file neutron/pecan_wsgi/hooks/query_parameters.py\n  Comments left for invalid file neutron/pecan_wsgi/hooks/userfilters.py\n  Comments left for invalid file neutron/plugins/ml2/db.py\n  Comments left for invalid file neutron/plugins/ml2/driver_context.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/agent/_common_agent.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/l2pop/mech_driver.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/l2pop/rpc_manager/l2population_rpc.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/linuxbridge/agent/extension_drivers/qos_driver.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/macvtap/mech_driver/mech_macvtap.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/mech_agent.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/mech_sriov/agent/eswitch_manager.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/mech_sriov/agent/extension_drivers/qos_driver.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/mech_sriov/agent/sriov_nic_agent.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/mech_sriov/mech_driver/mech_driver.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/extension_drivers/qos_driver.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/openflow/native/ofswitch.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/ovs_dvr_neutron_agent.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/agent/ovs_neutron_agent.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/openvswitch/mech_driver/mech_openvswitch.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/ovn/mech_driver/mech_driver.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/extensions/qos.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/impl_idl_ovn.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/maintenance.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_client.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovn_db_sync.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/ovn/mech_driver/ovsdb/ovsdb_monitor.py\n  Comments left for invalid file neutron/plugins/ml2/drivers/type_tunnel.py\n  Comments left for invalid file neutron/plugins/ml2/extensions/dns_integration.py\n  Comments left for invalid file neutron/plugins/ml2/extensions/subnet_dns_publish_fixed_ip.py\n  Comments left for invalid file neutron/plugins/ml2/managers.py\n  Comments left for invalid file neutron/plugins/ml2/plugin.py\n  Comments left for invalid file neutron/plugins/ml2/rpc.py\n  Comments left for invalid file neutron/policy.py\n  Comments left for invalid file neutron/privileged/agent/linux/ip_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/netlink_lib.py\n  Comments left for invalid file neutron/privileged/agent/linux/tc_lib.py\n  Comments left for invalid file neutron/profiling/profiled_decorator.py\n  Comments left for invalid file neutron/quota/__init__.py\n  Comments left for invalid file neutron/scheduler/base_resource_filter.py\n  Comments left for invalid file neutron/scheduler/dhcp_agent_scheduler.py\n  Comments left for invalid file neutron/scheduler/l3_agent_scheduler.py\n  Comments left for invalid file neutron/scheduler/l3_ovn_scheduler.py\n  Comments left for invalid file neutron/services/auto_allocate/db.py\n  Comments left for invalid file neutron/services/auto_allocate/plugin.py\n  Comments left for invalid file neutron/services/conntrack_helper/plugin.py\n  Comments left for invalid file neutron/services/externaldns/drivers/designate/driver.py\n  Comments left for invalid file neutron/services/l3_router/service_providers/driver_controller.py\n  Comments left for invalid file neutron/services/logapi/agent/l3/base.py\n  Comments left for invalid file neutron/services/logapi/common/db_api.py\n  Comments left for invalid file neutron/services/logapi/drivers/base.py\n  Comments left for invalid file neutron/services/logapi/drivers/openvswitch/ovs_firewall_log.py\n  Comments left for invalid file neutron/services/logapi/drivers/ovn/driver.py\n  Comments left for invalid file neutron/services/logapi/logging_plugin.py\n  Commen... (truncated)","accounts_in_message":[],"_revision_number":1},{"id":"4249fbca0151f397d3b09953c1ac36e4a8b2db04","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-08-09 19:09:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"882d7e4e2be234c2de3bcc18ee6a71dd281ed416","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-08-09 19:10:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"70c576edf194cef194e2c01b03c495c17b054801","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 19:49:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8a6893285094e169ef32cfac8bae4b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/626ec94ebf634303948459f424cdd716 : SUCCESS in 33m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e65b0ec66e27430a8260cde22fe293a3 : SUCCESS in 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed9ab90c000bba15a36fceb9bc1115a89543991f","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-09 19:54:34.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/50/852650/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"48c84df056a162229f25f834f465c0f6e321be22","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-09 20:55:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/852650/2 : SUCCESS in 1h 44m 30s","accounts_in_message":[],"_revision_number":2},{"id":"418bcd3a6d70836933257b971686686d6a52d2df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 20:57:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b27643477d94316a26d209c99db246e\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/f18295f735b74d40a94001c95367d8de : SUCCESS in 1h 25m 16s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/112f78c82f0746bc8c0381f7c6026f4b : SUCCESS in 1h 19m 06s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/e757f795be474cc7a48baa8550bd7acb : SUCCESS in 1h 05m 31s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/eb286be094bb464c955142fd5627cf9c : SUCCESS in 1h 12m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/756663b5de25491dbe25d4dbfa6f0897 : SUCCESS in 42m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e835a85bd1414766a0561514d17856a1 : SUCCESS in 38m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b82d4d579c124c6cacc0dc319cfb2755 : SUCCESS in 13m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0521a24ee37a42ec867396f0b57222d3 : SUCCESS in 30m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdd9d4cdf29e4f7b8c6b33d5416348f4 : SUCCESS in 26m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54e37d23f68d439994799caf90f4cb1c : SUCCESS in 28m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6c1315b71cf4212bc673b9c2c868fdb : SUCCESS in 9m 42s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/3d9d10d516c84f09ba1481dbdee92935 : FAILURE in 55m 02s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/2eabc5cdc7574b0e94db5c9c8179ed17 : SUCCESS in 1h 41m 08s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/6103b6d3a8a94ad1921c8229957a6f11 : SUCCESS in 1h 17m 25s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/61de627991a04c90a89f3851bf408675 : SUCCESS in 1h 07m 26s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/2c140f08dc8c41b8acca26ac252905e6 : SUCCESS in 1h 11m 15s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/e3f143b47c8e4878ab0837ecb69b2bd4 : SUCCESS in 1h 25m 07s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/93c8dda48a204c38a3313ab9078e3261 : SUCCESS in 1h 26m 01s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7d5953729934371b32c083e1a28d36e : SUCCESS in 1h 14m 40s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/ec64cf905fd64e858e54fe3fda716aef : SUCCESS in 1h 24m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8c88663deab78bb4b328e2f7c38a746b83411af","author":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"date":"2022-08-10 14:06:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"41deb8be90f56adca250ee85a01f3fe9c9da89cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 14:47:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35d96674af984ab8877e3d2ba3f6b7f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4980d85ecfdb4863a27f9639974ae32b : SUCCESS in 39m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c574f3f100e14b7fbf35b23191c86af5 : SUCCESS in 31m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"341182668b4a16d9a6f0def44ac4ca1ced1336f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 15:38:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ddfe168c0b143ea98d286b51078c87a\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/fe5fced138b5437884bba4315e8dbaaa : SUCCESS in 1h 24m 00s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/88e82b0c2c5243579b56e9fb522b1c7f : SUCCESS in 1h 19m 35s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/7189b748e1ee4b1d8a1d6565b34e6fe5 : SUCCESS in 1h 25m 39s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/5c859c70d6c0442895ccb61cff876127 : SUCCESS in 1h 22m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bd1f807afdcf485d8fc9ddafde119e8e : SUCCESS in 36m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75435df58574455eaeb78a49bd0b10bc : SUCCESS in 39m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2eff00a5eec4b3a8cc7e1fc86d8d146 : SUCCESS in 11m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50740b56f2fe44fba9412560b27b1232 : SUCCESS in 25m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36d0749d411b46ddb244e9561043eb1a : SUCCESS in 24m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/306caa5da59c49f0961a3ab1f2458211 : SUCCESS in 25m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a84b307da0497a8d8be65b9583458e : SUCCESS in 9m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/0b33544617b146ce815c9b5433624f6a : SUCCESS in 57m 05s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/9a8adfbb481c4f4aad7c362c6319b51b : SUCCESS in 1h 27m 33s\n- neutron-ovs-rally-task https://zuul.opendev.org/t/openstack/build/e65bb9ab603845b181caae0444352eef : SUCCESS in 1h 13m 22s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/69d5ae054a684930a88a4b29daf785f9 : SUCCESS in 50m 08s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/50601db5e8cf4a5396be461fbc77d963 : SUCCESS in 51m 51s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/6c8414b8150c4e94aa37fdc8f3585950 : SUCCESS in 1h 09m 11s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/28b38f018fa44b7a85ecd3893a45d813 : SUCCESS in 1h 08m 11s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21c3cdd04f2b42e1bb1289300b75b9d2 : SUCCESS in 1h 12m 51s\n- neutron-ovn-rally-task https://zuul.opendev.org/t/openstack/build/693fe9534cd741a48e0bcb9312d91f1e : SUCCESS in 1h 03m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86d7823b0321f24f301f070a6c6b2d33fce164cc","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-10 15:55:52.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/852650/2 : SUCCESS in 1h 48m 46s","accounts_in_message":[],"_revision_number":2},{"id":"8988f9dd13c9c2aa00a23f9b4964c87539180c45","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2022-08-16 04:24:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95b5f9e1b1929095affcfcb29e30b8ecdbdc281a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-16 08:30:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8a2ce8fc23f707cdd082a6300c75329ca2854c0d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-08-17 14:03:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fb188d75d0e6d983b59bc72347b51e15c71d3ff3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 14:04:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"207a84e8c8ba3e058c1169ecb5e89f471da3608b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 16:07:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1d02cb5e5a743e28ff52981275239b7\n\n- neutron-tempest-plugin-linuxbridge https://zuul.opendev.org/t/openstack/build/c4273dbf18fd4ac3941297291d79de71 : SUCCESS in 1h 42m 51s\n- neutron-tempest-plugin-openvswitch https://zuul.opendev.org/t/openstack/build/f7f15edc3e934b4eaa40b4e059c4ded1 : SUCCESS in 1h 28m 44s\n- neutron-tempest-plugin-ovn https://zuul.opendev.org/t/openstack/build/052dab58d7c54b8eb10b6f62eb54e93f : SUCCESS in 1h 28m 10s\n- neutron-tempest-plugin-openvswitch-iptables_hybrid https://zuul.opendev.org/t/openstack/build/0c542778c56449bca381e4cb9a2ea4af : SUCCESS in 1h 17m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6449e461a9a84cd5b4f1bce5e9ac4205 : SUCCESS in 12m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68918c1e998047c4b78a7fba21eef47d : SUCCESS in 27m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a48f3a599ac04273982e2eaad5318ee6 : SUCCESS in 27m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6623e49aff44a05800326ad5fcc9c29 : SUCCESS in 9m 54s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/6a93651feb70420fbb956ae29487b045 : SUCCESS in 58m 38s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c9902d0730674c2daad76f444c228f5c : SUCCESS in 1h 44m 37s\n- neutron-ovs-tempest-multinode-full https://zuul.opendev.org/t/openstack/build/87be8bf28a0a417f831cbfe486934713 : SUCCESS in 1h 14m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a3bb328d02b47359b0ea49c6fbc00bc : SUCCESS in 1h 13m 01s\n- neutron-ovs-grenade-dvr-multinode https://zuul.opendev.org/t/openstack/build/8b90ce245ce74190bba9056e8ce06537 : SUCCESS in 1h 12m 02s\n- neutron-ovn-tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81e2bda41d684c3b93db0e7939e87fb3 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":2},{"id":"48af5b0e6782a74a67af724fd613d5a113b1d8f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 16:07:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"36970636a5199da55f3fae6713f5ab5f8f86a8fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 16:08:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73ec732f1ebd44f0a8de1fae2392086b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87b287c3c11b44ba9d72b3c15591dd44 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b603093307295f47b0e45055ec21b29ac25b8660","revisions":{"1ad40ff3eb645f5c3a814d36491179cd3e1b4ae9":{"kind":"REWORK","_number":1,"created":"2022-08-09 17:29:51.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/50/852650/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/852650/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/852650/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/852650/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/852650/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/852650/1"}}},"commit":{"parents":[{"commit":"ac0f24735205612922be90e41c911056faf8ac14","subject":"Merge \"update the nova host aggregates links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac0f24735205612922be90e41c911056faf8ac14"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-08-09 17:17:54.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-08-09 17:17:54.000000000","tz":-240},"subject":"Fix some pylint indentation warnings","message":"Fix some pylint indentation warnings\n\nRunning with a stricter .pylintrc generates a lot of\nC0330 warnings (hanging/continued indentation). Fix\nthe ones in neutron/objects.\n\nTrivialfix\n\nChange-Id: I5e885e2b2a09d48197de30d8c6de5d74c10f5987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ad40ff3eb645f5c3a814d36491179cd3e1b4ae9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/1ad40ff3eb645f5c3a814d36491179cd3e1b4ae9"}]},"branch":"refs/heads/master"},"b603093307295f47b0e45055ec21b29ac25b8660":{"kind":"REWORK","_number":2,"created":"2022-08-09 19:10:25.000000000","uploader":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"ref":"refs/changes/50/852650/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/852650/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/852650/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/852650/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/852650/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/852650/2"}}},"commit":{"parents":[{"commit":"ac0f24735205612922be90e41c911056faf8ac14","subject":"Merge \"update the nova host aggregates links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/ac0f24735205612922be90e41c911056faf8ac14"}]}],"author":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-08-09 17:17:54.000000000","tz":-240},"committer":{"name":"Brian Haley","email":"haleyb.dev@gmail.com","date":"2022-08-09 19:10:01.000000000","tz":-240},"subject":"Fix some pylint indentation warnings","message":"Fix some pylint indentation warnings\n\nRunning with a stricter .pylintrc generates a lot of\nC0330 warnings (hanging/continued indentation). Fix\nthe ones in neutron/objects.\n\nTrivialfix\n\nChange-Id: I5e885e2b2a09d48197de30d8c6de5d74c10f5987\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b603093307295f47b0e45055ec21b29ac25b8660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/b603093307295f47b0e45055ec21b29ac25b8660"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
