)]}'
{"id":"openstack%2Fneutron~852841","triplet_id":"openstack%2Fneutron~master~I3bf3f80967ccffc02f3b937725c7363461dab29f","project":"openstack/neutron","branch":"master","topic":"bug/1985049","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-08-23 10:32:10.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-08-23 10:32:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3bf3f80967ccffc02f3b937725c7363461dab29f","subject":"Migrate \"download_gerrit_change\" to use \"cliff\"","status":"MERGED","created":"2022-08-11 09:58:58.000000000","updated":"2022-08-23 10:33:36.000000000","submitted":"2022-08-23 10:32:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"852841-bug/1985049","meta_rev_id":"e25a1164b0ca9adbb75d98331673a8ccbc6c19db","_number":852841,"virtual_id_number":852841,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 10:32:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":2,"date":"2022-08-22 23:30:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-08-22 15:00:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":1,"date":"2022-08-22 23:30:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-11 11:27:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-11 11:45:56.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-11 12:53:40.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"CC"},{"updated":"2022-08-22 15:00:47.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-08-22 23:30:24.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"ed7d9d310a63c5c2aa8772d4b94582c8868f8be6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-11 09:58:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a63648b3a809029f6d6eedf8b8bd25b29e494cef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 11:27:26.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6616d7c9826408aa092a97892e202c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b94f781175544491b60ba1fe946452bc : FAILURE in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acc79d7688034ef39a04a72e444f1fc7 : SUCCESS in 9m 08s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/023f095be8364f1e8326f55b63dbfa8d : SUCCESS in 53m 24s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/228f6e200b154d4e84d2b0e5c27952c0 : SUCCESS in 1h 21m 55s\n\nWarning:\n  Comments left for invalid file tools/migrate_names.py","accounts_in_message":[],"_revision_number":1},{"id":"13177085f6c6ca66fd720edd38aced11115b527a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-11 11:45:56.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/852841/1 : SUCCESS in 1h 45m 51s","accounts_in_message":[],"_revision_number":1},{"id":"941840f845dda7a255fa6b8a15551c37b7d2aadf","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-08-11 12:53:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d08f59bfaca0f02c545bc4a4c9accfc5a0d2245","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-11 13:26:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bca81d11f23c685f700f911c0132f9149a2d4e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-11 14:08:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"56097cadcaa7c87a56709fa330d3f7abcaa7f974","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-08-11 15:54:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6aa999a31f85a918962bdf969ebb6794e1387c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-11 16:04:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9553b596cc684653aeb4cc786b94fdce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36136117fd204b72a297928da9734e9e : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb174f9461a5412ca3d930069d2cca07 : SUCCESS in 9m 51s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/1a022dc9c5b24b9aa775056e1a99415d : SUCCESS in 1h 03m 42s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/50dc40e3b68d46e1bae53444b0e8fa81 : SUCCESS in 1h 54m 53s","accounts_in_message":[],"_revision_number":2},{"id":"d0e61f0a8d6aa9d9bfe1bedaf9e8067d4af0f70c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-11 16:31:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- neutron http://cloudbase-ci.com/neutron/852841/2 : SUCCESS in 2h 22m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8c414a5d7ee8ee6bb5e6d27da409f0a0005a46ed","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-08-22 15:00:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3be8cd273d6b3721540ac5526dbdc854037d4204","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-08-22 23:30:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8026d93cf45718627af612729243d99f7775294f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-22 23:31:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"acf590bd05e0f7b03431897aa171f96e3d9cc8ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 01:25:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/31573b9e4bed46a1a39e564c3be6d0e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b811d30a65242b5a501ba346fe4b987 : SUCCESS in 13m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb87ac864c0a47c5b9af7e86dece31e8 : SUCCESS in 8m 36s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/99b0ce678af7406c89375260a24060da : POST_FAILURE in 1h 20m 46s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/8262f8ff6c8b4c4c987bde96001ce1c3 : SUCCESS in 1h 19m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1b356418a98f2f7470cd1bb6d42ec15b60faf6d8","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-08-23 07:36:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b4cfe5eb4e989baa7f31aa0a6f96fb173ae4052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 09:04:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f42dbe7f82f348018fe7f180307b1c3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14fb1deea7aa4ea4b3b78dc0b84d37b5 : SUCCESS in 11m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2302f21e87c648e4a449d6d48ed2dc27 : SUCCESS in 9m 15s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/a1264c65c4c34fce871dea2c0b0620c8 : SUCCESS in 1h 01m 35s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/0a546479cf774b79a09333015c33d98d : SUCCESS in 1h 24m 33s","accounts_in_message":[],"_revision_number":2},{"id":"48f48715055b2bc2a755f9c0629962bf1cd00df0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 09:04:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"44a247acfe893b1486e5eecb640cccc15a368294","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-23 09:26:01.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/852841/2 : FAILURE in 1h 49m 16s","accounts_in_message":[],"_revision_number":2},{"id":"4e122fa96ffdf543eba645a04e7fe54a4f134d8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 10:32:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d648f38f82b44e0ad60091a0d7efa00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ab637fcc2554598bce60bee28085042 : SUCCESS in 11m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25661a910aea400dba3e6ce123a97aa9 : SUCCESS in 9m 11s\n- neutron-functional-with-uwsgi https://zuul.opendev.org/t/openstack/build/f5dc6ee51959457f83321da405c6c421 : SUCCESS in 53m 15s\n- neutron-fullstack-with-uwsgi https://zuul.opendev.org/t/openstack/build/c983541a2af04834866a7502a3cb801d : SUCCESS in 1h 25m 55s","accounts_in_message":[],"_revision_number":2},{"id":"64a2883448f846dee36a5ef1df8f7c26942ccaae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 10:32:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e25a1164b0ca9adbb75d98331673a8ccbc6c19db","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 10:33:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f5436ca68f642ab9f04792f5cc9ae10\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9fc6b2a2bfc445bab8af43ca369c8b7 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b20fa66b7bc321e6be1475ffa8856eb043064df","revisions":{"76f9060ddf7c9c1bf413e8325ef3d18b4fb5f994":{"kind":"REWORK","_number":1,"created":"2022-08-11 09:58:58.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/852841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/852841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/852841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/852841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/852841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/852841/1"}}},"commit":{"parents":[{"commit":"9ee0ff6790f6f01342d47060dc7a198d53343664","subject":"Migrate \"migrate_names\" to use \"cliff\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/9ee0ff6790f6f01342d47060dc7a198d53343664"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-10 03:49:01.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-10 03:50:10.000000000","tz":120},"subject":"Migrate \"download_gerrit_change\" to use \"cliff\"","message":"Migrate \"download_gerrit_change\" to use \"cliff\"\n\nPartial-Bug: #1985049\nChange-Id: I3bf3f80967ccffc02f3b937725c7363461dab29f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76f9060ddf7c9c1bf413e8325ef3d18b4fb5f994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76f9060ddf7c9c1bf413e8325ef3d18b4fb5f994"}]},"branch":"refs/heads/master"},"0b20fa66b7bc321e6be1475ffa8856eb043064df":{"kind":"REWORK","_number":2,"created":"2022-08-11 14:08:17.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/41/852841/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/41/852841/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/852841/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/852841/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/41/852841/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/41/852841/2"}}},"commit":{"parents":[{"commit":"8bf3eba1014bba4896c4654e932e7f7b8281925c","subject":"Migrate \"migrate_names\" to use \"cliff\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8bf3eba1014bba4896c4654e932e7f7b8281925c"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-10 03:49:01.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-08-10 07:59:32.000000000","tz":120},"subject":"Migrate \"download_gerrit_change\" to use \"cliff\"","message":"Migrate \"download_gerrit_change\" to use \"cliff\"\n\nPartial-Bug: #1985049\nChange-Id: I3bf3f80967ccffc02f3b937725c7363461dab29f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b20fa66b7bc321e6be1475ffa8856eb043064df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/0b20fa66b7bc321e6be1475ffa8856eb043064df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
