)]}'
{"id":"openstack%2Fneutron~853050","triplet_id":"openstack%2Fneutron~master~I3bda1f744f41ce82c4c3f2a4c9fc31582cce2253","project":"openstack/neutron","branch":"master","topic":"bug/1986418","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2022-08-15 13:53:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3bda1f744f41ce82c4c3f2a4c9fc31582cce2253","subject":"Fix deprecation warnings about oslo.context","status":"MERGED","created":"2022-08-13 09:04:18.000000000","updated":"2022-08-15 13:54:57.000000000","submitted":"2022-08-15 13:53:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853050-bug/1986418","meta_rev_id":"a9ee57a984b7b1f93b763bc5011ab1bab5a55b4c","_number":853050,"virtual_id_number":853050,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-08-13 11:21:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-15 13:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-15 13:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2022-08-15 07:46:10.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-15 13:22:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}],"REVIEWER":[{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-13 09:22:24.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-13 09:51:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-13 09:56:28.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-15 07:46:10.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-08-15 13:22:51.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"}],"messages":[{"id":"dc2ec3bfe5b407284d24f1d79ffdba7678f7fa79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-08-13 09:04:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d58b77b28f704a6609d9dcf7e6224cbc81a67b46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-08-13 09:07:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"eb5fa655ba74b169cadeb5b1a9f86878a1933ab8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-08-13 09:08:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c09e0934dc4c4bd8240e9db677112e2b40f024a2","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-13 09:22:24.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/853050/3 : FAILURE in 13m 44s","accounts_in_message":[],"_revision_number":3},{"id":"0ba2473ba136a02bdd5a1e942cc62ee6398dcb75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 09:51:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be146ab2ef99400fa720d6792490a74f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a080a842e6eb421d90f72418a1712593 : SUCCESS in 39m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f78f35fecd164c70ba8872a1980ad82f : SUCCESS in 12m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3454e3f3291a42faa24a248f1446d7d1 : SUCCESS in 27m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71c56df7460140c38339f89ab85af26a : SUCCESS in 26m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f22da54fffbf4995b2f036bcfe3b345a : SUCCESS in 27m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e413d332da6439fa1f97455bf32c0c9 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":3},{"id":"5c9d052df7ac17cbef394339bdb8e190b2c32fed","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-13 09:56:28.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/50/853050/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"6cd81135be574f48537de8be0ef8b4eb44ea3185","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 09:57:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f19b6befdcd407f89f0fb213a549ff1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fdc11cd3c354640a067ea56eed9f600 : SUCCESS in 47m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/777f620100ee460ea723f61c389f93a6 : SUCCESS in 31m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cb36b3a1ad88f023306e639d0ef309a6c80859a6","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-13 10:40:35.000000000","message":"Patch Set 2: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/50/853050/2 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":2},{"id":"7e12ee8d53861a69e383bf8da165a7832c699568","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-13 11:21:56.000000000","message":"Patch Set 3: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/50/853050/3 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":3},{"id":"cbf3f4749a63903b92f1eb218d5efaff50b02408","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-08-15 07:46:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2607bfd84e9fcc450d1f5439991e2642585349e1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-08-15 13:22:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e2b810d0ddb47b111af18bdc35ce10d1dca1bea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 13:23:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"66a72c4df95371b649aa46cb47b3d26ae3f70384","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 13:53:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ca22b0d378f4c08ad8e782cf99fc358\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a1341328494a37bd90984135f60bdc : SUCCESS in 13m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fded842ee5b8484db948d4548a06a018 : SUCCESS in 25m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c07fffc282e443f197e17c2027f601b2 : SUCCESS in 26m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea4bc2bae4548b0b2498842a0e27600 : SUCCESS in 9m 19s","accounts_in_message":[],"_revision_number":3},{"id":"05f3e930082ab20027dcb620ae772498052174a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 13:53:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a9ee57a984b7b1f93b763bc5011ab1bab5a55b4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 13:54:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6670ca3a83c4e2eaa31e0a8f161044c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb41d9b74f0d428eb4aab752f6010eb2 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a641b2508de41046666fff8f4d1e7a65aa80b1fe","revisions":{"8d438136cfce58ea2258c0611ab001ae3972336d":{"kind":"REWORK","_number":1,"created":"2022-08-13 09:04:18.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/50/853050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/853050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/853050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/853050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/853050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/853050/1"}}},"commit":{"parents":[{"commit":"f77ef29b870f374af417b414800797fc896b913f","subject":"Merge \"[OVN] Fix updating network segmentation ID\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f77ef29b870f374af417b414800797fc896b913f"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-13 08:03:32.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-13 08:40:20.000000000","tz":540},"subject":"Fix deprecation warnings about oslo.context","message":"Fix deprecation warnings about oslo.context\n\nFix the following deprecation warnings.\n\n  DeprecationWarning: Using the \u0027user\u0027 argument is deprecated in version \u00272.18\u0027\n  and will be removed in version \u00273.0\u0027, please use the \u0027user_id\u0027 argument\n  instead\n\n  DeprecationWarning: Property \u0027user\u0027 has moved to \u0027user_id\u0027 in version \u00272.6\u0027\n  and will be removed in version \u00273.0\u0027\n\nChange-Id: I3bda1f744f41ce82c4c3f2a4c9fc31582cce2253\nCloses-Bug: 1986418\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d438136cfce58ea2258c0611ab001ae3972336d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/8d438136cfce58ea2258c0611ab001ae3972336d"}]},"branch":"refs/heads/master"},"4479645f8cf599c6ca9fd58f38e2f52adad213b4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-08-13 09:07:37.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/50/853050/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/853050/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/853050/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/853050/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/853050/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/853050/2"}}},"commit":{"parents":[{"commit":"f77ef29b870f374af417b414800797fc896b913f","subject":"Merge \"[OVN] Fix updating network segmentation ID\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f77ef29b870f374af417b414800797fc896b913f"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-13 08:03:32.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-13 09:07:37.000000000","tz":0},"subject":"Fix deprecation warnings about oslo.context","message":"Fix deprecation warnings about oslo.context\n\nFix the following deprecation warnings.\n\n  DeprecationWarning: Using the \u0027user\u0027 argument is deprecated\n  in version \u00272.18\u0027 and will be removed in version \u00273.0\u0027,\n  please use the \u0027user_id\u0027 argument instead\n\n  DeprecationWarning: Property \u0027user\u0027 has moved to \u0027user_id\u0027 in version \u00272.6\u0027\n  and will be removed in version \u00273.0\u0027\n\nChange-Id: I3bda1f744f41ce82c4c3f2a4c9fc31582cce2253\nCloses-Bug: 1986418\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4479645f8cf599c6ca9fd58f38e2f52adad213b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/4479645f8cf599c6ca9fd58f38e2f52adad213b4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a641b2508de41046666fff8f4d1e7a65aa80b1fe":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-08-13 09:08:00.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/50/853050/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/50/853050/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/853050/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/853050/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/50/853050/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/50/853050/3"}}},"commit":{"parents":[{"commit":"f77ef29b870f374af417b414800797fc896b913f","subject":"Merge \"[OVN] Fix updating network segmentation ID\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f77ef29b870f374af417b414800797fc896b913f"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-13 08:03:32.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-13 09:08:00.000000000","tz":0},"subject":"Fix deprecation warnings about oslo.context","message":"Fix deprecation warnings about oslo.context\n\nFix the following deprecation warnings.\n\n  DeprecationWarning: Using the \u0027user\u0027 argument is deprecated\n  in version \u00272.18\u0027 and will be removed in version \u00273.0\u0027,\n  please use the \u0027user_id\u0027 argument instead\n\n  DeprecationWarning: Property \u0027user\u0027 has moved to \u0027user_id\u0027\n  in version \u00272.6\u0027 and will be removed in version \u00273.0\u0027\n\nChange-Id: I3bda1f744f41ce82c4c3f2a4c9fc31582cce2253\nCloses-Bug: 1986418\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a641b2508de41046666fff8f4d1e7a65aa80b1fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/a641b2508de41046666fff8f4d1e7a65aa80b1fe"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
