)]}'
{"id":"openstack%2Fneutron~853053","triplet_id":"openstack%2Fneutron~master~I23540114120f6ea52754116cfaaeac35e09543b4","project":"openstack/neutron","branch":"master","topic":"bug/1986428","attention_set":{},"removed_from_attention_set":{"7634":{"account":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"last_update":"2022-08-15 19:32:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23540114120f6ea52754116cfaaeac35e09543b4","subject":"Fix deprecation warnings in pkg_resources","status":"MERGED","created":"2022-08-13 13:54:20.000000000","updated":"2022-08-15 19:33:46.000000000","submitted":"2022-08-15 19:32:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853053-bug/1986428","meta_rev_id":"81d2fc71719f6e93dc1ce765ec7d5ed47dab630c","_number":853053,"virtual_id_number":853053,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"recommended":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"all":[{"value":1,"date":"2022-08-13 14:43:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-15 19:32:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-15 18:00:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":2,"date":"2022-08-15 07:44:50.000000000","_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-15 18:00:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-13 14:08:52.000000000","updated_by":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"reviewer":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"state":"CC"},{"updated":"2022-08-13 14:34:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-13 14:43:50.000000000","updated_by":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-15 07:44:50.000000000","updated_by":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"reviewer":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"state":"REVIEWER"},{"updated":"2022-08-15 18:00:47.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"}],"messages":[{"id":"7b39951d51f4b9bdb8915ed30e838a12fc874c53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2022-08-13 13:54:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bf02cec072698090fd14d47476ff33e6be1af77d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26622,"name":"Cloudbase Neutron Hyper-V CI","email":"neutron_hyperv_ci@cloudbasesolutions.com","username":"neutron_hyperv_ci"},"date":"2022-08-13 14:08:52.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Neutron Hyper-V CI, add a review comment with run-Cloudbase Neutron Hyper-V CI\n\n- neutron http://cloudbase-ci.com/neutron/853053/1 : FAILURE in 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c6425bf83e0c8d1c535c49306256d9af93b61f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 14:34:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2fb23c559274eae95ce2bc5023270c4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/012963b6f850410b9b3e0240a4063de6 : SUCCESS in 35m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2bab2f573f24004ab3bb0145d0a3a22 : SUCCESS in 13m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f108319f7f44f2fa176365973d4a6b7 : SUCCESS in 32m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c32fa59c54b44d01b6ad9244c1a49a6c : SUCCESS in 32m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7b3f4065f1f4cc4990df100fe42cb2e : SUCCESS in 29m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1fb1bfbf2594801be57fe51f8bba165 : SUCCESS in 9m 28s","accounts_in_message":[],"_revision_number":1},{"id":"138f9bd118c79cb52bdbb1edaabdaf34fc40ffc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-13 14:37:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c79705705a6458b974d596132e24367\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3c5a037737b4890b8678ff2baa14753 : SUCCESS in 40m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e30c658b9a84bf5afc796a32b390341 : SUCCESS in 40m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c951ef274ba8aca809bfed14dfb9af82581212b","author":{"_account_id":9845,"name":"Arista CI","email":"arista-openstack-test@aristanetworks.com","username":"arista-test","tags":["SERVICE_USER"]},"date":"2022-08-13 14:43:50.000000000","message":"Patch Set 1: Verified+1\n\nArista third party testing PASSED [ https://arista-openstack-ci.arista.com:8000/August-2022/Arista_Tempest_Testrefs/changes/53/853053/1 ]\n\nFor testing details or any questions please check https://wiki.openstack.org/wiki/Arista-third-party-testing","accounts_in_message":[],"_revision_number":1},{"id":"9d0bf8f9259c07920879f1d66caf9b3df05453ff","author":{"_account_id":5948,"name":"Oleg Bondarev","email":"obondarev@mirantis.com","username":"obondarev"},"date":"2022-08-15 07:44:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5e1df5d6feb15f0ff01a4dce1642b7316922719","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-08-15 18:00:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acc1679752a43238cb86d256ee843c3af6c261d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 18:01:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5bfc0a17c48420f61e0a90ad80851266358701d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 19:32:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2d024e1d1a74d12919c7226ab09fb29\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a927882e08674f03a986e84cafea4037 : SUCCESS in 11m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6d1ab6a134d452db4565113be5c8c8f : SUCCESS in 24m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84f3dda741cb447a9e0d6c71aa149493 : SUCCESS in 26m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c3e0ea5e279450992892688be5e9600 : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":1},{"id":"9113887f47711a354772987d95a6ae1368eeaad3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 19:32:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"81d2fc71719f6e93dc1ce765ec7d5ed47dab630c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 19:33:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc4b82803e7840aaa1474dc6765c680d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9192c368ea094d038e852d2c118f815a : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76cf6b4a9e9a8c8f7b01f44d787b66b9b894331b","revisions":{"76cf6b4a9e9a8c8f7b01f44d787b66b9b894331b":{"kind":"REWORK","_number":1,"created":"2022-08-13 13:54:20.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/53/853053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/neutron","ref":"refs/changes/53/853053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/853053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/853053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/neutron refs/changes/53/853053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/neutron refs/changes/53/853053/1"}}},"commit":{"parents":[{"commit":"f77ef29b870f374af417b414800797fc896b913f","subject":"Merge \"[OVN] Fix updating network segmentation ID\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/f77ef29b870f374af417b414800797fc896b913f"}]}],"author":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-13 13:30:51.000000000","tz":540},"committer":{"name":"Takashi Natsume","email":"takanattie@gmail.com","date":"2022-08-13 13:53:21.000000000","tz":540},"subject":"Fix deprecation warnings in pkg_resources","message":"Fix deprecation warnings in pkg_resources\n\nFix the following deprecation warnings.\n\n  PkgResourcesDeprecationWarning:\n  \u003cMagicMock name\u003d\u0027execute().split().__getitem__().__getitem__()\u0027\n  id\u003d\u0027140417024565696\u0027\u003e is an invalid version and will not be\n  supported in a future release\n\n  DeprecationWarning: Creating a LegacyVersion has been\n  deprecated and will be removed in the next major release\n\nChange-Id: I23540114120f6ea52754116cfaaeac35e09543b4\nCloses-Bug: 1986428\nSigned-off-by: Takashi Natsume \u003ctakanattie@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76cf6b4a9e9a8c8f7b01f44d787b66b9b894331b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/neutron/commit/76cf6b4a9e9a8c8f7b01f44d787b66b9b894331b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
